)]}'
{"id":"openstack%2Fswift~654671","triplet_id":"openstack%2Fswift~master~I973c2e1bb8969cf6bffece8ce68881c393efbaef","project":"openstack/swift","branch":"master","topic":"py3_symlink","hashtags":[],"change_id":"I973c2e1bb8969cf6bffece8ce68881c393efbaef","subject":"py3: port symlink middleware","status":"MERGED","created":"2019-04-23 02:18:03.000000000","updated":"2019-05-31 00:57:39.000000000","submitted":"2019-05-31 00:56:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654671-1559264178646-c1ac9d4a","meta_rev_id":"82281c223d2a862cd3688d2e7d267ba211cda90a","_number":654671,"virtual_id_number":654671,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":25314,"name":"黄海滔","email":"huanght@awcloud.com","username":"huanght"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":30427,"name":"Michelle King","email":"wangyan@awcloud.com"},{"value":2,"date":"2019-05-31 00:56:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":25314,"name":"黄海滔","email":"huanght@awcloud.com","username":"huanght"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":30427,"name":"Michelle King","email":"wangyan@awcloud.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-30 18:56:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":25314,"name":"黄海滔","email":"huanght@awcloud.com","username":"huanght"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":30427,"name":"Michelle King","email":"wangyan@awcloud.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-30 18:56:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25314,"name":"黄海滔","email":"huanght@awcloud.com","username":"huanght"},{"_account_id":30427,"name":"Michelle King","email":"wangyan@awcloud.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-15 02:09:37.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2019-05-30 06:04:44.000000000","updated_by":{"_account_id":25314,"name":"黄海滔","email":"huanght@awcloud.com","username":"huanght"},"reviewer":{"_account_id":25314,"name":"黄海滔","email":"huanght@awcloud.com","username":"huanght"},"state":"REVIEWER"},{"updated":"2019-05-30 06:13:13.000000000","updated_by":{"_account_id":30427,"name":"Michelle King","email":"wangyan@awcloud.com"},"reviewer":{"_account_id":30427,"name":"Michelle King","email":"wangyan@awcloud.com"},"state":"REVIEWER"},{"updated":"2019-05-30 08:08:31.000000000","updated_by":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"reviewer":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"state":"REVIEWER"},{"updated":"2019-05-30 18:56:48.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2019-05-31 00:56:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dda62125bbc6e24d18900b55960bb7b3a2c7aab0","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2019-04-23 02:18:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e9c8f715e97f30c139bb48a8656eab5ee8471613","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 03:49:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/71/654671/1/check/openstack-tox-docs/4e998f5/html/ : SUCCESS in 4m 06s\n- build-openstack-releasenotes http://logs.openstack.org/71/654671/1/check/build-openstack-releasenotes/4d3911b/html/ : SUCCESS in 6m 36s\n- swift-tox-py27 http://logs.openstack.org/71/654671/1/check/swift-tox-py27/5662035/ : SUCCESS in 14m 19s\n- swift-tox-py35 http://logs.openstack.org/71/654671/1/check/swift-tox-py35/8c18c37/ : SUCCESS in 9m 23s (non-voting)\n- swift-tox-py36 http://logs.openstack.org/71/654671/1/check/swift-tox-py36/57829f5/ : SUCCESS in 8m 03s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/71/654671/1/check/swift-tox-py37/c4b4489/ : SUCCESS in 8m 57s\n- swift-tox-func http://logs.openstack.org/71/654671/1/check/swift-tox-func/52e4a4f/ : SUCCESS in 9m 16s\n- swift-tox-func-encryption http://logs.openstack.org/71/654671/1/check/swift-tox-func-encryption/7ba0b0b/ : SUCCESS in 10m 54s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/71/654671/1/check/swift-tox-func-domain-remap-staticweb/7d17890/ : SUCCESS in 15m 52s\n- swift-tox-func-ec http://logs.openstack.org/71/654671/1/check/swift-tox-func-ec/3fcf211/ : SUCCESS in 12m 21s\n- swift-tox-func-s3api http://logs.openstack.org/71/654671/1/check/swift-tox-func-s3api/bc9ee03/ : SUCCESS in 10m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/71/654671/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/18df5f0/ : SUCCESS in 11m 58s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/71/654671/1/check/swift-probetests-centos-7/ec6c9eb/ : SUCCESS in 30m 19s\n- swift-dsvm-functional http://logs.openstack.org/71/654671/1/check/swift-dsvm-functional/afae25c/ : SUCCESS in 26m 22s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/71/654671/1/check/swift-dsvm-functional-ipv6/6e0a0ba/ : SUCCESS in 26m 04s\n- swift-tox-lower-constraints http://logs.openstack.org/71/654671/1/check/swift-tox-lower-constraints/71dc3f0/ : SUCCESS in 11m 57s\n- openstack-tox-pep8 http://logs.openstack.org/71/654671/1/check/openstack-tox-pep8/c9d1248/ : FAILURE in 5m 02s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/71/654671/1/check/swift-multinode-rolling-upgrade/dff9665/ : SUCCESS in 16m 26s","accounts_in_message":[],"_revision_number":1},{"id":"66be9a194d95acbfd0a13d4689ddab5975b91e1e","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2019-04-23 05:55:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"513b2357b70fd4e3ba9c0aabab4c0760c7d67193","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 06:37:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/71/654671/2/check/openstack-tox-docs/21a502c/html/ : SUCCESS in 4m 38s\n- build-openstack-releasenotes http://logs.openstack.org/71/654671/2/check/build-openstack-releasenotes/87d0c41/html/ : SUCCESS in 7m 14s\n- swift-tox-py27 http://logs.openstack.org/71/654671/2/check/swift-tox-py27/510c66b/ : FAILURE in 19m 04s\n- swift-tox-py35 http://logs.openstack.org/71/654671/2/check/swift-tox-py35/fc1a3b9/ : SUCCESS in 8m 08s (non-voting)\n- swift-tox-py36 http://logs.openstack.org/71/654671/2/check/swift-tox-py36/09da335/ : FAILURE in 39m 49s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/71/654671/2/check/swift-tox-py37/71badc3/ : SUCCESS in 8m 35s\n- swift-tox-func http://logs.openstack.org/71/654671/2/check/swift-tox-func/08c9a01/ : SUCCESS in 8m 26s\n- swift-tox-func-encryption http://logs.openstack.org/71/654671/2/check/swift-tox-func-encryption/318592f/ : SUCCESS in 10m 42s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/71/654671/2/check/swift-tox-func-domain-remap-staticweb/76a61ae/ : SUCCESS in 10m 04s\n- swift-tox-func-ec http://logs.openstack.org/71/654671/2/check/swift-tox-func-ec/1dafaf2/ : SUCCESS in 10m 05s\n- swift-tox-func-s3api http://logs.openstack.org/71/654671/2/check/swift-tox-func-s3api/f090dee/ : SUCCESS in 10m 31s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/71/654671/2/check/swift-tox-func-s3api-ceph-s3tests-tempauth/7c6b1a1/ : SUCCESS in 11m 54s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/71/654671/2/check/swift-probetests-centos-7/bdf1063/ : SUCCESS in 28m 58s\n- swift-dsvm-functional http://logs.openstack.org/71/654671/2/check/swift-dsvm-functional/ff35570/ : SUCCESS in 27m 27s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/71/654671/2/check/swift-dsvm-functional-ipv6/1e5a693/ : SUCCESS in 27m 15s\n- swift-tox-lower-constraints http://logs.openstack.org/71/654671/2/check/swift-tox-lower-constraints/8013890/ : SUCCESS in 11m 58s\n- openstack-tox-pep8 http://logs.openstack.org/71/654671/2/check/openstack-tox-pep8/b9706c0/ : SUCCESS in 5m 42s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/71/654671/2/check/swift-multinode-rolling-upgrade/30f5296/ : SUCCESS in 16m 40s","accounts_in_message":[],"_revision_number":2},{"id":"1d589795fd36266293c557e0b0acbaff278cad02","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-05-15 02:09:37.000000000","message":"Patch Set 2: Code-Review+2\n\nI still think that header dictionaries with WSGI strings is absolutely insane, but whatever.","accounts_in_message":[],"_revision_number":2},{"id":"8b7bf7eabc6e6c51dac7a6693c550907bbc710e5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-17 20:57:33.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"3c8715e85d8c4fd5a84a2dc26dbf9ca1f33e5f62","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 21:42:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/71/654671/2/check/openstack-tox-docs/9a019fa/html/ : SUCCESS in 3m 51s\n- build-openstack-releasenotes http://logs.openstack.org/71/654671/2/check/build-openstack-releasenotes/9b4a547/html/ : SUCCESS in 6m 57s\n- swift-build-image http://logs.openstack.org/71/654671/2/check/swift-build-image/22e8bfa/ : SUCCESS in 8m 00s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/71/654671/2/check/swift-tox-py27/0f12001/ : SUCCESS in 13m 02s\n- swift-tox-py36 http://logs.openstack.org/71/654671/2/check/swift-tox-py36/470fe2a/ : FAILURE in 5m 55s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/71/654671/2/check/swift-tox-py37/1ccfc76/ : SUCCESS in 13m 56s\n- swift-tox-func http://logs.openstack.org/71/654671/2/check/swift-tox-func/852c5c5/ : SUCCESS in 9m 39s\n- swift-tox-func-encryption http://logs.openstack.org/71/654671/2/check/swift-tox-func-encryption/7ac9f2b/ : SUCCESS in 15m 14s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/71/654671/2/check/swift-tox-func-domain-remap-staticweb/ce4e4c1/ : SUCCESS in 10m 58s\n- swift-tox-func-ec http://logs.openstack.org/71/654671/2/check/swift-tox-func-ec/415d72c/ : SUCCESS in 9m 32s\n- swift-tox-func-s3api http://logs.openstack.org/71/654671/2/check/swift-tox-func-s3api/1e61487/ : SUCCESS in 13m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/71/654671/2/check/swift-tox-func-s3api-ceph-s3tests-tempauth/a73efd6/ : SUCCESS in 16m 50s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/71/654671/2/check/swift-probetests-centos-7/c630493/ : SUCCESS in 29m 07s\n- swift-dsvm-functional http://logs.openstack.org/71/654671/2/check/swift-dsvm-functional/83803bc/ : SUCCESS in 25m 34s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/71/654671/2/check/swift-dsvm-functional-ipv6/fbd7922/ : SUCCESS in 31m 07s\n- swift-tox-lower-constraints http://logs.openstack.org/71/654671/2/check/swift-tox-lower-constraints/8e8e2f2/ : SUCCESS in 14m 39s\n- openstack-tox-pep8 http://logs.openstack.org/71/654671/2/check/openstack-tox-pep8/f8c01db/ : SUCCESS in 4m 45s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/71/654671/2/check/swift-multinode-rolling-upgrade/c6e968a/ : FAILURE in 8m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7fc134f09db98592749f9c6a54d371fa371c178b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-18 01:33:09.000000000","message":"Patch Set 2:\n\n(4 comments)\n\nStill messing with func tests, but I\u0027ve got a couple concerns...","accounts_in_message":[],"_revision_number":2},{"id":"2cf9dadae792e67394177c77ec12fd2e4bf4330f","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2019-05-20 06:22:18.000000000","message":"Patch Set 2:\n\n(2 comments)\n\nNew patch is coming soon.","accounts_in_message":[],"_revision_number":2},{"id":"a974850bfb724cdd982823c9cba22eb12ab78eec","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2019-05-20 12:17:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"03cd9ce5e0d5f347ee6e62ad0ba9473983bf3e31","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 16:47:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/71/654671/3/check/openstack-tox-docs/7f97edf/html/ : SUCCESS in 4m 06s\n- build-openstack-releasenotes http://logs.openstack.org/71/654671/3/check/build-openstack-releasenotes/019ccdf/html/ : SUCCESS in 7m 20s\n- swift-build-image http://logs.openstack.org/71/654671/3/check/swift-build-image/6809f62/ : SUCCESS in 8m 07s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/71/654671/3/check/swift-tox-py27/9fd4cd4/ : SUCCESS in 11m 47s\n- swift-tox-py36 http://logs.openstack.org/71/654671/3/check/swift-tox-py36/20b7423/ : SUCCESS in 15m 52s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/71/654671/3/check/swift-tox-py37/4920501/ : SUCCESS in 11m 03s\n- swift-tox-func http://logs.openstack.org/71/654671/3/check/swift-tox-func/8dcb8c6/ : SUCCESS in 9m 38s\n- swift-tox-func-encryption http://logs.openstack.org/71/654671/3/check/swift-tox-func-encryption/931b53c/ : SUCCESS in 14m 02s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/71/654671/3/check/swift-tox-func-domain-remap-staticweb/8e9476f/ : SUCCESS in 9m 47s\n- swift-tox-func-ec http://logs.openstack.org/71/654671/3/check/swift-tox-func-ec/6987198/ : SUCCESS in 9m 32s\n- swift-tox-func-s3api http://logs.openstack.org/71/654671/3/check/swift-tox-func-s3api/03198db/ : SUCCESS in 10m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/71/654671/3/check/swift-tox-func-s3api-ceph-s3tests-tempauth/5ae1449/ : SUCCESS in 18m 37s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/71/654671/3/check/swift-probetests-centos-7/24a7557/ : SUCCESS in 30m 50s\n- swift-dsvm-functional http://logs.openstack.org/71/654671/3/check/swift-dsvm-functional/d0ffd85/ : SUCCESS in 28m 38s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/71/654671/3/check/swift-dsvm-functional-ipv6/ba285d2/ : SUCCESS in 28m 28s\n- swift-tox-lower-constraints http://logs.openstack.org/71/654671/3/check/swift-tox-lower-constraints/0e668f7/ : SUCCESS in 11m 45s\n- openstack-tox-pep8 http://logs.openstack.org/71/654671/3/check/openstack-tox-pep8/3a61eae/ : SUCCESS in 5m 31s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/71/654671/3/check/swift-multinode-rolling-upgrade/fae4e6b/ : FAILURE in 8m 59s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f19b073e6c3e5fb12b62addd378ce43863953e1c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-20 20:26:38.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"5211a82bc86e70bc08379f6192aa4715ffca6889","author":{"_account_id":25314,"name":"黄海滔","email":"huanght@awcloud.com","username":"huanght"},"date":"2019-05-30 06:04:44.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"ad63bb52ed4fbfa7c8129743e183d18a344c4de5","author":{"_account_id":30427,"name":"Michelle King","email":"wangyan@awcloud.com"},"date":"2019-05-30 06:13:13.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"6ac7f7f11f925859dfba5712248584918291bac7","author":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"date":"2019-05-30 08:08:31.000000000","message":"Patch Set 3: Code-Review+2\n\n(3 comments)\n\nLooks fine. I have small comments but not significant and it\u0027s nice to have rather than block the py3 works.\n\nThe reason why add only +2 (not yet +A) is I\u0027d like to wait Tim\u0027s opinion.","accounts_in_message":[],"_revision_number":3},{"id":"db2bb52a8092ecb400d78bc50ce6fece74d9c3b5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-30 18:53:59.000000000","message":"Patch Set 3:\n\nOK, I\u0027ve had a chance to play with fixing up the unit tests to reflect real inputs and get the func tests running -- and the only changes I wanted in the middleware are to drop the TypeError handling. I think I agree with Kota, better to get this in sooner rather than later so we can continue making progress. I\u0027ll\n\n- rebase (as my follow-up depends on the already-landed slo port),\n- fix up the commit message,\n- submit my follow-up for the tests, and\n- approve\n\nshortly.","accounts_in_message":[],"_revision_number":3},{"id":"27c47b7b9b40f020dbd0b283eb80d2c2a9f72213","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-30 18:56:32.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8827404e96ddea3edba66dcf6f83b43b00f88e7a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-30 18:56:48.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"288be7696d802579edbc1e188326547132c75a5a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 22:23:53.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/71/654671/4/check/build-openstack-api-ref/1ee8e59/html/ : SUCCESS in 3m 21s\n- openstack-tox-docs http://logs.openstack.org/71/654671/4/check/openstack-tox-docs/dd277c5/html/ : SUCCESS in 3m 59s\n- build-openstack-releasenotes http://logs.openstack.org/71/654671/4/check/build-openstack-releasenotes/6b3a409/html/ : SUCCESS in 6m 38s\n- swift-build-image http://logs.openstack.org/71/654671/4/check/swift-build-image/96ffdde/ : SUCCESS in 10m 20s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/71/654671/4/check/swift-tox-py27/fc402a7/ : SUCCESS in 18m 14s\n- swift-tox-py36 http://logs.openstack.org/71/654671/4/check/swift-tox-py36/d8e7ba2/ : SUCCESS in 10m 31s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/71/654671/4/check/swift-tox-py37/76fcf03/ : SUCCESS in 14m 33s\n- swift-tox-func http://logs.openstack.org/71/654671/4/check/swift-tox-func/3f244f3/ : SUCCESS in 14m 57s\n- swift-tox-func-py37 http://logs.openstack.org/71/654671/4/check/swift-tox-func-py37/a54b32e/ : SUCCESS in 7m 13s\n- swift-tox-func-encryption http://logs.openstack.org/71/654671/4/check/swift-tox-func-encryption/89632db/ : SUCCESS in 8m 55s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/71/654671/4/check/swift-tox-func-domain-remap-staticweb/fd79ae1/ : SUCCESS in 9m 36s\n- swift-tox-func-ec http://logs.openstack.org/71/654671/4/check/swift-tox-func-ec/3bd276e/ : SUCCESS in 8m 55s\n- swift-tox-func-s3api http://logs.openstack.org/71/654671/4/check/swift-tox-func-s3api/f3f2dbf/ : SUCCESS in 11m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/71/654671/4/check/swift-tox-func-s3api-ceph-s3tests-tempauth/7d696d8/ : SUCCESS in 12m 57s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/71/654671/4/check/swift-probetests-centos-7/e8dc1b7/ : SUCCESS in 33m 05s\n- swift-dsvm-functional http://logs.openstack.org/71/654671/4/check/swift-dsvm-functional/3983124/ : SUCCESS in 24m 40s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/71/654671/4/check/swift-dsvm-functional-ipv6/52c3e2f/ : SUCCESS in 26m 12s\n- swift-tox-lower-constraints http://logs.openstack.org/71/654671/4/check/swift-tox-lower-constraints/02b0dfb/ : SUCCESS in 12m 18s\n- openstack-tox-pep8 http://logs.openstack.org/71/654671/4/check/openstack-tox-pep8/544fff4/ : SUCCESS in 5m 59s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/71/654671/4/check/swift-multinode-rolling-upgrade/eea579e/ : FAILURE in 9m 01s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"efd359fb0b84ffff2020c53cb24e79a7c668da89","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 22:24:17.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"f7936484ed54d5c0d5e6a0ea97aff65625228b06","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 00:56:18.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/71/654671/4/gate/build-openstack-api-ref/83a2e99/html/ : SUCCESS in 3m 42s\n- openstack-tox-docs http://logs.openstack.org/71/654671/4/gate/openstack-tox-docs/66ea89e/html/ : SUCCESS in 4m 59s\n- build-openstack-releasenotes http://logs.openstack.org/71/654671/4/gate/build-openstack-releasenotes/738cc08/html/ : SUCCESS in 7m 23s\n- swift-upload-image http://logs.openstack.org/71/654671/4/gate/swift-upload-image/37f5aa9/ : SUCCESS in 9m 14s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/71/654671/4/gate/swift-tox-py27/7e638ec/ : SUCCESS in 33m 37s\n- swift-tox-py37 http://logs.openstack.org/71/654671/4/gate/swift-tox-py37/91d7eea/ : SUCCESS in 36m 10s\n- swift-tox-func http://logs.openstack.org/71/654671/4/gate/swift-tox-func/f1a02f8/ : SUCCESS in 8m 28s\n- swift-tox-func-py37 http://logs.openstack.org/71/654671/4/gate/swift-tox-func-py37/43b5a17/ : SUCCESS in 7m 14s\n- swift-tox-func-encryption http://logs.openstack.org/71/654671/4/gate/swift-tox-func-encryption/013f355/ : SUCCESS in 9m 01s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/71/654671/4/gate/swift-tox-func-domain-remap-staticweb/1a9936d/ : SUCCESS in 9m 44s\n- swift-tox-func-ec http://logs.openstack.org/71/654671/4/gate/swift-tox-func-ec/8af5013/ : SUCCESS in 9m 09s\n- swift-tox-func-s3api http://logs.openstack.org/71/654671/4/gate/swift-tox-func-s3api/62b510a/ : SUCCESS in 13m 02s\n- swift-probetests-centos-7 http://logs.openstack.org/71/654671/4/gate/swift-probetests-centos-7/46c1fc7/ : SUCCESS in 31m 53s\n- swift-dsvm-functional http://logs.openstack.org/71/654671/4/gate/swift-dsvm-functional/3990536/ : SUCCESS in 27m 47s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/71/654671/4/gate/swift-dsvm-functional-ipv6/1fc0c42/ : SUCCESS in 29m 47s\n- swift-tox-lower-constraints http://logs.openstack.org/71/654671/4/gate/swift-tox-lower-constraints/f9b397e/ : SUCCESS in 34m 37s\n- openstack-tox-pep8 http://logs.openstack.org/71/654671/4/gate/openstack-tox-pep8/2c6e7c9/ : SUCCESS in 6m 08s","accounts_in_message":[],"_revision_number":4},{"id":"747eb125c09fdc086717cf072a5ef6508f7e8ff3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 00:56:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"dd97402bce29cbbb0403dfd7e29a62063225a39e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 00:57:39.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- swift-promote-image http://logs.openstack.org/71/654671/4/promote/swift-promote-image/2850261/ : SUCCESS in 57s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"4f9595f1130f6c45f8e3e08db55dddc31844be93","revisions":{"44ccd537c5e249538ec1951c2b5dbf0ed20c9058":{"kind":"REWORK","_number":1,"created":"2019-04-23 02:18:03.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/71/654671/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/654671/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/654671/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/654671/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/654671/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/654671/1"}}},"commit":{"parents":[{"commit":"89eced960c5bf5c2e14b6245c70b615dc23d45a6","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/89eced960c5bf5c2e14b6245c70b615dc23d45a6"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2019-04-23 02:15:32.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2019-04-23 02:15:32.000000000","tz":600},"subject":"py3: port symlink middleware","message":"py3: port symlink middleware\n\nThis patch ports the symlink middleware to py3.\nThe middlewre inself seemsa to be fine and all changes\nare in the symlink unit tests.\n\nChange-Id: I973c2e1bb8969cf6bffece8ce68881c393efbaef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/44ccd537c5e249538ec1951c2b5dbf0ed20c9058"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/44ccd537c5e249538ec1951c2b5dbf0ed20c9058"}]},"branch":"refs/heads/master"},"20a080e925ed402ecccb09b9e8d995d4bca2d003":{"kind":"REWORK","_number":2,"created":"2019-04-23 05:55:36.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/71/654671/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/654671/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/654671/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/654671/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/654671/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/654671/2"}}},"commit":{"parents":[{"commit":"89eced960c5bf5c2e14b6245c70b615dc23d45a6","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/89eced960c5bf5c2e14b6245c70b615dc23d45a6"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2019-04-23 02:15:32.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2019-04-23 05:55:14.000000000","tz":600},"subject":"py3: port symlink middleware","message":"py3: port symlink middleware\n\nThis patch ports the symlink middleware to py3.\nThe middlewre inself seemsa to be fine and all changes\nare in the symlink unit tests.\n\nChange-Id: I973c2e1bb8969cf6bffece8ce68881c393efbaef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/20a080e925ed402ecccb09b9e8d995d4bca2d003"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/20a080e925ed402ecccb09b9e8d995d4bca2d003"}]},"branch":"refs/heads/master"},"af38a489a2b9a514876c5ba1bfbc21f915ddeeb9":{"kind":"REWORK","_number":3,"created":"2019-05-20 12:17:59.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/71/654671/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/654671/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/654671/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/654671/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/654671/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/654671/3"}}},"commit":{"parents":[{"commit":"89eced960c5bf5c2e14b6245c70b615dc23d45a6","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/89eced960c5bf5c2e14b6245c70b615dc23d45a6"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2019-04-23 02:15:32.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2019-05-20 12:17:31.000000000","tz":600},"subject":"py3: port symlink middleware","message":"py3: port symlink middleware\n\nThis patch ports the symlink middleware to py3.\nThe middlewre inself seemsa to be fine and all changes\nare in the symlink unit tests.\n\nChange-Id: I973c2e1bb8969cf6bffece8ce68881c393efbaef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af38a489a2b9a514876c5ba1bfbc21f915ddeeb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af38a489a2b9a514876c5ba1bfbc21f915ddeeb9"}]},"branch":"refs/heads/master"},"4f9595f1130f6c45f8e3e08db55dddc31844be93":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":4,"created":"2019-05-30 18:56:32.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/71/654671/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/654671/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/654671/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/654671/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/654671/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/654671/4"}}},"commit":{"parents":[{"commit":"2be7dcd850be9ec8684d3094fcad92ec8cbbd748","subject":"Merge \"Modify the version path URL of the configuration file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2be7dcd850be9ec8684d3094fcad92ec8cbbd748"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2019-04-23 02:15:32.000000000","tz":600},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-05-30 18:44:42.000000000","tz":-420},"subject":"py3: port symlink middleware","message":"py3: port symlink middleware\n\nThis patch ports the symlink middleware to py3.\nThe middleware itself seems to be mostly fine and most changes\nare in the symlink unit tests.\n\nChange-Id: I973c2e1bb8969cf6bffece8ce68881c393efbaef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4f9595f1130f6c45f8e3e08db55dddc31844be93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4f9595f1130f6c45f8e3e08db55dddc31844be93"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
