)]}'
{"id":"openstack%2Fswift~655630","triplet_id":"openstack%2Fswift~feature%2Flosf~I8c988890173f8e038f3285ac63537f5bf8698937","project":"openstack/swift","branch":"feature/losf","topic":"bug/1774238","hashtags":[],"change_id":"I8c988890173f8e038f3285ac63537f5bf8698937","subject":"Merge remote-tracking branch \u0027remotes/origin/master\u0027 into merge-master","status":"MERGED","created":"2019-04-25 08:22:50.000000000","updated":"2019-04-29 05:37:56.000000000","submitted":"2019-04-29 05:37:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"655630-1556516276138-1b918c92","meta_rev_id":"6d6d10acf369556d37812949ce281f56c2ae4c29","_number":655630,"virtual_id_number":655630,"owner":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-04-26 03:12:12.000000000","_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"value":2,"date":"2019-04-29 05:37:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-04-25 16:48:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-04-25 16:48:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-25 16:48:19.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2019-04-29 05:37:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dd9221d6d668107f4da3c5659dc41dff4a2516dc","author":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"date":"2019-04-25 08:22:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"271bba1fa3f1d1d309ce01277ca7b4300964990a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 12:55:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/30/655630/1/check/openstack-tox-docs/2428338/html/ : SUCCESS in 4m 23s\n- swift-tox-py27 http://logs.openstack.org/30/655630/1/check/swift-tox-py27/e9efcc7/ : SUCCESS in 18m 33s\n- swift-tox-py35 http://logs.openstack.org/30/655630/1/check/swift-tox-py35/c4e21f3/ : SUCCESS in 13m 00s (non-voting)\n- swift-tox-py36 http://logs.openstack.org/30/655630/1/check/swift-tox-py36/8e47d0a/ : SUCCESS in 7m 42s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/30/655630/1/check/swift-tox-py37/5898383/ : SUCCESS in 9m 53s\n- swift-tox-func http://logs.openstack.org/30/655630/1/check/swift-tox-func/b5bbaac/ : SUCCESS in 8m 42s\n- swift-tox-func-encryption http://logs.openstack.org/30/655630/1/check/swift-tox-func-encryption/dda6289/ : SUCCESS in 9m 07s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/30/655630/1/check/swift-tox-func-domain-remap-staticweb/1a5ceb7/ : SUCCESS in 8m 42s\n- swift-tox-func-ec http://logs.openstack.org/30/655630/1/check/swift-tox-func-ec/9412fbc/ : SUCCESS in 11m 25s\n- swift-tox-func-s3api http://logs.openstack.org/30/655630/1/check/swift-tox-func-s3api/7a81e38/ : SUCCESS in 10m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/30/655630/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/41ab9de/ : SUCCESS in 12m 14s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/30/655630/1/check/swift-probetests-centos-7/75a2dff/ : SUCCESS in 32m 18s\n- swift-dsvm-functional http://logs.openstack.org/30/655630/1/check/swift-dsvm-functional/a865d07/ : SUCCESS in 26m 55s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/30/655630/1/check/swift-dsvm-functional-ipv6/942e5c0/ : SUCCESS in 26m 03s\n- swift-tox-lower-constraints http://logs.openstack.org/30/655630/1/check/swift-tox-lower-constraints/04d6f1f/ : SUCCESS in 12m 31s\n- openstack-tox-pep8 http://logs.openstack.org/30/655630/1/check/openstack-tox-pep8/d4c4434/ : SUCCESS in 4m 56s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/30/655630/1/check/swift-multinode-rolling-upgrade/3ec36a7/ : SUCCESS in 15m 27s","accounts_in_message":[],"_revision_number":1},{"id":"faeaa0c2c7e6b3ed1dcc1ee4e5f847be2fb72854","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-04-25 16:48:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"621474fa494066f885df5a9ff673c3493b399541","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 16:48:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"926b0fd3a6ade53f197cfd3b205483a7a7986d2e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 02:33:44.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/30/655630/1/gate/openstack-tox-docs/b5c4b29/html/ : SUCCESS in 4m 16s\n- swift-tox-py27 http://logs.openstack.org/30/655630/1/gate/swift-tox-py27/cd3879a/ : SUCCESS in 13m 04s\n- swift-tox-py37 http://logs.openstack.org/30/655630/1/gate/swift-tox-py37/36d4a56/ : SUCCESS in 8m 31s\n- swift-tox-func http://logs.openstack.org/30/655630/1/gate/swift-tox-func/19fa956/ : SUCCESS in 7m 49s\n- swift-tox-func-encryption http://logs.openstack.org/30/655630/1/gate/swift-tox-func-encryption/0849460/ : SUCCESS in 8m 44s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/30/655630/1/gate/swift-tox-func-domain-remap-staticweb/7baba66/ : SUCCESS in 8m 47s\n- swift-tox-func-ec http://logs.openstack.org/30/655630/1/gate/swift-tox-func-ec/e79eb9f/ : SUCCESS in 8m 59s\n- swift-tox-func-s3api http://logs.openstack.org/30/655630/1/gate/swift-tox-func-s3api/415f8fb/ : SUCCESS in 9m 52s\n- swift-probetests-centos-7 http://logs.openstack.org/30/655630/1/gate/swift-probetests-centos-7/9a1d310/ : FAILURE in 29m 56s\n- swift-dsvm-functional http://logs.openstack.org/30/655630/1/gate/swift-dsvm-functional/6858404/ : SUCCESS in 24m 45s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/30/655630/1/gate/swift-dsvm-functional-ipv6/9c53a6c/ : SUCCESS in 25m 52s\n- swift-tox-lower-constraints http://logs.openstack.org/30/655630/1/gate/swift-tox-lower-constraints/270b9cb/ : FAILURE in 11m 50s\n- openstack-tox-pep8 http://logs.openstack.org/30/655630/1/gate/openstack-tox-pep8/dba8db1/ : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":1},{"id":"a1f33aa84a9c9ca80fed6ccef9de266d4d4ab64d","author":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"date":"2019-04-26 03:12:12.000000000","message":"Patch Set 1:\n\nreverify","accounts_in_message":[],"_revision_number":1},{"id":"81eb9cc028227ba562616c391e38ca637b9cd759","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-04-29 04:07:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"aef5465a6769ffe6467f6478c051927338447f10","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 04:59:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/30/655630/1/check/openstack-tox-docs/4a26e2d/html/ : SUCCESS in 3m 24s\n- swift-tox-py27 http://logs.openstack.org/30/655630/1/check/swift-tox-py27/6333a50/ : SUCCESS in 10m 47s\n- swift-tox-py35 http://logs.openstack.org/30/655630/1/check/swift-tox-py35/4852632/ : SUCCESS in 9m 31s (non-voting)\n- swift-tox-py36 http://logs.openstack.org/30/655630/1/check/swift-tox-py36/8a02920/ : SUCCESS in 9m 25s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/30/655630/1/check/swift-tox-py37/b3fee8a/ : SUCCESS in 12m 10s\n- swift-tox-func http://logs.openstack.org/30/655630/1/check/swift-tox-func/da756f3/ : SUCCESS in 8m 44s\n- swift-tox-func-encryption http://logs.openstack.org/30/655630/1/check/swift-tox-func-encryption/892c834/ : SUCCESS in 17m 06s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/30/655630/1/check/swift-tox-func-domain-remap-staticweb/e25f067/ : SUCCESS in 10m 25s\n- swift-tox-func-ec http://logs.openstack.org/30/655630/1/check/swift-tox-func-ec/aef5aba/ : SUCCESS in 8m 45s\n- swift-tox-func-s3api http://logs.openstack.org/30/655630/1/check/swift-tox-func-s3api/22381cf/ : SUCCESS in 9m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/30/655630/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/3b58024/ : SUCCESS in 13m 00s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/30/655630/1/check/swift-probetests-centos-7/d5d1b76/ : SUCCESS in 30m 20s\n- swift-dsvm-functional http://logs.openstack.org/30/655630/1/check/swift-dsvm-functional/8caec4d/ : SUCCESS in 22m 34s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/30/655630/1/check/swift-dsvm-functional-ipv6/4ab44a4/ : SUCCESS in 25m 15s\n- swift-tox-lower-constraints http://logs.openstack.org/30/655630/1/check/swift-tox-lower-constraints/f649e4d/ : SUCCESS in 11m 05s\n- openstack-tox-pep8 http://logs.openstack.org/30/655630/1/check/openstack-tox-pep8/54c0e1a/ : SUCCESS in 4m 57s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/30/655630/1/check/swift-multinode-rolling-upgrade/08bc1e4/ : SUCCESS in 17m 46s","accounts_in_message":[],"_revision_number":1},{"id":"beb612f3a1e42724cc7472b35efe060139e4f065","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 05:00:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a74257f3d1824b14428c49bbf8283c85c49ed457","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 05:37:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/30/655630/1/gate/openstack-tox-docs/927d700/html/ : SUCCESS in 3m 35s\n- swift-tox-py27 http://logs.openstack.org/30/655630/1/gate/swift-tox-py27/2fa9127/ : SUCCESS in 12m 39s\n- swift-tox-py37 http://logs.openstack.org/30/655630/1/gate/swift-tox-py37/b00c180/ : SUCCESS in 8m 07s\n- swift-tox-func http://logs.openstack.org/30/655630/1/gate/swift-tox-func/a993f94/ : SUCCESS in 8m 42s\n- swift-tox-func-encryption http://logs.openstack.org/30/655630/1/gate/swift-tox-func-encryption/2f15d23/ : SUCCESS in 8m 54s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/30/655630/1/gate/swift-tox-func-domain-remap-staticweb/1e1359b/ : SUCCESS in 7m 49s\n- swift-tox-func-ec http://logs.openstack.org/30/655630/1/gate/swift-tox-func-ec/dcaca00/ : SUCCESS in 8m 31s\n- swift-tox-func-s3api http://logs.openstack.org/30/655630/1/gate/swift-tox-func-s3api/26ba098/ : SUCCESS in 9m 24s\n- swift-probetests-centos-7 http://logs.openstack.org/30/655630/1/gate/swift-probetests-centos-7/271fc83/ : SUCCESS in 32m 28s\n- swift-dsvm-functional http://logs.openstack.org/30/655630/1/gate/swift-dsvm-functional/3b2804b/ : SUCCESS in 25m 40s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/30/655630/1/gate/swift-dsvm-functional-ipv6/ba2c499/ : SUCCESS in 28m 44s\n- swift-tox-lower-constraints http://logs.openstack.org/30/655630/1/gate/swift-tox-lower-constraints/8120d52/ : SUCCESS in 11m 24s\n- openstack-tox-pep8 http://logs.openstack.org/30/655630/1/gate/openstack-tox-pep8/63b99fd/ : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":1},{"id":"2c79f42be47da634e2f1b845621801f011bccebf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 05:37:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d391957fd70a02866615c9c5edc070dbde2b23ea","revisions":{"d391957fd70a02866615c9c5edc070dbde2b23ea":{"kind":"REWORK","_number":1,"created":"2019-04-25 08:22:50.000000000","uploader":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"ref":"refs/changes/30/655630/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/30/655630/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/655630/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/655630/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/655630/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/30/655630/1"}}},"commit":{"parents":[{"commit":"648cf7a9aa9ff8948c7321020fac74ad4c916d92","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/648cf7a9aa9ff8948c7321020fac74ad4c916d92"}]},{"commit":"32bf43990c3c9ce771fca5968242f102155467c1","subject":"Replace git.openstack.org URLs with opendev.org URLs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/32bf43990c3c9ce771fca5968242f102155467c1"}]}],"author":{"name":"Kota Tsuyuzaki","email":"tsuyuzaki.kota@lab.ntt.co.jp","date":"2019-04-25 08:18:43.000000000","tz":540},"committer":{"name":"Kota Tsuyuzaki","email":"tsuyuzaki.kota@lab.ntt.co.jp","date":"2019-04-25 08:18:43.000000000","tz":540},"subject":"Merge remote-tracking branch \u0027remotes/origin/master\u0027 into merge-master","message":"Merge remote-tracking branch \u0027remotes/origin/master\u0027 into merge-master\n\nConflicts resolved:\n\t.zuul.yaml\n\trequirements.txt\n\tswift/obj/replicator.py\n\nChange-Id: I8c988890173f8e038f3285ac63537f5bf8698937\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d391957fd70a02866615c9c5edc070dbde2b23ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d391957fd70a02866615c9c5edc070dbde2b23ea"}]},"branch":"refs/heads/feature/losf"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
