)]}'
{"id":"openstack%2Fswift~656812","triplet_id":"openstack%2Fswift~master~Icade637b0d384dcc87aaa31da3861cd2646d4ccd","project":"openstack/swift","branch":"master","topic":"bug/1827394","hashtags":[],"change_id":"Icade637b0d384dcc87aaa31da3861cd2646d4ccd","subject":"enforce account quotas on empty accounts","status":"MERGED","created":"2019-05-02 16:56:30.000000000","updated":"2020-02-01 02:24:30.000000000","submitted":"2020-02-01 02:21:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"656812-1580523712316-87017a1f","meta_rev_id":"f3d1fecb4c1e3ee33e94a2b3dd9f780058720d65","_number":656812,"virtual_id_number":656812,"owner":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-01 02:21:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-31 17:41:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-31 17:41:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-31 17:41:19.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2020-02-01 02:21:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8c753deb6a86a8780a5e0f5316a6bc568215791d","author":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"date":"2019-05-02 16:56:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a9ab8767ff7dc762516d0f172076d0cbc5fc989f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-02 18:48:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/12/656812/1/check/openstack-tox-docs/9ff757e/html/ : SUCCESS in 5m 01s\n- swift-tox-py27 http://logs.openstack.org/12/656812/1/check/swift-tox-py27/edfb0b0/ : SUCCESS in 10m 48s\n- swift-tox-py35 http://logs.openstack.org/12/656812/1/check/swift-tox-py35/dc4d9fe/ : SUCCESS in 8m 03s (non-voting)\n- swift-tox-py36 http://logs.openstack.org/12/656812/1/check/swift-tox-py36/53a9f89/ : SUCCESS in 9m 59s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/12/656812/1/check/swift-tox-py37/d959534/ : SUCCESS in 9m 49s\n- swift-tox-func http://logs.openstack.org/12/656812/1/check/swift-tox-func/7b92c98/ : SUCCESS in 7m 51s\n- swift-tox-func-encryption http://logs.openstack.org/12/656812/1/check/swift-tox-func-encryption/15ff4f8/ : SUCCESS in 8m 22s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/12/656812/1/check/swift-tox-func-domain-remap-staticweb/ad808d4/ : SUCCESS in 10m 21s\n- swift-tox-func-ec http://logs.openstack.org/12/656812/1/check/swift-tox-func-ec/38cc5ed/ : SUCCESS in 10m 52s\n- swift-tox-func-s3api http://logs.openstack.org/12/656812/1/check/swift-tox-func-s3api/52c55b1/ : SUCCESS in 12m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/12/656812/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/a468649/ : SUCCESS in 11m 10s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/12/656812/1/check/swift-probetests-centos-7/3f9a90b/ : SUCCESS in 33m 36s\n- swift-dsvm-functional http://logs.openstack.org/12/656812/1/check/swift-dsvm-functional/df66a4d/ : SUCCESS in 27m 25s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/12/656812/1/check/swift-dsvm-functional-ipv6/e213d95/ : SUCCESS in 26m 31s\n- swift-tox-lower-constraints http://logs.openstack.org/12/656812/1/check/swift-tox-lower-constraints/1f013ec/ : SUCCESS in 13m 51s\n- openstack-tox-pep8 http://logs.openstack.org/12/656812/1/check/openstack-tox-pep8/2410dc0/ : SUCCESS in 5m 27s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/12/656812/1/check/swift-multinode-rolling-upgrade/b3a291a/ : SUCCESS in 15m 00s","accounts_in_message":[],"_revision_number":1},{"id":"53e2ffabf3a4b9df074b81d825e6726376a7f3b4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-02 19:25:19.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"a084927def17050f3c45b27f6b6140777b0dc8f1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-31 17:40:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c44095f040f3a6c71a3b466a57596a1369235a83","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-31 17:41:19.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nComparing against container quotas, this all seems very reasonable. I think my prior concern was really an artifact of our general quota system.","accounts_in_message":[],"_revision_number":2},{"id":"7c62bc2e6940155ab6e8b12a5f0f08af8a57a1ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 19:04:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f38456da90e44aafa5cc7caf9b3465d0 : SUCCESS in 14m 13s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/bcad8c69560a48aa953eef0f3e183af5 : FAILURE in 1h 20m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9dbcdca949044fba888066b002036ca4 : SUCCESS in 1h 04m 10s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/532457824fbe4094ac7ae834c867ed79 : SUCCESS in 13m 17s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/febe6965ce5a431bacfbe006c27543cb : SUCCESS in 10m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e3561d1733844fab83d77cca30c31ff5 : SUCCESS in 23m 35s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e90872e8e8ae4927ac02c634f8e972df : SUCCESS in 24m 45s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/55a730f0f5ab40e39400b7314662d7fd : SUCCESS in 25m 27s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/00916e4455a34ba18be8cba3c847e954 : SUCCESS in 21m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/61b09c95ed8940ab8b1e3f04529ed523 : SUCCESS in 16m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8955975c1245422099bc14bad69bca14 : SUCCESS in 16m 55s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/9fd76bebedcb4a87be1f4dd9edf8045b : SUCCESS in 18m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f40b866f1dd94ea98d83b2ff4a95043e : SUCCESS in 18m 28s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/3df30ccbaf3046e8ad98bc802c53c183 : SUCCESS in 33m 17s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/9579c4df22cf46448c719ae5a1330937 : SUCCESS in 25m 53s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/6575e109b6d54ab496edda54243c5aa8 : SUCCESS in 12m 31s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/28c4d67f86974fbc9ed133407126fc9a : SUCCESS in 17m 08s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/746010ee835a4b519746dad9f0b80f5c : SUCCESS in 13m 17s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/d8b1d627df2c4605afc465042b115a4e : SUCCESS in 23m 25s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8378366d71684b0fbb6bea3fb2b02cf1 : SUCCESS in 21m 58s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/eb39013fcf7f4d8ca09c3b010a889bcc : SUCCESS in 54m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/149fb45ca8124a8387f4da5413a8aa09 : SUCCESS in 37m 08s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/f3617f13cbc7498c835ff79d0c2ee521 : SUCCESS in 34m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3b574e17d22d48d4ae90dd75749a61ba : SUCCESS in 30m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3fa0756d9e4640a982cf39a9fb4f12a5 : SUCCESS in 17m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/527069e5bbc346b5a8d3c7862db451f7 : SUCCESS in 7m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/08a911e3a9c34eb68b66546d8d43d26b : SUCCESS in 24m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/db430e1ab55b4393a0c9192a58613e3c : SUCCESS in 1h 13m 38s","accounts_in_message":[],"_revision_number":2},{"id":"07d20a710841aaee2261416515df90c8ee835cca","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-31 20:11:28.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a8b4fc437cc8f5412f552e80634a7f15a1ea851d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 21:26:11.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cde2e085b544e8396a5bb327f667df5 : SUCCESS in 9m 45s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/77f9147ba8b64627907973924c2d3435 : SUCCESS in 1h 11m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f570da150bee45829915882d916f0207 : SUCCESS in 1h 01m 35s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/86bc5a87f8a240d8948ea786f6f70ee7 : SUCCESS in 9m 35s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/495557240eab42b3aac33e88599de74c : SUCCESS in 10m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0945fa1e7a7f45ac9fd6442800a3213e : SUCCESS in 13m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9abd0eeca85f48b69ecc19d381fba3d5 : SUCCESS in 12m 14s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/fa09cb49ffb64d27b8bc8a0560dfcc9f : SUCCESS in 14m 57s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/bb21b004fd2f4c5aaf0785ede7e858b5 : SUCCESS in 12m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2ef1abcf864f4ee88d0598eba184fd21 : SUCCESS in 10m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c0aabc47de454d1ba2a59b564365b763 : SUCCESS in 15m 38s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/d1cd23f5bfb541bd9578a9f7b7f6d684 : SUCCESS in 11m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7b7e6b62c2c24e1d954039ee18284d5f : SUCCESS in 10m 37s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/ed815a174e1149948ad429c5190c1985 : SUCCESS in 15m 56s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/9e7df1d24a5941ef9b777ad221a4382c : SUCCESS in 11m 35s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/cd95db80392144a294c381f0273df9d1 : SUCCESS in 12m 40s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/c7c6c979a56246c0b396ba0702c93a44 : SUCCESS in 10m 46s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/e76d2045ebc54b5d8b194b993f485e20 : SUCCESS in 11m 33s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/f523bf451f7f4d31b43ea5c639553000 : SUCCESS in 16m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e684ab40d265405ba1650857587fff7e : SUCCESS in 18m 20s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6127d82c47f644c88be22cf8f55641ca : SUCCESS in 48m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9d0c4fd73ecd40ad80d04cdba9fce3c8 : SUCCESS in 49m 01s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/0c6b9de303924a98b5b5a11131ed1079 : TIMED_OUT in 51m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e86ef6cefb2e46cab60ce00497b01ce2 : SUCCESS in 32m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ae96df5fb6f643349a560e4452982f34 : SUCCESS in 21m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8d1422a24ec4cba93551cdaff05e5dd : SUCCESS in 9m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/280c180318c242769e05c8d8f5f2c474 : SUCCESS in 27m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b894bdddfc054b3f9697198dc90ba670 : SUCCESS in 1h 11m 33s","accounts_in_message":[],"_revision_number":2},{"id":"0060dcef96bdeac48b3ff122f83fc858f209bed1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-31 21:40:04.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"89f24c50e1f92d502f562156ef4b9ea5b9f524c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 22:44:42.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56b802676bf04b2984f28d685ff6d7bd : SUCCESS in 10m 10s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f6a50075d410422499e11b82cfd8dfb5 : SUCCESS in 58m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/148b2ecb65dc467eb413d118b5276966 : SUCCESS in 54m 44s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/71ec5f1d2ad641be9a8ccc43b57a324d : SUCCESS in 8m 16s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3f852f5f12c24ef78613e3fba6bc5d4e : SUCCESS in 9m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6eed294fd5a84618a981b561eb8703e3 : SUCCESS in 12m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/982c67c6412d4294b707f67be2515097 : SUCCESS in 14m 00s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c00bbc785b75485095aa9bb5a599c784 : SUCCESS in 14m 03s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/91019b153e7c46f993b49146bc79de3c : SUCCESS in 13m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b1e70708fcf3453abc191904c0a43308 : SUCCESS in 12m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e9afae3c6a6e41ad8e3d745ccee9c4b9 : SUCCESS in 13m 04s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/bf2f4910205b40cf87b30023905ca7ae : SUCCESS in 9m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bfc281d87bd145499066e4a91807808b : SUCCESS in 14m 11s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/0479a249af6e4cbca855d6fa8c2a17fc : SUCCESS in 18m 23s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/ede28edea01a44478ddd347fe5bc4520 : SUCCESS in 9m 50s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/c01105f57afd4aad96c9e1d41e9cb07e : SUCCESS in 27m 38s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/95a46c1e9e4342fa8e76b9aa232b8399 : SUCCESS in 9m 32s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/9ae06594ca09433a90cf40ade3fe1c8a : SUCCESS in 18m 24s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/00e9b06dce1947fabf9d7bb3c1c32c10 : SUCCESS in 25m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e495cbe0ddbd474e82f449bf34701d8f : SUCCESS in 17m 47s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/80e39e9e036c4fb1af90fa652633c09b : FAILURE in 53m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/95964a0ae7414cf08093963be4148b7c : SUCCESS in 31m 47s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/0266da081d9f452eb6243c11155b2d13 : SUCCESS in 43m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/20847caf852243e798d46a978b3a6082 : SUCCESS in 42m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7340f97402394dad82dfc0e97ea28645 : SUCCESS in 11m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8f74558c86243c99998c3ed34b375c6 : SUCCESS in 6m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8d12326124c348318f6a924bd54c1032 : SUCCESS in 33m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5555d438303f4edba1f64c55423dd6ad : SUCCESS in 58m 37s","accounts_in_message":[],"_revision_number":2},{"id":"c3c39927111d28cb183df803db6a8d90625abe04","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-31 23:00:40.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"decc14c6eb3cc972c0af6e1887481aca25e6cc15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-01 00:17:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/538ec8771f99413c8d916a4053317b14 : SUCCESS in 7m 54s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9e8a4f297a1741b3ab960b9d805ec8f8 : SUCCESS in 1h 01m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4ebaa7bfb2b84e6f89cbdf2ede1f0c62 : SUCCESS in 45m 48s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b0996699f77441d3805a752922aea41e : SUCCESS in 16m 41s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2d08c36084d84d30baf7f3836859efbb : SUCCESS in 8m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b8e55c634a814830b90fd1df4d42ba76 : SUCCESS in 14m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b7852feb11b14ce2b96976488482424e : SUCCESS in 16m 18s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/eba244eae3b049b0a7af0bec8578d080 : SUCCESS in 15m 38s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/466abf69d0de4a8481f527bf52d4ed2a : SUCCESS in 12m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2ca09cc44a45420cbeb517b3c70993a2 : SUCCESS in 14m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d777d916e17a4e60a76170b03c9ded3c : SUCCESS in 16m 05s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/f3fee08d23bf47739b6e45fefa0899cb : SUCCESS in 10m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/78fb0178c80c468f884986341bf6a5c7 : SUCCESS in 10m 22s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/6565925ebf9644a0826e98878e352595 : SUCCESS in 19m 30s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/395f979124b04544a3f50bb2eb297794 : SUCCESS in 14m 49s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/88304603eeda4be08f165b37a2131f06 : SUCCESS in 14m 37s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/9f132c7d449549bba24a3c2b4db35b2f : SUCCESS in 13m 56s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/d0170932c03a44229bebf91c6e022537 : SUCCESS in 11m 20s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/8eb4a133451e4c4aade9d18cdc6cc9f0 : SUCCESS in 20m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/270cad0da19f48ed9a384f8267c45c13 : SUCCESS in 24m 56s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7fb0804d885f4772b9ecbcecca80d6ee : SUCCESS in 46m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/62eaecb72111480bbc33dcf719baf7a9 : SUCCESS in 31m 08s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/e5ab37d3e6254a95939875bd806b724d : SUCCESS in 32m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a8291a16eb1a4668ae5649ab59ee71c6 : SUCCESS in 35m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3aa05ed93a5b43c0b65077fdb8090e9a : SUCCESS in 14m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38d574607f8849f0bf6871d406b52bf9 : SUCCESS in 6m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c9eb2cb85d034927a2592c89c9503a64 : SUCCESS in 18m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e48cad73f9ec4fe38538c041e1e7e98f : SUCCESS in 1h 09m 52s","accounts_in_message":[],"_revision_number":2},{"id":"6c47c0543564a98acf62ee78e0e0e7134ac01014","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-01 00:17:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ec67b5a88a3faf7ecce900b525d893578a904674","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-01 02:21:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a62e39bf400a4c51bab8adde8736aabf : SUCCESS in 12m 31s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/77cfc70c4c4546ca8cd783d80545a2e9 : SUCCESS in 56m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/60bb3b68543c44a79022bd2deb0cc07b : SUCCESS in 1h 00m 01s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/1cbf90607e6146508e06f1f906f6a447 : SUCCESS in 11m 23s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/c011449464544d0bac1a9dea203f7432 : SUCCESS in 14m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8a7794754862445ba4ab23fbba2f4bbe : SUCCESS in 18m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/755e485a08604e71815476317baf4d97 : SUCCESS in 17m 08s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a4ecc4d1f1a5438397a7b0f6d464a58c : SUCCESS in 17m 12s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/92bd5442307846598d058a22a61196c7 : SUCCESS in 15m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8323b142065b449d82d112e29a36451c : SUCCESS in 14m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/17520851749b4df0a50e6186bb30dd87 : SUCCESS in 12m 47s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/d87b1059deca40f1abb18ea433f0d44c : SUCCESS in 12m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e339cd9fc3724d548f80b5ba50b4c5f5 : SUCCESS in 17m 48s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/de955da0bfaf4e02bbecb21cedb95f80 : SUCCESS in 18m 21s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/fbb76fa033c44eafbd6419f90447ca42 : SUCCESS in 14m 01s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/52e99995f0a648e3be8f472c90517675 : SUCCESS in 13m 31s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/d16647a100b54338bbb459528ca4fbf8 : SUCCESS in 14m 51s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/b2d474cec16e48758eeb45643eea8b4b : SUCCESS in 19m 46s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0382e0113b174b39ac624ed964c86e8f : SUCCESS in 44m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c81b80109995451cba288e958102497a : SUCCESS in 28m 44s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/fac0d4e447bb4a138f12c25b2c4ad0d6 : SUCCESS in 40m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/91ab1756c2124692ab38c8dab11fd186 : SUCCESS in 29m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/20921558b7374397bf1c4ab22dfd88cf : SUCCESS in 20m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07349c8bfd794a42b0ae79776be6afb7 : SUCCESS in 8m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5b05fb5f289b4e9ab285ccec2c448a17 : SUCCESS in 1h 07m 53s","accounts_in_message":[],"_revision_number":2},{"id":"b9fce0fce2c180331597e79f28163aafa12d9962","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-01 02:21:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"a65c40d06d96cfe17bee2b5216603bd404e4a65a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-01 02:24:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14c0b974d02b4ff9adf98bfe8a623c59 : SUCCESS in 1m 39s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/c1bae7d1c92e4f198d69a3c98e769b04 : SUCCESS in 48s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/c0ec61a6f3264e25bcc86593f2f7f629 : SUCCESS in 53s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0606f404fe92a3c56014feba55069a206ddb1cc8","revisions":{"e36b957237f14e051b6acaa96e8e342c8978bddf":{"kind":"REWORK","_number":1,"created":"2019-05-02 16:56:30.000000000","uploader":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"ref":"refs/changes/12/656812/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/656812/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/656812/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/656812/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/656812/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/656812/1"}}},"commit":{"parents":[{"commit":"f55167a735632cdc89fb6f83f97e4fe3ed09b7cc","subject":"Merge \"Increase node_timeout in gate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f55167a735632cdc89fb6f83f97e4fe3ed09b7cc"}]}],"author":{"name":"John Dickinson","email":"me@not.mn","date":"2019-05-02 16:53:44.000000000","tz":-360},"committer":{"name":"John Dickinson","email":"me@not.mn","date":"2019-05-02 16:56:04.000000000","tz":-360},"subject":"enforce account quotas on empty accounts","message":"enforce account quotas on empty accounts\n\nChange-Id: Icade637b0d384dcc87aaa31da3861cd2646d4ccd\nCloses-Bug: #1827394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e36b957237f14e051b6acaa96e8e342c8978bddf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e36b957237f14e051b6acaa96e8e342c8978bddf"}]},"branch":"refs/heads/master"},"0606f404fe92a3c56014feba55069a206ddb1cc8":{"kind":"REWORK","_number":2,"created":"2020-01-31 17:40:03.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/12/656812/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/656812/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/656812/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/656812/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/656812/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/656812/2"}}},"commit":{"parents":[{"commit":"5cd854b783ce80356ec62e709f0817cbe699c8cd","subject":"Authors/changelog for swift 2.24.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5cd854b783ce80356ec62e709f0817cbe699c8cd"}]}],"author":{"name":"John Dickinson","email":"me@not.mn","date":"2019-05-02 16:53:44.000000000","tz":-360},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-01-31 17:39:17.000000000","tz":-480},"subject":"enforce account quotas on empty accounts","message":"enforce account quotas on empty accounts\n\nChange-Id: Icade637b0d384dcc87aaa31da3861cd2646d4ccd\nCloses-Bug: #1827394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0606f404fe92a3c56014feba55069a206ddb1cc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0606f404fe92a3c56014feba55069a206ddb1cc8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
