)]}'
{"id":"openstack%2Fswift~657033","triplet_id":"openstack%2Fswift~master~Iadb3bed811907740f85bd9277ba09a4ca17fc025","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"Iadb3bed811907740f85bd9277ba09a4ca17fc025","subject":"Remove py35 unit test job","status":"MERGED","created":"2019-05-03 20:37:05.000000000","updated":"2019-05-04 22:38:47.000000000","submitted":"2019-05-04 22:38:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657033-1557009527207-7175b152","meta_rev_id":"a69f578d61a782c30ce0908b32b1c9aa2c686b7a","_number":657033,"virtual_id_number":657033,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-05-04 22:38:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-03 22:54:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-03 22:54:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-04 22:38:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0ffa4814bc8d4f4ce935c0d52d9bb463087bcb59","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-03 20:37:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"587b145af4516c335452ee4fc1014fdbace205d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 21:12:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/33/657033/1/check/openstack-tox-docs/5766691/html/ : SUCCESS in 4m 23s\n- build-openstack-releasenotes http://logs.openstack.org/33/657033/1/check/build-openstack-releasenotes/e57bc2d/html/ : SUCCESS in 8m 01s\n- swift-tox-py27 http://logs.openstack.org/33/657033/1/check/swift-tox-py27/6d2c64f/ : SUCCESS in 11m 01s\n- swift-tox-py36 http://logs.openstack.org/33/657033/1/check/swift-tox-py36/9338f69/ : SUCCESS in 9m 54s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/33/657033/1/check/swift-tox-py37/1fb5b20/ : SUCCESS in 10m 14s\n- swift-tox-func http://logs.openstack.org/33/657033/1/check/swift-tox-func/5771f65/ : SUCCESS in 8m 40s\n- swift-tox-func-encryption http://logs.openstack.org/33/657033/1/check/swift-tox-func-encryption/97ca0a2/ : SUCCESS in 9m 35s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/33/657033/1/check/swift-tox-func-domain-remap-staticweb/d157d05/ : SUCCESS in 10m 15s\n- swift-tox-func-ec http://logs.openstack.org/33/657033/1/check/swift-tox-func-ec/bb418bf/ : SUCCESS in 10m 08s\n- swift-tox-func-s3api http://logs.openstack.org/33/657033/1/check/swift-tox-func-s3api/5905f2a/ : SUCCESS in 12m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/33/657033/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/10020e4/ : SUCCESS in 15m 45s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/33/657033/1/check/swift-probetests-centos-7/0b5ca9a/ : FAILURE in 32m 54s\n- swift-dsvm-functional http://logs.openstack.org/33/657033/1/check/swift-dsvm-functional/c75cf97/ : SUCCESS in 28m 40s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/33/657033/1/check/swift-dsvm-functional-ipv6/276085d/ : SUCCESS in 26m 17s\n- swift-tox-lower-constraints http://logs.openstack.org/33/657033/1/check/swift-tox-lower-constraints/8917844/ : SUCCESS in 12m 01s\n- openstack-tox-pep8 http://logs.openstack.org/33/657033/1/check/openstack-tox-pep8/70856ea/ : SUCCESS in 4m 58s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/33/657033/1/check/swift-multinode-rolling-upgrade/be4b9c4/ : SUCCESS in 16m 40s","accounts_in_message":[],"_revision_number":1},{"id":"05072e7e70cd3b90131525fd379568689551e14d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-03 22:52:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"383cc4f068ef6752b655c2b8388b21dd0cebd51b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-03 22:54:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nPer discussion at the PTG, we\u0027ll no longer attempt to support py35. xenial still has two years of life in it, but we\u0027ll support it only via py27.","accounts_in_message":[],"_revision_number":1},{"id":"cd6a3e1a54f0132a67c3a6c667e058360fb86652","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 23:41:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/33/657033/1/check/openstack-tox-docs/8d4d25e/html/ : SUCCESS in 3m 56s\n- build-openstack-releasenotes http://logs.openstack.org/33/657033/1/check/build-openstack-releasenotes/cfae5af/html/ : SUCCESS in 6m 57s\n- swift-tox-py27 http://logs.openstack.org/33/657033/1/check/swift-tox-py27/62ac19b/ : SUCCESS in 10m 27s\n- swift-tox-py36 http://logs.openstack.org/33/657033/1/check/swift-tox-py36/a8c2fa9/ : SUCCESS in 7m 32s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/33/657033/1/check/swift-tox-py37/63449d0/ : SUCCESS in 8m 18s\n- swift-tox-func http://logs.openstack.org/33/657033/1/check/swift-tox-func/fd683c1/ : SUCCESS in 8m 51s\n- swift-tox-func-encryption http://logs.openstack.org/33/657033/1/check/swift-tox-func-encryption/25c8b77/ : SUCCESS in 8m 40s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/33/657033/1/check/swift-tox-func-domain-remap-staticweb/16fcc31/ : SUCCESS in 8m 26s\n- swift-tox-func-ec http://logs.openstack.org/33/657033/1/check/swift-tox-func-ec/16e8e9a/ : SUCCESS in 8m 33s\n- swift-tox-func-s3api http://logs.openstack.org/33/657033/1/check/swift-tox-func-s3api/e5fcfc6/ : SUCCESS in 9m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/33/657033/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/9f84747/ : SUCCESS in 11m 19s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/33/657033/1/check/swift-probetests-centos-7/807ef1c/ : FAILURE in 33m 05s\n- swift-dsvm-functional http://logs.openstack.org/33/657033/1/check/swift-dsvm-functional/a29923e/ : SUCCESS in 33m 06s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/33/657033/1/check/swift-dsvm-functional-ipv6/e978f85/ : TIMED_OUT in 47m 19s\n- swift-tox-lower-constraints http://logs.openstack.org/33/657033/1/check/swift-tox-lower-constraints/b1c91a6/ : SUCCESS in 11m 50s\n- openstack-tox-pep8 http://logs.openstack.org/33/657033/1/check/openstack-tox-pep8/b86ea38/ : SUCCESS in 5m 58s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/33/657033/1/check/swift-multinode-rolling-upgrade/6b63479/ : SUCCESS in 15m 39s","accounts_in_message":[],"_revision_number":1},{"id":"cabfe757ab34b21eac3aba1b62c6a9c781446b0c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-04 04:36:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"89dff1149d884037383c6c91704f985777551b83","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 05:11:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/33/657033/1/check/openstack-tox-docs/bc64b0b/html/ : SUCCESS in 3m 30s\n- build-openstack-releasenotes http://logs.openstack.org/33/657033/1/check/build-openstack-releasenotes/d69c26c/html/ : SUCCESS in 6m 07s\n- swift-tox-py27 http://logs.openstack.org/33/657033/1/check/swift-tox-py27/ab94b05/ : SUCCESS in 10m 14s\n- swift-tox-py36 http://logs.openstack.org/33/657033/1/check/swift-tox-py36/7de3de4/ : FAILURE in 9m 55s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/33/657033/1/check/swift-tox-py37/2380b02/ : SUCCESS in 7m 19s\n- swift-tox-func http://logs.openstack.org/33/657033/1/check/swift-tox-func/74692f8/ : FAILURE in 3m 02s\n- swift-tox-func-encryption http://logs.openstack.org/33/657033/1/check/swift-tox-func-encryption/0b1469e/ : FAILURE in 5m 21s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/33/657033/1/check/swift-tox-func-domain-remap-staticweb/374bff0/ : FAILURE in 3m 46s\n- swift-tox-func-ec http://logs.openstack.org/33/657033/1/check/swift-tox-func-ec/048f104/ : FAILURE in 4m 58s\n- swift-tox-func-s3api http://logs.openstack.org/33/657033/1/check/swift-tox-func-s3api/8b4407a/ : FAILURE in 3m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/33/657033/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/684716e/ : SUCCESS in 14m 21s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/33/657033/1/check/swift-probetests-centos-7/cd0895f/ : FAILURE in 30m 10s\n- swift-dsvm-functional http://logs.openstack.org/33/657033/1/check/swift-dsvm-functional/2cd1b02/ : FAILURE in 16m 32s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/33/657033/1/check/swift-dsvm-functional-ipv6/f72c6ed/ : FAILURE in 17m 10s\n- swift-tox-lower-constraints http://logs.openstack.org/33/657033/1/check/swift-tox-lower-constraints/3716b57/ : SUCCESS in 12m 55s\n- openstack-tox-pep8 http://logs.openstack.org/33/657033/1/check/openstack-tox-pep8/b25447d/ : SUCCESS in 5m 03s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/33/657033/1/check/swift-multinode-rolling-upgrade/f9ec240/ : FAILURE in 6m 57s","accounts_in_message":[],"_revision_number":1},{"id":"a10b0bbe1c73365fa2fb0d5afd82a34c30f24d97","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-04 15:53:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a249a749d18da0ef98ba0a5b4b02111efe6f7931","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 16:39:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/33/657033/1/check/openstack-tox-docs/16b5602/html/ : SUCCESS in 4m 19s\n- build-openstack-releasenotes http://logs.openstack.org/33/657033/1/check/build-openstack-releasenotes/d52e491/html/ : SUCCESS in 7m 37s\n- swift-tox-py27 http://logs.openstack.org/33/657033/1/check/swift-tox-py27/9fff696/ : SUCCESS in 10m 48s\n- swift-tox-py36 http://logs.openstack.org/33/657033/1/check/swift-tox-py36/30e5c38/ : SUCCESS in 7m 54s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/33/657033/1/check/swift-tox-py37/1c2b7d6/ : SUCCESS in 8m 19s\n- swift-tox-func http://logs.openstack.org/33/657033/1/check/swift-tox-func/8f23d28/ : SUCCESS in 9m 38s\n- swift-tox-func-encryption http://logs.openstack.org/33/657033/1/check/swift-tox-func-encryption/a5dd820/ : SUCCESS in 10m 03s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/33/657033/1/check/swift-tox-func-domain-remap-staticweb/a00d41d/ : SUCCESS in 8m 33s\n- swift-tox-func-ec http://logs.openstack.org/33/657033/1/check/swift-tox-func-ec/52859c7/ : SUCCESS in 9m 11s\n- swift-tox-func-s3api http://logs.openstack.org/33/657033/1/check/swift-tox-func-s3api/2eb7525/ : SUCCESS in 12m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/33/657033/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/767a86e/ : SUCCESS in 12m 44s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/33/657033/1/check/swift-probetests-centos-7/d3c521c/ : SUCCESS in 39m 22s\n- swift-dsvm-functional http://logs.openstack.org/33/657033/1/check/swift-dsvm-functional/a7a1a27/ : SUCCESS in 27m 16s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/33/657033/1/check/swift-dsvm-functional-ipv6/dada0e8/ : SUCCESS in 27m 24s\n- swift-tox-lower-constraints http://logs.openstack.org/33/657033/1/check/swift-tox-lower-constraints/b4f8d7f/ : SUCCESS in 11m 27s\n- openstack-tox-pep8 http://logs.openstack.org/33/657033/1/check/openstack-tox-pep8/6544ec1/ : SUCCESS in 5m 28s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/33/657033/1/check/swift-multinode-rolling-upgrade/151026c/ : FAILURE in 8m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c1491af7a11bfb8f124b971eafc7c9961b6b226a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 16:40:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"10eb388445b66cc2ff8145c509b58aaad2113696","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 18:35:52.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/33/657033/1/gate/openstack-tox-docs/0f2f45f/html/ : SUCCESS in 4m 34s\n- build-openstack-releasenotes http://logs.openstack.org/33/657033/1/gate/build-openstack-releasenotes/1135522/html/ : SUCCESS in 6m 11s\n- swift-tox-py27 http://logs.openstack.org/33/657033/1/gate/swift-tox-py27/2b9ead2/ : SUCCESS in 19m 50s\n- swift-tox-py37 http://logs.openstack.org/33/657033/1/gate/swift-tox-py37/764ff7b/ : SUCCESS in 10m 45s\n- swift-tox-func http://logs.openstack.org/33/657033/1/gate/swift-tox-func/25a507a/ : SUCCESS in 24m 42s\n- swift-tox-func-encryption http://logs.openstack.org/33/657033/1/gate/swift-tox-func-encryption/dceaad1/ : SUCCESS in 24m 49s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/33/657033/1/gate/swift-tox-func-domain-remap-staticweb/4ca204d/ : SUCCESS in 17m 31s\n- swift-tox-func-ec http://logs.openstack.org/33/657033/1/gate/swift-tox-func-ec/c2cd395/ : SUCCESS in 18m 11s\n- swift-tox-func-s3api http://logs.openstack.org/33/657033/1/gate/swift-tox-func-s3api/342ed8b/ : SUCCESS in 24m 26s\n- swift-probetests-centos-7 http://logs.openstack.org/33/657033/1/gate/swift-probetests-centos-7/6304e7e/ : TIMED_OUT in 1h 00m 54s\n- swift-dsvm-functional http://logs.openstack.org/33/657033/1/gate/swift-dsvm-functional/9793830/ : SUCCESS in 37m 53s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/33/657033/1/gate/swift-dsvm-functional-ipv6/959269c/ : SUCCESS in 24m 43s\n- swift-tox-lower-constraints http://logs.openstack.org/33/657033/1/gate/swift-tox-lower-constraints/aa5a5b4/ : SUCCESS in 24m 01s\n- openstack-tox-pep8 http://logs.openstack.org/33/657033/1/gate/openstack-tox-pep8/e85bcd1/ : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":1},{"id":"13b495b438a9fbad8a29b13fba6a2b2e793c378e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-04 18:36:57.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2a1f93e3d24723c816662f478ac7efef1d5cb64e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 19:20:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/33/657033/1/check/openstack-tox-docs/88da28f/html/ : SUCCESS in 3m 21s\n- build-openstack-releasenotes http://logs.openstack.org/33/657033/1/check/build-openstack-releasenotes/b78f947/html/ : SUCCESS in 6m 20s\n- swift-tox-py27 http://logs.openstack.org/33/657033/1/check/swift-tox-py27/8f58e4e/ : SUCCESS in 13m 46s\n- swift-tox-py36 http://logs.openstack.org/33/657033/1/check/swift-tox-py36/931e0ea/ : FAILURE in 25m 28s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/33/657033/1/check/swift-tox-py37/cc374fc/ : SUCCESS in 8m 46s\n- swift-tox-func http://logs.openstack.org/33/657033/1/check/swift-tox-func/64275c1/ : SUCCESS in 8m 22s\n- swift-tox-func-encryption http://logs.openstack.org/33/657033/1/check/swift-tox-func-encryption/a484912/ : SUCCESS in 27m 00s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/33/657033/1/check/swift-tox-func-domain-remap-staticweb/1fb600e/ : SUCCESS in 8m 49s\n- swift-tox-func-ec http://logs.openstack.org/33/657033/1/check/swift-tox-func-ec/26458c3/ : SUCCESS in 28m 17s\n- swift-tox-func-s3api http://logs.openstack.org/33/657033/1/check/swift-tox-func-s3api/354c030/ : SUCCESS in 11m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/33/657033/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/14655fe/ : SUCCESS in 11m 18s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/33/657033/1/check/swift-probetests-centos-7/75fc82a/ : SUCCESS in 30m 46s\n- swift-dsvm-functional http://logs.openstack.org/33/657033/1/check/swift-dsvm-functional/d172e73/ : SUCCESS in 21m 28s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/33/657033/1/check/swift-dsvm-functional-ipv6/d8867a1/ : SUCCESS in 36m 35s\n- swift-tox-lower-constraints http://logs.openstack.org/33/657033/1/check/swift-tox-lower-constraints/8b2d94f/ : SUCCESS in 32m 13s\n- openstack-tox-pep8 http://logs.openstack.org/33/657033/1/check/openstack-tox-pep8/a1e90a6/ : SUCCESS in 4m 53s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/33/657033/1/check/swift-multinode-rolling-upgrade/5679a17/ : FAILURE in 7m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7e24807cb7d97d0ffbc7635f964277af9d93b687","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 19:20:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1bbc6f21c38688211542d4de7a0b069dbdd121a2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 22:38:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/33/657033/1/gate/openstack-tox-docs/4c304fa/html/ : SUCCESS in 5m 09s\n- build-openstack-releasenotes http://logs.openstack.org/33/657033/1/gate/build-openstack-releasenotes/d2f3a03/html/ : SUCCESS in 7m 55s\n- swift-tox-py27 http://logs.openstack.org/33/657033/1/gate/swift-tox-py27/0a71e51/ : SUCCESS in 13m 32s\n- swift-tox-py37 http://logs.openstack.org/33/657033/1/gate/swift-tox-py37/b1dba39/ : SUCCESS in 8m 19s\n- swift-tox-func http://logs.openstack.org/33/657033/1/gate/swift-tox-func/a827f87/ : SUCCESS in 8m 28s\n- swift-tox-func-encryption http://logs.openstack.org/33/657033/1/gate/swift-tox-func-encryption/02ecd10/ : SUCCESS in 8m 31s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/33/657033/1/gate/swift-tox-func-domain-remap-staticweb/f0392f3/ : SUCCESS in 12m 12s\n- swift-tox-func-ec http://logs.openstack.org/33/657033/1/gate/swift-tox-func-ec/a2e1209/ : SUCCESS in 9m 04s\n- swift-tox-func-s3api http://logs.openstack.org/33/657033/1/gate/swift-tox-func-s3api/31ae0e4/ : SUCCESS in 9m 29s\n- swift-probetests-centos-7 http://logs.openstack.org/33/657033/1/gate/swift-probetests-centos-7/3617c3a/ : SUCCESS in 30m 40s\n- swift-dsvm-functional http://logs.openstack.org/33/657033/1/gate/swift-dsvm-functional/fa70d86/ : SUCCESS in 32m 11s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/33/657033/1/gate/swift-dsvm-functional-ipv6/751c02d/ : SUCCESS in 26m 48s\n- swift-tox-lower-constraints http://logs.openstack.org/33/657033/1/gate/swift-tox-lower-constraints/5037cf2/ : SUCCESS in 12m 03s\n- openstack-tox-pep8 http://logs.openstack.org/33/657033/1/gate/openstack-tox-pep8/3180eb2/ : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":1},{"id":"a69f578d61a782c30ce0908b32b1c9aa2c686b7a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 22:38:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a239f4ec355b48999cbdb8e54660624e9b44e835","revisions":{"a239f4ec355b48999cbdb8e54660624e9b44e835":{"kind":"REWORK","_number":1,"created":"2019-05-03 20:37:05.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/33/657033/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/33/657033/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/657033/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/657033/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/657033/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/33/657033/1"}}},"commit":{"parents":[{"commit":"32bf43990c3c9ce771fca5968242f102155467c1","subject":"Replace git.openstack.org URLs with opendev.org URLs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/32bf43990c3c9ce771fca5968242f102155467c1"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-05-03 20:36:42.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-05-03 20:36:42.000000000","tz":-420},"subject":"Remove py35 unit test job","message":"Remove py35 unit test job\n\nChange-Id: Iadb3bed811907740f85bd9277ba09a4ca17fc025\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a239f4ec355b48999cbdb8e54660624e9b44e835"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a239f4ec355b48999cbdb8e54660624e9b44e835"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
