)]}'
{"id":"openstack%2Fswift~657042","triplet_id":"openstack%2Fswift~feature%2Flosf~I5c3b2ca0c97afcb9e322459c7e150abce0a08877","project":"openstack/swift","branch":"feature/losf","topic":"alecuyer/get_volume_stats","hashtags":[],"change_id":"I5c3b2ca0c97afcb9e322459c7e150abce0a08877","subject":"Add GetVolumeStats RPC call","status":"ABANDONED","created":"2019-05-03 20:51:38.000000000","updated":"2021-10-30 05:19:28.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"571c0cebbd710e3f376214f2d42962a00756c017","_number":657042,"virtual_id_number":657042,"owner":{"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":1,"date":"2019-05-03 21:28:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2019-05-04 23:29:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-03 21:28:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-04 23:29:26.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"6678585ffcfbee2ab606bc0ee6d6258d752eb033","author":{"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},"date":"2019-05-03 20:51:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c792e51aa4102481c028c2af0d538c2be15238bd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 21:28:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/42/657042/1/check/openstack-tox-docs/bc55f60/html/ : SUCCESS in 3m 57s\n- swift-tox-py27 http://logs.openstack.org/42/657042/1/check/swift-tox-py27/6a2fa26/ : SUCCESS in 10m 22s\n- swift-tox-py35 http://logs.openstack.org/42/657042/1/check/swift-tox-py35/fae4eed/ : SUCCESS in 8m 02s (non-voting)\n- swift-tox-py36 http://logs.openstack.org/42/657042/1/check/swift-tox-py36/4da4d30/ : SUCCESS in 8m 15s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/42/657042/1/check/swift-tox-py37/7d9bc7a/ : SUCCESS in 7m 40s\n- swift-tox-func http://logs.openstack.org/42/657042/1/check/swift-tox-func/c5fbdb3/ : SUCCESS in 8m 00s\n- swift-tox-func-encryption http://logs.openstack.org/42/657042/1/check/swift-tox-func-encryption/53595c8/ : SUCCESS in 9m 12s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/42/657042/1/check/swift-tox-func-domain-remap-staticweb/d5c0196/ : SUCCESS in 7m 36s\n- swift-tox-func-ec http://logs.openstack.org/42/657042/1/check/swift-tox-func-ec/948c3fc/ : SUCCESS in 8m 34s\n- swift-tox-func-s3api http://logs.openstack.org/42/657042/1/check/swift-tox-func-s3api/abacb4b/ : SUCCESS in 10m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/42/657042/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/4e7f069/ : SUCCESS in 12m 12s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/42/657042/1/check/swift-probetests-centos-7/a51b6e0/ : SUCCESS in 31m 32s\n- swift-dsvm-functional http://logs.openstack.org/42/657042/1/check/swift-dsvm-functional/af86721/ : SUCCESS in 26m 20s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/42/657042/1/check/swift-dsvm-functional-ipv6/31eaf2a/ : SUCCESS in 27m 43s\n- swift-tox-lower-constraints http://logs.openstack.org/42/657042/1/check/swift-tox-lower-constraints/2480185/ : SUCCESS in 11m 52s\n- openstack-tox-pep8 http://logs.openstack.org/42/657042/1/check/openstack-tox-pep8/70f9914/ : SUCCESS in 6m 08s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/42/657042/1/check/swift-multinode-rolling-upgrade/79c6cc0/ : SUCCESS in 17m 39s","accounts_in_message":[],"_revision_number":1},{"id":"6933bcfa88d8df675802b5c667c82fd63542fb35","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-05-04 23:29:26.000000000","message":"Patch Set 1: Code-Review+1\n\n(3 comments)\n\nthis is maybe ok to expose as a one off command - this could doesn\u0027t actually make use of it anywhere - maybe we could expose it via a cli utility for operators and so I can functionally test\n\nspeaking of tests... uhhhh ;)","accounts_in_message":[],"_revision_number":1},{"id":"571c0cebbd710e3f376214f2d42962a00756c017","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-10-30 05:19:28.000000000","message":"Abandoned\n\nClosing out feature/losf","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a40676ac76e0c292e9528436f0ff5e088a90aa97","revisions":{"a40676ac76e0c292e9528436f0ff5e088a90aa97":{"kind":"REWORK","_number":1,"created":"2019-05-03 20:51:38.000000000","uploader":{"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},"ref":"refs/changes/42/657042/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/657042/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/657042/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/657042/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/657042/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/657042/1"}}},"commit":{"parents":[{"commit":"aa04c6c6f99c20f00e1719aedd98c2214ee05fd2","subject":"Do not use o_tmpfile_supported() in kvfile.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aa04c6c6f99c20f00e1719aedd98c2214ee05fd2"}]}],"author":{"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","date":"2018-11-16 15:16:27.000000000","tz":60},"committer":{"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","date":"2019-05-03 20:07:23.000000000","tz":-360},"subject":"Add GetVolumeStats RPC call","message":"Add GetVolumeStats RPC call\n\nThe GetVolumeStats RPC call returns the count of objects and quarantined\nobjects in a volume.\n\nThis will be used (at least) to remove empty volumes\n\nWhile there, add missing debug log of the GetKvState function\n\nChange-Id: I5c3b2ca0c97afcb9e322459c7e150abce0a08877\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a40676ac76e0c292e9528436f0ff5e088a90aa97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a40676ac76e0c292e9528436f0ff5e088a90aa97"}]},"branch":"refs/heads/feature/losf"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
