)]}'
{"id":"openstack%2Fswift~657195","triplet_id":"openstack%2Fswift~master~I2e233afff26878ba94e268c71f6779bcca7b641c","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I2e233afff26878ba94e268c71f6779bcca7b641c","subject":"py3: Make StoragePolicy objects hashable","status":"MERGED","created":"2019-05-05 04:02:09.000000000","updated":"2019-05-06 20:36:36.000000000","submitted":"2019-05-06 20:36:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657195-1557174996803-424ce972","meta_rev_id":"3e08f85f6db59748b04c55a6ae904154582d7818","_number":657195,"virtual_id_number":657195,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":2,"date":"2019-05-06 20:36:36.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-05 05:47:55.000000000","_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2019-05-06 19:11:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2019-05-06 19:11:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-06 19:11:40.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2019-05-06 20:36:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"33d8e2a43ff5bafdd9e943a7f3e9c02e8e7c8aea","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-05 04:02:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fa40c5ddefd03f057007532a358d4d2ae0126108","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-05 04:43:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/95/657195/1/check/openstack-tox-docs/be1e3fe/html/ : SUCCESS in 4m 02s\n- swift-tox-py27 http://logs.openstack.org/95/657195/1/check/swift-tox-py27/1ce5117/ : SUCCESS in 9m 17s\n- swift-tox-py36 http://logs.openstack.org/95/657195/1/check/swift-tox-py36/f74cf73/ : SUCCESS in 7m 44s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/95/657195/1/check/swift-tox-py37/c20598e/ : SUCCESS in 7m 53s\n- swift-tox-func http://logs.openstack.org/95/657195/1/check/swift-tox-func/9ad0c54/ : SUCCESS in 7m 58s\n- swift-tox-func-encryption http://logs.openstack.org/95/657195/1/check/swift-tox-func-encryption/4127a58/ : SUCCESS in 8m 49s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/95/657195/1/check/swift-tox-func-domain-remap-staticweb/fdf170f/ : SUCCESS in 8m 02s\n- swift-tox-func-ec http://logs.openstack.org/95/657195/1/check/swift-tox-func-ec/2e89f97/ : SUCCESS in 8m 32s\n- swift-tox-func-s3api http://logs.openstack.org/95/657195/1/check/swift-tox-func-s3api/2b4ff4d/ : SUCCESS in 8m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/95/657195/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/4fba553/ : SUCCESS in 11m 04s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/95/657195/1/check/swift-probetests-centos-7/b0e14d4/ : SUCCESS in 31m 10s\n- swift-dsvm-functional http://logs.openstack.org/95/657195/1/check/swift-dsvm-functional/630c2ec/ : SUCCESS in 26m 51s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/95/657195/1/check/swift-dsvm-functional-ipv6/d0b6528/ : SUCCESS in 24m 48s\n- swift-tox-lower-constraints http://logs.openstack.org/95/657195/1/check/swift-tox-lower-constraints/262916b/ : FAILURE in 37m 49s\n- openstack-tox-pep8 http://logs.openstack.org/95/657195/1/check/openstack-tox-pep8/70bbee3/ : SUCCESS in 4m 27s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/95/657195/1/check/swift-multinode-rolling-upgrade/c972398/ : FAILURE in 7m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9f49fcaff2c164c9bd2c622a5d9c2e9a4f7e7a40","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-05 04:50:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"434b0153b0a1b8d9204b37822c63793ef4d1e417","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-05 05:45:59.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/95/657195/1/check/openstack-tox-docs/7f15965/html/ : SUCCESS in 4m 03s\n- swift-tox-py27 http://logs.openstack.org/95/657195/1/check/swift-tox-py27/1061061/ : FAILURE in 12m 37s\n- swift-tox-py36 http://logs.openstack.org/95/657195/1/check/swift-tox-py36/1f4a9ca/ : SUCCESS in 9m 26s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/95/657195/1/check/swift-tox-py37/77b493f/ : SUCCESS in 9m 53s\n- swift-tox-func http://logs.openstack.org/95/657195/1/check/swift-tox-func/923ea7f/ : SUCCESS in 9m 35s\n- swift-tox-func-encryption http://logs.openstack.org/95/657195/1/check/swift-tox-func-encryption/5fe119f/ : SUCCESS in 9m 48s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/95/657195/1/check/swift-tox-func-domain-remap-staticweb/ca28448/ : SUCCESS in 10m 00s\n- swift-tox-func-ec http://logs.openstack.org/95/657195/1/check/swift-tox-func-ec/85ac405/ : SUCCESS in 10m 41s\n- swift-tox-func-s3api http://logs.openstack.org/95/657195/1/check/swift-tox-func-s3api/65cb107/ : SUCCESS in 12m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/95/657195/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/6e17756/ : SUCCESS in 11m 01s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/95/657195/1/check/swift-probetests-centos-7/b113e1b/ : SUCCESS in 30m 25s\n- swift-dsvm-functional http://logs.openstack.org/95/657195/1/check/swift-dsvm-functional/94596df/ : SUCCESS in 24m 35s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/95/657195/1/check/swift-dsvm-functional-ipv6/7fb2d5e/ : SUCCESS in 25m 33s\n- swift-tox-lower-constraints http://logs.openstack.org/95/657195/1/check/swift-tox-lower-constraints/cddf67f/ : SUCCESS in 13m 17s\n- openstack-tox-pep8 http://logs.openstack.org/95/657195/1/check/openstack-tox-pep8/ee68bf4/ : SUCCESS in 5m 28s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/95/657195/1/check/swift-multinode-rolling-upgrade/47d0e1a/ : FAILURE in 9m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0017bdc863c14cc696d2f6fb5bc0d70c4bf04aeb","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-05 05:47:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"36ea5fdb760369f7682e4b551f88e5806ba1c1b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-05 06:23:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/95/657195/1/check/openstack-tox-docs/a1b0f75/html/ : SUCCESS in 3m 54s\n- swift-tox-py27 http://logs.openstack.org/95/657195/1/check/swift-tox-py27/8801b41/ : SUCCESS in 9m 36s\n- swift-tox-py36 http://logs.openstack.org/95/657195/1/check/swift-tox-py36/cc830c8/ : SUCCESS in 7m 35s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/95/657195/1/check/swift-tox-py37/662e54d/ : SUCCESS in 8m 18s\n- swift-tox-func http://logs.openstack.org/95/657195/1/check/swift-tox-func/5d0e9c8/ : SUCCESS in 8m 07s\n- swift-tox-func-encryption http://logs.openstack.org/95/657195/1/check/swift-tox-func-encryption/5b1a484/ : SUCCESS in 8m 40s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/95/657195/1/check/swift-tox-func-domain-remap-staticweb/ff987a2/ : SUCCESS in 8m 00s\n- swift-tox-func-ec http://logs.openstack.org/95/657195/1/check/swift-tox-func-ec/7caf3d1/ : SUCCESS in 8m 36s\n- swift-tox-func-s3api http://logs.openstack.org/95/657195/1/check/swift-tox-func-s3api/10ff927/ : SUCCESS in 10m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/95/657195/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/0122cae/ : SUCCESS in 11m 18s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/95/657195/1/check/swift-probetests-centos-7/770aca0/ : SUCCESS in 30m 12s\n- swift-dsvm-functional http://logs.openstack.org/95/657195/1/check/swift-dsvm-functional/4d26224/ : SUCCESS in 29m 30s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/95/657195/1/check/swift-dsvm-functional-ipv6/67cdce6/ : SUCCESS in 30m 04s\n- swift-tox-lower-constraints http://logs.openstack.org/95/657195/1/check/swift-tox-lower-constraints/dc3292b/ : SUCCESS in 12m 01s\n- openstack-tox-pep8 http://logs.openstack.org/95/657195/1/check/openstack-tox-pep8/84cc894/ : SUCCESS in 5m 01s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/95/657195/1/check/swift-multinode-rolling-upgrade/dce59b0/ : FAILURE in 11m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"75cfc667a18236e92537ad95771b2067a658df86","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-05-06 19:11:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b3962423c93552bacb0e470611fc76243c838032","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 19:13:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1bc40e9dfa174e436b2046e24b64070988677d86","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 20:36:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3e08f85f6db59748b04c55a6ae904154582d7818","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 20:36:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/95/657195/1/gate/openstack-tox-docs/69d5afc/html/ : SUCCESS in 4m 13s\n- swift-tox-py27 http://logs.openstack.org/95/657195/1/gate/swift-tox-py27/a27be44/ : SUCCESS in 13m 56s\n- swift-tox-py37 http://logs.openstack.org/95/657195/1/gate/swift-tox-py37/a686e2b/ : SUCCESS in 11m 22s\n- swift-tox-func http://logs.openstack.org/95/657195/1/gate/swift-tox-func/d6ecc26/ : SUCCESS in 17m 00s\n- swift-tox-func-encryption http://logs.openstack.org/95/657195/1/gate/swift-tox-func-encryption/ae71790/ : SUCCESS in 11m 48s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/95/657195/1/gate/swift-tox-func-domain-remap-staticweb/cd0eb26/ : SUCCESS in 16m 47s\n- swift-tox-func-ec http://logs.openstack.org/95/657195/1/gate/swift-tox-func-ec/301cc36/ : SUCCESS in 12m 04s\n- swift-tox-func-s3api http://logs.openstack.org/95/657195/1/gate/swift-tox-func-s3api/a2d11f1/ : SUCCESS in 8m 58s\n- swift-probetests-centos-7 http://logs.openstack.org/95/657195/1/gate/swift-probetests-centos-7/48d3ef8/ : SUCCESS in 32m 35s\n- swift-dsvm-functional http://logs.openstack.org/95/657195/1/gate/swift-dsvm-functional/f81b047/ : SUCCESS in 31m 21s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/95/657195/1/gate/swift-dsvm-functional-ipv6/bcbf81f/ : SUCCESS in 27m 50s\n- swift-tox-lower-constraints http://logs.openstack.org/95/657195/1/gate/swift-tox-lower-constraints/e458625/ : SUCCESS in 13m 12s\n- openstack-tox-pep8 http://logs.openstack.org/95/657195/1/gate/openstack-tox-pep8/e65e06f/ : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fed0204e869bc1a92dbd74319ed5dbef95f3ea5e","revisions":{"fed0204e869bc1a92dbd74319ed5dbef95f3ea5e":{"kind":"REWORK","_number":1,"created":"2019-05-05 04:02:09.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/95/657195/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/95/657195/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/657195/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/657195/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/657195/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/95/657195/1"}}},"commit":{"parents":[{"commit":"b8284538be4c17052c4c399a7fa09fd2abb50c49","subject":"py3: start porting for unit/proxy/test_server.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b8284538be4c17052c4c399a7fa09fd2abb50c49"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-05-05 04:01:29.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-05-05 04:01:29.000000000","tz":-420},"subject":"py3: Make StoragePolicy objects hashable","message":"py3: Make StoragePolicy objects hashable\n\nChange-Id: I2e233afff26878ba94e268c71f6779bcca7b641c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fed0204e869bc1a92dbd74319ed5dbef95f3ea5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fed0204e869bc1a92dbd74319ed5dbef95f3ea5e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
