)]}'
{"id":"openstack%2Fswift~657505","triplet_id":"openstack%2Fswift~master~Iaf14c1ab3e236e6774492ed711fccef066a16fca","project":"openstack/swift","branch":"master","topic":"bulk-alt","hashtags":[],"change_id":"Iaf14c1ab3e236e6774492ed711fccef066a16fca","subject":"py3: bulk -- alternate approach","status":"MERGED","created":"2019-05-07 00:01:32.000000000","updated":"2019-05-17 22:50:12.000000000","submitted":"2019-05-17 22:49:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657505-1558133350553-deb1f29f","meta_rev_id":"bafa7baf7b357458f815e8b2525bf99dfc0574f1","_number":657505,"virtual_id_number":657505,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":2,"date":"2019-05-17 22:49:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2019-05-14 03:20:49.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"value":1,"date":"2019-05-14 08:53:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":1,"date":"2019-05-17 20:54:30.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-14 08:53:45.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-05-17 20:54:30.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2019-05-17 22:49:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9fe366783af2e747b3017bc004ef9b37700e33df","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-07 00:01:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4bb7c630c45743c7330a8fb857a9d239bde5565c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 00:44:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/05/657505/1/check/openstack-tox-docs/ebb4562/html/ : SUCCESS in 4m 02s\n- build-openstack-releasenotes http://logs.openstack.org/05/657505/1/check/build-openstack-releasenotes/8bc2d05/html/ : SUCCESS in 6m 44s\n- swift-tox-py27 http://logs.openstack.org/05/657505/1/check/swift-tox-py27/8a187bd/ : SUCCESS in 20m 45s\n- swift-tox-py36 http://logs.openstack.org/05/657505/1/check/swift-tox-py36/f1a8fe8/ : SUCCESS in 19m 53s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/05/657505/1/check/swift-tox-py37/97f9950/ : SUCCESS in 17m 28s\n- swift-tox-func http://logs.openstack.org/05/657505/1/check/swift-tox-func/5bd57cc/ : SUCCESS in 20m 11s\n- swift-tox-func-encryption http://logs.openstack.org/05/657505/1/check/swift-tox-func-encryption/89cdc70/ : SUCCESS in 9m 04s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/05/657505/1/check/swift-tox-func-domain-remap-staticweb/b40725d/ : SUCCESS in 20m 31s\n- swift-tox-func-ec http://logs.openstack.org/05/657505/1/check/swift-tox-func-ec/6f8faea/ : SUCCESS in 7m 39s\n- swift-tox-func-s3api http://logs.openstack.org/05/657505/1/check/swift-tox-func-s3api/82c2b88/ : SUCCESS in 12m 30s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/05/657505/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/60ce1b8/ : SUCCESS in 11m 48s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/05/657505/1/check/swift-probetests-centos-7/a71b42c/ : SUCCESS in 30m 15s\n- swift-dsvm-functional http://logs.openstack.org/05/657505/1/check/swift-dsvm-functional/039fa27/ : SUCCESS in 40m 48s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/05/657505/1/check/swift-dsvm-functional-ipv6/a804ed1/ : SUCCESS in 27m 30s\n- swift-tox-lower-constraints http://logs.openstack.org/05/657505/1/check/swift-tox-lower-constraints/d024eb1/ : FAILURE in 24m 06s\n- openstack-tox-pep8 http://logs.openstack.org/05/657505/1/check/openstack-tox-pep8/a2423fb/ : FAILURE in 4m 54s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/05/657505/1/check/swift-multinode-rolling-upgrade/b8392b0/ : FAILURE in 8m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9996be57ae37c2513ead1d9385a825d97192612c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-07 16:26:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"28106a067eefa8740ee66fe8f8105ba85143acd8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 18:30:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/05/657505/2/check/openstack-tox-docs/7466e8b/html/ : SUCCESS in 3m 48s\n- build-openstack-releasenotes http://logs.openstack.org/05/657505/2/check/build-openstack-releasenotes/5a2e059/html/ : SUCCESS in 6m 58s\n- swift-tox-py27 http://logs.openstack.org/05/657505/2/check/swift-tox-py27/e6034db/ : SUCCESS in 12m 48s\n- swift-tox-py36 http://logs.openstack.org/05/657505/2/check/swift-tox-py36/b47e866/ : SUCCESS in 8m 17s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/05/657505/2/check/swift-tox-py37/b94724e/ : FAILURE in 9m 45s\n- swift-tox-func http://logs.openstack.org/05/657505/2/check/swift-tox-func/1415cb2/ : SUCCESS in 8m 41s\n- swift-tox-func-encryption http://logs.openstack.org/05/657505/2/check/swift-tox-func-encryption/8fb48de/ : SUCCESS in 9m 21s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/05/657505/2/check/swift-tox-func-domain-remap-staticweb/a76df06/ : SUCCESS in 8m 12s\n- swift-tox-func-ec http://logs.openstack.org/05/657505/2/check/swift-tox-func-ec/2c5e89d/ : SUCCESS in 11m 24s\n- swift-tox-func-s3api http://logs.openstack.org/05/657505/2/check/swift-tox-func-s3api/a6589e9/ : SUCCESS in 9m 42s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/05/657505/2/check/swift-tox-func-s3api-ceph-s3tests-tempauth/c382dcb/ : POST_FAILURE in 6m 44s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/05/657505/2/check/swift-probetests-centos-7/ebbc811/ : SUCCESS in 29m 19s\n- swift-dsvm-functional http://logs.openstack.org/05/657505/2/check/swift-dsvm-functional/ca97935/ : SUCCESS in 25m 21s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/05/657505/2/check/swift-dsvm-functional-ipv6/fe6688b/ : SUCCESS in 26m 55s\n- swift-tox-lower-constraints http://logs.openstack.org/05/657505/2/check/swift-tox-lower-constraints/d13ebb6/ : SUCCESS in 13m 59s\n- openstack-tox-pep8 http://logs.openstack.org/05/657505/2/check/openstack-tox-pep8/657b427/ : SUCCESS in 5m 19s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/05/657505/2/check/swift-multinode-rolling-upgrade/b2bf81d/ : FAILURE in 8m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b3db90c8e64c599cc517b0591a31dc41c1c3b6f7","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-05-09 03:13:50.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"55371db80b693d9da16196546f9f1329cab15d2b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-09 15:14:56.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nrecheck\n\nlog_lines in test_multiphase_put_drains_extra_commit_junk_disconnect again","accounts_in_message":[],"_revision_number":2},{"id":"9140a5603dd88081027c5f6970eafb322d64e52b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-10 15:03:15.000000000","message":"Patch Set 2:\n\nrecheck\n\nlog_lines again","accounts_in_message":[],"_revision_number":2},{"id":"e69ade941f56e0c8cb6ea404d3be15e2fecded6b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 16:48:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/05/657505/2/check/openstack-tox-docs/66ea4ef/html/ : SUCCESS in 4m 39s\n- build-openstack-releasenotes http://logs.openstack.org/05/657505/2/check/build-openstack-releasenotes/17d96f8/html/ : SUCCESS in 7m 04s\n- swift-build-image http://logs.openstack.org/05/657505/2/check/swift-build-image/f25acbf/ : SUCCESS in 8m 35s\n- swift-tox-py27 http://logs.openstack.org/05/657505/2/check/swift-tox-py27/5689fc6/ : SUCCESS in 9m 57s\n- swift-tox-py36 http://logs.openstack.org/05/657505/2/check/swift-tox-py36/8116ade/ : SUCCESS in 8m 41s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/05/657505/2/check/swift-tox-py37/e9c0ec3/ : SUCCESS in 9m 42s\n- swift-tox-func http://logs.openstack.org/05/657505/2/check/swift-tox-func/da9dae5/ : SUCCESS in 8m 30s\n- swift-tox-func-encryption http://logs.openstack.org/05/657505/2/check/swift-tox-func-encryption/3537143/ : SUCCESS in 9m 33s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/05/657505/2/check/swift-tox-func-domain-remap-staticweb/386484c/ : SUCCESS in 14m 15s\n- swift-tox-func-ec http://logs.openstack.org/05/657505/2/check/swift-tox-func-ec/839ec29/ : SUCCESS in 10m 22s\n- swift-tox-func-s3api http://logs.openstack.org/05/657505/2/check/swift-tox-func-s3api/39a8ab2/ : SUCCESS in 9m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/05/657505/2/check/swift-tox-func-s3api-ceph-s3tests-tempauth/901e23f/ : SUCCESS in 12m 15s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/05/657505/2/check/swift-probetests-centos-7/7422bc7/ : SUCCESS in 34m 40s\n- swift-dsvm-functional http://logs.openstack.org/05/657505/2/check/swift-dsvm-functional/433bb87/ : SUCCESS in 29m 45s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/05/657505/2/check/swift-dsvm-functional-ipv6/8dd21e8/ : SUCCESS in 28m 15s\n- swift-tox-lower-constraints http://logs.openstack.org/05/657505/2/check/swift-tox-lower-constraints/e6f6950/ : SUCCESS in 13m 27s\n- openstack-tox-pep8 http://logs.openstack.org/05/657505/2/check/openstack-tox-pep8/8a10fe4/ : SUCCESS in 5m 28s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/05/657505/2/check/swift-multinode-rolling-upgrade/08c352c/ : FAILURE in 8m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8a6a40576f3d4786e463fdfed3bbf376064748e3","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-13 22:35:33.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"be9570c3335edeeff3547b69af022bd62ea24b34","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 00:00:23.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/05/657505/3/check/openstack-tox-docs/752b972/html/ : SUCCESS in 4m 23s\n- build-openstack-releasenotes http://logs.openstack.org/05/657505/3/check/build-openstack-releasenotes/1b55ff8/html/ : SUCCESS in 7m 26s\n- swift-build-image http://logs.openstack.org/05/657505/3/check/swift-build-image/2fdfe0a/ : SUCCESS in 8m 47s\n- swift-tox-py27 http://logs.openstack.org/05/657505/3/check/swift-tox-py27/63939fc/ : FAILURE in 10m 10s\n- swift-tox-py36 http://logs.openstack.org/05/657505/3/check/swift-tox-py36/a40f238/ : SUCCESS in 7m 56s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/05/657505/3/check/swift-tox-py37/49518d2/ : SUCCESS in 9m 25s\n- swift-tox-func http://logs.openstack.org/05/657505/3/check/swift-tox-func/f04d018/ : FAILURE in 8m 36s\n- swift-tox-func-encryption http://logs.openstack.org/05/657505/3/check/swift-tox-func-encryption/a51f3df/ : FAILURE in 8m 22s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/05/657505/3/check/swift-tox-func-domain-remap-staticweb/662ecbf/ : FAILURE in 11m 38s\n- swift-tox-func-ec http://logs.openstack.org/05/657505/3/check/swift-tox-func-ec/e92f530/ : FAILURE in 13m 32s\n- swift-tox-func-s3api http://logs.openstack.org/05/657505/3/check/swift-tox-func-s3api/377b08a/ : FAILURE in 9m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/05/657505/3/check/swift-tox-func-s3api-ceph-s3tests-tempauth/eb4ccc4/ : SUCCESS in 11m 58s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/05/657505/3/check/swift-probetests-centos-7/3cb03e2/ : SUCCESS in 28m 39s\n- swift-dsvm-functional http://logs.openstack.org/05/657505/3/check/swift-dsvm-functional/a08ec45/ : FAILURE in 21m 32s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/05/657505/3/check/swift-dsvm-functional-ipv6/97f639e/ : FAILURE in 23m 00s\n- swift-tox-lower-constraints http://logs.openstack.org/05/657505/3/check/swift-tox-lower-constraints/ad61976/ : FAILURE in 10m 56s\n- openstack-tox-pep8 http://logs.openstack.org/05/657505/3/check/openstack-tox-pep8/3ae4ccf/ : FAILURE in 5m 04s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/05/657505/3/check/swift-multinode-rolling-upgrade/2dbe6b5/ : FAILURE in 7m 33s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0de40ea4ed90dfce0a13579b5de76d5a4574810c","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-05-14 01:31:05.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3920bc3f379b9dd21d077f4ee058765a5a69f416","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 02:08:38.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/05/657505/4/check/openstack-tox-docs/5049e37/html/ : SUCCESS in 3m 41s\n- build-openstack-releasenotes http://logs.openstack.org/05/657505/4/check/build-openstack-releasenotes/3ea435d/html/ : SUCCESS in 7m 00s\n- swift-build-image http://logs.openstack.org/05/657505/4/check/swift-build-image/96ec5f0/ : SUCCESS in 9m 28s\n- swift-tox-py27 http://logs.openstack.org/05/657505/4/check/swift-tox-py27/ad5fa68/ : SUCCESS in 27m 04s\n- swift-tox-py36 http://logs.openstack.org/05/657505/4/check/swift-tox-py36/e193662/ : FAILURE in 10m 52s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/05/657505/4/check/swift-tox-py37/03734f4/ : FAILURE in 28m 07s\n- swift-tox-func http://logs.openstack.org/05/657505/4/check/swift-tox-func/0946953/ : SUCCESS in 7m 51s\n- swift-tox-func-encryption http://logs.openstack.org/05/657505/4/check/swift-tox-func-encryption/8690931/ : SUCCESS in 22m 59s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/05/657505/4/check/swift-tox-func-domain-remap-staticweb/a7d7f67/ : SUCCESS in 8m 01s\n- swift-tox-func-ec http://logs.openstack.org/05/657505/4/check/swift-tox-func-ec/a103376/ : SUCCESS in 8m 38s\n- swift-tox-func-s3api http://logs.openstack.org/05/657505/4/check/swift-tox-func-s3api/1e47d52/ : SUCCESS in 36m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/05/657505/4/check/swift-tox-func-s3api-ceph-s3tests-tempauth/1d14ff3/ : SUCCESS in 18m 57s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/05/657505/4/check/swift-probetests-centos-7/1bf598b/ : SUCCESS in 32m 05s\n- swift-dsvm-functional http://logs.openstack.org/05/657505/4/check/swift-dsvm-functional/428d207/ : SUCCESS in 22m 09s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/05/657505/4/check/swift-dsvm-functional-ipv6/b7024c1/ : SUCCESS in 25m 35s\n- swift-tox-lower-constraints http://logs.openstack.org/05/657505/4/check/swift-tox-lower-constraints/fd659e3/ : SUCCESS in 16m 24s\n- openstack-tox-pep8 http://logs.openstack.org/05/657505/4/check/openstack-tox-pep8/84a01b2/ : SUCCESS in 4m 37s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/05/657505/4/check/swift-multinode-rolling-upgrade/7f2a613/ : FAILURE in 7m 55s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e6c009a1809b35e9e47ba2a1fa889bd1fb5bbeb6","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-05-14 02:21:03.000000000","message":"Patch Set 4:\n\nrecheck\n\nagain test_multiphase_put_drains_extra_commit_junk_disconnect","accounts_in_message":[],"_revision_number":4},{"id":"26644b6ddfbd8ff568b00747e238396b68d0f605","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 03:05:00.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/05/657505/4/check/openstack-tox-docs/ec48513/html/ : SUCCESS in 3m 28s\n- build-openstack-releasenotes http://logs.openstack.org/05/657505/4/check/build-openstack-releasenotes/c92e158/html/ : SUCCESS in 5m 55s\n- swift-build-image http://logs.openstack.org/05/657505/4/check/swift-build-image/ebc10f7/ : SUCCESS in 7m 48s\n- swift-tox-py27 http://logs.openstack.org/05/657505/4/check/swift-tox-py27/8c60062/ : SUCCESS in 10m 16s\n- swift-tox-py36 http://logs.openstack.org/05/657505/4/check/swift-tox-py36/9536bd0/ : SUCCESS in 11m 38s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/05/657505/4/check/swift-tox-py37/1afbd48/ : SUCCESS in 20m 32s\n- swift-tox-func http://logs.openstack.org/05/657505/4/check/swift-tox-func/6424831/ : SUCCESS in 21m 22s\n- swift-tox-func-encryption http://logs.openstack.org/05/657505/4/check/swift-tox-func-encryption/6e95f44/ : SUCCESS in 19m 47s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/05/657505/4/check/swift-tox-func-domain-remap-staticweb/7c1432d/ : SUCCESS in 21m 13s\n- swift-tox-func-ec http://logs.openstack.org/05/657505/4/check/swift-tox-func-ec/d80b464/ : SUCCESS in 20m 50s\n- swift-tox-func-s3api http://logs.openstack.org/05/657505/4/check/swift-tox-func-s3api/0943e98/ : SUCCESS in 12m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/05/657505/4/check/swift-tox-func-s3api-ceph-s3tests-tempauth/b0e1263/ : SUCCESS in 10m 54s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/05/657505/4/check/swift-probetests-centos-7/121bb69/ : SUCCESS in 30m 34s\n- swift-dsvm-functional http://logs.openstack.org/05/657505/4/check/swift-dsvm-functional/e456c82/ : SUCCESS in 24m 49s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/05/657505/4/check/swift-dsvm-functional-ipv6/0ff5630/ : SUCCESS in 27m 04s\n- swift-tox-lower-constraints http://logs.openstack.org/05/657505/4/check/swift-tox-lower-constraints/4fdac3d/ : SUCCESS in 14m 23s\n- openstack-tox-pep8 http://logs.openstack.org/05/657505/4/check/openstack-tox-pep8/f56995b/ : SUCCESS in 5m 00s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/05/657505/4/check/swift-multinode-rolling-upgrade/143994c/ : FAILURE in 7m 07s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c0e8c33c101e0fc2e02046b12e91721943abc249","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-05-14 03:20:49.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"7511d7e9c1774c2085ed4d4fefe6ab35214d5ea9","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-05-14 08:53:45.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"b81ae275d4fdb457b50d647c812173a059058889","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-05-17 20:54:30.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"cb30794afa8d4d008b8016a338bbae79c91d06a1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 20:54:51.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"01c1bca3dafb12646299405c55898f418b56c806","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 22:49:10.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/05/657505/4/gate/openstack-tox-docs/24660b7/html/ : SUCCESS in 5m 35s\n- build-openstack-releasenotes http://logs.openstack.org/05/657505/4/gate/build-openstack-releasenotes/7b9752e/html/ : SUCCESS in 7m 16s\n- swift-upload-image http://logs.openstack.org/05/657505/4/gate/swift-upload-image/63fced6/ : SUCCESS in 8m 24s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/05/657505/4/gate/swift-tox-py27/997e500/ : SUCCESS in 10m 21s\n- swift-tox-py37 http://logs.openstack.org/05/657505/4/gate/swift-tox-py37/7b9d9c2/ : SUCCESS in 9m 45s\n- swift-tox-func http://logs.openstack.org/05/657505/4/gate/swift-tox-func/b24b99e/ : SUCCESS in 8m 36s\n- swift-tox-func-encryption http://logs.openstack.org/05/657505/4/gate/swift-tox-func-encryption/8b8c1cf/ : SUCCESS in 8m 26s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/05/657505/4/gate/swift-tox-func-domain-remap-staticweb/c3e9f61/ : SUCCESS in 8m 54s\n- swift-tox-func-ec http://logs.openstack.org/05/657505/4/gate/swift-tox-func-ec/ef9a244/ : SUCCESS in 13m 18s\n- swift-tox-func-s3api http://logs.openstack.org/05/657505/4/gate/swift-tox-func-s3api/1623e95/ : SUCCESS in 9m 36s\n- swift-probetests-centos-7 http://logs.openstack.org/05/657505/4/gate/swift-probetests-centos-7/fd08201/ : SUCCESS in 30m 12s\n- swift-dsvm-functional http://logs.openstack.org/05/657505/4/gate/swift-dsvm-functional/e08f11c/ : SUCCESS in 32m 10s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/05/657505/4/gate/swift-dsvm-functional-ipv6/890718e/ : SUCCESS in 31m 58s\n- swift-tox-lower-constraints http://logs.openstack.org/05/657505/4/gate/swift-tox-lower-constraints/b1d8d34/ : SUCCESS in 12m 19s\n- openstack-tox-pep8 http://logs.openstack.org/05/657505/4/gate/openstack-tox-pep8/6929a2f/ : SUCCESS in 6m 17s","accounts_in_message":[],"_revision_number":4},{"id":"1494dc4eb67a1c407040a7a4fd6e874ac88c9ba9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 22:49:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"04d9347e8cc355cda9a9aab93d7c8a6b2188dcd6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 22:50:12.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- swift-promote-image http://logs.openstack.org/05/657505/4/promote/swift-promote-image/988ac1a/ : FAILURE in 42s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"5cd2a2ac2f0552247ada19cc02ec0ca5389375bf","revisions":{"04255872e5b36b39ca8a749750e687b5c4b026be":{"kind":"REWORK","_number":1,"created":"2019-05-07 00:01:32.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/05/657505/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/657505/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/657505/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/657505/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/657505/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/657505/1"}}},"commit":{"parents":[{"commit":"fd73353bb1a04714f6aabc185a4b88ca23d2b1a6","subject":"Merge \"Enable to configure object-expirer in object-server.conf\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fd73353bb1a04714f6aabc185a4b88ca23d2b1a6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-05-05 19:38:57.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-05-07 00:01:02.000000000","tz":-420},"subject":"py3: bulk -- alternate approach","message":"py3: bulk -- alternate approach\n\nChange-Id: Iaf14c1ab3e236e6774492ed711fccef066a16fca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/04255872e5b36b39ca8a749750e687b5c4b026be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/04255872e5b36b39ca8a749750e687b5c4b026be"}]},"branch":"refs/heads/master"},"eb0a31fcf9d6f29554afc8b768beb2b2be7fd543":{"kind":"REWORK","_number":2,"created":"2019-05-07 16:26:50.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/05/657505/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/657505/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/657505/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/657505/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/657505/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/657505/2"}}},"commit":{"parents":[{"commit":"fd73353bb1a04714f6aabc185a4b88ca23d2b1a6","subject":"Merge \"Enable to configure object-expirer in object-server.conf\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fd73353bb1a04714f6aabc185a4b88ca23d2b1a6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-05-05 19:38:57.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-05-07 16:08:38.000000000","tz":-420},"subject":"py3: bulk -- alternate approach","message":"py3: bulk -- alternate approach\n\nRelated-Change: I3603247e0c36299de2107aa8e494a3f87647696f\nChange-Id: Iaf14c1ab3e236e6774492ed711fccef066a16fca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eb0a31fcf9d6f29554afc8b768beb2b2be7fd543"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eb0a31fcf9d6f29554afc8b768beb2b2be7fd543"}]},"branch":"refs/heads/master"},"e8e22ad863e3ce8c27ef78126104ade2772c6aec":{"kind":"REWORK","_number":3,"created":"2019-05-13 22:35:33.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/05/657505/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/657505/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/657505/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/657505/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/657505/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/657505/3"}}},"commit":{"parents":[{"commit":"fd73353bb1a04714f6aabc185a4b88ca23d2b1a6","subject":"Merge \"Enable to configure object-expirer in object-server.conf\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fd73353bb1a04714f6aabc185a4b88ca23d2b1a6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-05-05 19:38:57.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-05-13 22:34:52.000000000","tz":-420},"subject":"py3: bulk -- alternate approach","message":"py3: bulk -- alternate approach\n\nRelated-Change: I3603247e0c36299de2107aa8e494a3f87647696f\nChange-Id: Iaf14c1ab3e236e6774492ed711fccef066a16fca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e8e22ad863e3ce8c27ef78126104ade2772c6aec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e8e22ad863e3ce8c27ef78126104ade2772c6aec"}]},"branch":"refs/heads/master"},"5cd2a2ac2f0552247ada19cc02ec0ca5389375bf":{"kind":"REWORK","_number":4,"created":"2019-05-14 01:31:05.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/05/657505/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/657505/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/657505/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/657505/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/657505/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/657505/4"}}},"commit":{"parents":[{"commit":"fd73353bb1a04714f6aabc185a4b88ca23d2b1a6","subject":"Merge \"Enable to configure object-expirer in object-server.conf\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fd73353bb1a04714f6aabc185a4b88ca23d2b1a6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-05-05 19:38:57.000000000","tz":-420},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2019-05-14 01:08:50.000000000","tz":-300},"subject":"py3: bulk -- alternate approach","message":"py3: bulk -- alternate approach\n\nRelated-Change: I3603247e0c36299de2107aa8e494a3f87647696f\nChange-Id: Iaf14c1ab3e236e6774492ed711fccef066a16fca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5cd2a2ac2f0552247ada19cc02ec0ca5389375bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5cd2a2ac2f0552247ada19cc02ec0ca5389375bf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
