)]}'
{"id":"openstack%2Fswift~657700","triplet_id":"openstack%2Fswift~master~I8287db75b4f19581203360c646e72f64fe45f170","project":"openstack/swift","branch":"master","topic":"657705","hashtags":[],"change_id":"I8287db75b4f19581203360c646e72f64fe45f170","subject":"py3: finish porting proxy/test_server.py","status":"MERGED","created":"2019-05-07 22:45:18.000000000","updated":"2019-05-14 04:43:40.000000000","submitted":"2019-05-14 04:43:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657700-1557809020022-6085ff31","meta_rev_id":"e134ecf5b8df2ac15e21e4466cc0d5b64490f02a","_number":657700,"virtual_id_number":657700,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},{"value":2,"date":"2019-05-14 04:43:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-09 19:42:50.000000000","_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2019-05-14 03:37:13.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":1,"date":"2019-05-13 03:35:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":1,"date":"2019-05-14 03:37:13.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-13 03:35:02.000000000","updated_by":{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},"reviewer":{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},"state":"REVIEWER"},{"updated":"2019-05-14 03:37:13.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2019-05-14 04:43:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3eb31dabfe69cc68eeee306a90dab3baaae9a581","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-07 22:45:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dd6a6c62687f22a6918a6ab19369cd5c4c127740","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-07 23:13:30.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"e901274c34fe3f6aed11d97450dcb348137e9cdc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 00:10:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/00/657700/2/check/openstack-tox-docs/b3c8702/html/ : SUCCESS in 3m 45s\n- build-openstack-releasenotes http://logs.openstack.org/00/657700/2/check/build-openstack-releasenotes/df83b10/html/ : SUCCESS in 7m 21s\n- swift-tox-py27 http://logs.openstack.org/00/657700/2/check/swift-tox-py27/b356c06/ : SUCCESS in 11m 02s\n- swift-tox-py36 http://logs.openstack.org/00/657700/2/check/swift-tox-py36/9f0c793/ : FAILURE in 8m 54s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/00/657700/2/check/swift-tox-py37/162d0ba/ : FAILURE in 9m 54s\n- swift-tox-func http://logs.openstack.org/00/657700/2/check/swift-tox-func/eec15d9/ : SUCCESS in 9m 20s\n- swift-tox-func-encryption http://logs.openstack.org/00/657700/2/check/swift-tox-func-encryption/d04e240/ : SUCCESS in 9m 58s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/00/657700/2/check/swift-tox-func-domain-remap-staticweb/1d0e1af/ : SUCCESS in 8m 22s\n- swift-tox-func-ec http://logs.openstack.org/00/657700/2/check/swift-tox-func-ec/d314d47/ : SUCCESS in 9m 05s\n- swift-tox-func-s3api http://logs.openstack.org/00/657700/2/check/swift-tox-func-s3api/1cf1a97/ : SUCCESS in 10m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/00/657700/2/check/swift-tox-func-s3api-ceph-s3tests-tempauth/44a5af0/ : SUCCESS in 11m 24s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/00/657700/2/check/swift-probetests-centos-7/61a8d82/ : SUCCESS in 31m 28s\n- swift-dsvm-functional http://logs.openstack.org/00/657700/2/check/swift-dsvm-functional/d4cf948/ : SUCCESS in 26m 34s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/00/657700/2/check/swift-dsvm-functional-ipv6/c4ce60d/ : SUCCESS in 25m 58s\n- swift-tox-lower-constraints http://logs.openstack.org/00/657700/2/check/swift-tox-lower-constraints/6a7d8d9/ : SUCCESS in 11m 33s\n- openstack-tox-pep8 http://logs.openstack.org/00/657700/2/check/openstack-tox-pep8/5f336f2/ : SUCCESS in 4m 52s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/00/657700/2/check/swift-multinode-rolling-upgrade/6cc8a4c/ : FAILURE in 8m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e89b81b21cdda48fbddef72b8f200d766b119483","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-05-08 21:51:40.000000000","message":"Patch Set 2: Code-Review-1\n\nMake sure to look at parse_headers_string(). It parses a binary strng, so tries to store headers with bytes for names (keys in HeadersDict). As a result, nothing compares with literals like \u0027Content-Length\u0027. I\u0027d prefer leaving literal key names wihtout b\u0027\u0027 and converting in parse_headers_string().","accounts_in_message":[],"_revision_number":2},{"id":"4804ef6ef730f74e7032f01907b26cae342ae39c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-08 22:36:22.000000000","message":"Patch Set 2:\n\n\u003e Make sure to look at parse_headers_string(). ... I\u0027d prefer leaving literal key names wihtout b\u0027\u0027 and converting in parse_headers_string().\n\nCheck; done in the parent patch: https://review.opendev.org/#/c/657699/2/test/unit/proxy/test_server.py@146\n\nI agree, definitely better to keep the strings as native strings if at all possible.\n\nThere\u0027s still a bit of work needed around https://review.opendev.org/#/c/657700/2/test/unit/proxy/test_server.py@10199 to ensure the proxy under test is using the SwiftHttpProtocol so it correctly handles unquoted paths. Then I\u0027m down to just one unexplained failure...","accounts_in_message":[],"_revision_number":2},{"id":"e96b340b9772e03f4f0aa4253ec3a88f802aec8a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-08 22:59:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d32186bfc865f316b3e04260767c803aa0d1b4dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 23:37:09.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/00/657700/3/check/openstack-tox-docs/8e7f1e2/html/ : SUCCESS in 4m 07s\n- build-openstack-releasenotes http://logs.openstack.org/00/657700/3/check/build-openstack-releasenotes/9bea56b/html/ : SUCCESS in 7m 30s\n- swift-build-image http://logs.openstack.org/00/657700/3/check/swift-build-image/e431e4c/ : SUCCESS in 7m 58s\n- swift-tox-py27 http://logs.openstack.org/00/657700/3/check/swift-tox-py27/efc8b64/ : SUCCESS in 12m 32s\n- swift-tox-py36 http://logs.openstack.org/00/657700/3/check/swift-tox-py36/1875a7b/ : FAILURE in 9m 34s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/00/657700/3/check/swift-tox-py37/df8877b/ : FAILURE in 11m 05s\n- swift-tox-func http://logs.openstack.org/00/657700/3/check/swift-tox-func/679b0e5/ : SUCCESS in 8m 43s\n- swift-tox-func-encryption http://logs.openstack.org/00/657700/3/check/swift-tox-func-encryption/cbd37d4/ : SUCCESS in 9m 04s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/00/657700/3/check/swift-tox-func-domain-remap-staticweb/3644a6f/ : SUCCESS in 8m 31s\n- swift-tox-func-ec http://logs.openstack.org/00/657700/3/check/swift-tox-func-ec/7e78fdb/ : SUCCESS in 9m 12s\n- swift-tox-func-s3api http://logs.openstack.org/00/657700/3/check/swift-tox-func-s3api/f7e4425/ : SUCCESS in 9m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/00/657700/3/check/swift-tox-func-s3api-ceph-s3tests-tempauth/cb54b45/ : SUCCESS in 11m 42s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/00/657700/3/check/swift-probetests-centos-7/b0ffef9/ : SUCCESS in 29m 05s\n- swift-dsvm-functional http://logs.openstack.org/00/657700/3/check/swift-dsvm-functional/be17576/ : SUCCESS in 24m 31s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/00/657700/3/check/swift-dsvm-functional-ipv6/385790c/ : SUCCESS in 32m 01s\n- swift-tox-lower-constraints http://logs.openstack.org/00/657700/3/check/swift-tox-lower-constraints/31b6fe7/ : SUCCESS in 11m 04s\n- openstack-tox-pep8 http://logs.openstack.org/00/657700/3/check/openstack-tox-pep8/f13fa18/ : SUCCESS in 5m 11s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/00/657700/3/check/swift-multinode-rolling-upgrade/d140aa5/ : FAILURE in 8m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c7e2604d9bfe21ea0316ba5e5b0eacdd3747ea38","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-09 00:47:49.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a910e2581b217d7986bd8035861da2e6a4aff733","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 01:31:13.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/00/657700/4/check/openstack-tox-docs/f0b493d/html/ : SUCCESS in 3m 56s\n- build-openstack-releasenotes http://logs.openstack.org/00/657700/4/check/build-openstack-releasenotes/63cc80c/html/ : SUCCESS in 7m 00s\n- swift-build-image http://logs.openstack.org/00/657700/4/check/swift-build-image/b1e9c69/ : SUCCESS in 9m 57s\n- swift-tox-py27 http://logs.openstack.org/00/657700/4/check/swift-tox-py27/c332d07/ : FAILURE in 15m 52s\n- swift-tox-py36 http://logs.openstack.org/00/657700/4/check/swift-tox-py36/a5760d1/ : FAILURE in 32m 08s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/00/657700/4/check/swift-tox-py37/4f7d7c8/ : SUCCESS in 11m 52s\n- swift-tox-func http://logs.openstack.org/00/657700/4/check/swift-tox-func/3896127/ : SUCCESS in 7m 53s\n- swift-tox-func-encryption http://logs.openstack.org/00/657700/4/check/swift-tox-func-encryption/723acb1/ : SUCCESS in 8m 36s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/00/657700/4/check/swift-tox-func-domain-remap-staticweb/21aa0e3/ : SUCCESS in 8m 17s\n- swift-tox-func-ec http://logs.openstack.org/00/657700/4/check/swift-tox-func-ec/ee1e619/ : SUCCESS in 8m 46s\n- swift-tox-func-s3api http://logs.openstack.org/00/657700/4/check/swift-tox-func-s3api/7a3c91b/ : SUCCESS in 9m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/00/657700/4/check/swift-tox-func-s3api-ceph-s3tests-tempauth/3f95297/ : SUCCESS in 11m 27s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/00/657700/4/check/swift-probetests-centos-7/75ea116/ : SUCCESS in 29m 42s\n- swift-dsvm-functional http://logs.openstack.org/00/657700/4/check/swift-dsvm-functional/0f25155/ : SUCCESS in 25m 49s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/00/657700/4/check/swift-dsvm-functional-ipv6/dd4fcca/ : SUCCESS in 26m 04s\n- swift-tox-lower-constraints http://logs.openstack.org/00/657700/4/check/swift-tox-lower-constraints/7e5ca54/ : SUCCESS in 15m 53s\n- openstack-tox-pep8 http://logs.openstack.org/00/657700/4/check/openstack-tox-pep8/856932f/ : SUCCESS in 4m 55s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/00/657700/4/check/swift-multinode-rolling-upgrade/55f9747/ : FAILURE in 9m 16s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2c6366916a9bc014ab1983b78569ddc215027c03","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-09 15:05:14.000000000","message":"Patch Set 4:\n\nrecheck\n\nlog_lines in test_multiphase_put_drains_extra_commit_junk_disconnect again","accounts_in_message":[],"_revision_number":4},{"id":"2a000f1d7c6e36faa806ea69fec471472888d093","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 19:14:04.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/00/657700/4/check/openstack-tox-docs/983c238/html/ : SUCCESS in 4m 06s\n- build-openstack-releasenotes http://logs.openstack.org/00/657700/4/check/build-openstack-releasenotes/22d2b15/html/ : SUCCESS in 7m 20s\n- swift-build-image http://logs.openstack.org/00/657700/4/check/swift-build-image/debfc16/ : SUCCESS in 8m 06s\n- swift-tox-py27 http://logs.openstack.org/00/657700/4/check/swift-tox-py27/f26a6ea/ : SUCCESS in 12m 02s\n- swift-tox-py36 http://logs.openstack.org/00/657700/4/check/swift-tox-py36/abeb3e0/ : SUCCESS in 9m 13s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/00/657700/4/check/swift-tox-py37/1c4af7d/ : SUCCESS in 9m 04s\n- swift-tox-func http://logs.openstack.org/00/657700/4/check/swift-tox-func/c1d0bef/ : SUCCESS in 7m 44s\n- swift-tox-func-encryption http://logs.openstack.org/00/657700/4/check/swift-tox-func-encryption/4069909/ : SUCCESS in 8m 46s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/00/657700/4/check/swift-tox-func-domain-remap-staticweb/6c4e4c4/ : SUCCESS in 7m 51s\n- swift-tox-func-ec http://logs.openstack.org/00/657700/4/check/swift-tox-func-ec/fbeeb15/ : SUCCESS in 8m 34s\n- swift-tox-func-s3api http://logs.openstack.org/00/657700/4/check/swift-tox-func-s3api/b2af3fd/ : FAILURE in 18m 46s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/00/657700/4/check/swift-tox-func-s3api-ceph-s3tests-tempauth/695a0f4/ : SUCCESS in 11m 10s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/00/657700/4/check/swift-probetests-centos-7/131e8ac/ : SUCCESS in 30m 46s\n- swift-dsvm-functional http://logs.openstack.org/00/657700/4/check/swift-dsvm-functional/89cb641/ : SUCCESS in 26m 17s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/00/657700/4/check/swift-dsvm-functional-ipv6/f926b70/ : SUCCESS in 29m 14s\n- swift-tox-lower-constraints http://logs.openstack.org/00/657700/4/check/swift-tox-lower-constraints/bf1b539/ : FAILURE in 12m 57s\n- openstack-tox-pep8 http://logs.openstack.org/00/657700/4/check/openstack-tox-pep8/5146b91/ : SUCCESS in 5m 24s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/00/657700/4/check/swift-multinode-rolling-upgrade/547149f/ : FAILURE in 11m 32s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"dd27574ff7bec0f3efec36e7ec6ca5f9c6183bf0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-09 19:42:50.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"a096cc356f49c35ef8238fcfd2f7bc19524bdcac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 22:11:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/00/657700/4/check/openstack-tox-docs/1ed4290/html/ : SUCCESS in 3m 55s\n- build-openstack-releasenotes http://logs.openstack.org/00/657700/4/check/build-openstack-releasenotes/ccf7ddb/html/ : SUCCESS in 6m 52s\n- swift-build-image http://logs.openstack.org/00/657700/4/check/swift-build-image/fe4269a/ : SUCCESS in 9m 06s\n- swift-tox-py27 http://logs.openstack.org/00/657700/4/check/swift-tox-py27/57e7b3f/ : SUCCESS in 11m 40s\n- swift-tox-py36 http://logs.openstack.org/00/657700/4/check/swift-tox-py36/58b0915/ : SUCCESS in 9m 35s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/00/657700/4/check/swift-tox-py37/1750647/ : SUCCESS in 13m 41s\n- swift-tox-func http://logs.openstack.org/00/657700/4/check/swift-tox-func/1c81c80/ : SUCCESS in 10m 43s\n- swift-tox-func-encryption http://logs.openstack.org/00/657700/4/check/swift-tox-func-encryption/dab3714/ : SUCCESS in 10m 03s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/00/657700/4/check/swift-tox-func-domain-remap-staticweb/2269dcf/ : SUCCESS in 8m 17s\n- swift-tox-func-ec http://logs.openstack.org/00/657700/4/check/swift-tox-func-ec/842cfa2/ : SUCCESS in 8m 49s\n- swift-tox-func-s3api http://logs.openstack.org/00/657700/4/check/swift-tox-func-s3api/e37868d/ : SUCCESS in 9m 42s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/00/657700/4/check/swift-tox-func-s3api-ceph-s3tests-tempauth/8f8e9c0/ : SUCCESS in 12m 59s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/00/657700/4/check/swift-probetests-centos-7/f01d43a/ : SUCCESS in 32m 23s\n- swift-dsvm-functional http://logs.openstack.org/00/657700/4/check/swift-dsvm-functional/04ea42a/ : SUCCESS in 26m 17s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/00/657700/4/check/swift-dsvm-functional-ipv6/36d3e6a/ : SUCCESS in 25m 36s\n- swift-tox-lower-constraints http://logs.openstack.org/00/657700/4/check/swift-tox-lower-constraints/6785c90/ : SUCCESS in 12m 01s\n- openstack-tox-pep8 http://logs.openstack.org/00/657700/4/check/openstack-tox-pep8/7cd2623/ : SUCCESS in 5m 16s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/00/657700/4/check/swift-multinode-rolling-upgrade/e858f47/ : FAILURE in 8m 41s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"159142dc1f6c6b4ae386212a10324b47ab8a07e0","author":{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},"date":"2019-05-13 03:35:02.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"857ce5a800299869ea8cdcac58cd673e04e630db","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-05-14 03:37:13.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"001951a2961afc8304bb4abbae5455310bf54af6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 03:37:29.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"4ad8bc93c5b2b3e0db701751b55186f9113aa733","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 04:43:39.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/00/657700/4/gate/openstack-tox-docs/6c7642b/html/ : SUCCESS in 4m 06s\n- build-openstack-releasenotes http://logs.openstack.org/00/657700/4/gate/build-openstack-releasenotes/f10c9fa/html/ : SUCCESS in 7m 09s\n- swift-upload-image http://logs.openstack.org/00/657700/4/gate/swift-upload-image/7618de1/ : SUCCESS in 9m 16s\n- swift-tox-py27 http://logs.openstack.org/00/657700/4/gate/swift-tox-py27/ccc9ca8/ : SUCCESS in 10m 26s\n- swift-tox-py37 http://logs.openstack.org/00/657700/4/gate/swift-tox-py37/e591d12/ : SUCCESS in 9m 35s\n- swift-tox-func http://logs.openstack.org/00/657700/4/gate/swift-tox-func/a87f92d/ : SUCCESS in 8m 23s\n- swift-tox-func-encryption http://logs.openstack.org/00/657700/4/gate/swift-tox-func-encryption/3680f46/ : SUCCESS in 8m 50s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/00/657700/4/gate/swift-tox-func-domain-remap-staticweb/7232ffa/ : SUCCESS in 9m 49s\n- swift-tox-func-ec http://logs.openstack.org/00/657700/4/gate/swift-tox-func-ec/2e04590/ : SUCCESS in 9m 39s\n- swift-tox-func-s3api http://logs.openstack.org/00/657700/4/gate/swift-tox-func-s3api/b747f44/ : SUCCESS in 9m 57s\n- swift-probetests-centos-7 http://logs.openstack.org/00/657700/4/gate/swift-probetests-centos-7/f552200/ : SUCCESS in 30m 05s\n- swift-dsvm-functional http://logs.openstack.org/00/657700/4/gate/swift-dsvm-functional/fa39f9d/ : SUCCESS in 27m 33s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/00/657700/4/gate/swift-dsvm-functional-ipv6/8caf06e/ : SUCCESS in 28m 58s\n- swift-tox-lower-constraints http://logs.openstack.org/00/657700/4/gate/swift-tox-lower-constraints/0b5037c/ : SUCCESS in 12m 44s\n- openstack-tox-pep8 http://logs.openstack.org/00/657700/4/gate/openstack-tox-pep8/08ab2a7/ : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":4},{"id":"29289274c91038d677c68a96d6e47164334ae60c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 04:43:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"8b3d0a6c64d9efd6a7af753972a910144a494134","revisions":{"4bed8489bdada4c3f9f8a66212ebc5b1f8ee75ce":{"kind":"REWORK","_number":1,"created":"2019-05-07 22:45:18.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/00/657700/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/657700/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/657700/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/657700/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/657700/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/657700/1"}}},"commit":{"parents":[{"commit":"f94e7a05ba46c4ca152b41cf5f24a4e18d734f7f","subject":"py3: port unit/proxy/test_server.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f94e7a05ba46c4ca152b41cf5f24a4e18d734f7f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-05-07 22:43:34.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-05-07 22:44:50.000000000","tz":-420},"subject":"py3: finish porting proxy/test_server.py","message":"py3: finish porting proxy/test_server.py\n\nChange-Id: I8287db75b4f19581203360c646e72f64fe45f170\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4bed8489bdada4c3f9f8a66212ebc5b1f8ee75ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4bed8489bdada4c3f9f8a66212ebc5b1f8ee75ce"}]},"branch":"refs/heads/master"},"501b923aecfdcc209d4757d3ceb102d328713992":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-05-07 23:13:30.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/00/657700/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/657700/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/657700/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/657700/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/657700/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/657700/2"}}},"commit":{"parents":[{"commit":"8d1893f41b9169302b59ede3adca27c61bfe8bb8","subject":"py3: port unit/proxy/test_server.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8d1893f41b9169302b59ede3adca27c61bfe8bb8"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-05-07 22:43:34.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-05-07 23:06:52.000000000","tz":-420},"subject":"py3: finish porting proxy/test_server.py","message":"py3: finish porting proxy/test_server.py\n\nChange-Id: I8287db75b4f19581203360c646e72f64fe45f170\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/501b923aecfdcc209d4757d3ceb102d328713992"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/501b923aecfdcc209d4757d3ceb102d328713992"}]},"branch":"refs/heads/master"},"eb750900eb6c93bbb36de2de0953b3421fe23e7d":{"kind":"REWORK","_number":3,"created":"2019-05-08 22:59:11.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/00/657700/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/657700/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/657700/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/657700/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/657700/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/657700/3"}}},"commit":{"parents":[{"commit":"259224f00903443ca5154a2cf697abea13947fa1","subject":"py3: port unit/proxy/test_server.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/259224f00903443ca5154a2cf697abea13947fa1"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-05-07 22:43:34.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-05-08 22:59:06.000000000","tz":-420},"subject":"py3: finish porting proxy/test_server.py","message":"py3: finish porting proxy/test_server.py\n\nChange-Id: I8287db75b4f19581203360c646e72f64fe45f170\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eb750900eb6c93bbb36de2de0953b3421fe23e7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eb750900eb6c93bbb36de2de0953b3421fe23e7d"}]},"branch":"refs/heads/master"},"8b3d0a6c64d9efd6a7af753972a910144a494134":{"kind":"REWORK","_number":4,"created":"2019-05-09 00:47:49.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/00/657700/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/657700/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/657700/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/657700/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/657700/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/657700/4"}}},"commit":{"parents":[{"commit":"259224f00903443ca5154a2cf697abea13947fa1","subject":"py3: port unit/proxy/test_server.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/259224f00903443ca5154a2cf697abea13947fa1"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-05-07 22:43:34.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-05-09 00:47:40.000000000","tz":-420},"subject":"py3: finish porting proxy/test_server.py","message":"py3: finish porting proxy/test_server.py\n\nChange-Id: I8287db75b4f19581203360c646e72f64fe45f170\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8b3d0a6c64d9efd6a7af753972a910144a494134"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8b3d0a6c64d9efd6a7af753972a910144a494134"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
