)]}'
{"id":"openstack%2Fswift~658444","triplet_id":"openstack%2Fswift~master~If93325f2651a02f98f9d480c10bf7b849cc9617e","project":"openstack/swift","branch":"master","topic":"658444","hashtags":[],"change_id":"If93325f2651a02f98f9d480c10bf7b849cc9617e","subject":"Make ceph tests more portable","status":"MERGED","created":"2019-05-10 18:23:33.000000000","updated":"2019-09-25 00:34:10.000000000","submitted":"2019-09-25 00:31:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658444-1569371511767-17048323","meta_rev_id":"1adc9a6a52381f7759410a8c92848aef17f9824d","_number":658444,"virtual_id_number":658444,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-25 00:31:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-19 22:33:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2019-09-24 22:57:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-19 22:33:40.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2019-09-25 00:31:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e4dd0144c3fffe40825520ef7c23495d06b21455","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-05-10 18:23:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8de73d5867ba5ace4189ea32856a5f2dee16334f","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-05-10 18:28:54.000000000","message":"Patch Set 1:\n\n(4 comments)\n\nI\u0027m really curious to see what the gate does with these changes...","accounts_in_message":[],"_revision_number":1},{"id":"5eb7031a669919e1447b09e457c33f2e82d9beb4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 19:36:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/44/658444/1/check/openstack-tox-docs/8717d18/html/ : SUCCESS in 3m 59s\n- swift-build-image http://logs.openstack.org/44/658444/1/check/swift-build-image/5548062/ : SUCCESS in 8m 02s\n- swift-tox-py27 http://logs.openstack.org/44/658444/1/check/swift-tox-py27/62c1ed1/ : SUCCESS in 10m 55s\n- swift-tox-py36 http://logs.openstack.org/44/658444/1/check/swift-tox-py36/a8c950b/ : SUCCESS in 8m 16s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/44/658444/1/check/swift-tox-py37/7884847/ : SUCCESS in 8m 53s\n- swift-tox-func http://logs.openstack.org/44/658444/1/check/swift-tox-func/a9bacb3/ : SUCCESS in 8m 44s\n- swift-tox-func-encryption http://logs.openstack.org/44/658444/1/check/swift-tox-func-encryption/5df48fb/ : SUCCESS in 9m 18s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/44/658444/1/check/swift-tox-func-domain-remap-staticweb/51a6117/ : SUCCESS in 8m 49s\n- swift-tox-func-ec http://logs.openstack.org/44/658444/1/check/swift-tox-func-ec/767601c/ : SUCCESS in 8m 09s\n- swift-tox-func-s3api http://logs.openstack.org/44/658444/1/check/swift-tox-func-s3api/cd5049b/ : SUCCESS in 12m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/44/658444/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/8b45a5e/ : POST_FAILURE in 3m 46s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/44/658444/1/check/swift-probetests-centos-7/779e6e3/ : FAILURE in 33m 05s\n- swift-dsvm-functional http://logs.openstack.org/44/658444/1/check/swift-dsvm-functional/d3df545/ : SUCCESS in 27m 53s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/44/658444/1/check/swift-dsvm-functional-ipv6/bad4f42/ : SUCCESS in 27m 42s\n- swift-tox-lower-constraints http://logs.openstack.org/44/658444/1/check/swift-tox-lower-constraints/d055651/ : SUCCESS in 15m 42s\n- openstack-tox-pep8 http://logs.openstack.org/44/658444/1/check/openstack-tox-pep8/eea7a8e/ : SUCCESS in 5m 26s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/44/658444/1/check/swift-multinode-rolling-upgrade/ecb2a82/ : FAILURE in 8m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5bbbe88325b50429e86e9990fb727d847c022ff6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-10 22:37:09.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"2b5c65bf3a739ed8d8fda2ed16b6d13f2b9e8f38","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-14 16:51:37.000000000","message":"Patch Set 1:\n\nKnow what I realized I really want now? Run these with the dsvm setup, so we get Keystone coverage, too. We used to have that for swift3, but I never got around to doing it here...","accounts_in_message":[],"_revision_number":1},{"id":"89ade778b8d9a72260fb91dc7abc5b7b11faaf9e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-09-13 22:33:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3c0b59b65599e7a2338ea39cf64da8f641166ee4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-09-13 22:34:42.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"9c2ae4560948a2d2700db0e36ea12a55b01c9b79","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-09-13 23:23:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"89e8fb1039d1bd84e5f6cde60ad60d0209ffc3bc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 04:39:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8f02365923b4b67a621254a6c202ffb : SUCCESS in 4m 23s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/dfc236d2498145c49535351dd8aa4525 : SUCCESS in 1h 00m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7d6386512a174cd1b749608c9714e0f0 : SUCCESS in 55m 42s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1139d280f68b42708cc5ab6bd41c4577 : SUCCESS in 9m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bd508b6322ff49cca4cadfa8fa2afd28 : SUCCESS in 12m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/861cae9d11464fddb76f94e9be44ed46 : SUCCESS in 11m 52s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/92b7e0d8a1a144939c3f8ea2ecd4fcc4 : SUCCESS in 14m 01s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/d3137708a4064d418dc379432a74c1bd : SUCCESS in 9m 49s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/dacfeca54fe54767a24ac39f2b4d2311 : SUCCESS in 14m 49s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/b12d5f520714498f86b7dcf22bc83f91 : SUCCESS in 9m 11s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/9479227eb2594e14bbbd6889dd172e83 : SUCCESS in 11m 13s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/fd81fba7db84494f8a5895776fa2f4c1 : SUCCESS in 13m 18s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/3141c870ddec4e04b761358ceb9aa49d : SUCCESS in 10m 13s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/b2cd23af72e84ba5a0f33916b6c065db : SUCCESS in 13m 06s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/4a66716b73bd496fa763bcd4e0ecb81d : SUCCESS in 10m 25s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/46689662c42548f69ee0d6b7aaad0d42 : SUCCESS in 10m 28s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/aaa65dfdcdd54c94af9c468a2c925f52 : SUCCESS in 14m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f068e62b0c364b79893771422a6393e8 : SUCCESS in 18m 00s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2c99229c61f8441d9e96262b31dc62d3 : SUCCESS in 32m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fc4ac19fce874560958ea306afe27d16 : SUCCESS in 31m 14s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/d8dd5acc9c9744bf86fd0499dab56d8a : SUCCESS in 30m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/eb389b203d664b0a936c0601d9433ecf : SUCCESS in 27m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/730734d26cba486cb7114bc57e9be4c9 : SUCCESS in 11m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89d831679bd34b57b43c24652109cb1f : SUCCESS in 5m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/639dbaac65894ebb987105a413569bd4 : SUCCESS in 16m 41s","accounts_in_message":[],"_revision_number":2},{"id":"a017525b2868225951b26f21c2588ff59798d583","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-09-19 22:33:40.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"350f79f83f2a399a8dcebec58801c3dba7204339","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-09-24 22:57:47.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0b02d43156efdab38c7b55e4ad0805b812daa16f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 22:58:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7947c80b05840ad72ef613576fe244a02d40cc60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 00:31:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74e50698fb9a4196a33f9a4495306f1f : SUCCESS in 5m 10s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/25dedc49962e490580e4b75d3efdc63a : SUCCESS in 1h 19m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c3340f49a9084197bebd0e84e23a53a3 : SUCCESS in 53m 57s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/ccd36d497c924977b774cb67362acc9a : SUCCESS in 12m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/38e0a4d164044a4c81374f88a258d858 : SUCCESS in 12m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f4624cd56a4e4d33b3b4d248dcecc59d : SUCCESS in 11m 38s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f6a35dd75c33488383eb4cad05a8e9e4 : SUCCESS in 12m 11s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/f07b4c0e22fd43eaa7a9f1113a4e684a : SUCCESS in 10m 14s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/f3a9ca7e14c544709173720de9de05f9 : SUCCESS in 9m 45s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/02baf2b1054e46f4ad30d31442035833 : SUCCESS in 10m 03s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/cdc8f4c5265c49bc898ca0f8f39d3aa5 : SUCCESS in 10m 28s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/56b225a90a2e431288b4145ed679389d : SUCCESS in 13m 31s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/ba7aec456e6a479a831bd301f074f6aa : SUCCESS in 10m 55s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/22bf5165bde5461482f9a46baaeac2b9 : SUCCESS in 11m 09s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/11ffe82f993046719ed60370b9aa665e : SUCCESS in 10m 04s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/6b35428c28024b4a9b78bdb7a16250c6 : SUCCESS in 13m 51s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a444211367224f6793fb5242474c4427 : SUCCESS in 31m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/47abc9e2efeb4cb9bf778a85c915d2a7 : SUCCESS in 28m 51s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/668f2330eda94a8399796cd4bd8227ef : SUCCESS in 29m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/059ab1c9bf6b4a83b1598e47c5af4eff : SUCCESS in 30m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/635cdc3b50944bfda3ed33aa60356709 : SUCCESS in 11m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4593a1ec2e242fcb8b117d489908afb : SUCCESS in 5m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/db25081c0bf84a3182ee2c77d9dcfc1e : SUCCESS in 1h 00m 37s","accounts_in_message":[],"_revision_number":2},{"id":"2ba0827bc9ca2d8bc008936b9c51b5cf3b5c5667","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 00:31:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"b824dcf43b8c8cb7a7bf79042dbd2cc4d8f482e8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 00:34:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08049366f12742e684270f97c765bd4b : SUCCESS in 1m 10s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/1b51f1fa6c2c4908bf16e4eb6577350a : SUCCESS in 43s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"db8b0b6bc46a67b03af415d4e5e1429cc7d73bba","revisions":{"3e4154b87a206efed3a6825550633e3557e255a6":{"kind":"REWORK","_number":1,"created":"2019-05-10 18:23:33.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/44/658444/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/44/658444/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/658444/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/658444/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/658444/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/44/658444/1"}}},"commit":{"parents":[{"commit":"032cf3b3b4599bd720131c8f02cc6ed8a155683d","subject":"Update docker secret.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/032cf3b3b4599bd720131c8f02cc6ed8a155683d"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2019-05-10 18:15:42.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2019-05-10 18:16:33.000000000","tz":-300},"subject":"Make ceph tests more portable","message":"Make ceph tests more portable\n\nChange-Id: If93325f2651a02f98f9d480c10bf7b849cc9617e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3e4154b87a206efed3a6825550633e3557e255a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3e4154b87a206efed3a6825550633e3557e255a6"}]},"branch":"refs/heads/master"},"db8b0b6bc46a67b03af415d4e5e1429cc7d73bba":{"kind":"REWORK","_number":2,"created":"2019-09-13 22:33:23.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/44/658444/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/44/658444/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/658444/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/658444/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/658444/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/44/658444/2"}}},"commit":{"parents":[{"commit":"5cb53838a63791109697c91f742079b6dfb25263","subject":"Merge \"py3: (mostly) port probe tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5cb53838a63791109697c91f742079b6dfb25263"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2019-05-10 18:15:42.000000000","tz":-300},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-09-13 22:32:58.000000000","tz":-420},"subject":"Make ceph tests more portable","message":"Make ceph tests more portable\n\nChange-Id: If93325f2651a02f98f9d480c10bf7b849cc9617e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/db8b0b6bc46a67b03af415d4e5e1429cc7d73bba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/db8b0b6bc46a67b03af415d4e5e1429cc7d73bba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
