)]}'
{"id":"openstack%2Fswift~659108","triplet_id":"openstack%2Fswift~feature%2Flosf~I983ca653701e34010bb3e75a78d3d69cb6cb21ee","project":"openstack/swift","branch":"feature/losf","topic":"alecuyer/no_time","hashtags":[],"change_id":"I983ca653701e34010bb3e75a78d3d69cb6cb21ee","subject":"Use eventlet.sleep rather than time.sleep in objectrpcmanager.py","status":"MERGED","created":"2019-05-14 15:49:38.000000000","updated":"2019-05-14 23:04:54.000000000","submitted":"2019-05-14 23:04:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659108-1557875094198-0de1d12d","meta_rev_id":"6e573459b50de7fec74f8557ed8fc96fb2c2f258","_number":659108,"virtual_id_number":659108,"owner":{"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-05-14 23:04:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-14 19:54:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-14 19:54:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-14 19:54:03.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2019-05-14 23:04:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a916df2ed966d1aa9adda034b4f38e3affce77f6","author":{"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},"date":"2019-05-14 15:49:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8a003f0cf9b9892cee9b150c3d24cf2a4ac7392b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 16:29:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/08/659108/1/check/openstack-tox-docs/688df21/html/ : SUCCESS in 4m 11s\n- swift-tox-py27 http://logs.openstack.org/08/659108/1/check/swift-tox-py27/7b339fa/ : SUCCESS in 13m 50s\n- swift-tox-py35 http://logs.openstack.org/08/659108/1/check/swift-tox-py35/78bd9b9/ : SUCCESS in 9m 52s (non-voting)\n- swift-tox-py36 http://logs.openstack.org/08/659108/1/check/swift-tox-py36/c7208a3/ : SUCCESS in 8m 07s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/08/659108/1/check/swift-tox-py37/46c6ac7/ : SUCCESS in 10m 36s\n- swift-tox-func http://logs.openstack.org/08/659108/1/check/swift-tox-func/ca3fa06/ : SUCCESS in 9m 29s\n- swift-tox-func-encryption http://logs.openstack.org/08/659108/1/check/swift-tox-func-encryption/24126b0/ : SUCCESS in 9m 53s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/08/659108/1/check/swift-tox-func-domain-remap-staticweb/823ed2b/ : SUCCESS in 8m 03s\n- swift-tox-func-ec http://logs.openstack.org/08/659108/1/check/swift-tox-func-ec/3be6b73/ : SUCCESS in 9m 49s\n- swift-tox-func-s3api http://logs.openstack.org/08/659108/1/check/swift-tox-func-s3api/5a9c943/ : SUCCESS in 10m 20s\n- swift-tox-func-losf http://logs.openstack.org/08/659108/1/check/swift-tox-func-losf/59495f2/ : SUCCESS in 13m 16s (non-voting)\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/08/659108/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/7e16e30/ : SUCCESS in 13m 00s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/08/659108/1/check/swift-probetests-centos-7/7eb5882/ : SUCCESS in 33m 22s\n- swift-dsvm-functional http://logs.openstack.org/08/659108/1/check/swift-dsvm-functional/d3fcef5/ : SUCCESS in 27m 37s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/08/659108/1/check/swift-dsvm-functional-ipv6/4d48572/ : SUCCESS in 25m 54s\n- swift-tox-lower-constraints http://logs.openstack.org/08/659108/1/check/swift-tox-lower-constraints/4964e6a/ : SUCCESS in 11m 46s\n- openstack-tox-pep8 http://logs.openstack.org/08/659108/1/check/openstack-tox-pep8/28bb2dc/ : SUCCESS in 5m 42s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/08/659108/1/check/swift-multinode-rolling-upgrade/9e09a53/ : FAILURE in 8m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a489e119dec020b1ddf56539b332ccf1d7b5bc9e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-14 19:54:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThat must surely be better.","accounts_in_message":[],"_revision_number":1},{"id":"3ed091f601aa153d8ae2cd4303f6c9fec114883c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 19:54:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e265563010845695c93fc5ffc98d0a45ce7cf0e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 23:04:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/08/659108/1/gate/openstack-tox-docs/c3fb27a/html/ : SUCCESS in 4m 04s\n- swift-tox-py27 http://logs.openstack.org/08/659108/1/gate/swift-tox-py27/82aa872/ : SUCCESS in 9m 52s\n- swift-tox-py37 http://logs.openstack.org/08/659108/1/gate/swift-tox-py37/635bdae/ : SUCCESS in 8m 12s\n- swift-tox-func http://logs.openstack.org/08/659108/1/gate/swift-tox-func/ecd8bf8/ : SUCCESS in 8m 28s\n- swift-tox-func-encryption http://logs.openstack.org/08/659108/1/gate/swift-tox-func-encryption/425d37e/ : SUCCESS in 10m 14s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/08/659108/1/gate/swift-tox-func-domain-remap-staticweb/77ee867/ : SUCCESS in 8m 51s\n- swift-tox-func-ec http://logs.openstack.org/08/659108/1/gate/swift-tox-func-ec/4e1897e/ : SUCCESS in 10m 09s\n- swift-tox-func-s3api http://logs.openstack.org/08/659108/1/gate/swift-tox-func-s3api/742ff11/ : SUCCESS in 12m 04s\n- swift-probetests-centos-7 http://logs.openstack.org/08/659108/1/gate/swift-probetests-centos-7/a507659/ : SUCCESS in 31m 53s\n- swift-dsvm-functional http://logs.openstack.org/08/659108/1/gate/swift-dsvm-functional/8d02bd7/ : SUCCESS in 24m 31s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/08/659108/1/gate/swift-dsvm-functional-ipv6/983c2e6/ : SUCCESS in 25m 39s\n- swift-tox-lower-constraints http://logs.openstack.org/08/659108/1/gate/swift-tox-lower-constraints/ebb68a9/ : SUCCESS in 11m 10s\n- openstack-tox-pep8 http://logs.openstack.org/08/659108/1/gate/openstack-tox-pep8/579cf09/ : SUCCESS in 6m 11s","accounts_in_message":[],"_revision_number":1},{"id":"200d2c9cd51f9ca4a9426cd231ddc38f8dc73c8d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 23:04:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"287dd340f32077002a50fd2a2a967eab902c5ac1","revisions":{"287dd340f32077002a50fd2a2a967eab902c5ac1":{"kind":"REWORK","_number":1,"created":"2019-05-14 15:49:38.000000000","uploader":{"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},"ref":"refs/changes/08/659108/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/659108/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/659108/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/659108/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/659108/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/659108/1"}}},"commit":{"parents":[{"commit":"9d980e2a11f359338e27da74893546447db8c032","subject":"Setup losf functional test with saio setup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9d980e2a11f359338e27da74893546447db8c032"}]}],"author":{"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","date":"2019-05-14 15:49:10.000000000","tz":120},"committer":{"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","date":"2019-05-14 15:49:36.000000000","tz":120},"subject":"Use eventlet.sleep rather than time.sleep in objectrpcmanager.py","message":"Use eventlet.sleep rather than time.sleep in objectrpcmanager.py\n\nChange-Id: I983ca653701e34010bb3e75a78d3d69cb6cb21ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/287dd340f32077002a50fd2a2a967eab902c5ac1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/287dd340f32077002a50fd2a2a967eab902c5ac1"}]},"branch":"refs/heads/feature/losf"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
