)]}'
{"id":"openstack%2Fswift~659390","triplet_id":"openstack%2Fswift~master~Ie73f9df460f9014014168d6e60767e66b24d30b0","project":"openstack/swift","branch":"master","topic":"py3-obj","hashtags":[],"change_id":"Ie73f9df460f9014014168d6e60767e66b24d30b0","subject":"py3: port diskfile","status":"MERGED","created":"2019-05-15 22:35:07.000000000","updated":"2019-05-24 02:39:59.000000000","submitted":"2019-05-24 02:38:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659390-1558665489239-7d69f64c","meta_rev_id":"61d36c1e5d263981a0f8549d5af7b16cd29f3f7e","_number":659390,"virtual_id_number":659390,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":2,"date":"2019-05-24 02:38:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-22 20:26:40.000000000","_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2019-05-23 17:27:20.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":1,"date":"2019-05-23 17:27:20.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-23 17:27:20.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2019-05-24 02:38:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"62f53064319fd40091819c2b02aa748f3d6ca7ff","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-15 22:35:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d57d3ba2ba07bbe0318867645354c0ec2472ce9c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-15 22:50:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b452046e5adae0f6f236055e12734f51fe08a50b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 23:40:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/90/659390/2/check/openstack-tox-docs/3fc4118/html/ : SUCCESS in 4m 06s\n- build-openstack-releasenotes http://logs.openstack.org/90/659390/2/check/build-openstack-releasenotes/5f3fcff/html/ : SUCCESS in 7m 02s\n- swift-build-image swift-build-image : FAILURE\n- swift-tox-py27 http://logs.openstack.org/90/659390/2/check/swift-tox-py27/19b81c9/ : SUCCESS in 9m 57s\n- swift-tox-py36 http://logs.openstack.org/90/659390/2/check/swift-tox-py36/44c7643/ : SUCCESS in 9m 34s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/90/659390/2/check/swift-tox-py37/2d3cdf1/ : SUCCESS in 10m 44s\n- swift-tox-func http://logs.openstack.org/90/659390/2/check/swift-tox-func/f242898/ : SUCCESS in 9m 04s\n- swift-tox-func-encryption http://logs.openstack.org/90/659390/2/check/swift-tox-func-encryption/4f34e33/ : SUCCESS in 9m 00s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/90/659390/2/check/swift-tox-func-domain-remap-staticweb/ad3209a/ : SUCCESS in 8m 29s\n- swift-tox-func-ec http://logs.openstack.org/90/659390/2/check/swift-tox-func-ec/f37c399/ : SUCCESS in 8m 54s\n- swift-tox-func-s3api http://logs.openstack.org/90/659390/2/check/swift-tox-func-s3api/968c1b4/ : SUCCESS in 9m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/90/659390/2/check/swift-tox-func-s3api-ceph-s3tests-tempauth/c42ddc1/ : SUCCESS in 11m 50s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/90/659390/2/check/swift-probetests-centos-7/dcbf4c5/ : FAILURE in 28m 49s\n- swift-dsvm-functional http://logs.openstack.org/90/659390/2/check/swift-dsvm-functional/cf2bbd8/ : SUCCESS in 28m 24s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/90/659390/2/check/swift-dsvm-functional-ipv6/c02b1c2/ : SUCCESS in 42m 25s\n- swift-tox-lower-constraints http://logs.openstack.org/90/659390/2/check/swift-tox-lower-constraints/24a68b3/ : SUCCESS in 11m 38s\n- openstack-tox-pep8 http://logs.openstack.org/90/659390/2/check/openstack-tox-pep8/d7b729c/ : SUCCESS in 5m 30s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/90/659390/2/check/swift-multinode-rolling-upgrade/8e0271d/ : FAILURE in 7m 51s (non-voting)\n\nWarning:\n  Requirements [\u0027docker-image\u0027] not met by build a9326f1c430948b9844c7925a9db347a","accounts_in_message":[],"_revision_number":2},{"id":"5937962082519c39b1a34b96e8f682524f67fe7e","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-05-16 02:12:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ff2b18077a39593a3c27ae58268044c2b34575fb","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-16 04:29:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b39607b48717aa010b6a0a22e2b049777754f98a","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-05-16 04:41:51.000000000","message":"Patch Set 2: Code-Review+2\n\nwell, then","accounts_in_message":[],"_revision_number":2},{"id":"9a7307e3460f59d58ab7e0bf4eec4abd24b23fe8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-16 04:48:09.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ebc2a53e588b6c34d575f4f5175f609ad2de6e34","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 05:22:39.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/90/659390/2/check/openstack-tox-docs/aa5b4dd/html/ : SUCCESS in 4m 29s\n- build-openstack-releasenotes http://logs.openstack.org/90/659390/2/check/build-openstack-releasenotes/683ae86/html/ : SUCCESS in 6m 41s\n- swift-build-image swift-build-image : FAILURE\n- swift-tox-py27 http://logs.openstack.org/90/659390/2/check/swift-tox-py27/31a4ea8/ : SUCCESS in 10m 49s\n- swift-tox-py36 http://logs.openstack.org/90/659390/2/check/swift-tox-py36/d156a9f/ : SUCCESS in 15m 28s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/90/659390/2/check/swift-tox-py37/d166d7a/ : SUCCESS in 10m 04s\n- swift-tox-func http://logs.openstack.org/90/659390/2/check/swift-tox-func/f04f8e9/ : SUCCESS in 8m 14s\n- swift-tox-func-encryption http://logs.openstack.org/90/659390/2/check/swift-tox-func-encryption/16cb2f6/ : SUCCESS in 15m 22s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/90/659390/2/check/swift-tox-func-domain-remap-staticweb/ec4a818/ : SUCCESS in 8m 14s\n- swift-tox-func-ec http://logs.openstack.org/90/659390/2/check/swift-tox-func-ec/09632f0/ : SUCCESS in 8m 58s\n- swift-tox-func-s3api http://logs.openstack.org/90/659390/2/check/swift-tox-func-s3api/e7dc4cf/ : SUCCESS in 13m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/90/659390/2/check/swift-tox-func-s3api-ceph-s3tests-tempauth/a5533ee/ : SUCCESS in 11m 12s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/90/659390/2/check/swift-probetests-centos-7/bffe13e/ : SUCCESS in 29m 14s\n- swift-dsvm-functional http://logs.openstack.org/90/659390/2/check/swift-dsvm-functional/2ea3593/ : SUCCESS in 26m 27s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/90/659390/2/check/swift-dsvm-functional-ipv6/c91968c/ : SUCCESS in 27m 02s\n- swift-tox-lower-constraints http://logs.openstack.org/90/659390/2/check/swift-tox-lower-constraints/f9b3cc0/ : SUCCESS in 10m 51s\n- openstack-tox-pep8 http://logs.openstack.org/90/659390/2/check/openstack-tox-pep8/d79e847/ : SUCCESS in 5m 02s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/90/659390/2/check/swift-multinode-rolling-upgrade/dae999d/ : FAILURE in 7m 43s (non-voting)\n\nWarning:\n  Requirements [\u0027docker-image\u0027] not met by build a9326f1c430948b9844c7925a9db347a","accounts_in_message":[],"_revision_number":2},{"id":"a37ceddacb6206e2459a5e5b995d0a7b764bb340","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-17 15:08:39.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"202bf0933e9150216f7332bac22afdefe0daa366","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 16:08:34.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/90/659390/2/check/openstack-tox-docs/18a411a/html/ : SUCCESS in 4m 15s\n- build-openstack-releasenotes http://logs.openstack.org/90/659390/2/check/build-openstack-releasenotes/6ed7317/html/ : SUCCESS in 7m 11s\n- swift-build-image http://logs.openstack.org/90/659390/2/check/swift-build-image/a8abe93/ : SUCCESS in 8m 25s\n- swift-tox-py27 http://logs.openstack.org/90/659390/2/check/swift-tox-py27/08967fc/ : SUCCESS in 11m 50s\n- swift-tox-py36 http://logs.openstack.org/90/659390/2/check/swift-tox-py36/a8ccc86/ : SUCCESS in 9m 39s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/90/659390/2/check/swift-tox-py37/fdcdc5c/ : SUCCESS in 9m 57s\n- swift-tox-func http://logs.openstack.org/90/659390/2/check/swift-tox-func/2b73da2/ : SUCCESS in 10m 06s\n- swift-tox-func-encryption http://logs.openstack.org/90/659390/2/check/swift-tox-func-encryption/e954e5f/ : SUCCESS in 7m 48s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/90/659390/2/check/swift-tox-func-domain-remap-staticweb/075bc5d/ : SUCCESS in 8m 59s\n- swift-tox-func-ec http://logs.openstack.org/90/659390/2/check/swift-tox-func-ec/856ca55/ : SUCCESS in 9m 21s\n- swift-tox-func-s3api http://logs.openstack.org/90/659390/2/check/swift-tox-func-s3api/88861d9/ : SUCCESS in 11m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/90/659390/2/check/swift-tox-func-s3api-ceph-s3tests-tempauth/c3bf2a6/ : SUCCESS in 15m 01s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/90/659390/2/check/swift-probetests-centos-7/1361fe5/ : FAILURE in 28m 02s\n- swift-dsvm-functional http://logs.openstack.org/90/659390/2/check/swift-dsvm-functional/c1d8c7a/ : SUCCESS in 20m 52s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/90/659390/2/check/swift-dsvm-functional-ipv6/6b70ead/ : SUCCESS in 25m 22s\n- swift-tox-lower-constraints http://logs.openstack.org/90/659390/2/check/swift-tox-lower-constraints/cd080d5/ : SUCCESS in 11m 24s\n- openstack-tox-pep8 http://logs.openstack.org/90/659390/2/check/openstack-tox-pep8/345d75c/ : SUCCESS in 5m 19s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/90/659390/2/check/swift-multinode-rolling-upgrade/2754fa4/ : FAILURE in 9m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"69ca6e935b0b98d7e86985479e3c70890eea1d76","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-17 16:30:47.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"31a2ca760f2e85f36159c86f60c1f631c1adb69f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 17:29:25.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/90/659390/2/check/openstack-tox-docs/08b2014/html/ : SUCCESS in 3m 38s\n- build-openstack-releasenotes http://logs.openstack.org/90/659390/2/check/build-openstack-releasenotes/6ccb3f6/html/ : SUCCESS in 6m 11s\n- swift-build-image http://logs.openstack.org/90/659390/2/check/swift-build-image/c29482d/ : SUCCESS in 8m 40s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/90/659390/2/check/swift-tox-py27/404ce5a/ : SUCCESS in 13m 23s\n- swift-tox-py36 http://logs.openstack.org/90/659390/2/check/swift-tox-py36/0457c78/ : SUCCESS in 13m 25s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/90/659390/2/check/swift-tox-py37/abf9f2f/ : SUCCESS in 12m 59s\n- swift-tox-func http://logs.openstack.org/90/659390/2/check/swift-tox-func/fc0282c/ : SUCCESS in 10m 23s\n- swift-tox-func-encryption http://logs.openstack.org/90/659390/2/check/swift-tox-func-encryption/89ef783/ : SUCCESS in 11m 19s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/90/659390/2/check/swift-tox-func-domain-remap-staticweb/e40b5c5/ : SUCCESS in 8m 43s\n- swift-tox-func-ec http://logs.openstack.org/90/659390/2/check/swift-tox-func-ec/7cf0866/ : FAILURE in 12m 14s\n- swift-tox-func-s3api http://logs.openstack.org/90/659390/2/check/swift-tox-func-s3api/e3e55d2/ : SUCCESS in 9m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/90/659390/2/check/swift-tox-func-s3api-ceph-s3tests-tempauth/f74eb63/ : SUCCESS in 17m 39s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/90/659390/2/check/swift-probetests-centos-7/5bc3be6/ : SUCCESS in 33m 38s\n- swift-dsvm-functional http://logs.openstack.org/90/659390/2/check/swift-dsvm-functional/ed68dbd/ : SUCCESS in 34m 26s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/90/659390/2/check/swift-dsvm-functional-ipv6/8da0875/ : SUCCESS in 32m 33s\n- swift-tox-lower-constraints http://logs.openstack.org/90/659390/2/check/swift-tox-lower-constraints/c25e9e9/ : SUCCESS in 12m 09s\n- openstack-tox-pep8 http://logs.openstack.org/90/659390/2/check/openstack-tox-pep8/33ee94a/ : SUCCESS in 6m 19s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/90/659390/2/check/swift-multinode-rolling-upgrade/6849d2f/ : FAILURE in 9m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c8023eb0ec97087f88d1e6a8a69313064ca7779c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-17 19:19:51.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"023fe1d7ad1e797367ae6ef8f24649674175add5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 19:55:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/90/659390/2/check/openstack-tox-docs/9140940/html/ : SUCCESS in 3m 49s\n- build-openstack-releasenotes http://logs.openstack.org/90/659390/2/check/build-openstack-releasenotes/e990451/html/ : SUCCESS in 6m 20s\n- swift-build-image http://logs.openstack.org/90/659390/2/check/swift-build-image/5bfb777/ : SUCCESS in 8m 26s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/90/659390/2/check/swift-tox-py27/8a96df9/ : SUCCESS in 9m 38s\n- swift-tox-py36 http://logs.openstack.org/90/659390/2/check/swift-tox-py36/690e38c/ : SUCCESS in 19m 05s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/90/659390/2/check/swift-tox-py37/aef114f/ : SUCCESS in 11m 21s\n- swift-tox-func http://logs.openstack.org/90/659390/2/check/swift-tox-func/843b31a/ : SUCCESS in 8m 21s\n- swift-tox-func-encryption http://logs.openstack.org/90/659390/2/check/swift-tox-func-encryption/a99c12c/ : SUCCESS in 8m 09s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/90/659390/2/check/swift-tox-func-domain-remap-staticweb/443c786/ : SUCCESS in 8m 41s\n- swift-tox-func-ec http://logs.openstack.org/90/659390/2/check/swift-tox-func-ec/943e584/ : SUCCESS in 9m 26s\n- swift-tox-func-s3api http://logs.openstack.org/90/659390/2/check/swift-tox-func-s3api/918202e/ : SUCCESS in 9m 56s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/90/659390/2/check/swift-tox-func-s3api-ceph-s3tests-tempauth/fe6a6e8/ : SUCCESS in 11m 54s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/90/659390/2/check/swift-probetests-centos-7/24606bd/ : SUCCESS in 32m 28s\n- swift-dsvm-functional http://logs.openstack.org/90/659390/2/check/swift-dsvm-functional/03a6b1d/ : SUCCESS in 27m 11s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/90/659390/2/check/swift-dsvm-functional-ipv6/8f62363/ : SUCCESS in 26m 01s\n- swift-tox-lower-constraints http://logs.openstack.org/90/659390/2/check/swift-tox-lower-constraints/b6f7a91/ : SUCCESS in 14m 45s\n- openstack-tox-pep8 http://logs.openstack.org/90/659390/2/check/openstack-tox-pep8/ceac329/ : SUCCESS in 5m 10s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/90/659390/2/check/swift-multinode-rolling-upgrade/190d005/ : FAILURE in 9m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6ff2137fab6b50261e59df50428a48a9a30224af","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-22 20:26:06.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"172059e7d08f91c54f5b8d5ef5843417c4e766e1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-22 20:26:40.000000000","message":"Patch Set 3:\n\nJust rebased to resolve tox.ini merge conflict","accounts_in_message":[],"_revision_number":3},{"id":"faa201bc1d700b738baab7157c53e2d153f82690","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 06:18:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/90/659390/3/check/openstack-tox-docs/689fb01/html/ : SUCCESS in 4m 08s\n- build-openstack-releasenotes http://logs.openstack.org/90/659390/3/check/build-openstack-releasenotes/4d8d236/html/ : SUCCESS in 7m 58s\n- swift-build-image http://logs.openstack.org/90/659390/3/check/swift-build-image/f59b4cb/ : SUCCESS in 8m 42s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/90/659390/3/check/swift-tox-py27/13c2057/ : SUCCESS in 10m 37s\n- swift-tox-py36 http://logs.openstack.org/90/659390/3/check/swift-tox-py36/bc30729/ : SUCCESS in 10m 41s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/90/659390/3/check/swift-tox-py37/dc9c7a6/ : SUCCESS in 10m 43s\n- swift-tox-func http://logs.openstack.org/90/659390/3/check/swift-tox-func/42932ef/ : SUCCESS in 8m 37s\n- swift-tox-func-py37 http://logs.openstack.org/90/659390/3/check/swift-tox-func-py37/bcee959/ : SUCCESS in 10m 22s\n- swift-tox-func-encryption http://logs.openstack.org/90/659390/3/check/swift-tox-func-encryption/3fe08c6/ : SUCCESS in 9m 45s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/90/659390/3/check/swift-tox-func-domain-remap-staticweb/811a10f/ : SUCCESS in 13m 48s\n- swift-tox-func-ec http://logs.openstack.org/90/659390/3/check/swift-tox-func-ec/709d10c/ : SUCCESS in 12m 52s\n- swift-tox-func-s3api http://logs.openstack.org/90/659390/3/check/swift-tox-func-s3api/c50b95c/ : SUCCESS in 11m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/90/659390/3/check/swift-tox-func-s3api-ceph-s3tests-tempauth/2dfd17e/ : SUCCESS in 12m 39s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/90/659390/3/check/swift-probetests-centos-7/5ae2b4b/ : SUCCESS in 30m 01s\n- swift-dsvm-functional http://logs.openstack.org/90/659390/3/check/swift-dsvm-functional/7c980ae/ : SUCCESS in 25m 53s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/90/659390/3/check/swift-dsvm-functional-ipv6/44014ff/ : SUCCESS in 27m 23s\n- swift-tox-lower-constraints http://logs.openstack.org/90/659390/3/check/swift-tox-lower-constraints/b58d1e3/ : SUCCESS in 15m 42s\n- openstack-tox-pep8 http://logs.openstack.org/90/659390/3/check/openstack-tox-pep8/1ae1fab/ : SUCCESS in 6m 35s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/90/659390/3/check/swift-multinode-rolling-upgrade/7501f73/ : FAILURE in 8m 23s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"dc9ac20d70baa38233a9c94ca04ec6387c302414","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-05-23 17:27:20.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"0204781a1c73f68d8a68f0b7ae65fcf00a14ac5e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 21:46:25.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a11ef87354e0865cbbc2294e2f5c3214d8041be7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 02:38:08.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/90/659390/3/gate/openstack-tox-docs/29cdab5/html/ : SUCCESS in 5m 07s\n- build-openstack-releasenotes http://logs.openstack.org/90/659390/3/gate/build-openstack-releasenotes/d55be0e/html/ : SUCCESS in 7m 21s\n- swift-upload-image http://logs.openstack.org/90/659390/3/gate/swift-upload-image/6951110/ : SUCCESS in 9m 32s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/90/659390/3/gate/swift-tox-py27/ed663fb/ : SUCCESS in 10m 28s\n- swift-tox-py37 http://logs.openstack.org/90/659390/3/gate/swift-tox-py37/47cb463/ : SUCCESS in 10m 48s\n- swift-tox-func http://logs.openstack.org/90/659390/3/gate/swift-tox-func/4ea51d9/ : SUCCESS in 8m 48s\n- swift-tox-func-py37 http://logs.openstack.org/90/659390/3/gate/swift-tox-func-py37/750717f/ : SUCCESS in 7m 48s\n- swift-tox-func-encryption http://logs.openstack.org/90/659390/3/gate/swift-tox-func-encryption/05f2dbc/ : SUCCESS in 9m 11s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/90/659390/3/gate/swift-tox-func-domain-remap-staticweb/f19e7a8/ : SUCCESS in 8m 47s\n- swift-tox-func-ec http://logs.openstack.org/90/659390/3/gate/swift-tox-func-ec/270846e/ : SUCCESS in 10m 24s\n- swift-tox-func-s3api http://logs.openstack.org/90/659390/3/gate/swift-tox-func-s3api/0d7f5c1/ : SUCCESS in 11m 17s\n- swift-probetests-centos-7 http://logs.openstack.org/90/659390/3/gate/swift-probetests-centos-7/24f842e/ : SUCCESS in 35m 37s\n- swift-dsvm-functional http://logs.openstack.org/90/659390/3/gate/swift-dsvm-functional/de91cb2/ : SUCCESS in 26m 49s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/90/659390/3/gate/swift-dsvm-functional-ipv6/9ad35d2/ : SUCCESS in 27m 51s\n- swift-tox-lower-constraints http://logs.openstack.org/90/659390/3/gate/swift-tox-lower-constraints/80e7cf0/ : SUCCESS in 11m 13s\n- openstack-tox-pep8 http://logs.openstack.org/90/659390/3/gate/openstack-tox-pep8/5bd3fc3/ : SUCCESS in 6m 04s","accounts_in_message":[],"_revision_number":3},{"id":"283fdc87d38c8aff1aead0c79d5bf5d186ab69a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 02:38:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"c3f2b699deab047283c617c1873aec884a9d8834","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 02:39:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- swift-promote-image http://logs.openstack.org/90/659390/3/promote/swift-promote-image/b81a8d6/ : SUCCESS in 42s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"40bd9f16a348236a8710b98b66115f659c47354b","revisions":{"6a674def06c37a5b3706ac8378fab94158f2b481":{"kind":"REWORK","_number":1,"created":"2019-05-15 22:35:07.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/90/659390/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/659390/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/659390/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/659390/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/659390/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/659390/1"}}},"commit":{"parents":[{"commit":"a82590365eaf87ba35120e4cab46283a1393ada4","subject":"Clean up test_write_read_metadata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a82590365eaf87ba35120e4cab46283a1393ada4"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-05-15 22:29:08.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-05-15 22:29:08.000000000","tz":-420},"subject":"py3: port diskfile","message":"py3: port diskfile\n\nChange-Id: Ie73f9df460f9014014168d6e60767e66b24d30b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6a674def06c37a5b3706ac8378fab94158f2b481"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6a674def06c37a5b3706ac8378fab94158f2b481"}]},"branch":"refs/heads/master"},"7974e68213197791e833da01f1ff236f4f7f8de7":{"kind":"REWORK","_number":2,"created":"2019-05-15 22:50:29.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/90/659390/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/659390/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/659390/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/659390/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/659390/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/659390/2"}}},"commit":{"parents":[{"commit":"a82590365eaf87ba35120e4cab46283a1393ada4","subject":"Clean up test_write_read_metadata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a82590365eaf87ba35120e4cab46283a1393ada4"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-05-15 22:29:08.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-05-15 22:50:26.000000000","tz":-420},"subject":"py3: port diskfile","message":"py3: port diskfile\n\nChange-Id: Ie73f9df460f9014014168d6e60767e66b24d30b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7974e68213197791e833da01f1ff236f4f7f8de7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7974e68213197791e833da01f1ff236f4f7f8de7"}]},"branch":"refs/heads/master"},"40bd9f16a348236a8710b98b66115f659c47354b":{"kind":"REWORK","_number":3,"created":"2019-05-22 20:26:06.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/90/659390/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/659390/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/659390/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/659390/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/659390/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/659390/3"}}},"commit":{"parents":[{"commit":"8868e2cd5062541f2d9b3d5c2091e4122a830f83","subject":"Clean up test_write_read_metadata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8868e2cd5062541f2d9b3d5c2091e4122a830f83"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-05-15 22:29:08.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-05-22 20:25:54.000000000","tz":-420},"subject":"py3: port diskfile","message":"py3: port diskfile\n\nChange-Id: Ie73f9df460f9014014168d6e60767e66b24d30b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/40bd9f16a348236a8710b98b66115f659c47354b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/40bd9f16a348236a8710b98b66115f659c47354b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
