)]}'
{"id":"openstack%2Fswift~659614","triplet_id":"openstack%2Fswift~master~I19d23570ff3a084d288de1308e059cfd8134d6ad","project":"openstack/swift","branch":"master","topic":"gerrit-master","hashtags":[],"change_id":"I19d23570ff3a084d288de1308e059cfd8134d6ad","subject":"Specify pickle protocol in REPLICATE()","status":"MERGED","created":"2019-05-16 16:14:58.000000000","updated":"2019-05-23 09:26:56.000000000","submitted":"2019-05-23 09:25:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659614-1558603546138-0763d481","meta_rev_id":"8ec7f21c2e1950e72fb0c55c9900553b93a1856b","_number":659614,"virtual_id_number":659614,"owner":{"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},{"value":0,"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},{"value":2,"date":"2019-05-23 09:25:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},{"value":0,"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-22 20:13:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},{"value":0,"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-22 20:13:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-22 02:29:02.000000000","updated_by":{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},"reviewer":{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},"state":"REVIEWER"},{"updated":"2019-05-22 20:13:40.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2019-05-23 09:25:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"77412150c97375aea6a152e30f5a610181f8b067","author":{"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},"date":"2019-05-16 16:14:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb447594152368b363dba3d146c24afad6d83219","author":{"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},"date":"2019-05-16 16:18:17.000000000","message":"Patch Set 1:\n\nFollowing our discussion at the PTG, I think we want object servers to work with differing python versions?\n\nAs you suggested I have setup a small cluster, with py2 nodes, and a single py3 node. That\u0027s the first issue I came across (also the BoundedSemaphore issue, but I think you know about it).\n\npickle protocol 2 is supported from python 2.3 and higher. I don\u0027t know if we have a lower bound on the py2 version for swift?","accounts_in_message":[],"_revision_number":1},{"id":"62aea96887d00e6c451a780cd92862132300117c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 18:37:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/14/659614/1/check/openstack-tox-docs/7c984ce/html/ : SUCCESS in 4m 11s\n- swift-build-image http://logs.openstack.org/14/659614/1/check/swift-build-image/399b68e/ : SUCCESS in 9m 36s\n- swift-tox-py27 http://logs.openstack.org/14/659614/1/check/swift-tox-py27/5aa23e5/ : SUCCESS in 9m 48s\n- swift-tox-py36 http://logs.openstack.org/14/659614/1/check/swift-tox-py36/091a4a9/ : SUCCESS in 10m 39s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/14/659614/1/check/swift-tox-py37/5a49745/ : SUCCESS in 10m 35s\n- swift-tox-func http://logs.openstack.org/14/659614/1/check/swift-tox-func/239f2be/ : SUCCESS in 16m 00s\n- swift-tox-func-encryption http://logs.openstack.org/14/659614/1/check/swift-tox-func-encryption/074350b/ : SUCCESS in 12m 28s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/14/659614/1/check/swift-tox-func-domain-remap-staticweb/bbae61e/ : SUCCESS in 9m 32s\n- swift-tox-func-ec http://logs.openstack.org/14/659614/1/check/swift-tox-func-ec/5d61d0a/ : SUCCESS in 10m 42s\n- swift-tox-func-s3api http://logs.openstack.org/14/659614/1/check/swift-tox-func-s3api/23b0672/ : SUCCESS in 9m 57s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/14/659614/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/ac47c1e/ : SUCCESS in 12m 12s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/14/659614/1/check/swift-probetests-centos-7/f6389f3/ : SUCCESS in 31m 01s\n- swift-dsvm-functional http://logs.openstack.org/14/659614/1/check/swift-dsvm-functional/2a74af3/ : SUCCESS in 31m 25s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/14/659614/1/check/swift-dsvm-functional-ipv6/917df96/ : SUCCESS in 27m 04s\n- swift-tox-lower-constraints http://logs.openstack.org/14/659614/1/check/swift-tox-lower-constraints/4083acb/ : SUCCESS in 12m 05s\n- openstack-tox-pep8 http://logs.openstack.org/14/659614/1/check/openstack-tox-pep8/45793eb/ : FAILURE in 5m 07s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/14/659614/1/check/swift-multinode-rolling-upgrade/e7d020c/ : FAILURE in 8m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d46ee6b341bd861694af98e59dc99e883d10926c","author":{"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},"date":"2019-05-17 07:41:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9d9be32699d3dd6ba124af5a89d18f46372779c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 08:17:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/14/659614/2/check/openstack-tox-docs/9128a48/html/ : SUCCESS in 5m 25s\n- swift-build-image http://logs.openstack.org/14/659614/2/check/swift-build-image/d77f318/ : SUCCESS in 10m 32s\n- swift-tox-py27 http://logs.openstack.org/14/659614/2/check/swift-tox-py27/f75f824/ : SUCCESS in 11m 05s\n- swift-tox-py36 http://logs.openstack.org/14/659614/2/check/swift-tox-py36/89cc622/ : FAILURE in 21m 55s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/14/659614/2/check/swift-tox-py37/784a035/ : SUCCESS in 12m 21s\n- swift-tox-func http://logs.openstack.org/14/659614/2/check/swift-tox-func/b3ea257/ : SUCCESS in 8m 56s\n- swift-tox-func-encryption http://logs.openstack.org/14/659614/2/check/swift-tox-func-encryption/89fcab3/ : SUCCESS in 9m 01s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/14/659614/2/check/swift-tox-func-domain-remap-staticweb/1e242c8/ : SUCCESS in 7m 49s\n- swift-tox-func-ec http://logs.openstack.org/14/659614/2/check/swift-tox-func-ec/35c0325/ : SUCCESS in 8m 32s\n- swift-tox-func-s3api http://logs.openstack.org/14/659614/2/check/swift-tox-func-s3api/2a0d5d0/ : SUCCESS in 9m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/14/659614/2/check/swift-tox-func-s3api-ceph-s3tests-tempauth/cf87901/ : SUCCESS in 10m 48s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/14/659614/2/check/swift-probetests-centos-7/e6e53d1/ : SUCCESS in 30m 03s\n- swift-dsvm-functional http://logs.openstack.org/14/659614/2/check/swift-dsvm-functional/4774ea9/ : SUCCESS in 28m 03s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/14/659614/2/check/swift-dsvm-functional-ipv6/e0c89dd/ : SUCCESS in 26m 30s\n- swift-tox-lower-constraints http://logs.openstack.org/14/659614/2/check/swift-tox-lower-constraints/22df68b/ : SUCCESS in 11m 50s\n- openstack-tox-pep8 http://logs.openstack.org/14/659614/2/check/openstack-tox-pep8/24d5c30/ : SUCCESS in 5m 09s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/14/659614/2/check/swift-multinode-rolling-upgrade/90f7334/ : FAILURE in 8m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fcf9f32c628a4e0feb8ac665dedbe179b8cb8930","author":{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},"date":"2019-05-22 02:29:02.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"65da3155457588ddba9ebf1b8bed18e18eafc770","author":{"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},"date":"2019-05-22 12:11:31.000000000","message":"Patch Set 2:\n\nOne thing, I don\u0027t know how long the pickle 2 format will be supported.","accounts_in_message":[],"_revision_number":2},{"id":"c8b994f02409bc1ea8b7d83ad4b769e57206bbf0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-22 16:22:39.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(2 comments)\n\n\u003e One thing, I don\u0027t know how long the pickle 2 format will be supported.\n\nGiven how we use pickle to record not only db pending entries but also object metadata, that\u0027s a truly terrifying proposition. Fortunately, I see no pending deprecation or telegraphed plans for removal of *any* protocol version on https://docs.python.org/3/library/pickle.html, so I think (hope?) we\u0027re in the clear. They haven\u0027t even dropped support for the pre-2.3 protocols!","accounts_in_message":[],"_revision_number":2},{"id":"55d0fb7c9bc8290c0c769763f68358376f020432","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 16:23:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"25781fa3852f41a740ede194087f1acba43c2d48","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-22 16:30:57.000000000","message":"Patch Set 2:\n\nWhoa! I just realize, while py 3.0+ defaults to protocol three, py2 defaulted to *zero* -- you have to *opt in* to the binary format. From https://docs.python.org/2.7/library/pickle.html :\n\n\u003e If a protocol is not specified, protocol 0 is used.\n\nThe binary format should be fine, though.","accounts_in_message":[],"_revision_number":2},{"id":"b28d89fb37d47248a6ac944f0654fcae7f8756ed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 20:02:41.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/14/659614/2/gate/openstack-tox-docs/20234b4/html/ : SUCCESS in 6m 33s\n- swift-upload-image http://logs.openstack.org/14/659614/2/gate/swift-upload-image/d0f0652/ : SUCCESS in 12m 31s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/14/659614/2/gate/swift-tox-py27/04c2bf0/ : SUCCESS in 13m 24s\n- swift-tox-py37 http://logs.openstack.org/14/659614/2/gate/swift-tox-py37/9729312/ : SUCCESS in 14m 56s\n- swift-tox-func http://logs.openstack.org/14/659614/2/gate/swift-tox-func/f464afa/ : SUCCESS in 12m 00s\n- swift-tox-func-py37 http://logs.openstack.org/14/659614/2/gate/swift-tox-func-py37/91345e1/ : SUCCESS in 11m 20s\n- swift-tox-func-encryption http://logs.openstack.org/14/659614/2/gate/swift-tox-func-encryption/1819c10/ : SUCCESS in 11m 34s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/14/659614/2/gate/swift-tox-func-domain-remap-staticweb/768ae94/ : SUCCESS in 10m 06s\n- swift-tox-func-ec http://logs.openstack.org/14/659614/2/gate/swift-tox-func-ec/5eabb58/ : SUCCESS in 26m 30s\n- swift-tox-func-s3api http://logs.openstack.org/14/659614/2/gate/swift-tox-func-s3api/513d22f/ : SUCCESS in 15m 28s\n- swift-probetests-centos-7 http://logs.openstack.org/14/659614/2/gate/swift-probetests-centos-7/13add26/ : FAILURE in 26m 27s\n- swift-dsvm-functional http://logs.openstack.org/14/659614/2/gate/swift-dsvm-functional/0c85c49/ : SUCCESS in 29m 40s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/14/659614/2/gate/swift-dsvm-functional-ipv6/a6c4249/ : SUCCESS in 31m 12s\n- swift-tox-lower-constraints http://logs.openstack.org/14/659614/2/gate/swift-tox-lower-constraints/961ceac/ : SUCCESS in 14m 43s\n- openstack-tox-pep8 http://logs.openstack.org/14/659614/2/gate/openstack-tox-pep8/fcd5044/ : SUCCESS in 7m 16s","accounts_in_message":[],"_revision_number":2},{"id":"270f4a500faedd02b7a9a2717b774ff3f05c4e95","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-22 20:12:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b69b8a216dfaf0e8c788e3a8d6d674c6cb0a5af1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-22 20:13:40.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nSquashed in the (already approved) follow-up since we hit the gate failure.","accounts_in_message":[],"_revision_number":3},{"id":"00c577feb9054f8febeff49a8ce9e79ac9e68f8c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 06:02:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/14/659614/3/check/openstack-tox-docs/cb90ce2/html/ : SUCCESS in 4m 32s\n- swift-build-image http://logs.openstack.org/14/659614/3/check/swift-build-image/9ff9187/ : SUCCESS in 8m 26s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/14/659614/3/check/swift-tox-py27/2ef1030/ : SUCCESS in 13m 42s\n- swift-tox-py36 http://logs.openstack.org/14/659614/3/check/swift-tox-py36/f1ea743/ : SUCCESS in 12m 11s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/14/659614/3/check/swift-tox-py37/38faf76/ : SUCCESS in 12m 09s\n- swift-tox-func http://logs.openstack.org/14/659614/3/check/swift-tox-func/2267951/ : SUCCESS in 23m 02s\n- swift-tox-func-py37 http://logs.openstack.org/14/659614/3/check/swift-tox-func-py37/780600f/ : SUCCESS in 8m 53s\n- swift-tox-func-encryption http://logs.openstack.org/14/659614/3/check/swift-tox-func-encryption/deeb4a7/ : SUCCESS in 9m 05s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/14/659614/3/check/swift-tox-func-domain-remap-staticweb/c475306/ : SUCCESS in 10m 08s\n- swift-tox-func-ec http://logs.openstack.org/14/659614/3/check/swift-tox-func-ec/3226773/ : SUCCESS in 14m 00s\n- swift-tox-func-s3api http://logs.openstack.org/14/659614/3/check/swift-tox-func-s3api/83562fc/ : SUCCESS in 15m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/14/659614/3/check/swift-tox-func-s3api-ceph-s3tests-tempauth/35a121e/ : SUCCESS in 12m 16s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/14/659614/3/check/swift-probetests-centos-7/607f083/ : SUCCESS in 29m 51s\n- swift-dsvm-functional http://logs.openstack.org/14/659614/3/check/swift-dsvm-functional/b4bb138/ : SUCCESS in 33m 19s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/14/659614/3/check/swift-dsvm-functional-ipv6/5f23017/ : SUCCESS in 26m 26s\n- swift-tox-lower-constraints http://logs.openstack.org/14/659614/3/check/swift-tox-lower-constraints/335b369/ : SUCCESS in 11m 30s\n- openstack-tox-pep8 http://logs.openstack.org/14/659614/3/check/openstack-tox-pep8/9ba5470/ : SUCCESS in 5m 17s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/14/659614/3/check/swift-multinode-rolling-upgrade/9b0bc16/ : FAILURE in 7m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3a16405ea26dc29fc909a2f72f5d2c833fa17bb1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 06:03:03.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"76664cd5cfb59852005581e0f2cabb18bd1887e5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 09:25:45.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/14/659614/3/gate/openstack-tox-docs/34dcdfe/html/ : SUCCESS in 3m 59s\n- swift-upload-image http://logs.openstack.org/14/659614/3/gate/swift-upload-image/3297dbb/ : SUCCESS in 9m 09s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/14/659614/3/gate/swift-tox-py27/450eb80/ : SUCCESS in 10m 52s\n- swift-tox-py37 http://logs.openstack.org/14/659614/3/gate/swift-tox-py37/aaa9494/ : SUCCESS in 9m 58s\n- swift-tox-func http://logs.openstack.org/14/659614/3/gate/swift-tox-func/f3b1b52/ : SUCCESS in 8m 53s\n- swift-tox-func-py37 http://logs.openstack.org/14/659614/3/gate/swift-tox-func-py37/eba71d5/ : SUCCESS in 7m 04s\n- swift-tox-func-encryption http://logs.openstack.org/14/659614/3/gate/swift-tox-func-encryption/deb88b9/ : SUCCESS in 8m 29s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/14/659614/3/gate/swift-tox-func-domain-remap-staticweb/faa2e84/ : SUCCESS in 9m 01s\n- swift-tox-func-ec http://logs.openstack.org/14/659614/3/gate/swift-tox-func-ec/871b14e/ : SUCCESS in 9m 34s\n- swift-tox-func-s3api http://logs.openstack.org/14/659614/3/gate/swift-tox-func-s3api/bd7dea7/ : SUCCESS in 12m 46s\n- swift-probetests-centos-7 http://logs.openstack.org/14/659614/3/gate/swift-probetests-centos-7/f537104/ : SUCCESS in 34m 04s\n- swift-dsvm-functional http://logs.openstack.org/14/659614/3/gate/swift-dsvm-functional/e211ad3/ : SUCCESS in 25m 43s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/14/659614/3/gate/swift-dsvm-functional-ipv6/e39dc13/ : SUCCESS in 27m 03s\n- swift-tox-lower-constraints http://logs.openstack.org/14/659614/3/gate/swift-tox-lower-constraints/7c911e8/ : SUCCESS in 11m 51s\n- openstack-tox-pep8 http://logs.openstack.org/14/659614/3/gate/openstack-tox-pep8/0927bbe/ : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":3},{"id":"6c86a799a8824047b2a00eb7abae1c068fd20ccb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 09:25:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"f6e8f7b4314bb197a93b85f36873647fc8f35822","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 09:26:56.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- swift-promote-image http://logs.openstack.org/14/659614/3/promote/swift-promote-image/b2d9e7c/ : SUCCESS in 46s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4927b1f29c4d993cb29786312a9efc00850c10cb","revisions":{"ddcc0f4dabb9adbdaac81ebe3b1a14109314bb31":{"kind":"REWORK","_number":1,"created":"2019-05-16 16:14:58.000000000","uploader":{"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},"ref":"refs/changes/14/659614/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/659614/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/659614/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/659614/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/659614/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/659614/1"}}},"commit":{"parents":[{"commit":"51e415b749cc85757805053931679c1cd61730f3","subject":"py3: port proxy/test_mem_server.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/51e415b749cc85757805053931679c1cd61730f3"}]}],"author":{"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","date":"2019-05-16 16:08:03.000000000","tz":120},"committer":{"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","date":"2019-05-16 16:14:39.000000000","tz":120},"subject":"Specify pickle protocol in REPLICATE()","message":"Specify pickle protocol in REPLICATE()\n\nThe default pickle protocol in python3 is version 3. This is not\nreadable by a python2 interpreter.\n\nForce the use of version 2 in the object server REPLICATE() function,\nfor compatibility with python 2.\n\nChange-Id: I19d23570ff3a084d288de1308e059cfd8134d6ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ddcc0f4dabb9adbdaac81ebe3b1a14109314bb31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ddcc0f4dabb9adbdaac81ebe3b1a14109314bb31"}]},"branch":"refs/heads/master"},"0d186af49692815c1683d87ca8f27905ffd715e4":{"kind":"REWORK","_number":2,"created":"2019-05-17 07:41:36.000000000","uploader":{"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},"ref":"refs/changes/14/659614/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/659614/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/659614/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/659614/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/659614/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/659614/2"}}},"commit":{"parents":[{"commit":"51e415b749cc85757805053931679c1cd61730f3","subject":"py3: port proxy/test_mem_server.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/51e415b749cc85757805053931679c1cd61730f3"}]}],"author":{"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","date":"2019-05-16 16:08:03.000000000","tz":120},"committer":{"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","date":"2019-05-17 07:41:23.000000000","tz":120},"subject":"Specify pickle protocol in REPLICATE()","message":"Specify pickle protocol in REPLICATE()\n\nThe default pickle protocol in python3 is version 3. This is not\nreadable by a python2 interpreter.\n\nForce the use of version 2 in the object server REPLICATE() function,\nfor compatibility with python 2.\n\nChange-Id: I19d23570ff3a084d288de1308e059cfd8134d6ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0d186af49692815c1683d87ca8f27905ffd715e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0d186af49692815c1683d87ca8f27905ffd715e4"}]},"branch":"refs/heads/master"},"4927b1f29c4d993cb29786312a9efc00850c10cb":{"kind":"REWORK","_number":3,"created":"2019-05-22 20:12:43.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/14/659614/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/659614/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/659614/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/659614/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/659614/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/659614/3"}}},"commit":{"parents":[{"commit":"b4c8cf192a39004645e2608d4cfc0ac5dcf3b516","subject":"Merge \"Fix quarantine when object path is not a directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b4c8cf192a39004645e2608d4cfc0ac5dcf3b516"}]}],"author":{"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","date":"2019-05-16 16:08:03.000000000","tz":120},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-05-22 20:12:32.000000000","tz":-420},"subject":"Specify pickle protocol in REPLICATE()","message":"Specify pickle protocol in REPLICATE()\n\nThe default pickle protocol in python3 is version 3. This is not\nreadable by a python2 interpreter.\n\nForce the use of version 2 in the object server REPLICATE() function,\nfor compatibility with python 2.\n\nChange-Id: I19d23570ff3a084d288de1308e059cfd8134d6ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4927b1f29c4d993cb29786312a9efc00850c10cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4927b1f29c4d993cb29786312a9efc00850c10cb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
