)]}'
{"id":"openstack%2Fswift~659851","triplet_id":"openstack%2Fswift~master~I7d0345a712f71e76034910fa7bb2bbaf42b4d2ff","project":"openstack/swift","branch":"master","topic":"py3-func-tests","hashtags":[],"change_id":"I7d0345a712f71e76034910fa7bb2bbaf42b4d2ff","subject":"Be explicit that a code path is just for py2","status":"MERGED","created":"2019-05-17 17:30:32.000000000","updated":"2019-05-18 17:50:22.000000000","submitted":"2019-05-18 17:49:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659851-1558201763858-deb1f29f","meta_rev_id":"cc986b8308c04f4e5dfaffa96c8a947350f03e81","_number":659851,"virtual_id_number":659851,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":2,"date":"2019-05-18 17:49:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-18 16:38:30.000000000","_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2019-05-17 18:00:38.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":1,"date":"2019-05-17 18:00:38.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-17 18:00:38.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2019-05-18 17:49:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"70e4373f6bc08a4b312ffde1fe4c963110a5e580","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-17 17:30:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"364d4f70a5095fc1c8d05b23bda9e4e5d1d60bb4","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-05-17 18:00:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c85cc6a288b941f78abb25feaff99858c653f6d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 18:02:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/51/659851/1/check/openstack-tox-docs/e1fe010/html/ : SUCCESS in 4m 15s\n- swift-tox-func http://logs.openstack.org/51/659851/1/check/swift-tox-func/32342d9/ : SUCCESS in 8m 40s\n- swift-tox-func-encryption http://logs.openstack.org/51/659851/1/check/swift-tox-func-encryption/dc77d2a/ : SUCCESS in 10m 55s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/51/659851/1/check/swift-tox-func-domain-remap-staticweb/8dfba54/ : SUCCESS in 9m 29s\n- swift-tox-func-ec http://logs.openstack.org/51/659851/1/check/swift-tox-func-ec/66e0070/ : SUCCESS in 11m 28s\n- swift-tox-func-s3api http://logs.openstack.org/51/659851/1/check/swift-tox-func-s3api/418d035/ : SUCCESS in 8m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/51/659851/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/457295c/ : SUCCESS in 11m 39s (non-voting)\n- swift-dsvm-functional http://logs.openstack.org/51/659851/1/check/swift-dsvm-functional/40d3824/ : SUCCESS in 26m 39s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/51/659851/1/check/swift-dsvm-functional-ipv6/0c3f72d/ : SUCCESS in 25m 48s\n- openstack-tox-pep8 http://logs.openstack.org/51/659851/1/check/openstack-tox-pep8/2c9c1fd/ : SUCCESS in 5m 39s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/51/659851/1/check/swift-multinode-rolling-upgrade/0191a04/ : FAILURE in 9m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ef20e5c2f8dcf9d74df8a13abc922bf22975a594","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 20:48:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9909c13dd25071fb063de0bb939cbd8bd88cc370","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 22:49:09.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"4852b140b6d2426ffbbcb9434bed81fbf09facb7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-18 16:38:30.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6ea63e89c310b79b11103ea675f1826d2f496fcc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-18 17:15:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/51/659851/1/check/openstack-tox-docs/843dd69/html/ : SUCCESS in 3m 55s\n- swift-tox-func http://logs.openstack.org/51/659851/1/check/swift-tox-func/d43fb15/ : SUCCESS in 17m 37s\n- swift-tox-func-encryption http://logs.openstack.org/51/659851/1/check/swift-tox-func-encryption/d276bd0/ : SUCCESS in 17m 33s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/51/659851/1/check/swift-tox-func-domain-remap-staticweb/8587efe/ : SUCCESS in 17m 51s\n- swift-tox-func-ec http://logs.openstack.org/51/659851/1/check/swift-tox-func-ec/4c52205/ : SUCCESS in 18m 46s\n- swift-tox-func-s3api http://logs.openstack.org/51/659851/1/check/swift-tox-func-s3api/67c5af2/ : SUCCESS in 20m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/51/659851/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/b35f6c2/ : SUCCESS in 23m 18s (non-voting)\n- swift-dsvm-functional http://logs.openstack.org/51/659851/1/check/swift-dsvm-functional/9a1f7d4/ : SUCCESS in 33m 07s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/51/659851/1/check/swift-dsvm-functional-ipv6/caf39e1/ : SUCCESS in 25m 48s\n- openstack-tox-pep8 http://logs.openstack.org/51/659851/1/check/openstack-tox-pep8/86a669b/ : SUCCESS in 6m 35s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/51/659851/1/check/swift-multinode-rolling-upgrade/82fd8f1/ : FAILURE in 8m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"63650e9c2aa919ae2219fe5348e3bc020920c573","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-18 17:16:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"395bbfc7a0b33606d3b1f2924ae9a6d979c732ee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-18 17:49:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"947ca0fe229998634cf93ab76022f0ab0585177b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-18 17:49:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/51/659851/1/gate/openstack-tox-docs/c4c76af/html/ : SUCCESS in 3m 43s\n- swift-upload-image http://logs.openstack.org/51/659851/1/gate/swift-upload-image/8d99025/ : SUCCESS in 7m 40s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/51/659851/1/gate/swift-tox-py27/bacc95b/ : SUCCESS in 12m 12s\n- swift-tox-py37 http://logs.openstack.org/51/659851/1/gate/swift-tox-py37/2a56c18/ : SUCCESS in 9m 48s\n- swift-tox-func http://logs.openstack.org/51/659851/1/gate/swift-tox-func/aa62c3b/ : SUCCESS in 8m 34s\n- swift-tox-func-encryption http://logs.openstack.org/51/659851/1/gate/swift-tox-func-encryption/d59d879/ : SUCCESS in 9m 43s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/51/659851/1/gate/swift-tox-func-domain-remap-staticweb/ecbee50/ : SUCCESS in 8m 48s\n- swift-tox-func-ec http://logs.openstack.org/51/659851/1/gate/swift-tox-func-ec/7f20acb/ : SUCCESS in 9m 09s\n- swift-tox-func-s3api http://logs.openstack.org/51/659851/1/gate/swift-tox-func-s3api/3e0debc/ : SUCCESS in 10m 01s\n- swift-dsvm-functional http://logs.openstack.org/51/659851/1/gate/swift-dsvm-functional/7814a44/ : SUCCESS in 27m 21s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/51/659851/1/gate/swift-dsvm-functional-ipv6/abf2eef/ : SUCCESS in 32m 03s\n- openstack-tox-pep8 http://logs.openstack.org/51/659851/1/gate/openstack-tox-pep8/7e9846b/ : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":1},{"id":"f5101d163f2c98cfee9e376906d609cf40b95bf4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-18 17:50:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- swift-promote-image http://logs.openstack.org/51/659851/1/promote/swift-promote-image/6124c71/ : SUCCESS in 40s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ad034023f30f4d00e9c21bf3e095d3c04b4b7609","revisions":{"ad034023f30f4d00e9c21bf3e095d3c04b4b7609":{"kind":"REWORK","_number":1,"created":"2019-05-17 17:30:32.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/51/659851/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/51/659851/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/659851/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/659851/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/659851/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/51/659851/1"}}},"commit":{"parents":[{"commit":"506279235d12f1590bd928de51d991b7fb425e6e","subject":"Get functional/tests.py running under py3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/506279235d12f1590bd928de51d991b7fb425e6e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-05-17 17:30:21.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-05-17 17:30:21.000000000","tz":-420},"subject":"Be explicit that a code path is just for py2","message":"Be explicit that a code path is just for py2\n\nChange-Id: I7d0345a712f71e76034910fa7bb2bbaf42b4d2ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ad034023f30f4d00e9c21bf3e095d3c04b4b7609"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ad034023f30f4d00e9c21bf3e095d3c04b4b7609"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
