)]}'
{"id":"openstack%2Fswift~660836","triplet_id":"openstack%2Fswift~master~Ic70eb61b32b196d356c9a3b40a22807456c7b0f2","project":"openstack/swift","branch":"master","topic":"async-delete","hashtags":[],"change_id":"Ic70eb61b32b196d356c9a3b40a22807456c7b0f2","subject":"Add more tests for async expirer","status":"MERGED","created":"2019-05-22 21:28:04.000000000","updated":"2019-05-23 09:27:57.000000000","submitted":"2019-05-23 09:25:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660836-1558603550057-2a496124","meta_rev_id":"af55b34aa0f721c213b2149b8cc1316f2f1130d1","_number":660836,"virtual_id_number":660836,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-05-23 09:25:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-22 21:48:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-22 21:48:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-22 21:48:12.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2019-05-23 09:25:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d60cc56262d743ace4aebcafa198652f154e384e","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-05-22 21:28:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9f4f4d087f42eeb8cd9c30eb8c2c46b0fd0be4d5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-22 21:48:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(3 comments)\n\nYeah, Swift is better with these.","accounts_in_message":[],"_revision_number":1},{"id":"539bde83e4064b15bd6d97fe7eeffc3c90cbffd9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 06:25:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/36/660836/1/check/openstack-tox-docs/0f00fa1/html/ : SUCCESS in 4m 00s\n- swift-build-image http://logs.openstack.org/36/660836/1/check/swift-build-image/d1d4fa7/ : SUCCESS in 8m 14s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/36/660836/1/check/swift-tox-py27/f710634/ : SUCCESS in 14m 54s\n- swift-tox-py36 http://logs.openstack.org/36/660836/1/check/swift-tox-py36/19b6e30/ : SUCCESS in 13m 50s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/36/660836/1/check/swift-tox-py37/8c8fcfe/ : SUCCESS in 9m 05s\n- swift-tox-func http://logs.openstack.org/36/660836/1/check/swift-tox-func/52f6e36/ : SUCCESS in 8m 20s\n- swift-tox-func-py37 http://logs.openstack.org/36/660836/1/check/swift-tox-func-py37/0b11f1c/ : SUCCESS in 8m 01s\n- swift-tox-func-encryption http://logs.openstack.org/36/660836/1/check/swift-tox-func-encryption/7aca4ac/ : SUCCESS in 8m 51s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/36/660836/1/check/swift-tox-func-domain-remap-staticweb/f29d998/ : SUCCESS in 8m 27s\n- swift-tox-func-ec http://logs.openstack.org/36/660836/1/check/swift-tox-func-ec/ac2f67c/ : SUCCESS in 9m 49s\n- swift-tox-func-s3api http://logs.openstack.org/36/660836/1/check/swift-tox-func-s3api/e9e5b35/ : SUCCESS in 11m 01s\n- swift-dsvm-functional http://logs.openstack.org/36/660836/1/check/swift-dsvm-functional/2540890/ : SUCCESS in 27m 20s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/36/660836/1/check/swift-dsvm-functional-ipv6/9b14012/ : SUCCESS in 25m 43s\n- swift-tox-lower-constraints http://logs.openstack.org/36/660836/1/check/swift-tox-lower-constraints/0135f08/ : SUCCESS in 11m 54s\n- openstack-tox-pep8 http://logs.openstack.org/36/660836/1/check/openstack-tox-pep8/9f5698e/ : SUCCESS in 5m 38s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/36/660836/1/check/swift-multinode-rolling-upgrade/2eca488/ : FAILURE in 9m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1aad9aded421b11f41227e076f2eba14703e12ff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 06:26:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"be86d78ab30c36e8fd1ddee271d0dbc6d0d280bd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 09:25:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/36/660836/1/gate/openstack-tox-docs/da52838/html/ : SUCCESS in 4m 01s\n- swift-upload-image http://logs.openstack.org/36/660836/1/gate/swift-upload-image/28a2cb3/ : SUCCESS in 11m 30s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/36/660836/1/gate/swift-tox-py27/a1086e5/ : SUCCESS in 9m 38s\n- swift-tox-py37 http://logs.openstack.org/36/660836/1/gate/swift-tox-py37/fec5f30/ : SUCCESS in 10m 30s\n- swift-tox-func http://logs.openstack.org/36/660836/1/gate/swift-tox-func/0957f02/ : SUCCESS in 8m 17s\n- swift-tox-func-py37 http://logs.openstack.org/36/660836/1/gate/swift-tox-func-py37/f4cf438/ : SUCCESS in 8m 36s\n- swift-tox-func-encryption http://logs.openstack.org/36/660836/1/gate/swift-tox-func-encryption/8b99c84/ : SUCCESS in 10m 00s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/36/660836/1/gate/swift-tox-func-domain-remap-staticweb/5901099/ : SUCCESS in 10m 02s\n- swift-tox-func-ec http://logs.openstack.org/36/660836/1/gate/swift-tox-func-ec/cc95ddf/ : SUCCESS in 19m 42s\n- swift-tox-func-s3api http://logs.openstack.org/36/660836/1/gate/swift-tox-func-s3api/56dbb86/ : SUCCESS in 33m 45s\n- swift-dsvm-functional http://logs.openstack.org/36/660836/1/gate/swift-dsvm-functional/b953cf8/ : SUCCESS in 37m 21s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/36/660836/1/gate/swift-dsvm-functional-ipv6/59da1c4/ : SUCCESS in 27m 57s\n- swift-tox-lower-constraints http://logs.openstack.org/36/660836/1/gate/swift-tox-lower-constraints/0122ade/ : SUCCESS in 12m 16s\n- openstack-tox-pep8 http://logs.openstack.org/36/660836/1/gate/openstack-tox-pep8/4d73b12/ : SUCCESS in 7m 00s","accounts_in_message":[],"_revision_number":1},{"id":"024a186da9a94455689db3c783e50a756a43d795","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 09:25:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"bc4d0d238604d272bf55535c23c30693ad2b61bd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 09:27:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- swift-promote-image http://logs.openstack.org/36/660836/1/promote/swift-promote-image/5e33fd0/ : SUCCESS in 50s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"04c8015b889e3ddcbf111f3b0fa99bbbe12b8624","revisions":{"04c8015b889e3ddcbf111f3b0fa99bbbe12b8624":{"kind":"REWORK","_number":1,"created":"2019-05-22 21:28:04.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/36/660836/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/36/660836/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/660836/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/660836/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/660836/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/36/660836/1"}}},"commit":{"parents":[{"commit":"83d01619910c34f2e3fafeda20bc4f7e7e8f482f","subject":"Add operator tool to async-delete some or all objects in a container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/83d01619910c34f2e3fafeda20bc4f7e7e8f482f"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2019-05-22 21:23:42.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2019-05-22 21:24:05.000000000","tz":-300},"subject":"Add more tests for async expirer","message":"Add more tests for async expirer\n\nRelated-Change-Id: Ia13ee5da3d1b5c536eccaadc7a6fdcd997374443\nChange-Id: Ic70eb61b32b196d356c9a3b40a22807456c7b0f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/04c8015b889e3ddcbf111f3b0fa99bbbe12b8624"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/04c8015b889e3ddcbf111f3b0fa99bbbe12b8624"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
