)]}'
{"id":"openstack%2Fswift~660869","triplet_id":"openstack%2Fswift~master~Ib1bc4c59ed4c7624782b8b1cdea20847c14c90bf","project":"openstack/swift","branch":"master","topic":"fix_url","hashtags":[],"change_id":"Ib1bc4c59ed4c7624782b8b1cdea20847c14c90bf","subject":"Modify the version path URL of the configuration file","status":"MERGED","created":"2019-05-23 03:01:33.000000000","updated":"2019-05-29 23:34:44.000000000","submitted":"2019-05-29 23:32:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660869-1559172744081-45b1259d","meta_rev_id":"9c89e6054ddaca9fe2930544d1962a6596eda110","_number":660869,"virtual_id_number":660869,"owner":{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-05-29 23:32:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-29 22:17:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-29 22:17:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-29 22:17:08.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2019-05-29 23:32:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3b878371577a159004ba8a0f98527fd6828848d5","author":{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},"date":"2019-05-23 03:01:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a4edfee2b9d6bf6a0ee445811e8986454d56bc98","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 08:05:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/69/660869/1/check/openstack-tox-docs/33c1f5c/html/ : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":1},{"id":"7be1e0c2f9dd607746b340ff4b2a2b79d59a90f9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-29 22:17:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"03fe9bea9bb942e24b12f92a0cbda6911c8da4e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 22:18:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b5bbfeaf3953ebb8951d628fcebe294179e71521","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 23:32:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/69/660869/1/gate/openstack-tox-docs/999d371/html/ : SUCCESS in 4m 27s\n- swift-upload-image http://logs.openstack.org/69/660869/1/gate/swift-upload-image/926bd39/ : SUCCESS in 9m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"445ee5e4889714ce54603411a7bc62c04c50ca54","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 23:32:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"82be169967e7d492d344c2ff8f5966b29b5655e3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 23:34:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- swift-promote-image http://logs.openstack.org/69/660869/1/promote/swift-promote-image/6eb60e1/ : SUCCESS in 46s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2c4f284055e1031f17f97229056616017385d8bc","revisions":{"2c4f284055e1031f17f97229056616017385d8bc":{"kind":"REWORK","_number":1,"created":"2019-05-23 03:01:33.000000000","uploader":{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},"ref":"refs/changes/69/660869/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/69/660869/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/660869/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/660869/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/660869/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/69/660869/1"}}},"commit":{"parents":[{"commit":"1adde2fba52b248440c0fb51ddacfefb3d7543ae","subject":"Merge \"Controller install Debian missing proxy-server.conf configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1adde2fba52b248440c0fb51ddacfefb3d7543ae"}]}],"author":{"name":"zengjia","email":"zengjia@awcloud.com","date":"2019-05-23 02:53:31.000000000","tz":480},"committer":{"name":"zengjia","email":"zengjia@awcloud.com","date":"2019-05-23 02:53:31.000000000","tz":480},"subject":"Modify the version path URL of the configuration file","message":"Modify the version path URL of the configuration file\n\nIn the main branch, I think it is necessary to modify\nthe configuration file path in the document as the\nmain branch path, and to change the path to a stable\nversion when the version is released.\n\nChange-Id: Ib1bc4c59ed4c7624782b8b1cdea20847c14c90bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2c4f284055e1031f17f97229056616017385d8bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2c4f284055e1031f17f97229056616017385d8bc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
