)]}'
{"id":"openstack%2Fswift~661296","triplet_id":"openstack%2Fswift~master~Icc583e3545763c9c15da9ce4e0544a3ecf4cb089","project":"openstack/swift","branch":"master","topic":"fix-hashes-py3","hashtags":[],"change_id":"Icc583e3545763c9c15da9ce4e0544a3ecf4cb089","subject":"WIP - Fix handling of hashes.invalid file with py3","status":"ABANDONED","created":"2019-05-24 13:59:05.000000000","updated":"2019-05-29 15:54:59.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f6f5a9a9454e2564d9a0b384626ab725fb6ddf2b","_number":661296,"virtual_id_number":661296,"owner":{"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-05-29 15:54:33.000000000","_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},{"value":1,"date":"2019-05-24 17:13:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-05-24 17:26:30.000000000","_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-24 17:13:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-24 17:26:30.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"66212dd6afacd8b7955f9a86605eac426ff62be1","author":{"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},"date":"2019-05-24 13:59:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1aabc2025690fbb184df2ef015e4a9dae5012a74","author":{"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},"date":"2019-05-24 14:05:32.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nI don\u0027t know if that\u0027s the proper way of handling this issue. I am sure Tim and Pete will have something to say about this ;)","accounts_in_message":[],"_revision_number":1},{"id":"beed7d51b0a081d36bda0c887fc3ea30eaadc210","author":{"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},"date":"2019-05-24 14:07:07.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"1a3576fe30e997108046eacfd06293095d8e6331","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 17:13:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/96/661296/2/check/openstack-tox-docs/b1f7437/html/ : SUCCESS in 5m 22s\n- swift-build-image http://logs.openstack.org/96/661296/2/check/swift-build-image/55df09c/ : SUCCESS in 8m 17s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/96/661296/2/check/swift-tox-py27/b476776/ : SUCCESS in 12m 16s\n- swift-tox-py36 http://logs.openstack.org/96/661296/2/check/swift-tox-py36/fa6742c/ : SUCCESS in 10m 13s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/96/661296/2/check/swift-tox-py37/88b456d/ : SUCCESS in 12m 33s\n- swift-tox-func http://logs.openstack.org/96/661296/2/check/swift-tox-func/1c0ec9c/ : SUCCESS in 9m 26s\n- swift-tox-func-py37 http://logs.openstack.org/96/661296/2/check/swift-tox-func-py37/24c3eed/ : SUCCESS in 9m 31s\n- swift-tox-func-encryption http://logs.openstack.org/96/661296/2/check/swift-tox-func-encryption/387e817/ : SUCCESS in 11m 12s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/96/661296/2/check/swift-tox-func-domain-remap-staticweb/b241c0e/ : SUCCESS in 9m 56s\n- swift-tox-func-ec http://logs.openstack.org/96/661296/2/check/swift-tox-func-ec/1761f86/ : SUCCESS in 9m 26s\n- swift-tox-func-s3api http://logs.openstack.org/96/661296/2/check/swift-tox-func-s3api/c6a265b/ : SUCCESS in 16m 36s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/96/661296/2/check/swift-tox-func-s3api-ceph-s3tests-tempauth/f09a9ce/ : SUCCESS in 20m 10s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/96/661296/2/check/swift-probetests-centos-7/860aca7/ : SUCCESS in 45m 08s\n- swift-dsvm-functional http://logs.openstack.org/96/661296/2/check/swift-dsvm-functional/5e1abbc/ : SUCCESS in 30m 13s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/96/661296/2/check/swift-dsvm-functional-ipv6/c36d15d/ : SUCCESS in 26m 04s\n- swift-tox-lower-constraints http://logs.openstack.org/96/661296/2/check/swift-tox-lower-constraints/1c84f57/ : SUCCESS in 18m 33s\n- openstack-tox-pep8 http://logs.openstack.org/96/661296/2/check/openstack-tox-pep8/c15196b/ : SUCCESS in 6m 14s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/96/661296/2/check/swift-multinode-rolling-upgrade/2c0a7a0/ : FAILURE in 8m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"161d63c2b0d9bff27f4f323f3d826d5d9afcbc30","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-24 17:26:30.000000000","message":"Patch Set 2:\n\n(2 comments)\n\nGood news: we found a py3 bug!\n\nBetter news: I think we may have already fixed it ;-)\n\nAny chance you could try out https://review.opendev.org/#/c/660542/ (which ports SSYNC)? I can think of few people who know SSYNC as inside and out as you and Romain (or who would have a more vested interest in ensuring that we both correctly port it for py3 *and* prevent regressions on py2).","accounts_in_message":[],"_revision_number":2},{"id":"2a161473af72956d64e011d97c9586c161d8f296","author":{"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},"date":"2019-05-29 15:54:33.000000000","message":"Patch Set 2:\n\nSorry it took a while to reply - will take a look at the SSYNC patch!","accounts_in_message":[],"_revision_number":2},{"id":"b1c74fe3bf1a1a9f6ec7b4159eb5d272aa9fc548","author":{"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},"date":"2019-05-29 15:54:59.000000000","message":"Abandoned\n\nfixed in a later commit","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d889758bd8f4788c50db417fc4ad69c8b47137bb","revisions":{"771f599050e9edbc0d958dc3765b8ae7cea7cefc":{"kind":"REWORK","_number":1,"created":"2019-05-24 13:59:05.000000000","uploader":{"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},"ref":"refs/changes/96/661296/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/96/661296/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/661296/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/661296/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/661296/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/96/661296/1"}}},"commit":{"parents":[{"commit":"657b022689ebe252a67e2ae46d233df79d6c0bdd","subject":"Merge \"py3: port diskfile\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/657b022689ebe252a67e2ae46d233df79d6c0bdd"}]}],"author":{"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","date":"2019-05-24 13:53:27.000000000","tz":120},"committer":{"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","date":"2019-05-24 13:57:19.000000000","tz":120},"subject":"WIP - Fix handling of hashes.invalid file with py3","message":"WIP - Fix handling of hashes.invalid file with py3\n\nWith python 3, we would write bytes rather than str to the\nhashes.invalid file. This will cause byte keys to appear in\nthe pickled dict in hashes.pkl, making __get_hashes() fail.\n\nThis patch handles hashes.invalid as a text file instead.\n\nFixes-bug: 1651530\n\nChange-Id: Icc583e3545763c9c15da9ce4e0544a3ecf4cb089\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/771f599050e9edbc0d958dc3765b8ae7cea7cefc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/771f599050e9edbc0d958dc3765b8ae7cea7cefc"}]},"branch":"refs/heads/master"},"d889758bd8f4788c50db417fc4ad69c8b47137bb":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-05-24 14:07:07.000000000","uploader":{"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},"ref":"refs/changes/96/661296/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/96/661296/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/661296/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/661296/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/661296/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/96/661296/2"}}},"commit":{"parents":[{"commit":"657b022689ebe252a67e2ae46d233df79d6c0bdd","subject":"Merge \"py3: port diskfile\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/657b022689ebe252a67e2ae46d233df79d6c0bdd"}]}],"author":{"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","date":"2019-05-24 13:53:27.000000000","tz":120},"committer":{"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","date":"2019-05-24 14:06:43.000000000","tz":120},"subject":"WIP - Fix handling of hashes.invalid file with py3","message":"WIP - Fix handling of hashes.invalid file with py3\n\nWith python 3, we would write bytes rather than str to the\nhashes.invalid file. This will cause byte keys to appear in\nthe pickled dict in hashes.pkl, making __get_hashes() fail.\n\nThis patch handles hashes.invalid as a text file instead.\n\nFixes-bug: 1830372\n\nChange-Id: Icc583e3545763c9c15da9ce4e0544a3ecf4cb089\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d889758bd8f4788c50db417fc4ad69c8b47137bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d889758bd8f4788c50db417fc4ad69c8b47137bb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
