)]}'
{"id":"openstack%2Fswift~661930","triplet_id":"openstack%2Fswift~master~I84b062d062ff49935feed0aee3e1963bb72eb5ea","project":"openstack/swift","branch":"master","topic":"bug/1830881","hashtags":[],"change_id":"I84b062d062ff49935feed0aee3e1963bb72eb5ea","subject":"Remove invalid dict entries from hashes.pkl","status":"MERGED","created":"2019-05-29 10:12:46.000000000","updated":"2019-06-17 20:42:23.000000000","submitted":"2019-06-17 20:38:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"661930-1560803925086-63dd4f34","meta_rev_id":"ac8c2938512557d05616f592dd2dbfbd33f1996b","_number":661930,"virtual_id_number":661930,"owner":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19234,"name":"Alexey Stupnikov","email":"aleksey.stupnikov@gmail.com","username":"astupnikov"},{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"date":"2019-06-17 14:59:13.000000000","_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":2,"date":"2019-06-17 20:38:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":19234,"name":"Alexey Stupnikov","email":"aleksey.stupnikov@gmail.com","username":"astupnikov"},{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-15 04:30:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":19234,"name":"Alexey Stupnikov","email":"aleksey.stupnikov@gmail.com","username":"astupnikov"},{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-15 04:30:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":19234,"name":"Alexey Stupnikov","email":"aleksey.stupnikov@gmail.com","username":"astupnikov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-29 13:52:01.000000000","updated_by":{"_account_id":19234,"name":"Alexey Stupnikov","email":"aleksey.stupnikov@gmail.com","username":"astupnikov"},"reviewer":{"_account_id":19234,"name":"Alexey Stupnikov","email":"aleksey.stupnikov@gmail.com","username":"astupnikov"},"state":"REVIEWER"},{"updated":"2019-06-15 04:30:42.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2019-06-17 14:59:13.000000000","updated_by":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"reviewer":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"state":"REVIEWER"},{"updated":"2019-06-17 20:38:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"aa3db71d7b2b7e66be997d23a72208ded8ee16c1","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2019-05-29 10:12:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a08208638e5c7141b4b5bb21dff83253e25fd398","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 13:52:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/30/661930/1/check/openstack-tox-docs/f709af7/html/ : SUCCESS in 4m 28s\n- swift-build-image http://logs.openstack.org/30/661930/1/check/swift-build-image/d02eca1/ : SUCCESS in 7m 21s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/30/661930/1/check/swift-tox-py27/516358a/ : SUCCESS in 10m 56s\n- swift-tox-py36 http://logs.openstack.org/30/661930/1/check/swift-tox-py36/d410b0c/ : FAILURE in 10m 23s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/30/661930/1/check/swift-tox-py37/6b61689/ : FAILURE in 12m 41s\n- swift-tox-func http://logs.openstack.org/30/661930/1/check/swift-tox-func/03d9cc9/ : SUCCESS in 10m 38s\n- swift-tox-func-py37 http://logs.openstack.org/30/661930/1/check/swift-tox-func-py37/5c9d02a/ : SUCCESS in 7m 06s\n- swift-tox-func-encryption http://logs.openstack.org/30/661930/1/check/swift-tox-func-encryption/73c82aa/ : SUCCESS in 8m 56s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/30/661930/1/check/swift-tox-func-domain-remap-staticweb/e51c14e/ : SUCCESS in 8m 54s\n- swift-tox-func-ec http://logs.openstack.org/30/661930/1/check/swift-tox-func-ec/a0da83c/ : SUCCESS in 10m 06s\n- swift-tox-func-s3api http://logs.openstack.org/30/661930/1/check/swift-tox-func-s3api/b2f7359/ : SUCCESS in 12m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/30/661930/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/1101e50/ : SUCCESS in 15m 13s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/30/661930/1/check/swift-probetests-centos-7/55b211d/ : FAILURE in 31m 01s\n- swift-dsvm-functional http://logs.openstack.org/30/661930/1/check/swift-dsvm-functional/0aa9f7d/ : SUCCESS in 25m 16s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/30/661930/1/check/swift-dsvm-functional-ipv6/39f9b64/ : SUCCESS in 29m 27s\n- swift-tox-lower-constraints http://logs.openstack.org/30/661930/1/check/swift-tox-lower-constraints/7979cbf/ : SUCCESS in 12m 36s\n- openstack-tox-pep8 http://logs.openstack.org/30/661930/1/check/openstack-tox-pep8/8b4e7ef/ : SUCCESS in 5m 27s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/30/661930/1/check/swift-multinode-rolling-upgrade/fdf8f98/ : FAILURE in 8m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"99a06990e0f7df584e305d9690d9dbc1f0bf8c51","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-29 17:17:33.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)\n\nMakes me wonder if we should be writing down an MD5 of the content when writing hashes.pkl, similar to what Sam did for object metadata in https://github.com/openstack/swift/commit/728b4ba ...\n\nGotta get py3 happy, and I\u0027m nervous about just dropping individual suffixes when we detect corruption instead of the whole thing, particularly given the narrow scope of corruption detection.","accounts_in_message":[],"_revision_number":1},{"id":"94bf03d4860b7cba6ab29633c84f7c7f634db29c","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2019-05-31 08:58:49.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"1a1768fcea7636838fb69d7628e67e66822b0ebc","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2019-05-31 09:07:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ca47ac501d22c6e2d19d4250215307e7b017e2bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 09:44:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/30/661930/2/check/openstack-tox-docs/1f48775/html/ : SUCCESS in 4m 00s\n- swift-build-image http://logs.openstack.org/30/661930/2/check/swift-build-image/4b9422c/ : SUCCESS in 7m 52s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/30/661930/2/check/swift-tox-py27/106868e/ : SUCCESS in 10m 24s\n- swift-tox-py36 http://logs.openstack.org/30/661930/2/check/swift-tox-py36/f65d756/ : SUCCESS in 9m 29s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/30/661930/2/check/swift-tox-py37/2005fa1/ : SUCCESS in 10m 23s\n- swift-tox-func http://logs.openstack.org/30/661930/2/check/swift-tox-func/124726d/ : SUCCESS in 7m 52s\n- swift-tox-func-py37 http://logs.openstack.org/30/661930/2/check/swift-tox-func-py37/61b0d0b/ : SUCCESS in 7m 02s\n- swift-tox-func-encryption http://logs.openstack.org/30/661930/2/check/swift-tox-func-encryption/a27f4ca/ : SUCCESS in 8m 55s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/30/661930/2/check/swift-tox-func-domain-remap-staticweb/028491f/ : SUCCESS in 8m 30s\n- swift-tox-func-ec http://logs.openstack.org/30/661930/2/check/swift-tox-func-ec/74569d6/ : SUCCESS in 10m 11s\n- swift-tox-func-s3api http://logs.openstack.org/30/661930/2/check/swift-tox-func-s3api/4876a38/ : SUCCESS in 11m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/30/661930/2/check/swift-tox-func-s3api-ceph-s3tests-tempauth/d78d616/ : SUCCESS in 12m 57s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/30/661930/2/check/swift-probetests-centos-7/6e6c6f2/ : SUCCESS in 31m 01s\n- swift-dsvm-functional http://logs.openstack.org/30/661930/2/check/swift-dsvm-functional/0cb2bf9/ : SUCCESS in 26m 42s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/30/661930/2/check/swift-dsvm-functional-ipv6/d7eea7e/ : SUCCESS in 28m 38s\n- swift-tox-lower-constraints http://logs.openstack.org/30/661930/2/check/swift-tox-lower-constraints/13d4004/ : SUCCESS in 11m 04s\n- openstack-tox-pep8 http://logs.openstack.org/30/661930/2/check/openstack-tox-pep8/e707012/ : SUCCESS in 5m 55s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/30/661930/2/check/swift-multinode-rolling-upgrade/9f297ca/ : FAILURE in 8m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"03321899749ae6511d0c5c20f51dd603a01fc323","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-15 04:30:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d809aec7629d3d7887a5343eda4bd533da83777d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-15 04:31:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c0fe8039f7dd4e1c7d712f60fb750e8d1c128858","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-15 09:37:17.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/30/661930/2/gate/openstack-tox-docs/94bd4b6/html/ : SUCCESS in 5m 58s\n- swift-upload-image http://logs.openstack.org/30/661930/2/gate/swift-upload-image/935f624/ : SUCCESS in 8m 11s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/30/661930/2/gate/swift-tox-py27/57af96f/ : SUCCESS in 13m 36s\n- swift-tox-py37 http://logs.openstack.org/30/661930/2/gate/swift-tox-py37/3b3b6cb/ : SUCCESS in 11m 16s\n- swift-tox-func http://logs.openstack.org/30/661930/2/gate/swift-tox-func/d93364a/ : SUCCESS in 9m 27s\n- swift-tox-func-py37 http://logs.openstack.org/30/661930/2/gate/swift-tox-func-py37/63de7eb/ : SUCCESS in 9m 39s\n- swift-tox-func-encryption http://logs.openstack.org/30/661930/2/gate/swift-tox-func-encryption/54fe01e/ : SUCCESS in 9m 34s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/30/661930/2/gate/swift-tox-func-domain-remap-staticweb/d675a80/ : SUCCESS in 9m 14s\n- swift-tox-func-ec http://logs.openstack.org/30/661930/2/gate/swift-tox-func-ec/b1d930c/ : SUCCESS in 10m 26s\n- swift-tox-func-s3api http://logs.openstack.org/30/661930/2/gate/swift-tox-func-s3api/4de661e/ : SUCCESS in 14m 37s\n- swift-probetests-centos-7 http://logs.openstack.org/30/661930/2/gate/swift-probetests-centos-7/e2ca09e/ : TIMED_OUT in 1h 01m 06s\n- swift-dsvm-functional http://logs.openstack.org/30/661930/2/gate/swift-dsvm-functional/5ae0319/ : SUCCESS in 27m 15s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/30/661930/2/gate/swift-dsvm-functional-ipv6/1c46197/ : SUCCESS in 27m 25s\n- swift-tox-lower-constraints http://logs.openstack.org/30/661930/2/gate/swift-tox-lower-constraints/7c52146/ : SUCCESS in 12m 24s\n- openstack-tox-pep8 http://logs.openstack.org/30/661930/2/gate/openstack-tox-pep8/759d9c9/ : SUCCESS in 6m 12s","accounts_in_message":[],"_revision_number":2},{"id":"9240acaf74df38c3d3f24e13caff93b7db3da9c4","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2019-06-17 14:59:13.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a8f165744882d0d4b60ad44fa1c0e1f71cdc1424","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 18:55:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/30/661930/2/check/openstack-tox-docs/f029fda/html/ : SUCCESS in 4m 30s\n- swift-build-image http://logs.openstack.org/30/661930/2/check/swift-build-image/35c8712/ : SUCCESS in 8m 35s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/30/661930/2/check/swift-tox-py27/61f0999/ : SUCCESS in 10m 33s\n- swift-tox-py36 http://logs.openstack.org/30/661930/2/check/swift-tox-py36/dba4c3e/ : SUCCESS in 10m 52s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/30/661930/2/check/swift-tox-py37/c82d68f/ : SUCCESS in 11m 32s\n- swift-tox-func http://logs.openstack.org/30/661930/2/check/swift-tox-func/eed55b3/ : SUCCESS in 9m 03s\n- swift-tox-func-py37 http://logs.openstack.org/30/661930/2/check/swift-tox-func-py37/1408968/ : SUCCESS in 10m 21s\n- swift-tox-func-encryption http://logs.openstack.org/30/661930/2/check/swift-tox-func-encryption/152f4de/ : SUCCESS in 10m 51s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/30/661930/2/check/swift-tox-func-domain-remap-staticweb/99b9868/ : SUCCESS in 8m 12s\n- swift-tox-func-ec http://logs.openstack.org/30/661930/2/check/swift-tox-func-ec/2d50b37/ : SUCCESS in 9m 28s\n- swift-tox-func-s3api http://logs.openstack.org/30/661930/2/check/swift-tox-func-s3api/5177e35/ : SUCCESS in 14m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/30/661930/2/check/swift-tox-func-s3api-ceph-s3tests-tempauth/0617910/ : SUCCESS in 13m 13s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/30/661930/2/check/swift-probetests-centos-7/1003ee9/ : SUCCESS in 37m 05s\n- swift-dsvm-functional http://logs.openstack.org/30/661930/2/check/swift-dsvm-functional/cfa63d9/ : SUCCESS in 27m 25s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/30/661930/2/check/swift-dsvm-functional-ipv6/610b363/ : SUCCESS in 32m 55s\n- swift-tox-lower-constraints http://logs.openstack.org/30/661930/2/check/swift-tox-lower-constraints/c86fb46/ : SUCCESS in 11m 31s\n- openstack-tox-pep8 http://logs.openstack.org/30/661930/2/check/openstack-tox-pep8/03881f7/ : SUCCESS in 5m 35s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/30/661930/2/check/swift-multinode-rolling-upgrade/1fa22a7/ : FAILURE in 8m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"08b02315591e907d7b678ae56420744acd41769d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 18:55:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c86de61783de2ac93f0912a6b4e64779c3aa7614","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 20:38:44.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/30/661930/2/gate/openstack-tox-docs/3ad0082/html/ : SUCCESS in 4m 49s\n- swift-upload-image http://logs.openstack.org/30/661930/2/gate/swift-upload-image/7e02d8d/ : SUCCESS in 8m 42s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/30/661930/2/gate/swift-tox-py27/9639235/ : SUCCESS in 10m 19s\n- swift-tox-py37 http://logs.openstack.org/30/661930/2/gate/swift-tox-py37/5dc1b97/ : SUCCESS in 12m 12s\n- swift-tox-func http://logs.openstack.org/30/661930/2/gate/swift-tox-func/9902a77/ : SUCCESS in 9m 19s\n- swift-tox-func-py37 http://logs.openstack.org/30/661930/2/gate/swift-tox-func-py37/432c933/ : SUCCESS in 9m 07s\n- swift-tox-func-encryption http://logs.openstack.org/30/661930/2/gate/swift-tox-func-encryption/af991bb/ : SUCCESS in 9m 07s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/30/661930/2/gate/swift-tox-func-domain-remap-staticweb/741389e/ : SUCCESS in 14m 06s\n- swift-tox-func-ec http://logs.openstack.org/30/661930/2/gate/swift-tox-func-ec/8100650/ : SUCCESS in 9m 38s\n- swift-tox-func-s3api http://logs.openstack.org/30/661930/2/gate/swift-tox-func-s3api/5771c8c/ : SUCCESS in 12m 26s\n- swift-probetests-centos-7 http://logs.openstack.org/30/661930/2/gate/swift-probetests-centos-7/7d9b2e0/ : SUCCESS in 34m 56s\n- swift-dsvm-functional http://logs.openstack.org/30/661930/2/gate/swift-dsvm-functional/3c79962/ : SUCCESS in 26m 25s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/30/661930/2/gate/swift-dsvm-functional-ipv6/508d456/ : SUCCESS in 29m 09s\n- swift-tox-lower-constraints http://logs.openstack.org/30/661930/2/gate/swift-tox-lower-constraints/113558e/ : SUCCESS in 13m 46s\n- openstack-tox-pep8 http://logs.openstack.org/30/661930/2/gate/openstack-tox-pep8/ef2fd34/ : SUCCESS in 6m 16s","accounts_in_message":[],"_revision_number":2},{"id":"e36fd1d71380a77d4fc44d6f53ad15f9ae21c854","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 20:38:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"ea0e008bcc07e148da241fe823ea9d5d9216a32d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 20:42:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- swift-promote-image http://logs.openstack.org/30/661930/2/promote/swift-promote-image/221ae00/ : SUCCESS in 47s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c9e78d15e1e05f23facf7c28758b442bb25bde68","revisions":{"9fe0ab84b77e9a5dcf0089d5d598bf365500f8dd":{"kind":"REWORK","_number":1,"created":"2019-05-29 10:12:46.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/30/661930/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/30/661930/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/661930/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/661930/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/661930/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/30/661930/1"}}},"commit":{"parents":[{"commit":"32d1b3246f6e5fd6176d2af1c22a3ac6d0fe6a17","subject":"Merge \"authors/changelog updates for 2.17.0 release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/32d1b3246f6e5fd6176d2af1c22a3ac6d0fe6a17"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2019-05-29 09:37:54.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2019-05-29 10:12:43.000000000","tz":120},"subject":"Remove invalid dict entries from hashes.pkl","message":"Remove invalid dict entries from hashes.pkl\n\nIf the data in a hashes.pkl is corrupted but still de-serialized without\nerrors, it will mess up the replication and gets never fixed. This\nhappens for example if one of the keys is a NULL byte.\n\nThis patch checks if the dict keys are valid strings, and removes\npossible corrupted keys if not.\n\nChange-Id: I84b062d062ff49935feed0aee3e1963bb72eb5ea\nCloses-Bug: 1830881\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9fe0ab84b77e9a5dcf0089d5d598bf365500f8dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9fe0ab84b77e9a5dcf0089d5d598bf365500f8dd"}]},"branch":"refs/heads/master"},"c9e78d15e1e05f23facf7c28758b442bb25bde68":{"kind":"REWORK","_number":2,"created":"2019-05-31 09:07:56.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/30/661930/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/30/661930/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/661930/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/661930/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/661930/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/30/661930/2"}}},"commit":{"parents":[{"commit":"32d1b3246f6e5fd6176d2af1c22a3ac6d0fe6a17","subject":"Merge \"authors/changelog updates for 2.17.0 release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/32d1b3246f6e5fd6176d2af1c22a3ac6d0fe6a17"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2019-05-29 09:37:54.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2019-05-31 09:07:42.000000000","tz":120},"subject":"Remove invalid dict entries from hashes.pkl","message":"Remove invalid dict entries from hashes.pkl\n\nIf the data in a hashes.pkl is corrupted but still de-serialized without\nerrors, it will mess up the replication and gets never fixed. This\nhappens for example if one of the keys is a NULL byte.\n\nThis patch checks if the dict keys in hashes.pkl are valid strings and\ninvalidates it if not.\n\nCloses-Bug: 1830881\nChange-Id: I84b062d062ff49935feed0aee3e1963bb72eb5ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c9e78d15e1e05f23facf7c28758b442bb25bde68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c9e78d15e1e05f23facf7c28758b442bb25bde68"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
