)]}'
{"id":"openstack%2Fswift~662115","triplet_id":"openstack%2Fswift~master~I8d6f66dfe846493972e433f70bad76a33d204562","project":"openstack/swift","branch":"master","topic":"py3-tempurl","hashtags":[],"change_id":"I8d6f66dfe846493972e433f70bad76a33d204562","subject":"Fix up how we memcache on py3","status":"MERGED","created":"2019-05-30 01:14:51.000000000","updated":"2019-06-19 20:09:56.000000000","submitted":"2019-06-19 20:08:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662115-1560974920787-91b8215b","meta_rev_id":"0ab6369ee0ba916b9dc790231a23933fb6eb4bb2","_number":662115,"virtual_id_number":662115,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":2,"date":"2019-06-19 20:08:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-19 04:53:14.000000000","_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2019-06-18 22:59:11.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":2,"date":"2019-06-18 17:38:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":1,"date":"2019-06-18 17:38:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-13 22:48:59.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2019-06-14 01:42:22.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-06-18 17:38:56.000000000","updated_by":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"reviewer":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"state":"REVIEWER"},{"updated":"2019-06-18 22:59:11.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2019-06-19 20:08:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e323a41fda8a1d888964e77080245a2bb54d57c2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-30 01:14:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ce21c999c5d7e2852bf0c1fdd247c19cfd7a7eb3","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-30 01:23:03.000000000","message":"Patch Set 1: Workflow-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"49da9e51e03b739fdabd7e38e23cf29042a50487","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 03:50:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/15/662115/1/check/openstack-tox-docs/f765d6a/html/ : SUCCESS in 4m 30s\n- swift-build-image http://logs.openstack.org/15/662115/1/check/swift-build-image/d47ffcf/ : SUCCESS in 8m 09s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/15/662115/1/check/swift-tox-py27/ccc177b/ : SUCCESS in 9m 54s\n- swift-tox-py36 http://logs.openstack.org/15/662115/1/check/swift-tox-py36/16a08ea/ : FAILURE in 8m 13s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/15/662115/1/check/swift-tox-py37/b5936dd/ : FAILURE in 8m 34s\n- swift-tox-func http://logs.openstack.org/15/662115/1/check/swift-tox-func/684b8c2/ : SUCCESS in 8m 26s\n- swift-tox-func-py37 http://logs.openstack.org/15/662115/1/check/swift-tox-func-py37/5a52357/ : FAILURE in 4m 10s\n- swift-tox-func-encryption http://logs.openstack.org/15/662115/1/check/swift-tox-func-encryption/feb4482/ : SUCCESS in 9m 45s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/15/662115/1/check/swift-tox-func-domain-remap-staticweb/1a6f8b2/ : SUCCESS in 9m 13s\n- swift-tox-func-ec http://logs.openstack.org/15/662115/1/check/swift-tox-func-ec/031f5b9/ : SUCCESS in 9m 24s\n- swift-tox-func-s3api http://logs.openstack.org/15/662115/1/check/swift-tox-func-s3api/a3e73a9/ : SUCCESS in 11m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/15/662115/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/7ae5413/ : SUCCESS in 12m 37s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/15/662115/1/check/swift-probetests-centos-7/96401cc/ : SUCCESS in 33m 02s\n- swift-dsvm-functional http://logs.openstack.org/15/662115/1/check/swift-dsvm-functional/c023323/ : SUCCESS in 26m 09s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/15/662115/1/check/swift-dsvm-functional-ipv6/d88496c/ : SUCCESS in 25m 05s\n- swift-tox-lower-constraints http://logs.openstack.org/15/662115/1/check/swift-tox-lower-constraints/aafedec/ : SUCCESS in 12m 49s\n- openstack-tox-pep8 http://logs.openstack.org/15/662115/1/check/openstack-tox-pep8/f97d7d1/ : SUCCESS in 5m 25s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/15/662115/1/check/swift-multinode-rolling-upgrade/6f64104/ : FAILURE in 8m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"04f90f394a030ec0527542e6b9afaa708ed6f1a4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-31 15:43:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c792504746ba321164c818f50463437a24799818","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-31 19:08:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"709ec56d3154eca2d38c2f3c16ae21010f631b7b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 23:13:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/15/662115/2/check/openstack-tox-docs/c02e3d4/html/ : SUCCESS in 4m 30s\n- swift-build-image http://logs.openstack.org/15/662115/2/check/swift-build-image/00ad85d/ : SUCCESS in 8m 17s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/15/662115/2/check/swift-tox-py27/ce7a0cc/ : SUCCESS in 11m 30s\n- swift-tox-py36 http://logs.openstack.org/15/662115/2/check/swift-tox-py36/2e66777/ : SUCCESS in 19m 55s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/15/662115/2/check/swift-tox-py37/e3eaf76/ : SUCCESS in 10m 21s\n- swift-tox-func http://logs.openstack.org/15/662115/2/check/swift-tox-func/9c5e17a/ : SUCCESS in 8m 11s\n- swift-tox-func-py37 http://logs.openstack.org/15/662115/2/check/swift-tox-func-py37/2c69423/ : SUCCESS in 8m 27s\n- swift-tox-func-encryption http://logs.openstack.org/15/662115/2/check/swift-tox-func-encryption/68e3cf7/ : SUCCESS in 9m 33s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/15/662115/2/check/swift-tox-func-domain-remap-staticweb/2ac50d7/ : FAILURE in 9m 54s\n- swift-tox-func-ec http://logs.openstack.org/15/662115/2/check/swift-tox-func-ec/78839ca/ : SUCCESS in 11m 13s\n- swift-tox-func-s3api http://logs.openstack.org/15/662115/2/check/swift-tox-func-s3api/9cfd15d/ : SUCCESS in 12m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/15/662115/2/check/swift-tox-func-s3api-ceph-s3tests-tempauth/93bf204/ : SUCCESS in 12m 11s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/15/662115/2/check/swift-probetests-centos-7/40fc70c/ : SUCCESS in 35m 25s\n- swift-dsvm-functional http://logs.openstack.org/15/662115/2/check/swift-dsvm-functional/7c53fee/ : SUCCESS in 26m 27s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/15/662115/2/check/swift-dsvm-functional-ipv6/90b8945/ : SUCCESS in 29m 20s\n- swift-tox-lower-constraints http://logs.openstack.org/15/662115/2/check/swift-tox-lower-constraints/067d04b/ : SUCCESS in 11m 54s\n- openstack-tox-pep8 http://logs.openstack.org/15/662115/2/check/openstack-tox-pep8/ca00711/ : SUCCESS in 5m 05s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/15/662115/2/check/swift-multinode-rolling-upgrade/a722ebc/ : FAILURE in 8m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cf2710425c8a4f03414b7eaa956a380dd253eff3","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-31 23:16:05.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"fa9695f81610309a28e9e06f4973e5548273b51c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-01 01:13:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/15/662115/2/check/openstack-tox-docs/898f5ba/html/ : SUCCESS in 4m 23s\n- swift-build-image http://logs.openstack.org/15/662115/2/check/swift-build-image/ff8587e/ : SUCCESS in 8m 32s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/15/662115/2/check/swift-tox-py27/a40f631/ : SUCCESS in 11m 04s\n- swift-tox-py36 http://logs.openstack.org/15/662115/2/check/swift-tox-py36/3d2247e/ : SUCCESS in 10m 22s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/15/662115/2/check/swift-tox-py37/58617b5/ : SUCCESS in 10m 58s\n- swift-tox-func http://logs.openstack.org/15/662115/2/check/swift-tox-func/31e5209/ : SUCCESS in 8m 42s\n- swift-tox-func-py37 http://logs.openstack.org/15/662115/2/check/swift-tox-func-py37/0731bbb/ : SUCCESS in 6m 45s\n- swift-tox-func-encryption http://logs.openstack.org/15/662115/2/check/swift-tox-func-encryption/ffa880d/ : SUCCESS in 9m 55s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/15/662115/2/check/swift-tox-func-domain-remap-staticweb/080b969/ : SUCCESS in 10m 05s\n- swift-tox-func-ec http://logs.openstack.org/15/662115/2/check/swift-tox-func-ec/e055148/ : SUCCESS in 8m 41s\n- swift-tox-func-s3api http://logs.openstack.org/15/662115/2/check/swift-tox-func-s3api/a72f1aa/ : SUCCESS in 12m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/15/662115/2/check/swift-tox-func-s3api-ceph-s3tests-tempauth/e6eccc9/ : SUCCESS in 13m 13s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/15/662115/2/check/swift-probetests-centos-7/c734637/ : SUCCESS in 33m 12s\n- swift-dsvm-functional http://logs.openstack.org/15/662115/2/check/swift-dsvm-functional/442a909/ : SUCCESS in 25m 46s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/15/662115/2/check/swift-dsvm-functional-ipv6/4b27cb6/ : SUCCESS in 27m 45s\n- swift-tox-lower-constraints http://logs.openstack.org/15/662115/2/check/swift-tox-lower-constraints/dcbd4bb/ : SUCCESS in 12m 09s\n- openstack-tox-pep8 http://logs.openstack.org/15/662115/2/check/openstack-tox-pep8/5305e77/ : SUCCESS in 5m 59s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/15/662115/2/check/swift-multinode-rolling-upgrade/a6c6484/ : FAILURE in 7m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2267c7b60ebfba87ae2fbb6202bd83d1c4353f8e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-06 18:57:53.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"ae6e38af1efc8096f195cace93ed26ba2fbc291b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 00:45:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/15/662115/3/check/openstack-tox-docs/a9764a9/html/ : SUCCESS in 5m 06s\n- swift-build-image http://logs.openstack.org/15/662115/3/check/swift-build-image/4ae9d87/ : SUCCESS in 9m 21s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/15/662115/3/check/swift-tox-py27/a6c9b07/ : SUCCESS in 11m 30s\n- swift-tox-py36 http://logs.openstack.org/15/662115/3/check/swift-tox-py36/6d79564/ : SUCCESS in 11m 11s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/15/662115/3/check/swift-tox-py37/c86a496/ : SUCCESS in 10m 37s\n- swift-tox-func http://logs.openstack.org/15/662115/3/check/swift-tox-func/27fd4c1/ : SUCCESS in 8m 39s\n- swift-tox-func-py37 http://logs.openstack.org/15/662115/3/check/swift-tox-func-py37/83fe91f/ : SUCCESS in 8m 01s\n- swift-tox-func-encryption http://logs.openstack.org/15/662115/3/check/swift-tox-func-encryption/f308188/ : SUCCESS in 10m 17s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/15/662115/3/check/swift-tox-func-domain-remap-staticweb/062259c/ : SUCCESS in 13m 14s\n- swift-tox-func-ec http://logs.openstack.org/15/662115/3/check/swift-tox-func-ec/93d69fb/ : SUCCESS in 16m 55s\n- swift-tox-func-s3api http://logs.openstack.org/15/662115/3/check/swift-tox-func-s3api/2ac492a/ : SUCCESS in 13m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/15/662115/3/check/swift-tox-func-s3api-ceph-s3tests-tempauth/31d426f/ : SUCCESS in 13m 15s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/15/662115/3/check/swift-probetests-centos-7/36fadd9/ : SUCCESS in 29m 46s\n- swift-dsvm-functional http://logs.openstack.org/15/662115/3/check/swift-dsvm-functional/0697808/ : SUCCESS in 28m 36s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/15/662115/3/check/swift-dsvm-functional-ipv6/1eefc38/ : SUCCESS in 32m 08s\n- swift-tox-lower-constraints http://logs.openstack.org/15/662115/3/check/swift-tox-lower-constraints/c785467/ : SUCCESS in 11m 34s\n- openstack-tox-pep8 http://logs.openstack.org/15/662115/3/check/openstack-tox-pep8/84928b1/ : SUCCESS in 5m 38s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/15/662115/3/check/swift-multinode-rolling-upgrade/ad9a12a/ : FAILURE in 8m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0573d72c049d5aeb6e401f5471ec2414d1cd3315","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-06-12 02:55:00.000000000","message":"Patch Set 3: Code-Review+2\n\naffects py2 but should be safe by my estimate","accounts_in_message":[],"_revision_number":3},{"id":"74deee9e567bb59fc95f38c2c5e29899153e98bd","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-13 16:40:22.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b26f41e1b6de32fd6543a624418116ddc13ac1ed","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-13 16:41:31.000000000","message":"Patch Set 4:\n\nRebased to resolve conflict with https://review.opendev.org/#/c/664710/","accounts_in_message":[],"_revision_number":4},{"id":"2a20f0fb8bb549becb93501fd6d9e1c42eb434e7","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-06-13 18:31:29.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"de22265fccff0ff42058651935573a219a5ffdbe","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-06-13 22:48:59.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"ccde41937a2bd26cf3e47eef65a9acb0086f7419","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-06-14 01:42:22.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"4efcd0ac16f613bce1774e758633647d27adf56c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 03:58:26.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/15/662115/4/check/openstack-tox-docs/7c78af9/html/ : SUCCESS in 3m 54s\n- swift-build-image http://logs.openstack.org/15/662115/4/check/swift-build-image/389c636/ : SUCCESS in 7m 50s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/15/662115/4/check/swift-tox-py27/b04f2f6/ : SUCCESS in 10m 35s\n- swift-tox-py36 http://logs.openstack.org/15/662115/4/check/swift-tox-py36/6076576/ : SUCCESS in 11m 59s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/15/662115/4/check/swift-tox-py37/80a83b9/ : SUCCESS in 18m 02s\n- swift-tox-func http://logs.openstack.org/15/662115/4/check/swift-tox-func/df897ec/ : SUCCESS in 8m 29s\n- swift-tox-func-py37 http://logs.openstack.org/15/662115/4/check/swift-tox-func-py37/cb832ce/ : SUCCESS in 8m 53s\n- swift-tox-func-encryption http://logs.openstack.org/15/662115/4/check/swift-tox-func-encryption/6055f7a/ : SUCCESS in 13m 12s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/15/662115/4/check/swift-tox-func-domain-remap-staticweb/beae837/ : SUCCESS in 8m 55s\n- swift-tox-func-ec http://logs.openstack.org/15/662115/4/check/swift-tox-func-ec/7392b7d/ : SUCCESS in 9m 38s\n- swift-tox-func-s3api http://logs.openstack.org/15/662115/4/check/swift-tox-func-s3api/1f93ace/ : SUCCESS in 12m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/15/662115/4/check/swift-tox-func-s3api-ceph-s3tests-tempauth/dfc61c4/ : SUCCESS in 18m 14s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/15/662115/4/check/swift-probetests-centos-7/0e8b898/ : SUCCESS in 36m 30s\n- swift-dsvm-functional http://logs.openstack.org/15/662115/4/check/swift-dsvm-functional/5c2f299/ : SUCCESS in 27m 59s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/15/662115/4/check/swift-dsvm-functional-ipv6/7a06403/ : SUCCESS in 29m 37s\n- swift-tox-lower-constraints http://logs.openstack.org/15/662115/4/check/swift-tox-lower-constraints/3956621/ : SUCCESS in 12m 25s\n- openstack-tox-pep8 http://logs.openstack.org/15/662115/4/check/openstack-tox-pep8/d52c51c/ : FAILURE in 4m 54s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/15/662115/4/check/swift-multinode-rolling-upgrade/4d13819/ : FAILURE in 9m 06s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"148781f6acdcdfa0c00aadd809af3f8a2b4f0e9a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-14 15:21:25.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"19714d66296bbddfe0193fde6f0aef509f32bdb2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 23:53:51.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/15/662115/5/check/openstack-tox-docs/163e57c/html/ : SUCCESS in 5m 20s\n- swift-build-image http://logs.openstack.org/15/662115/5/check/swift-build-image/f550d00/ : SUCCESS in 9m 30s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/15/662115/5/check/swift-tox-py27/99fe3f9/ : SUCCESS in 12m 32s\n- swift-tox-py36 http://logs.openstack.org/15/662115/5/check/swift-tox-py36/e16006e/ : SUCCESS in 10m 12s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/15/662115/5/check/swift-tox-py37/0d86957/ : SUCCESS in 10m 34s\n- swift-tox-func http://logs.openstack.org/15/662115/5/check/swift-tox-func/cae5e76/ : SUCCESS in 8m 22s\n- swift-tox-func-py37 http://logs.openstack.org/15/662115/5/check/swift-tox-func-py37/7c8ef23/ : SUCCESS in 8m 39s\n- swift-tox-func-encryption http://logs.openstack.org/15/662115/5/check/swift-tox-func-encryption/760cd55/ : SUCCESS in 11m 17s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/15/662115/5/check/swift-tox-func-domain-remap-staticweb/466dbaa/ : SUCCESS in 10m 40s\n- swift-tox-func-ec http://logs.openstack.org/15/662115/5/check/swift-tox-func-ec/ccd5bc4/ : SUCCESS in 11m 34s\n- swift-tox-func-s3api http://logs.openstack.org/15/662115/5/check/swift-tox-func-s3api/c66f5d6/ : SUCCESS in 16m 56s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/15/662115/5/check/swift-tox-func-s3api-ceph-s3tests-tempauth/6edd584/ : SUCCESS in 12m 42s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/15/662115/5/check/swift-probetests-centos-7/d602270/ : SUCCESS in 38m 22s\n- swift-dsvm-functional http://logs.openstack.org/15/662115/5/check/swift-dsvm-functional/cb40ee6/ : SUCCESS in 31m 14s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/15/662115/5/check/swift-dsvm-functional-ipv6/180a344/ : SUCCESS in 28m 12s\n- swift-tox-lower-constraints http://logs.openstack.org/15/662115/5/check/swift-tox-lower-constraints/f2298f5/ : FAILURE in 12m 53s\n- openstack-tox-pep8 http://logs.openstack.org/15/662115/5/check/openstack-tox-pep8/8ceac73/ : SUCCESS in 6m 14s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/15/662115/5/check/swift-multinode-rolling-upgrade/288fc29/ : FAILURE in 8m 33s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c6c771a396a25bbaf4fb92ab075d7b943f6ae268","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-15 04:14:52.000000000","message":"Patch Set 5:\n\nrecheck\n\nWell that\u0027s a new one:\n\n 2019-06-14 23:25:41.624261 | ubuntu-bionic | FAIL: test_rebalance_dispersion (test.unit.common.ring.test_builder.TestRingBuilderDispersion)\n 2019-06-14 23:25:41.624398 | ubuntu-bionic | ----------------------------------------------------------------------\n 2019-06-14 23:25:41.624474 | ubuntu-bionic | Traceback (most recent call last):\n 2019-06-14 23:25:41.624682 | ubuntu-bionic |   File \"/home/zuul/src/opendev.org/openstack/swift/test/unit/common/ring/test_builder.py\", line 4439, in test_rebalance_dispersion\n 2019-06-14 23:25:41.624729 | ubuntu-bionic |     expected)\n 2019-06-14 23:25:41.624930 | ubuntu-bionic |   File \"/home/zuul/src/opendev.org/openstack/swift/test/unit/common/ring/test_builder.py\", line 4395, in assertAlmostPartCount\n 2019-06-14 23:25:41.625012 | ubuntu-bionic |     \u0027to expected:\\n\u0027 + \u0027\\n\u0027.join(msgs))\n 2019-06-14 23:25:41.625124 | ubuntu-bionic | AssertionError: some part counts not close enough to expected:\n 2019-06-14 23:25:41.625201 | ubuntu-bionic | parts in 0 was 768 expected 768 (ok)\n 2019-06-14 23:25:41.625278 | ubuntu-bionic | parts in 1 was 380 expected 384 (!\u003d)\n 2019-06-14 23:25:41.625354 | ubuntu-bionic | parts in 2 was 388 expected 384 (!\u003d)","accounts_in_message":[],"_revision_number":5},{"id":"527569a6c555a2f35107b462b3ccaf865f7ca7ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-15 08:27:00.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/15/662115/5/check/openstack-tox-docs/fc16010/html/ : SUCCESS in 4m 45s\n- swift-build-image http://logs.openstack.org/15/662115/5/check/swift-build-image/71b175d/ : SUCCESS in 8m 15s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/15/662115/5/check/swift-tox-py27/bc7a5fd/ : SUCCESS in 11m 14s\n- swift-tox-py36 http://logs.openstack.org/15/662115/5/check/swift-tox-py36/a4e538d/ : FAILURE in 12m 37s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/15/662115/5/check/swift-tox-py37/55cb618/ : FAILURE in 11m 34s\n- swift-tox-func http://logs.openstack.org/15/662115/5/check/swift-tox-func/c693769/ : SUCCESS in 8m 33s\n- swift-tox-func-py37 http://logs.openstack.org/15/662115/5/check/swift-tox-func-py37/9aaee1d/ : SUCCESS in 8m 40s\n- swift-tox-func-encryption http://logs.openstack.org/15/662115/5/check/swift-tox-func-encryption/aeb357e/ : SUCCESS in 9m 21s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/15/662115/5/check/swift-tox-func-domain-remap-staticweb/6a00c4f/ : SUCCESS in 8m 52s\n- swift-tox-func-ec http://logs.openstack.org/15/662115/5/check/swift-tox-func-ec/26e625b/ : SUCCESS in 9m 35s\n- swift-tox-func-s3api http://logs.openstack.org/15/662115/5/check/swift-tox-func-s3api/5d6fa06/ : SUCCESS in 11m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/15/662115/5/check/swift-tox-func-s3api-ceph-s3tests-tempauth/6f89286/ : SUCCESS in 13m 30s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/15/662115/5/check/swift-probetests-centos-7/b1968b6/ : FAILURE in 33m 01s\n- swift-dsvm-functional http://logs.openstack.org/15/662115/5/check/swift-dsvm-functional/fde154a/ : SUCCESS in 30m 36s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/15/662115/5/check/swift-dsvm-functional-ipv6/07c25b4/ : SUCCESS in 29m 58s\n- swift-tox-lower-constraints http://logs.openstack.org/15/662115/5/check/swift-tox-lower-constraints/3fb1542/ : SUCCESS in 15m 37s\n- openstack-tox-pep8 http://logs.openstack.org/15/662115/5/check/openstack-tox-pep8/d84ffc3/ : SUCCESS in 7m 22s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/15/662115/5/check/swift-multinode-rolling-upgrade/1427614/ : FAILURE in 9m 27s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7af30c3c9348e4a46566e04eb54eb641fb7faac5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-17 14:58:48.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"0dca2c697f526a54a56c7c82d2b7d00b61dd141e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 18:38:11.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/15/662115/5/check/openstack-tox-docs/2d9da0b/html/ : SUCCESS in 4m 17s\n- swift-build-image http://logs.openstack.org/15/662115/5/check/swift-build-image/530b2df/ : SUCCESS in 9m 50s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/15/662115/5/check/swift-tox-py27/294b903/ : SUCCESS in 11m 31s\n- swift-tox-py36 http://logs.openstack.org/15/662115/5/check/swift-tox-py36/4a412ac/ : SUCCESS in 10m 59s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/15/662115/5/check/swift-tox-py37/75d57df/ : SUCCESS in 10m 18s\n- swift-tox-func http://logs.openstack.org/15/662115/5/check/swift-tox-func/51b30d9/ : SUCCESS in 13m 30s\n- swift-tox-func-py37 http://logs.openstack.org/15/662115/5/check/swift-tox-func-py37/40ca935/ : SUCCESS in 11m 01s\n- swift-tox-func-encryption http://logs.openstack.org/15/662115/5/check/swift-tox-func-encryption/d39f2fc/ : SUCCESS in 8m 59s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/15/662115/5/check/swift-tox-func-domain-remap-staticweb/19ab76b/ : SUCCESS in 8m 14s\n- swift-tox-func-ec http://logs.openstack.org/15/662115/5/check/swift-tox-func-ec/a0dc049/ : SUCCESS in 9m 44s\n- swift-tox-func-s3api http://logs.openstack.org/15/662115/5/check/swift-tox-func-s3api/e6bbb4c/ : SUCCESS in 17m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/15/662115/5/check/swift-tox-func-s3api-ceph-s3tests-tempauth/bb43bcb/ : SUCCESS in 14m 29s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/15/662115/5/check/swift-probetests-centos-7/7eed6ef/ : SUCCESS in 29m 57s\n- swift-dsvm-functional http://logs.openstack.org/15/662115/5/check/swift-dsvm-functional/f9cf548/ : SUCCESS in 25m 55s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/15/662115/5/check/swift-dsvm-functional-ipv6/32214b7/ : SUCCESS in 32m 20s\n- swift-tox-lower-constraints http://logs.openstack.org/15/662115/5/check/swift-tox-lower-constraints/cb1d347/ : SUCCESS in 11m 25s\n- openstack-tox-pep8 http://logs.openstack.org/15/662115/5/check/openstack-tox-pep8/81977c5/ : SUCCESS in 5m 46s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/15/662115/5/check/swift-multinode-rolling-upgrade/b9ca67a/ : FAILURE in 8m 33s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"063f114814bc85006ad59f0100f2e069cf0f0134","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2019-06-18 17:38:56.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nlgtm, ran unit tests under py3 and unit/func tests under py2.\nalso ran tests as documented in Tim\u0027s comment for patchset1","accounts_in_message":[],"_revision_number":5},{"id":"b82cfb6d28d8224f245b80245914f6e2ed873940","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 17:39:17.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"6d46ca2e95e76ea08002e89f98276de67d4cf4e5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 20:19:12.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/15/662115/5/gate/openstack-tox-docs/5e05da0/html/ : SUCCESS in 4m 11s\n- swift-upload-image http://logs.openstack.org/15/662115/5/gate/swift-upload-image/03367a9/ : SUCCESS in 8m 28s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/15/662115/5/gate/swift-tox-py27/385c2b2/ : SUCCESS in 12m 40s\n- swift-tox-py37 http://logs.openstack.org/15/662115/5/gate/swift-tox-py37/51d76fb/ : SUCCESS in 11m 38s\n- swift-tox-func http://logs.openstack.org/15/662115/5/gate/swift-tox-func/76fee46/ : SUCCESS in 12m 24s\n- swift-tox-func-py37 http://logs.openstack.org/15/662115/5/gate/swift-tox-func-py37/f888982/ : SUCCESS in 9m 27s\n- swift-tox-func-encryption http://logs.openstack.org/15/662115/5/gate/swift-tox-func-encryption/1df2241/ : SUCCESS in 8m 32s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/15/662115/5/gate/swift-tox-func-domain-remap-staticweb/cf24608/ : SUCCESS in 13m 19s\n- swift-tox-func-ec http://logs.openstack.org/15/662115/5/gate/swift-tox-func-ec/5dead83/ : SUCCESS in 11m 59s\n- swift-tox-func-s3api http://logs.openstack.org/15/662115/5/gate/swift-tox-func-s3api/b1247af/ : SUCCESS in 15m 29s\n- swift-probetests-centos-7 http://logs.openstack.org/15/662115/5/gate/swift-probetests-centos-7/d8cbce7/ : TIMED_OUT in 1h 00m 56s\n- swift-dsvm-functional http://logs.openstack.org/15/662115/5/gate/swift-dsvm-functional/6498c4e/ : SUCCESS in 26m 24s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/15/662115/5/gate/swift-dsvm-functional-ipv6/5633f3a/ : SUCCESS in 30m 50s\n- swift-tox-lower-constraints http://logs.openstack.org/15/662115/5/gate/swift-tox-lower-constraints/88fd04e/ : SUCCESS in 11m 11s\n- openstack-tox-pep8 http://logs.openstack.org/15/662115/5/gate/openstack-tox-pep8/85d6122/ : SUCCESS in 6m 13s","accounts_in_message":[],"_revision_number":5},{"id":"cc1076e991c4581ac072b00b5cc5e9e7266e4902","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-18 20:26:48.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"3da5a0f90c5c8c06ec2538829e5e0ec8f470e0ad","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-06-18 22:59:11.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"fa0341db3729b2789a1543ec41064d42a06d8074","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 04:50:12.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/15/662115/5/check/openstack-tox-docs/5547985/html/ : SUCCESS in 4m 16s\n- swift-build-image http://logs.openstack.org/15/662115/5/check/swift-build-image/ddc9614/ : SUCCESS in 7m 49s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/15/662115/5/check/swift-tox-py27/ae1d98c/ : SUCCESS in 11m 50s\n- swift-tox-py36 http://logs.openstack.org/15/662115/5/check/swift-tox-py36/51652f6/ : SUCCESS in 10m 34s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/15/662115/5/check/swift-tox-py37/7960a36/ : FAILURE in 11m 44s\n- swift-tox-func http://logs.openstack.org/15/662115/5/check/swift-tox-func/06a1979/ : SUCCESS in 8m 37s\n- swift-tox-func-py37 http://logs.openstack.org/15/662115/5/check/swift-tox-func-py37/eb3e98d/ : SUCCESS in 8m 57s\n- swift-tox-func-encryption http://logs.openstack.org/15/662115/5/check/swift-tox-func-encryption/25426cf/ : SUCCESS in 8m 44s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/15/662115/5/check/swift-tox-func-domain-remap-staticweb/141e0e9/ : SUCCESS in 7m 58s\n- swift-tox-func-ec http://logs.openstack.org/15/662115/5/check/swift-tox-func-ec/224c2af/ : SUCCESS in 9m 51s\n- swift-tox-func-s3api http://logs.openstack.org/15/662115/5/check/swift-tox-func-s3api/61d7283/ : SUCCESS in 11m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/15/662115/5/check/swift-tox-func-s3api-ceph-s3tests-tempauth/eca064c/ : SUCCESS in 13m 02s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/15/662115/5/check/swift-probetests-centos-7/38c69bd/ : SUCCESS in 34m 57s\n- swift-dsvm-functional http://logs.openstack.org/15/662115/5/check/swift-dsvm-functional/958cbae/ : SUCCESS in 29m 36s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/15/662115/5/check/swift-dsvm-functional-ipv6/bbd67b4/ : SUCCESS in 29m 14s\n- swift-tox-lower-constraints http://logs.openstack.org/15/662115/5/check/swift-tox-lower-constraints/636e284/ : SUCCESS in 11m 44s\n- openstack-tox-pep8 http://logs.openstack.org/15/662115/5/check/openstack-tox-pep8/08fe849/ : SUCCESS in 5m 19s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/15/662115/5/check/swift-multinode-rolling-upgrade/da6e6f0/ : FAILURE in 9m 42s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a0dbd069d99498754bc1f824a0e30a3511b6666b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-19 04:53:14.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"3a23643549be602d84603de8afb580551a3358be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 10:22:16.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/15/662115/5/check/openstack-tox-docs/7ca5e6b/html/ : SUCCESS in 5m 19s\n- swift-build-image http://logs.openstack.org/15/662115/5/check/swift-build-image/99c4b8b/ : SUCCESS in 9m 35s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/15/662115/5/check/swift-tox-py27/5b470f9/ : SUCCESS in 10m 20s\n- swift-tox-py36 http://logs.openstack.org/15/662115/5/check/swift-tox-py36/d555b4b/ : SUCCESS in 11m 10s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/15/662115/5/check/swift-tox-py37/4fd398d/ : SUCCESS in 10m 43s\n- swift-tox-func http://logs.openstack.org/15/662115/5/check/swift-tox-func/17b65cb/ : SUCCESS in 8m 40s\n- swift-tox-func-py37 http://logs.openstack.org/15/662115/5/check/swift-tox-func-py37/4d452ac/ : SUCCESS in 9m 30s\n- swift-tox-func-encryption http://logs.openstack.org/15/662115/5/check/swift-tox-func-encryption/500278b/ : SUCCESS in 9m 36s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/15/662115/5/check/swift-tox-func-domain-remap-staticweb/c7db679/ : SUCCESS in 8m 20s\n- swift-tox-func-ec http://logs.openstack.org/15/662115/5/check/swift-tox-func-ec/a12770a/ : SUCCESS in 8m 59s\n- swift-tox-func-s3api http://logs.openstack.org/15/662115/5/check/swift-tox-func-s3api/c983115/ : SUCCESS in 14m 24s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/15/662115/5/check/swift-tox-func-s3api-ceph-s3tests-tempauth/07cdd3f/ : SUCCESS in 18m 58s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/15/662115/5/check/swift-probetests-centos-7/8b52a21/ : SUCCESS in 31m 23s\n- swift-dsvm-functional http://logs.openstack.org/15/662115/5/check/swift-dsvm-functional/11c4f66/ : SUCCESS in 26m 04s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/15/662115/5/check/swift-dsvm-functional-ipv6/bd4263f/ : SUCCESS in 29m 11s\n- swift-tox-lower-constraints http://logs.openstack.org/15/662115/5/check/swift-tox-lower-constraints/ee33415/ : SUCCESS in 11m 09s\n- openstack-tox-pep8 http://logs.openstack.org/15/662115/5/check/openstack-tox-pep8/d5bbc18/ : SUCCESS in 6m 11s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/15/662115/5/check/swift-multinode-rolling-upgrade/f6a7663/ : FAILURE in 8m 02s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"8d6637a39156ec6b43ba4d0a073afce3a49f510a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 10:22:31.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"09d245cc0284bd326c7700616c7db170d52836aa","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2019-06-19 12:36:16.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"9220d49feee8c2daf1777759236f15c520006999","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 14:13:13.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/15/662115/5/gate/openstack-tox-docs/8187654/html/ : SUCCESS in 4m 57s\n- swift-upload-image http://logs.openstack.org/15/662115/5/gate/swift-upload-image/b75a2e4/ : SUCCESS in 9m 11s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/15/662115/5/gate/swift-tox-py27/79d6213/ : SUCCESS in 10m 13s\n- swift-tox-py37 http://logs.openstack.org/15/662115/5/gate/swift-tox-py37/bd8a466/ : SUCCESS in 14m 58s\n- swift-tox-func http://logs.openstack.org/15/662115/5/gate/swift-tox-func/83288b0/ : SUCCESS in 9m 18s\n- swift-tox-func-py37 http://logs.openstack.org/15/662115/5/gate/swift-tox-func-py37/432125b/ : SUCCESS in 8m 54s\n- swift-tox-func-encryption http://logs.openstack.org/15/662115/5/gate/swift-tox-func-encryption/b1c9d1b/ : FAILURE in 10m 27s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/15/662115/5/gate/swift-tox-func-domain-remap-staticweb/0aa65ef/ : SUCCESS in 9m 30s\n- swift-tox-func-ec http://logs.openstack.org/15/662115/5/gate/swift-tox-func-ec/1479141/ : SUCCESS in 10m 21s\n- swift-tox-func-s3api http://logs.openstack.org/15/662115/5/gate/swift-tox-func-s3api/22e5a28/ : SUCCESS in 24m 22s\n- swift-probetests-centos-7 http://logs.openstack.org/15/662115/5/gate/swift-probetests-centos-7/6f798a8/ : FAILURE in 31m 35s\n- swift-dsvm-functional http://logs.openstack.org/15/662115/5/gate/swift-dsvm-functional/ccf2b8b/ : SUCCESS in 27m 28s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/15/662115/5/gate/swift-dsvm-functional-ipv6/d76f2f7/ : SUCCESS in 26m 52s\n- swift-tox-lower-constraints http://logs.openstack.org/15/662115/5/gate/swift-tox-lower-constraints/b8b356f/ : SUCCESS in 13m 47s\n- openstack-tox-pep8 http://logs.openstack.org/15/662115/5/gate/openstack-tox-pep8/8b51be1/ : SUCCESS in 7m 18s","accounts_in_message":[],"_revision_number":5},{"id":"0c28297d60096146e35c26a269b3d13b4eaae46f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 19:01:45.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/15/662115/5/check/openstack-tox-docs/3541520/html/ : SUCCESS in 4m 44s\n- swift-build-image http://logs.openstack.org/15/662115/5/check/swift-build-image/b4ee45e/ : SUCCESS in 8m 37s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/15/662115/5/check/swift-tox-py27/13e9975/ : SUCCESS in 10m 59s\n- swift-tox-py36 http://logs.openstack.org/15/662115/5/check/swift-tox-py36/a2fd28a/ : SUCCESS in 11m 53s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/15/662115/5/check/swift-tox-py37/47ed16b/ : SUCCESS in 10m 02s\n- swift-tox-func http://logs.openstack.org/15/662115/5/check/swift-tox-func/e4b4223/ : SUCCESS in 17m 22s\n- swift-tox-func-py37 http://logs.openstack.org/15/662115/5/check/swift-tox-func-py37/cdf947e/ : SUCCESS in 9m 36s\n- swift-tox-func-encryption http://logs.openstack.org/15/662115/5/check/swift-tox-func-encryption/f119db5/ : SUCCESS in 9m 34s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/15/662115/5/check/swift-tox-func-domain-remap-staticweb/1a0f423/ : SUCCESS in 8m 43s\n- swift-tox-func-ec http://logs.openstack.org/15/662115/5/check/swift-tox-func-ec/c24418e/ : SUCCESS in 10m 09s\n- swift-tox-func-s3api http://logs.openstack.org/15/662115/5/check/swift-tox-func-s3api/64f2835/ : SUCCESS in 12m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/15/662115/5/check/swift-tox-func-s3api-ceph-s3tests-tempauth/46790b1/ : SUCCESS in 18m 17s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/15/662115/5/check/swift-probetests-centos-7/3d1a2cf/ : SUCCESS in 42m 19s\n- swift-dsvm-functional http://logs.openstack.org/15/662115/5/check/swift-dsvm-functional/6ca2e9e/ : SUCCESS in 26m 57s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/15/662115/5/check/swift-dsvm-functional-ipv6/73e2c86/ : SUCCESS in 25m 44s\n- swift-tox-lower-constraints http://logs.openstack.org/15/662115/5/check/swift-tox-lower-constraints/7b40b85/ : SUCCESS in 11m 14s\n- openstack-tox-pep8 http://logs.openstack.org/15/662115/5/check/openstack-tox-pep8/d457feb/ : SUCCESS in 6m 02s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/15/662115/5/check/swift-multinode-rolling-upgrade/9b44384/ : FAILURE in 9m 58s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"8be8d0b2f77dfdadec96f91b6c93c13e0cfd53b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 19:02:09.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"3b241a2d1f4653e954375e7222e979f21fd7eeef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 20:08:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"feb0fb1a4a044191a535cfb161bcf2e9a799d52d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 20:08:40.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/15/662115/5/gate/openstack-tox-docs/9f46136/html/ : SUCCESS in 4m 49s\n- swift-upload-image http://logs.openstack.org/15/662115/5/gate/swift-upload-image/8d74585/ : SUCCESS in 9m 41s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/15/662115/5/gate/swift-tox-py27/2732f67/ : SUCCESS in 9m 48s\n- swift-tox-py37 http://logs.openstack.org/15/662115/5/gate/swift-tox-py37/fec5142/ : SUCCESS in 11m 00s\n- swift-tox-func http://logs.openstack.org/15/662115/5/gate/swift-tox-func/870be29/ : SUCCESS in 8m 39s\n- swift-tox-func-py37 http://logs.openstack.org/15/662115/5/gate/swift-tox-func-py37/3a67666/ : SUCCESS in 9m 04s\n- swift-tox-func-encryption http://logs.openstack.org/15/662115/5/gate/swift-tox-func-encryption/f124769/ : SUCCESS in 9m 17s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/15/662115/5/gate/swift-tox-func-domain-remap-staticweb/cff4cc1/ : SUCCESS in 7m 59s\n- swift-tox-func-ec http://logs.openstack.org/15/662115/5/gate/swift-tox-func-ec/874fb3b/ : SUCCESS in 16m 10s\n- swift-tox-func-s3api http://logs.openstack.org/15/662115/5/gate/swift-tox-func-s3api/0e50985/ : SUCCESS in 14m 40s\n- swift-probetests-centos-7 http://logs.openstack.org/15/662115/5/gate/swift-probetests-centos-7/3df976f/ : SUCCESS in 31m 40s\n- swift-dsvm-functional http://logs.openstack.org/15/662115/5/gate/swift-dsvm-functional/900ead7/ : SUCCESS in 39m 22s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/15/662115/5/gate/swift-dsvm-functional-ipv6/1fd33db/ : SUCCESS in 29m 10s\n- swift-tox-lower-constraints http://logs.openstack.org/15/662115/5/gate/swift-tox-lower-constraints/c946fba/ : SUCCESS in 22m 01s\n- openstack-tox-pep8 http://logs.openstack.org/15/662115/5/gate/openstack-tox-pep8/9de8a0b/ : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":5},{"id":"351508ae6a717a990b6d6a0ec15699ce6d2a41c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 20:09:56.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- swift-promote-image http://logs.openstack.org/15/662115/5/promote/swift-promote-image/d821c80/ : SUCCESS in 48s (non-voting)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"ef8818a6399e295ebf0dc891ef483e4840a61475","revisions":{"5d3f4fdbc1265035f70b3815495892a48924a8c3":{"kind":"REWORK","_number":1,"created":"2019-05-30 01:14:51.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/15/662115/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/15/662115/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/662115/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/662115/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/662115/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/15/662115/1"}}},"commit":{"parents":[{"commit":"769c0fb7fb95d3f9bfe68196af41f40e403c87fa","subject":"Merge \"Rework private-request-method interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/769c0fb7fb95d3f9bfe68196af41f40e403c87fa"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-05-30 01:14:17.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-05-30 01:14:17.000000000","tz":-420},"subject":"WIP: fix up how we memcache on py3","message":"WIP: fix up how we memcache on py3\n\nPreviously, we stored the WSGI strings in memcached and returned them when\nresponding to get_account/container_info calls. This would lead to cache\ncorruption in a heterogenous py2/py3 cluster such as you would have during\na rolling upgrade.\n\nNow, only store and return native strings.\n\nSeparately, work around https://bugs.python.org/issue37093 which would\ntruncate headers, for example, on existing containers with metadata like\n\n    X-Container-Meta-α: Ω\n\nChange-Id: I8d6f66dfe846493972e433f70bad76a33d204562\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5d3f4fdbc1265035f70b3815495892a48924a8c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5d3f4fdbc1265035f70b3815495892a48924a8c3"}]},"branch":"refs/heads/master"},"2c505cf0ff9176625af16ccb357158902c898080":{"kind":"REWORK","_number":2,"created":"2019-05-31 19:08:29.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/15/662115/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/15/662115/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/662115/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/662115/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/662115/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/15/662115/2"}}},"commit":{"parents":[{"commit":"4f9595f1130f6c45f8e3e08db55dddc31844be93","subject":"py3: port symlink middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4f9595f1130f6c45f8e3e08db55dddc31844be93"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-05-30 01:14:17.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-05-31 19:03:58.000000000","tz":-420},"subject":"Fix up how we memcache on py3","message":"Fix up how we memcache on py3\n\nPreviously, we stored the WSGI strings in memcached and returned them when\nresponding to get_account/container_info calls. This would lead to cache\ncorruption in a heterogenous py2/py3 cluster such as you would have during\na rolling upgrade.\n\nNow, only store and return native strings.\n\nChange-Id: I8d6f66dfe846493972e433f70bad76a33d204562\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2c505cf0ff9176625af16ccb357158902c898080"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2c505cf0ff9176625af16ccb357158902c898080"}]},"branch":"refs/heads/master"},"46b07a3c1c292090c29f94413d740d3f3b0a2444":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-06-06 18:57:53.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/15/662115/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/15/662115/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/662115/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/662115/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/662115/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/15/662115/3"}}},"commit":{"parents":[{"commit":"05bfaf7ae1667b7a3f59543d17280715a5b33969","subject":"Merge \"py3: port ssync\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/05bfaf7ae1667b7a3f59543d17280715a5b33969"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-05-30 01:14:17.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-06-06 18:54:16.000000000","tz":-420},"subject":"Fix up how we memcache on py3","message":"Fix up how we memcache on py3\n\nPreviously, we stored the WSGI strings in memcached and returned them when\nresponding to get_account/container_info calls. This would lead to cache\ncorruption in a heterogenous py2/py3 cluster such as you would have during\na rolling upgrade.\n\nNow, only store and return native strings.\n\nChange-Id: I8d6f66dfe846493972e433f70bad76a33d204562\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/46b07a3c1c292090c29f94413d740d3f3b0a2444"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/46b07a3c1c292090c29f94413d740d3f3b0a2444"}]},"branch":"refs/heads/master"},"7da78db66f6c6ea2facdb15c671dcfc604dcd79d":{"kind":"REWORK","_number":4,"created":"2019-06-13 16:40:22.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/15/662115/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/15/662115/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/662115/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/662115/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/662115/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/15/662115/4"}}},"commit":{"parents":[{"commit":"aa2f1db1b71c1b2bf746b72515e3efd15598b6aa","subject":"Ensure get_*_info keys are native strings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aa2f1db1b71c1b2bf746b72515e3efd15598b6aa"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-05-30 01:14:17.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-06-13 16:39:13.000000000","tz":-420},"subject":"Fix up how we memcache on py3","message":"Fix up how we memcache on py3\n\nPreviously, we stored the WSGI strings in memcached and returned them when\nresponding to get_account/container_info calls. This would lead to cache\ncorruption in a heterogenous py2/py3 cluster such as you would have during\na rolling upgrade.\n\nNow, only store and return native strings.\n\nChange-Id: I8d6f66dfe846493972e433f70bad76a33d204562\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7da78db66f6c6ea2facdb15c671dcfc604dcd79d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7da78db66f6c6ea2facdb15c671dcfc604dcd79d"}]},"branch":"refs/heads/master"},"ef8818a6399e295ebf0dc891ef483e4840a61475":{"kind":"REWORK","_number":5,"created":"2019-06-14 15:21:25.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/15/662115/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/15/662115/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/662115/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/662115/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/662115/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/15/662115/5"}}},"commit":{"parents":[{"commit":"74e1f2e05339880b724161241f3acbb6af6ea769","subject":"Merge \"Update mailmap for Kota Tsuyuzaki\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/74e1f2e05339880b724161241f3acbb6af6ea769"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-05-30 01:14:17.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-06-14 15:20:36.000000000","tz":-420},"subject":"Fix up how we memcache on py3","message":"Fix up how we memcache on py3\n\nPreviously, we stored the WSGI strings in memcached and returned them when\nresponding to get_account/container_info calls. This would lead to cache\ncorruption in a heterogenous py2/py3 cluster such as you would have during\na rolling upgrade.\n\nNow, only store and return native strings.\n\nChange-Id: I8d6f66dfe846493972e433f70bad76a33d204562\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ef8818a6399e295ebf0dc891ef483e4840a61475"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ef8818a6399e295ebf0dc891ef483e4840a61475"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
