)]}'
{"id":"openstack%2Fswift~662546","triplet_id":"openstack%2Fswift~master~I0f03c211f35a9a49e047a5718a9907b515ca88d7","project":"openstack/swift","branch":"master","topic":"py2-tests-against-py3-servers","hashtags":[],"change_id":"I0f03c211f35a9a49e047a5718a9907b515ca88d7","subject":"py3: Be able to read and write non-ASCII headers","status":"MERGED","created":"2019-05-31 19:08:29.000000000","updated":"2019-07-10 09:36:59.000000000","submitted":"2019-07-10 09:35:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662546-1562751335181-10db0568","meta_rev_id":"7da8ce0890fbd0184bb97a19a93be6cb6f40911c","_number":662546,"virtual_id_number":662546,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":2,"date":"2019-07-10 09:35:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-10 07:29:19.000000000","_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"value":0,"date":"2019-07-03 18:37:51.000000000","_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"recommended":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2019-07-03 02:01:55.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":2,"date":"2019-07-06 03:00:00.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2019-07-10 03:29:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-06 03:00:00.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2019-07-10 03:29:18.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2019-07-10 07:29:19.000000000","updated_by":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"reviewer":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"state":"REVIEWER"},{"updated":"2019-07-10 09:35:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3449515e14dd620ed0fe17845fd284993d7f4ecc","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-31 19:08:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a2cdb39430a32fa11ae35d0f68135e2fb2d914ca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 22:44:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/46/662546/1/check/openstack-tox-docs/b59c3f4/html/ : SUCCESS in 4m 00s\n- swift-build-image http://logs.openstack.org/46/662546/1/check/swift-build-image/6cb1c21/ : SUCCESS in 9m 47s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/46/662546/1/check/swift-tox-py27/8313510/ : SUCCESS in 19m 54s\n- swift-tox-py36 http://logs.openstack.org/46/662546/1/check/swift-tox-py36/4513be4/ : SUCCESS in 17m 30s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/46/662546/1/check/swift-tox-py37/5933e7e/ : SUCCESS in 19m 46s\n- swift-tox-func http://logs.openstack.org/46/662546/1/check/swift-tox-func/34632fe/ : FAILURE in 14m 30s\n- swift-tox-func-py37 http://logs.openstack.org/46/662546/1/check/swift-tox-func-py37/2ecbc37/ : SUCCESS in 9m 10s\n- swift-tox-func-encryption http://logs.openstack.org/46/662546/1/check/swift-tox-func-encryption/f471785/ : SUCCESS in 8m 24s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/46/662546/1/check/swift-tox-func-domain-remap-staticweb/f7230c3/ : SUCCESS in 9m 58s\n- swift-tox-func-ec http://logs.openstack.org/46/662546/1/check/swift-tox-func-ec/7499796/ : SUCCESS in 9m 02s\n- swift-tox-func-s3api http://logs.openstack.org/46/662546/1/check/swift-tox-func-s3api/e822423/ : SUCCESS in 18m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/46/662546/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/bf18199/ : SUCCESS in 15m 30s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/46/662546/1/check/swift-probetests-centos-7/1eb1d0f/ : SUCCESS in 31m 23s\n- swift-dsvm-functional http://logs.openstack.org/46/662546/1/check/swift-dsvm-functional/b45403e/ : SUCCESS in 26m 02s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/46/662546/1/check/swift-dsvm-functional-ipv6/66487ed/ : SUCCESS in 26m 05s\n- swift-tox-lower-constraints http://logs.openstack.org/46/662546/1/check/swift-tox-lower-constraints/92c381c/ : SUCCESS in 11m 40s\n- openstack-tox-pep8 http://logs.openstack.org/46/662546/1/check/openstack-tox-pep8/5686042/ : SUCCESS in 5m 03s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/46/662546/1/check/swift-multinode-rolling-upgrade/3e20ee2/ : FAILURE in 8m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5ae28130bf0d3bb8215216da47f2f9548dc549e7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-31 22:54:51.000000000","message":"Patch Set 1:\n\nrecheck\n\nSmells like an eventual-consistency issue.","accounts_in_message":[],"_revision_number":1},{"id":"6eaefa8ccac9e01ea87402ea90c9c48dbfac17a2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-01 00:56:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/46/662546/1/check/openstack-tox-docs/0432f0a/html/ : SUCCESS in 4m 28s\n- swift-build-image http://logs.openstack.org/46/662546/1/check/swift-build-image/9b39e94/ : SUCCESS in 8m 30s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/46/662546/1/check/swift-tox-py27/479a29a/ : SUCCESS in 11m 46s\n- swift-tox-py36 http://logs.openstack.org/46/662546/1/check/swift-tox-py36/309b450/ : SUCCESS in 12m 31s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/46/662546/1/check/swift-tox-py37/d8e1e90/ : SUCCESS in 17m 37s\n- swift-tox-func http://logs.openstack.org/46/662546/1/check/swift-tox-func/0dac458/ : SUCCESS in 8m 48s\n- swift-tox-func-py37 http://logs.openstack.org/46/662546/1/check/swift-tox-func-py37/67e72fd/ : SUCCESS in 7m 13s\n- swift-tox-func-encryption http://logs.openstack.org/46/662546/1/check/swift-tox-func-encryption/7ca782c/ : SUCCESS in 9m 17s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/46/662546/1/check/swift-tox-func-domain-remap-staticweb/1220eff/ : SUCCESS in 8m 57s\n- swift-tox-func-ec http://logs.openstack.org/46/662546/1/check/swift-tox-func-ec/9b44f36/ : SUCCESS in 13m 39s\n- swift-tox-func-s3api http://logs.openstack.org/46/662546/1/check/swift-tox-func-s3api/b5693e0/ : SUCCESS in 11m 50s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/46/662546/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/e9f2d0c/ : SUCCESS in 12m 28s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/46/662546/1/check/swift-probetests-centos-7/e553f5c/ : SUCCESS in 32m 18s\n- swift-dsvm-functional http://logs.openstack.org/46/662546/1/check/swift-dsvm-functional/11f0e96/ : SUCCESS in 25m 10s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/46/662546/1/check/swift-dsvm-functional-ipv6/23ab740/ : SUCCESS in 25m 34s\n- swift-tox-lower-constraints http://logs.openstack.org/46/662546/1/check/swift-tox-lower-constraints/6b59bb5/ : SUCCESS in 11m 30s\n- openstack-tox-pep8 http://logs.openstack.org/46/662546/1/check/openstack-tox-pep8/b2e6ff5/ : SUCCESS in 5m 52s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/46/662546/1/check/swift-multinode-rolling-upgrade/595d1ed/ : FAILURE in 8m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4966b47028aea51f835a5cf0bfbc0d4c57a010a1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-06 18:17:38.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c4971125d2e812fe18ffa739cf27d86f0a9530f3","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-06 23:01:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d3f4fc3f1eb96bbb1cd65493aad1438c792d2c2b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 03:40:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/46/662546/2/check/openstack-tox-docs/e35129a/html/ : SUCCESS in 4m 38s\n- swift-build-image http://logs.openstack.org/46/662546/2/check/swift-build-image/4c03b61/ : TIMED_OUT in 31m 21s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/46/662546/2/check/swift-tox-py27/81af500/ : SUCCESS in 17m 54s\n- swift-tox-py36 http://logs.openstack.org/46/662546/2/check/swift-tox-py36/15cd1c3/ : FAILURE in 14m 47s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/46/662546/2/check/swift-tox-py37/18f5ac5/ : FAILURE in 10m 43s\n- swift-tox-func http://logs.openstack.org/46/662546/2/check/swift-tox-func/72b95f2/ : SUCCESS in 10m 56s\n- swift-tox-func-py37 http://logs.openstack.org/46/662546/2/check/swift-tox-func-py37/bd4932a/ : FAILURE in 4m 29s\n- swift-tox-func-encryption http://logs.openstack.org/46/662546/2/check/swift-tox-func-encryption/0defe6d/ : FAILURE in 9m 28s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/46/662546/2/check/swift-tox-func-domain-remap-staticweb/14f3465/ : SUCCESS in 10m 21s\n- swift-tox-func-ec http://logs.openstack.org/46/662546/2/check/swift-tox-func-ec/68e67c9/ : SUCCESS in 9m 17s\n- swift-tox-func-s3api http://logs.openstack.org/46/662546/2/check/swift-tox-func-s3api/718676e/ : SUCCESS in 18m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/46/662546/2/check/swift-tox-func-s3api-ceph-s3tests-tempauth/6b1d547/ : SUCCESS in 19m 05s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/46/662546/2/check/swift-probetests-centos-7/2397dbb/ : SUCCESS in 29m 53s\n- swift-dsvm-functional http://logs.openstack.org/46/662546/2/check/swift-dsvm-functional/cab38ce/ : SUCCESS in 25m 33s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/46/662546/2/check/swift-dsvm-functional-ipv6/d1e69ee/ : SUCCESS in 28m 00s\n- swift-tox-lower-constraints http://logs.openstack.org/46/662546/2/check/swift-tox-lower-constraints/cdc67b6/ : SUCCESS in 12m 24s\n- openstack-tox-pep8 http://logs.openstack.org/46/662546/2/check/openstack-tox-pep8/cc0885a/ : FAILURE in 5m 30s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/46/662546/2/check/swift-multinode-rolling-upgrade/0558048/ : FAILURE in 8m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"163d37ef00915087a6b11debe9ee28243f6e06f1","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-06-07 22:01:16.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\n(-1 for the question)","accounts_in_message":[],"_revision_number":2},{"id":"e2c773abe69320dd25cc945a8fe58a0f298aa444","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-07 22:15:37.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nI\u0027ll play with this some more, try to find a better way.","accounts_in_message":[],"_revision_number":2},{"id":"6a5e471ab79241976d1d3322726250ff9333f17e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-17 16:58:33.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"637b7a6ddb5f8562af3adb49780280652542fe72","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-17 20:10:31.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d858eb35e87511aabdef640372a619ed56c799ec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 00:47:33.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/46/662546/4/check/openstack-tox-docs/cea9ecc/html/ : SUCCESS in 4m 55s\n- swift-build-image http://logs.openstack.org/46/662546/4/check/swift-build-image/90efe41/ : POST_FAILURE in 11m 33s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/46/662546/4/check/swift-tox-py27/743d812/ : SUCCESS in 10m 32s\n- swift-tox-py36 http://logs.openstack.org/46/662546/4/check/swift-tox-py36/587ab80/ : SUCCESS in 10m 50s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/46/662546/4/check/swift-tox-py37/43e16a6/ : SUCCESS in 12m 00s\n- swift-tox-func http://logs.openstack.org/46/662546/4/check/swift-tox-func/24fd571/ : SUCCESS in 9m 02s\n- swift-tox-func-py37 http://logs.openstack.org/46/662546/4/check/swift-tox-func-py37/9d13786/ : SUCCESS in 9m 21s\n- swift-tox-func-encryption http://logs.openstack.org/46/662546/4/check/swift-tox-func-encryption/626213f/ : SUCCESS in 11m 49s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/46/662546/4/check/swift-tox-func-domain-remap-staticweb/f16635c/ : SUCCESS in 8m 58s\n- swift-tox-func-ec http://logs.openstack.org/46/662546/4/check/swift-tox-func-ec/1cf1662/ : SUCCESS in 9m 56s\n- swift-tox-func-s3api http://logs.openstack.org/46/662546/4/check/swift-tox-func-s3api/85e0669/ : SUCCESS in 11m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/46/662546/4/check/swift-tox-func-s3api-ceph-s3tests-tempauth/2628530/ : SUCCESS in 14m 22s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/46/662546/4/check/swift-probetests-centos-7/a9b216f/ : SUCCESS in 35m 53s\n- swift-dsvm-functional http://logs.openstack.org/46/662546/4/check/swift-dsvm-functional/dd0cceb/ : SUCCESS in 29m 12s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/46/662546/4/check/swift-dsvm-functional-ipv6/0b52bbe/ : SUCCESS in 28m 35s\n- swift-tox-lower-constraints http://logs.openstack.org/46/662546/4/check/swift-tox-lower-constraints/6f95c00/ : SUCCESS in 11m 18s\n- openstack-tox-pep8 http://logs.openstack.org/46/662546/4/check/openstack-tox-pep8/20d1097/ : SUCCESS in 5m 06s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/46/662546/4/check/swift-multinode-rolling-upgrade/8beeb64/ : FAILURE in 9m 13s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4d50ca16ffd82b2ac17084c5a4246b54a55e8cf8","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2019-07-02 07:14:04.000000000","message":"Patch Set 4: Code-Review+1\n\nlooks good, just gone down a test py2 func tests to a py3 swift rabbit hole. The SAIO I was using was a little old, and the object servers didn\u0027t start in py3.4 with this patch checked out. But it\u0027s rather old.\n\nBuilding a more recent OpenSuse SAIO. So just +1ing until I\u0027m done.","accounts_in_message":[],"_revision_number":4},{"id":"3143a5ea45674b208276e38cef9fb11c5b1658fc","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2019-07-02 11:35:12.000000000","message":"Patch Set 4:\n\nI\u0027m failing to start the object servers in py36 as well. Am I missing a patch that hasn\u0027t landed or simply missing something obvious.. it is getting late here.. so it\u0027s entirely possible :P\n\n  b\u0027Traceback (most recent call last):\\n  File \"/home/matt/venv/swift/py3/bin/swift-object-server\", line 7, in \u003cmodule\u003e\\n    exec(compile(f.read(), __file__, \\\u0027exec\\\u0027))\\n  File \"/home/matt/code/swift/bin/swift-object-server\", line 27, in \u003cmodule\u003e\\n    **options))\\n  File \"/home/matt/code/swift/swift/common/wsgi.py\", line 1079, in run_wsgi\\n    loadapp(conf_path, global_conf\u003dglobal_conf)\\n  File \"/home/matt/code/swift/swift/common/wsgi.py\", line 375, in loadapp\\n    ctx \u003d loadcontext(loadwsgi.APP, conf_file, global_conf\u003dglobal_conf)\\n  File \"/home/matt/code/swift/swift/common/wsgi.py\", line 359, in loadcontext\\n    global_conf\u003dglobal_conf)\\n  File \"/home/matt/venv/swift/py3/lib/python3.6/site-packages/paste/deploy/loadwsgi.py\", line 302, in loadcontext\\n    global_conf\u003dglobal_conf)\\n  File \"/home/matt/venv/swift/py3/lib/python3.6/site-packages/paste/deploy/loadwsgi.py\", line 326, in _loadconfig\\n    return loader.get_context(object_type, name, global_conf)\\n  File \"/home/matt/code/swift/swift/common/wsgi.py\", line 66, in get_context\\n    object_type, name\u003dname, global_conf\u003dglobal_conf)\\n  File \"/home/matt/venv/swift/py3/lib/python3.6/site-packages/paste/deploy/loadwsgi.py\", line 415, in get_context\\n    defaults \u003d self.parser.defaults()\\n  File \"/home/matt/venv/swift/py3/lib/python3.6/site-packages/paste/deploy/loadwsgi.py\", line 74, in defaults\\n    defaults[key] \u003d self.get(\\\u0027DEFAULT\\\u0027, key) or val\\n  File \"/usr/lib64/python3.6/configparser.py\", line 800, in get\\n    d)\\n  File \"/home/matt/venv/swift/py3/lib/python3.6/site-packages/paste/deploy/loadwsgi.py\", line 101, in before_get\\n    value, defaults)\\n  File \"/usr/lib64/python3.6/configparser.py\", line 394, in before_get\\n    self._interpolate_some(parser, option, L, value, section, defaults, 1)\\n  File \"/usr/lib64/python3.6/configparser.py\", line 411, in _interpolate_some\\n    p \u003d rest.find(\"%\")\\nAttributeError: Error in file /etc/swift/object-server/1.conf: \\\u0027list\\\u0027 object has no attribute \\\u0027find\\\u0027\\n\u0027\n\n^ I\u0027ll debug it properly in the morning. Too tired now.","accounts_in_message":[],"_revision_number":4},{"id":"2a6a3f8d556dc1737c6ea7fe7515451de271a22a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-07-02 18:49:15.000000000","message":"Patch Set 4:\n\nShould be fixed by https://review.opendev.org/#/c/665494/ (which is now landed on master, but I pushed this up before then). You could also work around it by setting replication_concurrency to zero -- sorry, this final push toward func tests means that we need a bunch of different patches all together to make it work. I think I still like having the separate patches though, since they each tell part of the story.","accounts_in_message":[],"_revision_number":4},{"id":"e0de71a35bd3b1c1eb28fd919ffc71ad7a58f0ac","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2019-07-03 02:01:40.000000000","message":"Patch Set 4:\n\nAhh yes a rebase to master seems to have fixed it. \n\nMaybe we should just rebase this patch. I\u0027ll give it a go.","accounts_in_message":[],"_revision_number":4},{"id":"09ca521f4ecc62e258ce9392e657ce12f87d6b10","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2019-07-03 02:01:55.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"773d204f4a8d83179aadd82a19637bc046813dcd","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2019-07-03 02:02:33.000000000","message":"Patch Set 5:\n\nRebased using the button in gerrit.. I\u0027ll assume it did the right thing :P","accounts_in_message":[],"_revision_number":5},{"id":"e80419edc02052ac05d7e2b40447487aedac145c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 02:49:18.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/46/662546/5/check/openstack-tox-docs/3a621b8/html/ : SUCCESS in 6m 24s\n- swift-build-image http://logs.openstack.org/46/662546/5/check/swift-build-image/5fcd68a/ : SUCCESS in 10m 16s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/46/662546/5/check/swift-tox-py27/5b0dec2/ : SUCCESS in 16m 37s\n- swift-tox-py36 http://logs.openstack.org/46/662546/5/check/swift-tox-py36/5ffc8d3/ : SUCCESS in 16m 52s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/46/662546/5/check/swift-tox-py37/f507bd1/ : FAILURE in 16m 47s\n- swift-tox-func http://logs.openstack.org/46/662546/5/check/swift-tox-func/50f4f30/ : SUCCESS in 17m 59s\n- swift-tox-func-py37 http://logs.openstack.org/46/662546/5/check/swift-tox-func-py37/c3538e2/ : SUCCESS in 16m 27s\n- swift-tox-func-encryption http://logs.openstack.org/46/662546/5/check/swift-tox-func-encryption/1d1d214/ : SUCCESS in 13m 47s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/46/662546/5/check/swift-tox-func-domain-remap-staticweb/3b5a24a/ : SUCCESS in 17m 44s\n- swift-tox-func-ec http://logs.openstack.org/46/662546/5/check/swift-tox-func-ec/3223ada/ : SUCCESS in 16m 00s\n- swift-tox-func-s3api http://logs.openstack.org/46/662546/5/check/swift-tox-func-s3api/f6f7897/ : SUCCESS in 19m 45s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/46/662546/5/check/swift-tox-func-s3api-ceph-s3tests-tempauth/c6c149e/ : SUCCESS in 13m 30s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/46/662546/5/check/swift-probetests-centos-7/4307586/ : SUCCESS in 33m 41s\n- swift-dsvm-functional http://logs.openstack.org/46/662546/5/check/swift-dsvm-functional/a3855c1/ : SUCCESS in 28m 48s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/46/662546/5/check/swift-dsvm-functional-ipv6/fc64089/ : SUCCESS in 28m 05s\n- swift-tox-lower-constraints http://logs.openstack.org/46/662546/5/check/swift-tox-lower-constraints/86091a5/ : SUCCESS in 14m 20s\n- openstack-tox-pep8 http://logs.openstack.org/46/662546/5/check/openstack-tox-pep8/43aaf61/ : SUCCESS in 5m 28s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/46/662546/5/check/swift-multinode-rolling-upgrade/c651f88/ : FAILURE in 9m 40s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"674f0b1dd7b294ad2d8f288061b8f3e2d3afb2af","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-07-03 03:12:28.000000000","message":"Patch Set 5:\n\nrecheck\n\nWell that\u0027s new:\n\n \u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n ERROR: test_v1_response (test.unit.common.middleware.test_list_endpoints.TestListEndpoints)\n ----------------------------------------------------------------------\n Traceback (most recent call last):\n   File \"/home/zuul/src/opendev.org/openstack/swift/test/unit/common/middleware/test_list_endpoints.py\", line 364, in test_v1_response\n     resp \u003d req.get_response(self.list_endpoints)\n   File \"/home/zuul/src/opendev.org/openstack/swift/swift/common/swob.py\", line 1122, in get_response\n     status, headers, app_iter \u003d self.call_application(application)\n   File \"/home/zuul/src/opendev.org/openstack/swift/swift/common/swob.py\", line 1106, in call_application\n     app_iter \u003d application(self.environ, start_response)\n   File \"/home/zuul/src/opendev.org/openstack/swift/swift/common/middleware/list_endpoints.py\", line 226, in __call__\n     account, container, obj)\n   File \"/home/zuul/src/opendev.org/openstack/swift/swift/common/ring/ring.py\", line 391, in get_nodes\n     part \u003d self.get_part(account, container, obj)\n   File \"/home/zuul/src/opendev.org/openstack/swift/swift/common/ring/ring.py\", line 342, in get_part\n     self._reload()\n   File \"/home/zuul/src/opendev.org/openstack/swift/swift/common/ring/ring.py\", line 208, in _reload\n     if force or self.has_changed():\n   File \"/home/zuul/src/opendev.org/openstack/swift/swift/common/ring/ring.py\", line 318, in has_changed\n     return getmtime(self.serialized_path) !\u003d self._mtime\n   File \"/home/zuul/src/opendev.org/openstack/swift/.tox/py37/lib/python3.7/genericpath.py\", line 55, in getmtime\n     return os.stat(filename).st_mtime\n FileNotFoundError: [Errno 2] No such file or directory: \u0027/home/zuul/xfstmp/tmpc14a1z6q/object.ring.gz\u0027\n\n(And a similar failure for test_v2_obj_response)","accounts_in_message":[],"_revision_number":5},{"id":"21091ea0ea7ad2ec8ae13ba4b2a7b8062efa7ab0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 03:48:06.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/46/662546/5/check/openstack-tox-docs/e5ec270/html/ : SUCCESS in 3m 45s\n- swift-build-image http://logs.openstack.org/46/662546/5/check/swift-build-image/47e5f15/ : SUCCESS in 8m 06s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/46/662546/5/check/swift-tox-py27/2f64071/ : SUCCESS in 9m 26s\n- swift-tox-py36 http://logs.openstack.org/46/662546/5/check/swift-tox-py36/035d164/ : SUCCESS in 9m 08s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/46/662546/5/check/swift-tox-py37/f259d2b/ : SUCCESS in 9m 19s\n- swift-tox-func http://logs.openstack.org/46/662546/5/check/swift-tox-func/9c24f4e/ : SUCCESS in 8m 43s\n- swift-tox-func-py37 http://logs.openstack.org/46/662546/5/check/swift-tox-func-py37/4da1f94/ : SUCCESS in 8m 46s\n- swift-tox-func-encryption http://logs.openstack.org/46/662546/5/check/swift-tox-func-encryption/196dfcd/ : SUCCESS in 7m 59s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/46/662546/5/check/swift-tox-func-domain-remap-staticweb/f716605/ : SUCCESS in 8m 39s\n- swift-tox-func-ec http://logs.openstack.org/46/662546/5/check/swift-tox-func-ec/8e4c7ff/ : SUCCESS in 9m 17s\n- swift-tox-func-s3api http://logs.openstack.org/46/662546/5/check/swift-tox-func-s3api/13011b3/ : FAILURE in 12m 56s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/46/662546/5/check/swift-tox-func-s3api-ceph-s3tests-tempauth/1ce63a6/ : SUCCESS in 12m 51s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/46/662546/5/check/swift-probetests-centos-7/77e6cb6/ : SUCCESS in 29m 58s\n- swift-dsvm-functional http://logs.openstack.org/46/662546/5/check/swift-dsvm-functional/4adec96/ : SUCCESS in 23m 49s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/46/662546/5/check/swift-dsvm-functional-ipv6/940c382/ : SUCCESS in 29m 25s\n- swift-tox-lower-constraints http://logs.openstack.org/46/662546/5/check/swift-tox-lower-constraints/987066d/ : SUCCESS in 13m 23s\n- openstack-tox-pep8 http://logs.openstack.org/46/662546/5/check/openstack-tox-pep8/3a7e5a1/ : SUCCESS in 5m 37s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/46/662546/5/check/swift-multinode-rolling-upgrade/874f765/ : FAILURE in 11m 05s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"64e7fe36fd542baec37310c071f595ec781505dc","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-07-03 03:59:16.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"af6bf186a3cb18180474b35d5151be386422c8bb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 04:52:21.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/46/662546/5/check/openstack-tox-docs/f31fece/html/ : SUCCESS in 3m 58s\n- swift-build-image http://logs.openstack.org/46/662546/5/check/swift-build-image/82db9c5/ : SUCCESS in 6m 59s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/46/662546/5/check/swift-tox-py27/e5dc49d/ : SUCCESS in 12m 16s\n- swift-tox-py36 http://logs.openstack.org/46/662546/5/check/swift-tox-py36/f019ffa/ : SUCCESS in 11m 12s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/46/662546/5/check/swift-tox-py37/d79f036/ : SUCCESS in 15m 02s\n- swift-tox-func http://logs.openstack.org/46/662546/5/check/swift-tox-func/00ffc59/ : SUCCESS in 8m 40s\n- swift-tox-func-py37 http://logs.openstack.org/46/662546/5/check/swift-tox-func-py37/8df8b9a/ : SUCCESS in 13m 36s\n- swift-tox-func-encryption http://logs.openstack.org/46/662546/5/check/swift-tox-func-encryption/8b1c2e6/ : SUCCESS in 8m 49s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/46/662546/5/check/swift-tox-func-domain-remap-staticweb/834b824/ : SUCCESS in 8m 46s\n- swift-tox-func-ec http://logs.openstack.org/46/662546/5/check/swift-tox-func-ec/ed12bbd/ : SUCCESS in 9m 19s\n- swift-tox-func-s3api http://logs.openstack.org/46/662546/5/check/swift-tox-func-s3api/6996d6a/ : SUCCESS in 11m 42s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/46/662546/5/check/swift-tox-func-s3api-ceph-s3tests-tempauth/4004117/ : SUCCESS in 12m 47s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/46/662546/5/check/swift-probetests-centos-7/e81291a/ : FAILURE in 31m 26s\n- swift-dsvm-functional http://logs.openstack.org/46/662546/5/check/swift-dsvm-functional/1e46b52/ : SUCCESS in 31m 58s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/46/662546/5/check/swift-dsvm-functional-ipv6/38320eb/ : SUCCESS in 26m 42s\n- swift-tox-lower-constraints http://logs.openstack.org/46/662546/5/check/swift-tox-lower-constraints/acd6ac5/ : SUCCESS in 10m 59s\n- openstack-tox-pep8 http://logs.openstack.org/46/662546/5/check/openstack-tox-pep8/5f64c5e/ : SUCCESS in 5m 10s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/46/662546/5/check/swift-multinode-rolling-upgrade/24de8f6/ : FAILURE in 7m 54s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"84fd633e8061d1b0e02d4da38559b9bb14735fcd","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-07-03 18:37:51.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"b5baa1d0ffc6f97611b1e82c7c9db209476b57d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 20:26:56.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/46/662546/5/check/openstack-tox-docs/4d65219/html/ : SUCCESS in 4m 17s\n- swift-build-image http://logs.openstack.org/46/662546/5/check/swift-build-image/65e6235/ : SUCCESS in 7m 59s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/46/662546/5/check/swift-tox-py27/030da5e/ : SUCCESS in 11m 45s\n- swift-tox-py36 http://logs.openstack.org/46/662546/5/check/swift-tox-py36/c662bf8/ : SUCCESS in 10m 44s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/46/662546/5/check/swift-tox-py37/f8048d4/ : SUCCESS in 11m 55s\n- swift-tox-func http://logs.openstack.org/46/662546/5/check/swift-tox-func/ad7569c/ : SUCCESS in 8m 24s\n- swift-tox-func-py37 http://logs.openstack.org/46/662546/5/check/swift-tox-func-py37/d6a9abb/ : SUCCESS in 11m 13s\n- swift-tox-func-encryption http://logs.openstack.org/46/662546/5/check/swift-tox-func-encryption/44e4a46/ : SUCCESS in 9m 02s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/46/662546/5/check/swift-tox-func-domain-remap-staticweb/55f7f2f/ : SUCCESS in 8m 45s\n- swift-tox-func-ec http://logs.openstack.org/46/662546/5/check/swift-tox-func-ec/7f8581d/ : SUCCESS in 9m 35s\n- swift-tox-func-s3api http://logs.openstack.org/46/662546/5/check/swift-tox-func-s3api/b6e939c/ : SUCCESS in 11m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/46/662546/5/check/swift-tox-func-s3api-ceph-s3tests-tempauth/a805602/ : SUCCESS in 12m 57s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/46/662546/5/check/swift-probetests-centos-7/e3f2d55/ : SUCCESS in 35m 29s\n- swift-dsvm-functional http://logs.openstack.org/46/662546/5/check/swift-dsvm-functional/de8f4b9/ : SUCCESS in 26m 24s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/46/662546/5/check/swift-dsvm-functional-ipv6/5d1f102/ : SUCCESS in 26m 03s\n- swift-tox-lower-constraints http://logs.openstack.org/46/662546/5/check/swift-tox-lower-constraints/c02876f/ : SUCCESS in 12m 46s\n- openstack-tox-pep8 http://logs.openstack.org/46/662546/5/check/openstack-tox-pep8/ce5f469/ : SUCCESS in 5m 44s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/46/662546/5/check/swift-multinode-rolling-upgrade/5e68b27/ : FAILURE in 8m 07s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2b1541cf91517a32d6d3df012c336c5921d07cf7","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-07-06 03:00:00.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"c216362205a2b97107f47ebcab6fd3e5f79c77a0","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2019-07-10 03:29:18.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"e6f5d2f54fabba714420f3abb1987eda3776d9ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 03:29:51.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"cba17babebace85b8d7bb461b2c1870c7a509fed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 04:18:29.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/46/662546/5/gate/openstack-tox-docs/9d36aa6/html/ : SUCCESS in 5m 56s\n- swift-upload-image http://logs.openstack.org/46/662546/5/gate/swift-upload-image/a3350a4/ : SUCCESS in 9m 07s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/46/662546/5/gate/swift-tox-py27/1a768ef/ : FAILURE in 11m 46s\n- swift-tox-py37 http://logs.openstack.org/46/662546/5/gate/swift-tox-py37/fcc0622/ : SUCCESS in 12m 30s\n- swift-tox-func http://logs.openstack.org/46/662546/5/gate/swift-tox-func/52f1e71/ : SUCCESS in 9m 04s\n- swift-tox-func-py37 http://logs.openstack.org/46/662546/5/gate/swift-tox-func-py37/479bb3e/ : SUCCESS in 9m 39s\n- swift-tox-func-encryption http://logs.openstack.org/46/662546/5/gate/swift-tox-func-encryption/857ffc4/ : SUCCESS in 8m 29s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/46/662546/5/gate/swift-tox-func-domain-remap-staticweb/9e0eed7/ : SUCCESS in 8m 56s\n- swift-tox-func-ec http://logs.openstack.org/46/662546/5/gate/swift-tox-func-ec/d6a42f4/ : SUCCESS in 9m 42s\n- swift-tox-func-s3api http://logs.openstack.org/46/662546/5/gate/swift-tox-func-s3api/d421119/ : SUCCESS in 12m 05s\n- swift-probetests-centos-7 http://logs.openstack.org/46/662546/5/gate/swift-probetests-centos-7/4d23737/ : FAILURE in 35m 14s\n- swift-dsvm-functional http://logs.openstack.org/46/662546/5/gate/swift-dsvm-functional/5c6ce81/ : SUCCESS in 30m 21s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/46/662546/5/gate/swift-dsvm-functional-ipv6/90edda9/ : SUCCESS in 33m 03s\n- swift-tox-lower-constraints http://logs.openstack.org/46/662546/5/gate/swift-tox-lower-constraints/5c81bae/ : SUCCESS in 11m 28s\n- openstack-tox-pep8 http://logs.openstack.org/46/662546/5/gate/openstack-tox-pep8/b357c90/ : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":5},{"id":"62dbf3a25e08a1dcb055209f69df04fc149da056","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2019-07-10 07:29:19.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"0a5a579f6572cdf21fbfc16dc03e28beff2bc2d8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 08:16:00.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/46/662546/5/check/openstack-tox-docs/de2f2b7/html/ : SUCCESS in 3m 55s\n- swift-build-image http://logs.openstack.org/46/662546/5/check/swift-build-image/d5e9e9e/ : SUCCESS in 8m 11s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/46/662546/5/check/swift-tox-py27/7053c38/ : SUCCESS in 11m 35s\n- swift-tox-py36 http://logs.openstack.org/46/662546/5/check/swift-tox-py36/18da397/ : SUCCESS in 10m 32s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/46/662546/5/check/swift-tox-py37/7404165/ : SUCCESS in 11m 15s\n- swift-tox-func http://logs.openstack.org/46/662546/5/check/swift-tox-func/c31cb1d/ : SUCCESS in 9m 48s\n- swift-tox-func-py37 http://logs.openstack.org/46/662546/5/check/swift-tox-func-py37/c0f3eac/ : SUCCESS in 12m 01s\n- swift-tox-func-encryption http://logs.openstack.org/46/662546/5/check/swift-tox-func-encryption/55bf6b8/ : SUCCESS in 11m 46s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/46/662546/5/check/swift-tox-func-domain-remap-staticweb/7bce9c3/ : SUCCESS in 9m 35s\n- swift-tox-func-ec http://logs.openstack.org/46/662546/5/check/swift-tox-func-ec/013ea02/ : SUCCESS in 8m 17s\n- swift-tox-func-s3api http://logs.openstack.org/46/662546/5/check/swift-tox-func-s3api/8c1b50a/ : SUCCESS in 12m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/46/662546/5/check/swift-tox-func-s3api-ceph-s3tests-tempauth/61443de/ : SUCCESS in 13m 54s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/46/662546/5/check/swift-probetests-centos-7/fcac10b/ : SUCCESS in 32m 41s\n- swift-dsvm-functional http://logs.openstack.org/46/662546/5/check/swift-dsvm-functional/0bbd45c/ : SUCCESS in 27m 42s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/46/662546/5/check/swift-dsvm-functional-ipv6/8a1c3bd/ : SUCCESS in 25m 54s\n- swift-tox-lower-constraints http://logs.openstack.org/46/662546/5/check/swift-tox-lower-constraints/70d099e/ : SUCCESS in 12m 40s\n- openstack-tox-pep8 http://logs.openstack.org/46/662546/5/check/openstack-tox-pep8/2283979/ : SUCCESS in 5m 54s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/46/662546/5/check/swift-multinode-rolling-upgrade/414c221/ : FAILURE in 10m 42s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a1a49151adc5e973ebb50f01aafbe9bc006668cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 08:16:23.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"8025d19d7f6e457047ba76f29dc5fa6136110f49","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 09:35:34.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/46/662546/5/gate/openstack-tox-docs/dfd3417/html/ : SUCCESS in 5m 29s\n- swift-upload-image http://logs.openstack.org/46/662546/5/gate/swift-upload-image/1340303/ : SUCCESS in 10m 03s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/46/662546/5/gate/swift-tox-py27/066e3b3/ : SUCCESS in 11m 14s\n- swift-tox-py37 http://logs.openstack.org/46/662546/5/gate/swift-tox-py37/63a26b1/ : SUCCESS in 10m 46s\n- swift-tox-func http://logs.openstack.org/46/662546/5/gate/swift-tox-func/ea3d35b/ : SUCCESS in 11m 50s\n- swift-tox-func-py37 http://logs.openstack.org/46/662546/5/gate/swift-tox-func-py37/250e16e/ : SUCCESS in 10m 37s\n- swift-tox-func-encryption http://logs.openstack.org/46/662546/5/gate/swift-tox-func-encryption/52e6c45/ : SUCCESS in 9m 52s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/46/662546/5/gate/swift-tox-func-domain-remap-staticweb/82476db/ : SUCCESS in 13m 23s\n- swift-tox-func-ec http://logs.openstack.org/46/662546/5/gate/swift-tox-func-ec/6f5e547/ : SUCCESS in 11m 46s\n- swift-tox-func-s3api http://logs.openstack.org/46/662546/5/gate/swift-tox-func-s3api/2c75f06/ : SUCCESS in 15m 43s\n- swift-probetests-centos-7 http://logs.openstack.org/46/662546/5/gate/swift-probetests-centos-7/673a438/ : SUCCESS in 35m 50s\n- swift-dsvm-functional http://logs.openstack.org/46/662546/5/gate/swift-dsvm-functional/e16138a/ : SUCCESS in 34m 19s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/46/662546/5/gate/swift-dsvm-functional-ipv6/7d997cb/ : SUCCESS in 30m 08s\n- swift-tox-lower-constraints http://logs.openstack.org/46/662546/5/gate/swift-tox-lower-constraints/363f9a0/ : SUCCESS in 12m 12s\n- openstack-tox-pep8 http://logs.openstack.org/46/662546/5/gate/openstack-tox-pep8/ec6b2dc/ : SUCCESS in 6m 15s","accounts_in_message":[],"_revision_number":5},{"id":"234be2e81e66387abc10f425904011fb759153a4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 09:35:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"f9eb524aaf7d3c98f61d485e55baa5f084057f8c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 09:36:59.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- swift-promote-image http://logs.openstack.org/46/662546/5/promote/swift-promote-image/a379d0a/ : SUCCESS in 1m 01s (non-voting)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"76fde89261e1940daadb720c41df1a3595314a97","revisions":{"2812934d060521a59701915c7f19e789de6a0bf6":{"kind":"REWORK","_number":1,"created":"2019-05-31 19:08:29.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/46/662546/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/46/662546/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/662546/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/662546/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/662546/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/46/662546/1"}}},"commit":{"parents":[{"commit":"2c505cf0ff9176625af16ccb357158902c898080","subject":"Fix up how we memcache on py3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2c505cf0ff9176625af16ccb357158902c898080"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-05-31 16:21:43.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-05-31 19:07:51.000000000","tz":-420},"subject":"py3: Stop truncating response headers","message":"py3: Stop truncating response headers\n\n... when there\u0027s a non-ASCII header name. Values were fine; just not\nnames. See https://bugs.python.org/issue37093 for more information.\n\nChange-Id: I0f03c211f35a9a49e047a5718a9907b515ca88d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2812934d060521a59701915c7f19e789de6a0bf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2812934d060521a59701915c7f19e789de6a0bf6"}]},"branch":"refs/heads/master"},"3cd558989fc7f292b2341af2eb0faa6abf563e5f":{"kind":"REWORK","_number":2,"created":"2019-06-06 23:01:18.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/46/662546/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/46/662546/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/662546/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/662546/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/662546/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/46/662546/2"}}},"commit":{"parents":[{"commit":"61e2d507bf68ffabffd73ab6fc51009825722a52","subject":"py3: Port the tempurl middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/61e2d507bf68ffabffd73ab6fc51009825722a52"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-05-31 16:21:43.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-06-06 22:59:31.000000000","tz":-420},"subject":"py3: Stop truncating response headers","message":"py3: Stop truncating response headers\n\n... when there\u0027s a non-ASCII header name. Values were fine; just not\nnames. See https://bugs.python.org/issue37093 for more information.\n\nChange-Id: I0f03c211f35a9a49e047a5718a9907b515ca88d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3cd558989fc7f292b2341af2eb0faa6abf563e5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3cd558989fc7f292b2341af2eb0faa6abf563e5f"}]},"branch":"refs/heads/master"},"b00b238e3c8cb6a167cc6b83a64128a48791e22e":{"kind":"REWORK","_number":3,"created":"2019-06-17 16:58:33.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/46/662546/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/46/662546/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/662546/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/662546/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/662546/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/46/662546/3"}}},"commit":{"parents":[{"commit":"bc19dd1649115d9c83e2c939bcca990acfe7867c","subject":"Merge \"Installing liberasurecode from Alpine Linux repos instead of building from source.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bc19dd1649115d9c83e2c939bcca990acfe7867c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-06-17 16:25:52.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-06-17 16:53:22.000000000","tz":-420},"subject":"py3: Be able to read and write non-ASCII headers","message":"py3: Be able to read and write non-ASCII headers\n\nApparently Python\u0027s stdlib got more picky about what a header should\nlook like. As a result, if an account, container, or object had a\nnon-ASCII metadata name (values were fine), the proxy-server wouldn\u0027t\nparse all of the headers. See https://bugs.python.org/issue37093 for\nmore information.\n\nThis presented several problems:\n- Since the non-ASCII header aborts parsing, we may lose important\n  HTTP-level information like Content-Length or Transfer-Encoding.\n- Since the offending header wouldn\u0027t get parsed, the client wouldn\u0027t\n  even know what the problem was.\n- Even if the client knew what the bad header was, it would have no way\n  to clear it, as the server uses the same logic to parse incoming\n  requests.\n\nSo, hack in our own header parsing if we detect that parsing was\naborted. Note that we also have to mangle bufferedhttp\u0027s putheader so we\ncan get non-ASCII headers to the backend servers.\n\nNow, we can run the test_unicode_metadata tests in\ntest/functional/test_account.py and test/functional/test_container.py\nunder py2 against services running under py3.\n\nChange-Id: I0f03c211f35a9a49e047a5718a9907b515ca88d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b00b238e3c8cb6a167cc6b83a64128a48791e22e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b00b238e3c8cb6a167cc6b83a64128a48791e22e"}]},"branch":"refs/heads/master"},"0474a932e7ec3d27d2cb69b5b8562ffc7d881950":{"kind":"REWORK","_number":4,"created":"2019-06-17 20:10:31.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/46/662546/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/46/662546/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/662546/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/662546/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/662546/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/46/662546/4"}}},"commit":{"parents":[{"commit":"bc19dd1649115d9c83e2c939bcca990acfe7867c","subject":"Merge \"Installing liberasurecode from Alpine Linux repos instead of building from source.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bc19dd1649115d9c83e2c939bcca990acfe7867c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-06-17 16:25:52.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-06-17 20:10:19.000000000","tz":-420},"subject":"py3: Be able to read and write non-ASCII headers","message":"py3: Be able to read and write non-ASCII headers\n\nApparently Python\u0027s stdlib got more picky about what a header should\nlook like. As a result, if an account, container, or object had a\nnon-ASCII metadata name (values were fine), the proxy-server wouldn\u0027t\nparse all of the headers. See https://bugs.python.org/issue37093 for\nmore information.\n\nThis presented several problems:\n- Since the non-ASCII header aborts parsing, we may lose important\n  HTTP-level information like Content-Length or Transfer-Encoding.\n- Since the offending header wouldn\u0027t get parsed, the client wouldn\u0027t\n  even know what the problem was.\n- Even if the client knew what the bad header was, it would have no way\n  to clear it, as the server uses the same logic to parse incoming\n  requests.\n\nSo, hack in our own header parsing if we detect that parsing was\naborted. Note that we also have to mangle bufferedhttp\u0027s putheader so we\ncan get non-ASCII headers to the backend servers.\n\nNow, we can run the test_unicode_metadata tests in\ntest/functional/test_account.py and test/functional/test_container.py\nunder py2 against services running under py3.\n\nChange-Id: I0f03c211f35a9a49e047a5718a9907b515ca88d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0474a932e7ec3d27d2cb69b5b8562ffc7d881950"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0474a932e7ec3d27d2cb69b5b8562ffc7d881950"}]},"branch":"refs/heads/master"},"76fde89261e1940daadb720c41df1a3595314a97":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-07-03 02:01:55.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/46/662546/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/46/662546/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/662546/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/662546/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/662546/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/46/662546/5"}}},"commit":{"parents":[{"commit":"bf3e2548b337ad149994eca596e6b67904cdf00a","subject":"Merge \"s3api: add unit test case for s3acl\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bf3e2548b337ad149994eca596e6b67904cdf00a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-06-17 16:25:52.000000000","tz":-420},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2019-07-03 02:01:55.000000000","tz":0},"subject":"py3: Be able to read and write non-ASCII headers","message":"py3: Be able to read and write non-ASCII headers\n\nApparently Python\u0027s stdlib got more picky about what a header should\nlook like. As a result, if an account, container, or object had a\nnon-ASCII metadata name (values were fine), the proxy-server wouldn\u0027t\nparse all of the headers. See https://bugs.python.org/issue37093 for\nmore information.\n\nThis presented several problems:\n- Since the non-ASCII header aborts parsing, we may lose important\n  HTTP-level information like Content-Length or Transfer-Encoding.\n- Since the offending header wouldn\u0027t get parsed, the client wouldn\u0027t\n  even know what the problem was.\n- Even if the client knew what the bad header was, it would have no way\n  to clear it, as the server uses the same logic to parse incoming\n  requests.\n\nSo, hack in our own header parsing if we detect that parsing was\naborted. Note that we also have to mangle bufferedhttp\u0027s putheader so we\ncan get non-ASCII headers to the backend servers.\n\nNow, we can run the test_unicode_metadata tests in\ntest/functional/test_account.py and test/functional/test_container.py\nunder py2 against services running under py3.\n\nChange-Id: I0f03c211f35a9a49e047a5718a9907b515ca88d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/76fde89261e1940daadb720c41df1a3595314a97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/76fde89261e1940daadb720c41df1a3595314a97"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
