)]}'
{"id":"openstack%2Fswift~662615","triplet_id":"openstack%2Fswift~master~Ic9a2844fffc6dba6a4d44a9712edbb5cd91f0470","project":"openstack/swift","branch":"master","topic":"docker-liberasurecode","hashtags":[],"change_id":"Ic9a2844fffc6dba6a4d44a9712edbb5cd91f0470","subject":"Installing liberasurecode from Alpine Linux repos instead of building from source.","status":"MERGED","created":"2019-06-02 05:27:53.000000000","updated":"2019-06-15 00:16:16.000000000","submitted":"2019-06-15 00:15:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662615-1560557705481-7895b45e","meta_rev_id":"650dfa6c2ba03ae652589c58ac02800792927e7e","_number":662615,"virtual_id_number":662615,"owner":{"_account_id":24168,"name":"Ehud Kaldor","email":"ehud@unfairfunction.org","username":"UnfairFunction"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"date":"2019-06-10 04:50:26.000000000","_account_id":24168,"name":"Ehud Kaldor","email":"ehud@unfairfunction.org","username":"UnfairFunction"},{"value":0,"date":"2019-06-14 13:24:43.000000000","_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":2,"date":"2019-06-15 00:15:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"recommended":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"all":[{"value":1,"date":"2019-06-14 11:59:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":24168,"name":"Ehud Kaldor","email":"ehud@unfairfunction.org","username":"UnfairFunction"},{"value":0,"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":1,"date":"2019-06-14 14:26:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-06 17:54:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":24168,"name":"Ehud Kaldor","email":"ehud@unfairfunction.org","username":"UnfairFunction"},{"value":0,"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-06 17:54:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24168,"name":"Ehud Kaldor","email":"ehud@unfairfunction.org","username":"UnfairFunction"},{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-06 17:54:58.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2019-06-14 11:59:40.000000000","updated_by":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"reviewer":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"state":"REVIEWER"},{"updated":"2019-06-14 13:24:43.000000000","updated_by":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"reviewer":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"state":"REVIEWER"},{"updated":"2019-06-14 14:26:07.000000000","updated_by":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"reviewer":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"state":"REVIEWER"},{"updated":"2019-06-15 00:15:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"527fb25988628433027b551dc0461948d16acad8","author":{"_account_id":24168,"name":"Ehud Kaldor","email":"ehud@unfairfunction.org","username":"UnfairFunction"},"date":"2019-06-02 05:27:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"23b2fdb16643a4cb7d0975f46d07c164f120a24f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-02 06:11:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/15/662615/1/check/openstack-tox-docs/9146fdd/html/ : SUCCESS in 3m 45s\n- swift-build-image http://logs.openstack.org/15/662615/1/check/swift-build-image/bea344d/ : SUCCESS in 6m 40s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/15/662615/1/check/swift-tox-py27/fb4d226/ : SUCCESS in 10m 39s\n- swift-tox-py36 http://logs.openstack.org/15/662615/1/check/swift-tox-py36/a5ec6e7/ : SUCCESS in 10m 06s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/15/662615/1/check/swift-tox-py37/cd94bb0/ : SUCCESS in 10m 08s\n- swift-tox-func http://logs.openstack.org/15/662615/1/check/swift-tox-func/c96bacd/ : SUCCESS in 8m 11s\n- swift-tox-func-py37 http://logs.openstack.org/15/662615/1/check/swift-tox-func-py37/80d2718/ : SUCCESS in 6m 30s\n- swift-tox-func-encryption http://logs.openstack.org/15/662615/1/check/swift-tox-func-encryption/0a2cd61/ : SUCCESS in 8m 09s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/15/662615/1/check/swift-tox-func-domain-remap-staticweb/737d1c2/ : SUCCESS in 8m 13s\n- swift-tox-func-ec http://logs.openstack.org/15/662615/1/check/swift-tox-func-ec/7aacd1b/ : SUCCESS in 14m 48s\n- swift-tox-func-s3api http://logs.openstack.org/15/662615/1/check/swift-tox-func-s3api/50fbdaf/ : SUCCESS in 10m 50s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/15/662615/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/879beea/ : SUCCESS in 12m 10s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/15/662615/1/check/swift-probetests-centos-7/7d85c22/ : SUCCESS in 31m 12s\n- swift-dsvm-functional http://logs.openstack.org/15/662615/1/check/swift-dsvm-functional/5157725/ : SUCCESS in 25m 59s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/15/662615/1/check/swift-dsvm-functional-ipv6/69cd87c/ : SUCCESS in 24m 59s\n- swift-tox-lower-constraints http://logs.openstack.org/15/662615/1/check/swift-tox-lower-constraints/bd27f2e/ : SUCCESS in 11m 11s\n- openstack-tox-pep8 http://logs.openstack.org/15/662615/1/check/openstack-tox-pep8/1d8e655/ : SUCCESS in 5m 23s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/15/662615/1/check/swift-multinode-rolling-upgrade/6bab2d0/ : FAILURE in 8m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3dfa9c2fe94aae7cb0cc905a8abf55791291c025","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-03 16:21:20.000000000","message":"Patch Set 1: Code-Review+2\n\n(2 comments)\n\nJust a couple questions. Threw up https://review.opendev.org/#/c/662799/ to try out dropping the build deps.","accounts_in_message":[],"_revision_number":1},{"id":"411eaf2b1dd5e4d316530239d58d2ab13c836a51","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2019-06-03 16:26:00.000000000","message":"Patch Set 1: Code-Review+2\n\nEhud, thanks for working on this. Will wait on Tim\u0027s question to +A","accounts_in_message":[],"_revision_number":1},{"id":"b70f6c0bf79c40d62d67c4ac17efe4de123887bc","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-05 16:20:17.000000000","message":"Patch Set 1: Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"87abc06a83d5fdaf1578ae2958514a09c714f2e5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 16:20:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"75e3f026f2d27a2cfccc4dd093e59a5fbf251769","author":{"_account_id":24168,"name":"Ehud Kaldor","email":"ehud@unfairfunction.org","username":"UnfairFunction"},"date":"2019-06-05 18:27:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fe69350008031f640e618236a67842756835c028","author":{"_account_id":24168,"name":"Ehud Kaldor","email":"ehud@unfairfunction.org","username":"UnfairFunction"},"date":"2019-06-05 18:32:34.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a3a2af5dd686c8a378f4b88b506339bd949a3b9c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 01:20:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/15/662615/3/check/openstack-tox-docs/3b6dfdd/html/ : SUCCESS in 3m 43s\n- swift-build-image http://logs.openstack.org/15/662615/3/check/swift-build-image/85f19a9/ : SUCCESS in 7m 21s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/15/662615/3/check/swift-tox-py27/db0b1f3/ : SUCCESS in 9m 55s\n- swift-tox-py36 http://logs.openstack.org/15/662615/3/check/swift-tox-py36/ee4655e/ : SUCCESS in 23m 25s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/15/662615/3/check/swift-tox-py37/0928f32/ : SUCCESS in 11m 11s\n- swift-tox-func http://logs.openstack.org/15/662615/3/check/swift-tox-func/bc47969/ : SUCCESS in 9m 15s\n- swift-tox-func-py37 http://logs.openstack.org/15/662615/3/check/swift-tox-func-py37/406950b/ : SUCCESS in 15m 23s\n- swift-tox-func-encryption http://logs.openstack.org/15/662615/3/check/swift-tox-func-encryption/91064aa/ : SUCCESS in 11m 54s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/15/662615/3/check/swift-tox-func-domain-remap-staticweb/71bd4b9/ : SUCCESS in 18m 44s\n- swift-tox-func-ec http://logs.openstack.org/15/662615/3/check/swift-tox-func-ec/ae18193/ : SUCCESS in 10m 46s\n- swift-tox-func-s3api http://logs.openstack.org/15/662615/3/check/swift-tox-func-s3api/a0a0738/ : SUCCESS in 11m 11s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/15/662615/3/check/swift-tox-func-s3api-ceph-s3tests-tempauth/f6f66a7/ : SUCCESS in 13m 14s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/15/662615/3/check/swift-probetests-centos-7/6c726b6/ : SUCCESS in 35m 31s\n- swift-dsvm-functional http://logs.openstack.org/15/662615/3/check/swift-dsvm-functional/fc91505/ : SUCCESS in 28m 53s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/15/662615/3/check/swift-dsvm-functional-ipv6/945b88a/ : SUCCESS in 25m 00s\n- swift-tox-lower-constraints http://logs.openstack.org/15/662615/3/check/swift-tox-lower-constraints/d2b4155/ : SUCCESS in 11m 51s\n- openstack-tox-pep8 http://logs.openstack.org/15/662615/3/check/openstack-tox-pep8/f06eb92/ : SUCCESS in 6m 54s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/15/662615/3/check/swift-multinode-rolling-upgrade/da6017b/ : FAILURE in 8m 41s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e789cf8a23a93484a3d23572a9d0d0bc71acb6ca","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-06 17:54:58.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"db5b6f7b1d6a9c590237c852004275b7522fe7f3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 17:55:22.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4d77df7fb88b1fc421ee5816a85228a25bab0bee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 20:40:40.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/15/662615/3/gate/openstack-tox-docs/f632211/html/ : SUCCESS in 4m 01s\n- swift-upload-image http://logs.openstack.org/15/662615/3/gate/swift-upload-image/e0fb130/ : SUCCESS in 9m 57s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/15/662615/3/gate/swift-tox-py27/ea1fc4c/ : SUCCESS in 10m 25s\n- swift-tox-py37 http://logs.openstack.org/15/662615/3/gate/swift-tox-py37/403dbfe/ : SUCCESS in 10m 14s\n- swift-tox-func http://logs.openstack.org/15/662615/3/gate/swift-tox-func/4ef4953/ : SUCCESS in 8m 51s\n- swift-tox-func-py37 http://logs.openstack.org/15/662615/3/gate/swift-tox-func-py37/b06ba38/ : SUCCESS in 9m 44s\n- swift-tox-func-encryption http://logs.openstack.org/15/662615/3/gate/swift-tox-func-encryption/9ed2c22/ : SUCCESS in 8m 31s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/15/662615/3/gate/swift-tox-func-domain-remap-staticweb/c349242/ : SUCCESS in 12m 58s\n- swift-tox-func-ec http://logs.openstack.org/15/662615/3/gate/swift-tox-func-ec/36b144e/ : SUCCESS in 9m 30s\n- swift-tox-func-s3api http://logs.openstack.org/15/662615/3/gate/swift-tox-func-s3api/356bcc9/ : SUCCESS in 11m 25s\n- swift-probetests-centos-7 http://logs.openstack.org/15/662615/3/gate/swift-probetests-centos-7/eb51977/ : SUCCESS in 32m 01s\n- swift-dsvm-functional http://logs.openstack.org/15/662615/3/gate/swift-dsvm-functional/f4e08be/ : TIMED_OUT in 47m 28s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/15/662615/3/gate/swift-dsvm-functional-ipv6/70e7ec4/ : SUCCESS in 28m 43s\n- swift-tox-lower-constraints http://logs.openstack.org/15/662615/3/gate/swift-tox-lower-constraints/a97b445/ : SUCCESS in 11m 17s\n- openstack-tox-pep8 http://logs.openstack.org/15/662615/3/gate/openstack-tox-pep8/c79c88b/ : SUCCESS in 6m 16s","accounts_in_message":[],"_revision_number":3},{"id":"fe95ab41db0a39c3fe52eac4021402f2fc1443ea","author":{"_account_id":24168,"name":"Ehud Kaldor","email":"ehud@unfairfunction.org","username":"UnfairFunction"},"date":"2019-06-10 04:50:26.000000000","message":"Patch Set 3:\n\n(3 comments)\n\ninteresting. Zuul failed it due to something my changes never touched...","accounts_in_message":[],"_revision_number":3},{"id":"cdcadb7ee2aa046b16d10d36bd8d49d913dec9a1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-10 15:33:38.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"4a95e4810cbc5a762b1199efec648f13dc2c1a34","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 20:33:13.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/15/662615/3/check/openstack-tox-docs/8b0e421/html/ : SUCCESS in 5m 30s\n- swift-build-image http://logs.openstack.org/15/662615/3/check/swift-build-image/1b7f022/ : SUCCESS in 8m 30s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/15/662615/3/check/swift-tox-py27/364ebae/ : SUCCESS in 11m 07s\n- swift-tox-py36 http://logs.openstack.org/15/662615/3/check/swift-tox-py36/6adbd1a/ : SUCCESS in 10m 50s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/15/662615/3/check/swift-tox-py37/94283c1/ : SUCCESS in 11m 04s\n- swift-tox-func http://logs.openstack.org/15/662615/3/check/swift-tox-func/4e2c05b/ : SUCCESS in 9m 03s\n- swift-tox-func-py37 http://logs.openstack.org/15/662615/3/check/swift-tox-func-py37/3ef3182/ : SUCCESS in 11m 33s\n- swift-tox-func-encryption http://logs.openstack.org/15/662615/3/check/swift-tox-func-encryption/8f0238e/ : SUCCESS in 9m 18s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/15/662615/3/check/swift-tox-func-domain-remap-staticweb/b323688/ : SUCCESS in 9m 38s\n- swift-tox-func-ec http://logs.openstack.org/15/662615/3/check/swift-tox-func-ec/bc899ef/ : SUCCESS in 11m 00s\n- swift-tox-func-s3api http://logs.openstack.org/15/662615/3/check/swift-tox-func-s3api/290c365/ : SUCCESS in 11m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/15/662615/3/check/swift-tox-func-s3api-ceph-s3tests-tempauth/9a75fd2/ : SUCCESS in 12m 23s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/15/662615/3/check/swift-probetests-centos-7/c74defd/ : SUCCESS in 29m 41s\n- swift-dsvm-functional http://logs.openstack.org/15/662615/3/check/swift-dsvm-functional/8fe8452/ : SUCCESS in 28m 20s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/15/662615/3/check/swift-dsvm-functional-ipv6/b760a08/ : SUCCESS in 27m 19s\n- swift-tox-lower-constraints http://logs.openstack.org/15/662615/3/check/swift-tox-lower-constraints/f2e70a5/ : FAILURE in 5m 38s\n- openstack-tox-pep8 http://logs.openstack.org/15/662615/3/check/openstack-tox-pep8/1c06aac/ : SUCCESS in 5m 34s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/15/662615/3/check/swift-multinode-rolling-upgrade/4dbae78/ : FAILURE in 9m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c27fd450364e404591815c74afacf70d61780161","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-13 16:34:52.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"e07b2999460dec35349453944980a288eef0504d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 03:47:20.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/15/662615/3/check/openstack-tox-docs/8182397/html/ : SUCCESS in 4m 08s\n- swift-build-image http://logs.openstack.org/15/662615/3/check/swift-build-image/9f63aef/ : SUCCESS in 8m 28s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/15/662615/3/check/swift-tox-py27/629a5a9/ : SUCCESS in 13m 59s\n- swift-tox-py36 http://logs.openstack.org/15/662615/3/check/swift-tox-py36/8056e85/ : SUCCESS in 12m 55s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/15/662615/3/check/swift-tox-py37/636c6c0/ : SUCCESS in 11m 03s\n- swift-tox-func http://logs.openstack.org/15/662615/3/check/swift-tox-func/bc2b4ef/ : FAILURE in 10m 35s\n- swift-tox-func-py37 http://logs.openstack.org/15/662615/3/check/swift-tox-func-py37/ffefddc/ : SUCCESS in 9m 11s\n- swift-tox-func-encryption http://logs.openstack.org/15/662615/3/check/swift-tox-func-encryption/b811ccd/ : SUCCESS in 9m 05s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/15/662615/3/check/swift-tox-func-domain-remap-staticweb/e1e0301/ : SUCCESS in 9m 06s\n- swift-tox-func-ec http://logs.openstack.org/15/662615/3/check/swift-tox-func-ec/4758f51/ : SUCCESS in 22m 17s\n- swift-tox-func-s3api http://logs.openstack.org/15/662615/3/check/swift-tox-func-s3api/64371a6/ : SUCCESS in 15m 57s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/15/662615/3/check/swift-tox-func-s3api-ceph-s3tests-tempauth/ae56499/ : SUCCESS in 13m 47s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/15/662615/3/check/swift-probetests-centos-7/0400a6c/ : SUCCESS in 33m 07s\n- swift-dsvm-functional http://logs.openstack.org/15/662615/3/check/swift-dsvm-functional/7c49fe6/ : SUCCESS in 26m 02s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/15/662615/3/check/swift-dsvm-functional-ipv6/381fcd5/ : SUCCESS in 28m 58s\n- swift-tox-lower-constraints http://logs.openstack.org/15/662615/3/check/swift-tox-lower-constraints/f80aea0/ : SUCCESS in 14m 51s\n- openstack-tox-pep8 http://logs.openstack.org/15/662615/3/check/openstack-tox-pep8/b60f4be/ : SUCCESS in 5m 41s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/15/662615/3/check/swift-multinode-rolling-upgrade/301603e/ : FAILURE in 9m 05s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fc2d3e4ac7e781801fdfa2ee1b190c6036b38599","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2019-06-14 07:30:26.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"5ee3838ad5cb51cec86df1d75b371096f6480367","author":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"date":"2019-06-14 11:59:40.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"663f5c2df61de4179496430a549c351c4f62a5c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 12:15:23.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/15/662615/3/check/openstack-tox-docs/e724b41/html/ : SUCCESS in 4m 14s\n- swift-build-image http://logs.openstack.org/15/662615/3/check/swift-build-image/c7a0193/ : SUCCESS in 10m 33s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/15/662615/3/check/swift-tox-py27/af1963b/ : SUCCESS in 11m 04s\n- swift-tox-py36 http://logs.openstack.org/15/662615/3/check/swift-tox-py36/9b888e0/ : SUCCESS in 10m 52s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/15/662615/3/check/swift-tox-py37/25cdfef/ : SUCCESS in 11m 06s\n- swift-tox-func http://logs.openstack.org/15/662615/3/check/swift-tox-func/3309be9/ : SUCCESS in 9m 55s\n- swift-tox-func-py37 http://logs.openstack.org/15/662615/3/check/swift-tox-func-py37/f2d8aca/ : SUCCESS in 9m 20s\n- swift-tox-func-encryption http://logs.openstack.org/15/662615/3/check/swift-tox-func-encryption/273bcbc/ : SUCCESS in 13m 58s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/15/662615/3/check/swift-tox-func-domain-remap-staticweb/599be3a/ : SUCCESS in 10m 52s\n- swift-tox-func-ec http://logs.openstack.org/15/662615/3/check/swift-tox-func-ec/e798589/ : SUCCESS in 9m 03s\n- swift-tox-func-s3api http://logs.openstack.org/15/662615/3/check/swift-tox-func-s3api/04e47ed/ : SUCCESS in 11m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/15/662615/3/check/swift-tox-func-s3api-ceph-s3tests-tempauth/1fa4ff1/ : SUCCESS in 15m 19s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/15/662615/3/check/swift-probetests-centos-7/524f521/ : SUCCESS in 30m 44s\n- swift-dsvm-functional http://logs.openstack.org/15/662615/3/check/swift-dsvm-functional/359f15f/ : SUCCESS in 27m 15s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/15/662615/3/check/swift-dsvm-functional-ipv6/687018c/ : TIMED_OUT in 47m 39s\n- swift-tox-lower-constraints http://logs.openstack.org/15/662615/3/check/swift-tox-lower-constraints/6240628/ : SUCCESS in 11m 15s\n- openstack-tox-pep8 http://logs.openstack.org/15/662615/3/check/openstack-tox-pep8/963b9eb/ : SUCCESS in 5m 46s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/15/662615/3/check/swift-multinode-rolling-upgrade/bd331f3/ : FAILURE in 8m 15s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0febcc1cd71579fd4e14e3afb48e5e1300581a4a","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2019-06-14 13:24:43.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"66744e70ea44268a07e1d69a2e086eda2122a552","author":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"date":"2019-06-14 14:26:07.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"e7bd875d6d1b10c37ccd0dc3f5ae759a1c396f02","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-14 15:18:29.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"01155ac61ad67e588896b2ea772c8bf2434cdc33","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 21:42:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/15/662615/3/check/openstack-tox-docs/f75e5bb/html/ : SUCCESS in 5m 52s\n- swift-build-image http://logs.openstack.org/15/662615/3/check/swift-build-image/bacaa62/ : SUCCESS in 9m 26s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/15/662615/3/check/swift-tox-py27/fc04f09/ : SUCCESS in 12m 07s\n- swift-tox-py36 http://logs.openstack.org/15/662615/3/check/swift-tox-py36/9c9f190/ : SUCCESS in 10m 37s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/15/662615/3/check/swift-tox-py37/3a75610/ : SUCCESS in 18m 17s\n- swift-tox-func http://logs.openstack.org/15/662615/3/check/swift-tox-func/e03b9cb/ : SUCCESS in 11m 53s\n- swift-tox-func-py37 http://logs.openstack.org/15/662615/3/check/swift-tox-func-py37/edd7435/ : SUCCESS in 9m 57s\n- swift-tox-func-encryption http://logs.openstack.org/15/662615/3/check/swift-tox-func-encryption/13ec90e/ : SUCCESS in 14m 59s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/15/662615/3/check/swift-tox-func-domain-remap-staticweb/7affb0e/ : SUCCESS in 11m 11s\n- swift-tox-func-ec http://logs.openstack.org/15/662615/3/check/swift-tox-func-ec/4683631/ : SUCCESS in 9m 21s\n- swift-tox-func-s3api http://logs.openstack.org/15/662615/3/check/swift-tox-func-s3api/c14b6d4/ : SUCCESS in 12m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/15/662615/3/check/swift-tox-func-s3api-ceph-s3tests-tempauth/9762252/ : SUCCESS in 13m 44s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/15/662615/3/check/swift-probetests-centos-7/b561a7b/ : SUCCESS in 29m 45s\n- swift-dsvm-functional http://logs.openstack.org/15/662615/3/check/swift-dsvm-functional/bd0f29f/ : SUCCESS in 46m 11s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/15/662615/3/check/swift-dsvm-functional-ipv6/eb88342/ : SUCCESS in 29m 02s\n- swift-tox-lower-constraints http://logs.openstack.org/15/662615/3/check/swift-tox-lower-constraints/8f0efe0/ : SUCCESS in 20m 10s\n- openstack-tox-pep8 http://logs.openstack.org/15/662615/3/check/openstack-tox-pep8/759e735/ : SUCCESS in 7m 42s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/15/662615/3/check/swift-multinode-rolling-upgrade/1646ee5/ : FAILURE in 8m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e412c13d8684f09099ec7ae3ddec8b4d318be2dd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 21:43:07.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"883e556661a011c04089d5df04ddc15f8b1a2d8c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-15 00:15:05.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/15/662615/3/gate/openstack-tox-docs/e43495f/html/ : SUCCESS in 4m 40s\n- swift-upload-image http://logs.openstack.org/15/662615/3/gate/swift-upload-image/c203ad0/ : SUCCESS in 9m 27s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/15/662615/3/gate/swift-tox-py27/ef54e66/ : SUCCESS in 10m 53s\n- swift-tox-py37 http://logs.openstack.org/15/662615/3/gate/swift-tox-py37/0eaa0ee/ : SUCCESS in 11m 13s\n- swift-tox-func http://logs.openstack.org/15/662615/3/gate/swift-tox-func/385754d/ : SUCCESS in 8m 50s\n- swift-tox-func-py37 http://logs.openstack.org/15/662615/3/gate/swift-tox-func-py37/28b526f/ : SUCCESS in 10m 37s\n- swift-tox-func-encryption http://logs.openstack.org/15/662615/3/gate/swift-tox-func-encryption/a09a42d/ : SUCCESS in 9m 31s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/15/662615/3/gate/swift-tox-func-domain-remap-staticweb/3ffbe19/ : SUCCESS in 12m 51s\n- swift-tox-func-ec http://logs.openstack.org/15/662615/3/gate/swift-tox-func-ec/45c7e8e/ : SUCCESS in 29m 54s\n- swift-tox-func-s3api http://logs.openstack.org/15/662615/3/gate/swift-tox-func-s3api/b9f9378/ : SUCCESS in 12m 18s\n- swift-probetests-centos-7 http://logs.openstack.org/15/662615/3/gate/swift-probetests-centos-7/60699c7/ : SUCCESS in 30m 19s\n- swift-dsvm-functional http://logs.openstack.org/15/662615/3/gate/swift-dsvm-functional/a36010b/ : SUCCESS in 31m 28s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/15/662615/3/gate/swift-dsvm-functional-ipv6/f39ed52/ : SUCCESS in 32m 13s\n- swift-tox-lower-constraints http://logs.openstack.org/15/662615/3/gate/swift-tox-lower-constraints/f2bddd5/ : SUCCESS in 12m 37s\n- openstack-tox-pep8 http://logs.openstack.org/15/662615/3/gate/openstack-tox-pep8/3c432b4/ : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":3},{"id":"647f225e98d76afe20ac4ab2600f2e0e1a2dd5a4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-15 00:15:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"a37c2966b622c8e7b6d35441b209f32e1c8dcb71","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-15 00:16:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- swift-promote-image http://logs.openstack.org/15/662615/3/promote/swift-promote-image/968d3e9/ : SUCCESS in 44s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"17fd725707c9a36fc4536bd3c955042931e314b3","revisions":{"3640338ffc96525ab614f862dd2a1674ad437e04":{"kind":"REWORK","_number":1,"created":"2019-06-02 05:27:53.000000000","uploader":{"_account_id":24168,"name":"Ehud Kaldor","email":"ehud@unfairfunction.org","username":"UnfairFunction"},"ref":"refs/changes/15/662615/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/15/662615/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/662615/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/662615/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/662615/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/15/662615/1"}}},"commit":{"parents":[{"commit":"769c0fb7fb95d3f9bfe68196af41f40e403c87fa","subject":"Merge \"Rework private-request-method interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/769c0fb7fb95d3f9bfe68196af41f40e403c87fa"}]}],"author":{"name":"Ehud Kaldor","email":"ehud@unfairfunction.org","date":"2019-06-02 05:27:40.000000000","tz":-420},"committer":{"name":"Ehud Kaldor","email":"ehud@unfairfunction.org","date":"2019-06-02 05:27:40.000000000","tz":-420},"subject":"Installing liberasurecode from Alpine Linux repos instead of building from source.","message":"Installing liberasurecode from Alpine Linux repos instead of building from source.\n\ncloses #1831136\n\nChange-Id: Ic9a2844fffc6dba6a4d44a9712edbb5cd91f0470\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3640338ffc96525ab614f862dd2a1674ad437e04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3640338ffc96525ab614f862dd2a1674ad437e04"}]},"branch":"refs/heads/master"},"f3321c4ecb65c7ea35b375fc979b7a891b2a3ec4":{"kind":"REWORK","_number":2,"created":"2019-06-05 18:27:49.000000000","uploader":{"_account_id":24168,"name":"Ehud Kaldor","email":"ehud@unfairfunction.org","username":"UnfairFunction"},"ref":"refs/changes/15/662615/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/15/662615/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/662615/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/662615/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/662615/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/15/662615/2"}}},"commit":{"parents":[{"commit":"769c0fb7fb95d3f9bfe68196af41f40e403c87fa","subject":"Merge \"Rework private-request-method interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/769c0fb7fb95d3f9bfe68196af41f40e403c87fa"}]}],"author":{"name":"Ehud Kaldor","email":"ehud@unfairfunction.org","date":"2019-06-05 18:27:34.000000000","tz":-420},"committer":{"name":"Ehud Kaldor","email":"ehud@unfairfunction.org","date":"2019-06-05 18:27:34.000000000","tz":-420},"subject":"Installing liberasurecode from Alpine Linux repos instead of building from source.","message":"Installing liberasurecode from Alpine Linux repos instead of building from source.\n\ncloses #1831136\n\nChange-Id: Ic9a2844fffc6dba6a4d44a9712edbb5cd91f0470\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f3321c4ecb65c7ea35b375fc979b7a891b2a3ec4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f3321c4ecb65c7ea35b375fc979b7a891b2a3ec4"}]},"branch":"refs/heads/master"},"17fd725707c9a36fc4536bd3c955042931e314b3":{"kind":"REWORK","_number":3,"created":"2019-06-05 18:32:34.000000000","uploader":{"_account_id":24168,"name":"Ehud Kaldor","email":"ehud@unfairfunction.org","username":"UnfairFunction"},"ref":"refs/changes/15/662615/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/15/662615/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/662615/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/662615/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/662615/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/15/662615/3"}}},"commit":{"parents":[{"commit":"769c0fb7fb95d3f9bfe68196af41f40e403c87fa","subject":"Merge \"Rework private-request-method interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/769c0fb7fb95d3f9bfe68196af41f40e403c87fa"}]}],"author":{"name":"Ehud Kaldor","email":"ehud@unfairfunction.org","date":"2019-06-05 18:29:28.000000000","tz":-420},"committer":{"name":"Ehud Kaldor","email":"ehud@unfairfunction.org","date":"2019-06-05 18:29:28.000000000","tz":-420},"subject":"Installing liberasurecode from Alpine Linux repos instead of building from source.","message":"Installing liberasurecode from Alpine Linux repos instead of building from source.\n\ncloses #1831136\n\nChange-Id: Ic9a2844fffc6dba6a4d44a9712edbb5cd91f0470\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/17fd725707c9a36fc4536bd3c955042931e314b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/17fd725707c9a36fc4536bd3c955042931e314b3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
