)]}'
{"id":"openstack%2Fswift~662735","triplet_id":"openstack%2Fswift~master~Ida50539e645ea7e2950ba668c7f031a8d10da787","project":"openstack/swift","branch":"master","topic":"bug/1665141","hashtags":[],"change_id":"Ida50539e645ea7e2950ba668c7f031a8d10da787","subject":"Do not sync suffixes when remote rejects reconstructor sync","status":"MERGED","created":"2019-06-03 11:14:29.000000000","updated":"2019-06-06 01:18:41.000000000","submitted":"2019-06-06 01:16:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662735-1559783811092-2d8656ec","meta_rev_id":"7b13c4e4e30708ae3450f9fb29762baf1e914af6","_number":662735,"virtual_id_number":662735,"owner":{"_account_id":30464,"name":"Kuan-Lin Chen","email":"kuanlinchen@synology.com","username":"kkchen"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-06-05 09:36:01.000000000","_account_id":30464,"name":"Kuan-Lin Chen","email":"kuanlinchen@synology.com","username":"kkchen"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"date":"2019-06-04 13:05:17.000000000","_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":2,"date":"2019-06-06 01:16:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":0,"_account_id":30464,"name":"Kuan-Lin Chen","email":"kuanlinchen@synology.com","username":"kkchen"},{"value":2,"date":"2019-06-04 03:06:10.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-03 15:33:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"_account_id":30464,"name":"Kuan-Lin Chen","email":"kuanlinchen@synology.com","username":"kkchen"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":1,"date":"2019-06-06 00:15:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30464,"name":"Kuan-Lin Chen","email":"kuanlinchen@synology.com","username":"kkchen"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-03 15:33:29.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2019-06-04 03:06:10.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2019-06-04 13:05:17.000000000","updated_by":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"reviewer":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"state":"REVIEWER"},{"updated":"2019-06-06 00:15:16.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2019-06-06 01:16:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f4d5aac1699527ccca17e89094fb0026baf4aaa8","author":{"_account_id":30464,"name":"Kuan-Lin Chen","email":"kuanlinchen@synology.com","username":"kkchen"},"date":"2019-06-03 11:14:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ff3a9a936000e70ed6e5c8ce8451243ee5401e3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 12:28:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/35/662735/1/check/openstack-tox-docs/559bf5f/ : FAILURE in 3m 34s\n- swift-build-image http://logs.openstack.org/35/662735/1/check/swift-build-image/1a450ed/ : SUCCESS in 9m 33s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/35/662735/1/check/swift-tox-py27/9521e6d/ : SUCCESS in 29m 11s\n- swift-tox-py36 http://logs.openstack.org/35/662735/1/check/swift-tox-py36/7b56efb/ : SUCCESS in 24m 10s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/35/662735/1/check/swift-tox-py37/28ab1e7/ : SUCCESS in 13m 46s\n- swift-tox-func http://logs.openstack.org/35/662735/1/check/swift-tox-func/1e591a6/ : SUCCESS in 8m 19s\n- swift-tox-func-py37 http://logs.openstack.org/35/662735/1/check/swift-tox-func-py37/a4c8bdf/ : FAILURE in 3m 03s\n- swift-tox-func-encryption http://logs.openstack.org/35/662735/1/check/swift-tox-func-encryption/03817ad/ : SUCCESS in 8m 46s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/35/662735/1/check/swift-tox-func-domain-remap-staticweb/a72a52e/ : SUCCESS in 7m 48s\n- swift-tox-func-ec http://logs.openstack.org/35/662735/1/check/swift-tox-func-ec/3a28162/ : FAILURE in 2m 35s\n- swift-tox-func-s3api http://logs.openstack.org/35/662735/1/check/swift-tox-func-s3api/d0938ec/ : FAILURE in 3m 00s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/35/662735/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/b558f2e/ : FAILURE in 4m 30s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/35/662735/1/check/swift-probetests-centos-7/73dd251/ : SUCCESS in 33m 01s\n- swift-dsvm-functional http://logs.openstack.org/35/662735/1/check/swift-dsvm-functional/eee439a/ : SUCCESS in 21m 09s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/35/662735/1/check/swift-dsvm-functional-ipv6/05aa0ea/ : SUCCESS in 26m 59s\n- swift-tox-lower-constraints http://logs.openstack.org/35/662735/1/check/swift-tox-lower-constraints/9361944/ : SUCCESS in 12m 01s\n- openstack-tox-pep8 http://logs.openstack.org/35/662735/1/check/openstack-tox-pep8/d4c824a/ : SUCCESS in 5m 51s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/35/662735/1/check/swift-multinode-rolling-upgrade/4a740eb/ : FAILURE in 10m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"81641fff69f1509d81854ea4694a4ece82cb90cb","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-03 15:33:29.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck\n\nI think this is reasonable -- in particular, I was reassured when I looked at https://review.opendev.org/#/c/435152/ for more context and saw\n\n\u003e SSYNC does not *need* to send a post-sync REPLICATE request. Any suffixes that are modified by the SSYNC protocol will call _finalize_put under the hood as it is syncing. It is however not harmful and potentially useful to go ahead refresh hashes after an SSYNC while the inodes of those suffixes are warm in the cache.\n\nin the commit message. I might bug Clay to take a look, too, though.","accounts_in_message":[],"_revision_number":1},{"id":"1c5c31bd3a1db01a2abb9ee36e04cbf871471cec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 18:45:06.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/35/662735/1/check/openstack-tox-docs/b7d2ef8/ : FAILURE in 4m 09s\n- swift-build-image http://logs.openstack.org/35/662735/1/check/swift-build-image/484a136/ : SUCCESS in 7m 33s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/35/662735/1/check/swift-tox-py27/72f9943/ : SUCCESS in 11m 21s\n- swift-tox-py36 http://logs.openstack.org/35/662735/1/check/swift-tox-py36/32e8778/ : SUCCESS in 10m 40s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/35/662735/1/check/swift-tox-py37/f7aea2b/ : SUCCESS in 11m 29s\n- swift-tox-func http://logs.openstack.org/35/662735/1/check/swift-tox-func/9cb0dc8/ : SUCCESS in 8m 50s\n- swift-tox-func-py37 http://logs.openstack.org/35/662735/1/check/swift-tox-func-py37/9b69e48/ : SUCCESS in 10m 56s\n- swift-tox-func-encryption http://logs.openstack.org/35/662735/1/check/swift-tox-func-encryption/f2fc338/ : SUCCESS in 8m 53s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/35/662735/1/check/swift-tox-func-domain-remap-staticweb/861d1a0/ : SUCCESS in 12m 58s\n- swift-tox-func-ec http://logs.openstack.org/35/662735/1/check/swift-tox-func-ec/a5072ea/ : SUCCESS in 10m 43s\n- swift-tox-func-s3api http://logs.openstack.org/35/662735/1/check/swift-tox-func-s3api/c0450c7/ : SUCCESS in 12m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/35/662735/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/d5dd21c/ : SUCCESS in 23m 01s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/35/662735/1/check/swift-probetests-centos-7/38f6d34/ : SUCCESS in 32m 29s\n- swift-dsvm-functional http://logs.openstack.org/35/662735/1/check/swift-dsvm-functional/be2a34d/ : SUCCESS in 27m 17s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/35/662735/1/check/swift-dsvm-functional-ipv6/f3791ba/ : SUCCESS in 26m 13s\n- swift-tox-lower-constraints http://logs.openstack.org/35/662735/1/check/swift-tox-lower-constraints/943db2c/ : SUCCESS in 12m 03s\n- openstack-tox-pep8 http://logs.openstack.org/35/662735/1/check/openstack-tox-pep8/d9a211e/ : SUCCESS in 5m 36s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/35/662735/1/check/swift-multinode-rolling-upgrade/bfbc067/ : FAILURE in 8m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ccb5114cc6968270ab7c4969d129ee7c73c12aa6","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-06-04 03:06:10.000000000","message":"Patch Set 1: Code-Review+2\n\nOkay, added Clay.\n\nDo not attempt more rechecks until this clears gate:\nhttps://review.opendev.org/662918","accounts_in_message":[],"_revision_number":1},{"id":"0d29b950660624e0809c427593728678a726df9e","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2019-06-04 13:05:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0fb9e27778d9c83e35ee048d657ec2122b2f5d63","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 18:07:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/35/662735/1/check/openstack-tox-docs/99522ca/html/ : SUCCESS in 3m 48s\n- swift-build-image http://logs.openstack.org/35/662735/1/check/swift-build-image/0dc2074/ : SUCCESS in 8m 08s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/35/662735/1/check/swift-tox-py27/4f2f678/ : SUCCESS in 10m 15s\n- swift-tox-py36 http://logs.openstack.org/35/662735/1/check/swift-tox-py36/f55ebdc/ : SUCCESS in 12m 06s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/35/662735/1/check/swift-tox-py37/adcae4e/ : SUCCESS in 10m 47s\n- swift-tox-func http://logs.openstack.org/35/662735/1/check/swift-tox-func/ef3f224/ : TIMED_OUT in 40m 44s\n- swift-tox-func-py37 http://logs.openstack.org/35/662735/1/check/swift-tox-func-py37/768230f/ : SUCCESS in 8m 21s\n- swift-tox-func-encryption http://logs.openstack.org/35/662735/1/check/swift-tox-func-encryption/b08bbc3/ : SUCCESS in 30m 12s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/35/662735/1/check/swift-tox-func-domain-remap-staticweb/22eb149/ : TIMED_OUT in 40m 51s\n- swift-tox-func-ec http://logs.openstack.org/35/662735/1/check/swift-tox-func-ec/17c1ff5/ : SUCCESS in 15m 46s\n- swift-tox-func-s3api http://logs.openstack.org/35/662735/1/check/swift-tox-func-s3api/cf8c0d4/ : SUCCESS in 11m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/35/662735/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/fb91ddb/ : SUCCESS in 13m 38s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/35/662735/1/check/swift-probetests-centos-7/d9f9b8a/ : SUCCESS in 29m 46s\n- swift-dsvm-functional http://logs.openstack.org/35/662735/1/check/swift-dsvm-functional/f5ccd38/ : SUCCESS in 24m 51s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/35/662735/1/check/swift-dsvm-functional-ipv6/f7bb32d/ : SUCCESS in 26m 51s\n- swift-tox-lower-constraints http://logs.openstack.org/35/662735/1/check/swift-tox-lower-constraints/e2874ec/ : SUCCESS in 12m 27s\n- openstack-tox-pep8 http://logs.openstack.org/35/662735/1/check/openstack-tox-pep8/a83de32/ : SUCCESS in 6m 56s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/35/662735/1/check/swift-multinode-rolling-upgrade/909a754/ : FAILURE in 8m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9d6d9715fdbe89f241da5a9a51be409d082db1d0","author":{"_account_id":30464,"name":"Kuan-Lin Chen","email":"kuanlinchen@synology.com","username":"kkchen"},"date":"2019-06-05 09:36:01.000000000","message":"Patch Set 1:\n\n`tox -efunc -vv` is failing but I can\u0027t reproduce it in my environment.\nAny idea?","accounts_in_message":[],"_revision_number":1},{"id":"cc77671161bace5d32549d8c56aed6bc71da78d0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-05 16:18:02.000000000","message":"Patch Set 1:\n\nrecheck\n\nSorry, the gate is way more resource constrained than most dev machines, which can cause timeouts or failures due to eventual consistency issues that would not normally be seen when working at the scale that the func tests exercise. You should be good -- If I don\u0027t hear anything from Clay in the next few days, I\u0027ll just go ahead and merge it.","accounts_in_message":[],"_revision_number":1},{"id":"94ac588e3d5070734b4904be5c3e10bd62ad6f63","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 21:29:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/35/662735/1/check/openstack-tox-docs/c9c60d4/html/ : SUCCESS in 4m 53s\n- swift-build-image http://logs.openstack.org/35/662735/1/check/swift-build-image/17f4485/ : SUCCESS in 9m 40s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/35/662735/1/check/swift-tox-py27/7ef2b67/ : SUCCESS in 10m 26s\n- swift-tox-py36 http://logs.openstack.org/35/662735/1/check/swift-tox-py36/996ac37/ : SUCCESS in 17m 48s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/35/662735/1/check/swift-tox-py37/b98ca83/ : SUCCESS in 18m 16s\n- swift-tox-func http://logs.openstack.org/35/662735/1/check/swift-tox-func/304e7bd/ : SUCCESS in 14m 45s\n- swift-tox-func-py37 http://logs.openstack.org/35/662735/1/check/swift-tox-func-py37/22dd648/ : SUCCESS in 7m 36s\n- swift-tox-func-encryption http://logs.openstack.org/35/662735/1/check/swift-tox-func-encryption/2a17602/ : SUCCESS in 10m 13s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/35/662735/1/check/swift-tox-func-domain-remap-staticweb/eeb7d2b/ : SUCCESS in 11m 23s\n- swift-tox-func-ec http://logs.openstack.org/35/662735/1/check/swift-tox-func-ec/bb00ec3/ : SUCCESS in 35m 53s\n- swift-tox-func-s3api http://logs.openstack.org/35/662735/1/check/swift-tox-func-s3api/6df59c6/ : SUCCESS in 37m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/35/662735/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/66dccf3/ : SUCCESS in 19m 25s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/35/662735/1/check/swift-probetests-centos-7/d91d1b9/ : SUCCESS in 38m 55s\n- swift-dsvm-functional http://logs.openstack.org/35/662735/1/check/swift-dsvm-functional/f4ab708/ : SUCCESS in 28m 51s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/35/662735/1/check/swift-dsvm-functional-ipv6/b8dae1a/ : SUCCESS in 28m 34s\n- swift-tox-lower-constraints http://logs.openstack.org/35/662735/1/check/swift-tox-lower-constraints/8d9af36/ : SUCCESS in 11m 52s\n- openstack-tox-pep8 http://logs.openstack.org/35/662735/1/check/openstack-tox-pep8/30fecdc/ : SUCCESS in 5m 32s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/35/662735/1/check/swift-multinode-rolling-upgrade/afce973/ : FAILURE in 9m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eb23434cc94bab5296a07d75f1360dc6b2a4e3a0","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-06-06 00:15:16.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6b1a610bed2955c5e606aadd7dccdd0be048bb02","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 00:15:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1dff39446ac6156f58808b0889683f31450d0326","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 01:16:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/35/662735/1/gate/openstack-tox-docs/0ca19a4/html/ : SUCCESS in 4m 12s\n- swift-upload-image http://logs.openstack.org/35/662735/1/gate/swift-upload-image/b0f94e3/ : SUCCESS in 9m 36s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/35/662735/1/gate/swift-tox-py27/d7580ec/ : SUCCESS in 10m 46s\n- swift-tox-py37 http://logs.openstack.org/35/662735/1/gate/swift-tox-py37/8acfd86/ : SUCCESS in 11m 12s\n- swift-tox-func http://logs.openstack.org/35/662735/1/gate/swift-tox-func/9defe60/ : SUCCESS in 8m 50s\n- swift-tox-func-py37 http://logs.openstack.org/35/662735/1/gate/swift-tox-func-py37/4d1ea03/ : SUCCESS in 8m 38s\n- swift-tox-func-encryption http://logs.openstack.org/35/662735/1/gate/swift-tox-func-encryption/509a4e0/ : SUCCESS in 8m 33s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/35/662735/1/gate/swift-tox-func-domain-remap-staticweb/b448c1b/ : SUCCESS in 8m 24s\n- swift-tox-func-ec http://logs.openstack.org/35/662735/1/gate/swift-tox-func-ec/c18995d/ : SUCCESS in 8m 55s\n- swift-tox-func-s3api http://logs.openstack.org/35/662735/1/gate/swift-tox-func-s3api/a7b62cb/ : SUCCESS in 26m 05s\n- swift-probetests-centos-7 http://logs.openstack.org/35/662735/1/gate/swift-probetests-centos-7/33194dc/ : SUCCESS in 34m 53s\n- swift-dsvm-functional http://logs.openstack.org/35/662735/1/gate/swift-dsvm-functional/e9ca711/ : SUCCESS in 27m 28s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/35/662735/1/gate/swift-dsvm-functional-ipv6/1cfa122/ : SUCCESS in 25m 26s\n- swift-tox-lower-constraints http://logs.openstack.org/35/662735/1/gate/swift-tox-lower-constraints/5bc9d9a/ : SUCCESS in 12m 15s\n- openstack-tox-pep8 http://logs.openstack.org/35/662735/1/gate/openstack-tox-pep8/c1309f7/ : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":1},{"id":"640f9abdb48c18f149d02b8c7e3ebeccb715af56","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 01:16:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1a1008e7ee774d6a0d9c0eb9deb90b04735764aa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 01:18:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- swift-promote-image http://logs.openstack.org/35/662735/1/promote/swift-promote-image/b6e693b/ : SUCCESS in 39s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"37fa12cd83849a3ae8374ff07861d1d710d53174","revisions":{"37fa12cd83849a3ae8374ff07861d1d710d53174":{"kind":"REWORK","_number":1,"created":"2019-06-03 11:14:29.000000000","uploader":{"_account_id":30464,"name":"Kuan-Lin Chen","email":"kuanlinchen@synology.com","username":"kkchen"},"ref":"refs/changes/35/662735/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/662735/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/662735/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/662735/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/662735/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/662735/1"}}},"commit":{"parents":[{"commit":"4f9595f1130f6c45f8e3e08db55dddc31844be93","subject":"py3: port symlink middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4f9595f1130f6c45f8e3e08db55dddc31844be93"}]}],"author":{"name":"Kuan-Lin Chen","email":"kuanlinchen@synology.com","date":"2019-06-03 10:39:51.000000000","tz":480},"committer":{"name":"Kuan-Lin Chen","email":"kuanlinchen@synology.com","date":"2019-06-03 10:39:51.000000000","tz":480},"subject":"Do not sync suffixes when remote rejects reconstructor sync","message":"Do not sync suffixes when remote rejects reconstructor sync\n\nThe commit a0fcca1e makes reconstructor not sync suffixes when remote\nreject reconstructor revert. However, the exact same logic should\nbe applied to SYNC job as well. REPLICATE requests aren\u0027t generally\nneeded when using SSYC (which the reconstructor always does).\n\nIf a ssync_sender fails to finish a sync the reconstructor should skip\nthe REPLICATE call entirely and move on to the next partition without\ncausing any useless remote IO.\n\nChange-Id: Ida50539e645ea7e2950ba668c7f031a8d10da787\nCloses-Bug: #1665141\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/37fa12cd83849a3ae8374ff07861d1d710d53174"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/37fa12cd83849a3ae8374ff07861d1d710d53174"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
