)]}'
{"id":"openstack%2Fswift~662799","triplet_id":"openstack%2Fswift~master~I05f38c6366e3da0aeed2b6451a8bd9d7dc531119","project":"openstack/swift","branch":"master","topic":"docker-liberasurecode","hashtags":[],"change_id":"I05f38c6366e3da0aeed2b6451a8bd9d7dc531119","subject":"Dockerfile cleanup","status":"MERGED","created":"2019-06-03 16:20:21.000000000","updated":"2019-06-19 03:23:55.000000000","submitted":"2019-06-19 03:21:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662799-1560914490799-6407a728","meta_rev_id":"2fe69bbf73de4b6c050793ca254dc6ef51b73b45","_number":662799,"virtual_id_number":662799,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-06-19 02:44:42.000000000","_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":2,"date":"2019-06-19 03:21:30.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":2,"date":"2019-06-18 12:37:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":1,"date":"2019-06-18 12:37:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-18 12:37:53.000000000","updated_by":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"reviewer":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"state":"REVIEWER"},{"updated":"2019-06-19 02:44:42.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2019-06-19 03:21:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7e77194298b5ceaacb582db87b284bd6113f3792","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-03 16:20:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a38af2e6d42d739f77b20a3a89dde763a4e2afa7","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2019-06-03 16:26:59.000000000","message":"Patch Set 1: Code-Review+1\n\ngood clean up. will wait on gate checks to +2 +A","accounts_in_message":[],"_revision_number":1},{"id":"e899bf12f9c481bec6c64952fd66c6d083c7b29e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 19:17:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/99/662799/1/check/openstack-tox-docs/38bb09e/ : FAILURE in 4m 04s\n- swift-build-image http://logs.openstack.org/99/662799/1/check/swift-build-image/7dc1624/ : SUCCESS in 7m 50s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/99/662799/1/check/swift-tox-py27/bc2fe2e/ : SUCCESS in 10m 36s\n- swift-tox-py36 http://logs.openstack.org/99/662799/1/check/swift-tox-py36/f0d2033/ : SUCCESS in 10m 19s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/99/662799/1/check/swift-tox-py37/c359dcb/ : SUCCESS in 14m 55s\n- swift-tox-func http://logs.openstack.org/99/662799/1/check/swift-tox-func/0258954/ : SUCCESS in 9m 24s\n- swift-tox-func-py37 http://logs.openstack.org/99/662799/1/check/swift-tox-func-py37/85710b7/ : SUCCESS in 8m 02s\n- swift-tox-func-encryption http://logs.openstack.org/99/662799/1/check/swift-tox-func-encryption/69e7a73/ : SUCCESS in 10m 35s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/99/662799/1/check/swift-tox-func-domain-remap-staticweb/afbe958/ : TIMED_OUT in 40m 48s\n- swift-tox-func-ec http://logs.openstack.org/99/662799/1/check/swift-tox-func-ec/b5c5094/ : SUCCESS in 9m 28s\n- swift-tox-func-s3api http://logs.openstack.org/99/662799/1/check/swift-tox-func-s3api/a3d79a6/ : SUCCESS in 12m 56s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/99/662799/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/1f88f92/ : SUCCESS in 16m 34s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/99/662799/1/check/swift-probetests-centos-7/4b29632/ : SUCCESS in 30m 12s\n- swift-dsvm-functional http://logs.openstack.org/99/662799/1/check/swift-dsvm-functional/35a2e1a/ : SUCCESS in 29m 33s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/99/662799/1/check/swift-dsvm-functional-ipv6/7bdcccf/ : SUCCESS in 27m 44s\n- swift-tox-lower-constraints http://logs.openstack.org/99/662799/1/check/swift-tox-lower-constraints/7b24aba/ : SUCCESS in 11m 45s\n- openstack-tox-pep8 http://logs.openstack.org/99/662799/1/check/openstack-tox-pep8/f8c9cf0/ : SUCCESS in 5m 14s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/99/662799/1/check/swift-multinode-rolling-upgrade/c24fd79/ : FAILURE in 9m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e7a757a653fc7516159841566dcc020f329cf237","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2019-06-04 13:05:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7ce7d25726eab0c4b7ed43f39e57a8266105e72f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 18:07:13.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/99/662799/1/check/openstack-tox-docs/f59fe39/html/ : SUCCESS in 4m 07s\n- swift-build-image http://logs.openstack.org/99/662799/1/check/swift-build-image/7158833/ : SUCCESS in 9m 28s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/99/662799/1/check/swift-tox-py27/845a3cb/ : TIMED_OUT in 40m 51s\n- swift-tox-py36 http://logs.openstack.org/99/662799/1/check/swift-tox-py36/8f57f1a/ : SUCCESS in 11m 15s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/99/662799/1/check/swift-tox-py37/9f2831b/ : SUCCESS in 12m 12s\n- swift-tox-func http://logs.openstack.org/99/662799/1/check/swift-tox-func/8aac75d/ : SUCCESS in 10m 32s\n- swift-tox-func-py37 http://logs.openstack.org/99/662799/1/check/swift-tox-func-py37/57b680f/ : SUCCESS in 7m 23s\n- swift-tox-func-encryption http://logs.openstack.org/99/662799/1/check/swift-tox-func-encryption/4ab0bdc/ : SUCCESS in 9m 17s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/99/662799/1/check/swift-tox-func-domain-remap-staticweb/05eebb0/ : SUCCESS in 8m 53s\n- swift-tox-func-ec http://logs.openstack.org/99/662799/1/check/swift-tox-func-ec/b24a15e/ : SUCCESS in 40m 50s\n- swift-tox-func-s3api http://logs.openstack.org/99/662799/1/check/swift-tox-func-s3api/3bb71ee/ : SUCCESS in 12m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/99/662799/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/8bb3150/ : SUCCESS in 12m 07s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/99/662799/1/check/swift-probetests-centos-7/f416d95/ : SUCCESS in 32m 40s\n- swift-dsvm-functional http://logs.openstack.org/99/662799/1/check/swift-dsvm-functional/7d64f5c/ : SUCCESS in 26m 25s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/99/662799/1/check/swift-dsvm-functional-ipv6/57a8a8a/ : SUCCESS in 27m 43s\n- swift-tox-lower-constraints http://logs.openstack.org/99/662799/1/check/swift-tox-lower-constraints/dd5b15a/ : SUCCESS in 11m 48s\n- openstack-tox-pep8 http://logs.openstack.org/99/662799/1/check/openstack-tox-pep8/49b1e05/ : SUCCESS in 6m 03s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/99/662799/1/check/swift-multinode-rolling-upgrade/75da42d/ : FAILURE in 8m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f3f363190b42831511a7d85f8a950e3649c36c58","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-04 23:13:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4581d23ca6b8f9100fa168571830e3a8c2b29681","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 01:46:27.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/99/662799/1/check/openstack-tox-docs/66e730a/html/ : SUCCESS in 4m 28s\n- swift-build-image http://logs.openstack.org/99/662799/1/check/swift-build-image/132e9ab/ : SUCCESS in 8m 28s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/99/662799/1/check/swift-tox-py27/f958013/ : SUCCESS in 28m 20s\n- swift-tox-py36 http://logs.openstack.org/99/662799/1/check/swift-tox-py36/a823235/ : SUCCESS in 10m 38s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/99/662799/1/check/swift-tox-py37/ec8b297/ : SUCCESS in 28m 30s\n- swift-tox-func http://logs.openstack.org/99/662799/1/check/swift-tox-func/742152c/ : SUCCESS in 11m 31s\n- swift-tox-func-py37 http://logs.openstack.org/99/662799/1/check/swift-tox-func-py37/892e9fd/ : SUCCESS in 7m 03s\n- swift-tox-func-encryption http://logs.openstack.org/99/662799/1/check/swift-tox-func-encryption/e87a235/ : SUCCESS in 10m 46s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/99/662799/1/check/swift-tox-func-domain-remap-staticweb/6d3e913/ : SUCCESS in 9m 10s\n- swift-tox-func-ec http://logs.openstack.org/99/662799/1/check/swift-tox-func-ec/9d7da7d/ : RETRY_LIMIT in 1m 35s\n- swift-tox-func-s3api http://logs.openstack.org/99/662799/1/check/swift-tox-func-s3api/88563c7/ : SUCCESS in 12m 18s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/99/662799/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/06cb069/ : SUCCESS in 14m 30s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/99/662799/1/check/swift-probetests-centos-7/21a72bd/ : FAILURE in 33m 21s\n- swift-dsvm-functional http://logs.openstack.org/99/662799/1/check/swift-dsvm-functional/ac0cc4d/ : SUCCESS in 24m 42s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/99/662799/1/check/swift-dsvm-functional-ipv6/722cf5a/ : RETRY_LIMIT in 2m 24s\n- swift-tox-lower-constraints http://logs.openstack.org/99/662799/1/check/swift-tox-lower-constraints/d3a6975/ : SUCCESS in 14m 01s\n- openstack-tox-pep8 http://logs.openstack.org/99/662799/1/check/openstack-tox-pep8/0113b28/ : SUCCESS in 5m 27s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/99/662799/1/check/swift-multinode-rolling-upgrade/4db598a/ : FAILURE in 8m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"128e4f5b6c1862e488d345eae83757ad874d5b55","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-05 16:19:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0e55424b4d5ff827f68e9adc79a06b7d27924f15","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-05 19:14:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"83cae2078ee67e60da7eda852882c6df55f0b22b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 01:25:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/99/662799/2/check/openstack-tox-docs/ede9876/html/ : SUCCESS in 3m 41s\n- swift-build-image http://logs.openstack.org/99/662799/2/check/swift-build-image/816a3e9/ : SUCCESS in 9m 13s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/99/662799/2/check/swift-tox-py27/fe99d2d/ : SUCCESS in 31m 01s\n- swift-tox-py36 http://logs.openstack.org/99/662799/2/check/swift-tox-py36/364f492/ : SUCCESS in 10m 36s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/99/662799/2/check/swift-tox-py37/b4b977d/ : SUCCESS in 10m 17s\n- swift-tox-func http://logs.openstack.org/99/662799/2/check/swift-tox-func/fc595e3/ : SUCCESS in 21m 03s\n- swift-tox-func-py37 http://logs.openstack.org/99/662799/2/check/swift-tox-func-py37/52b0946/ : SUCCESS in 8m 32s\n- swift-tox-func-encryption http://logs.openstack.org/99/662799/2/check/swift-tox-func-encryption/0137170/ : SUCCESS in 9m 51s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/99/662799/2/check/swift-tox-func-domain-remap-staticweb/ad9c304/ : SUCCESS in 8m 18s\n- swift-tox-func-ec http://logs.openstack.org/99/662799/2/check/swift-tox-func-ec/fe8fe65/ : SUCCESS in 8m 46s\n- swift-tox-func-s3api http://logs.openstack.org/99/662799/2/check/swift-tox-func-s3api/e20b0a5/ : SUCCESS in 10m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/99/662799/2/check/swift-tox-func-s3api-ceph-s3tests-tempauth/899ac20/ : SUCCESS in 14m 11s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/99/662799/2/check/swift-probetests-centos-7/0abe3f4/ : SUCCESS in 33m 35s\n- swift-dsvm-functional http://logs.openstack.org/99/662799/2/check/swift-dsvm-functional/224fd6f/ : SUCCESS in 23m 50s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/99/662799/2/check/swift-dsvm-functional-ipv6/831b895/ : SUCCESS in 27m 23s\n- swift-tox-lower-constraints http://logs.openstack.org/99/662799/2/check/swift-tox-lower-constraints/cb1bc42/ : SUCCESS in 11m 00s\n- openstack-tox-pep8 http://logs.openstack.org/99/662799/2/check/openstack-tox-pep8/852a051/ : SUCCESS in 5m 02s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/99/662799/2/check/swift-multinode-rolling-upgrade/5a1ee5e/ : FAILURE in 9m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1f69c40c1c0cab7a487c3a74f9fd595c9cb08e60","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2019-06-18 12:37:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f466ea925be9413eae857e983819c27652658f53","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 12:38:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a9e4de69c9410701082c24b5959808af14a34abe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 14:19:33.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/99/662799/2/gate/openstack-tox-docs/9f7d652/html/ : SUCCESS in 4m 09s\n- swift-upload-image http://logs.openstack.org/99/662799/2/gate/swift-upload-image/124d372/ : SUCCESS in 8m 35s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/99/662799/2/gate/swift-tox-py27/55dd00b/ : SUCCESS in 11m 19s\n- swift-tox-py37 http://logs.openstack.org/99/662799/2/gate/swift-tox-py37/910287f/ : SUCCESS in 11m 22s\n- swift-tox-func http://logs.openstack.org/99/662799/2/gate/swift-tox-func/a23f68c/ : RETRY_LIMIT in 2m 15s\n- swift-tox-func-py37 http://logs.openstack.org/99/662799/2/gate/swift-tox-func-py37/bb0b9dc/ : SUCCESS in 8m 44s\n- swift-tox-func-encryption http://logs.openstack.org/99/662799/2/gate/swift-tox-func-encryption/76d2498/ : SUCCESS in 13m 27s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/99/662799/2/gate/swift-tox-func-domain-remap-staticweb/9cb30d7/ : RETRY_LIMIT in 1m 47s\n- swift-tox-func-ec http://logs.openstack.org/99/662799/2/gate/swift-tox-func-ec/8b514f2/ : SUCCESS in 9m 18s\n- swift-tox-func-s3api http://logs.openstack.org/99/662799/2/gate/swift-tox-func-s3api/9054eea/ : SUCCESS in 11m 49s\n- swift-probetests-centos-7 http://logs.openstack.org/99/662799/2/gate/swift-probetests-centos-7/47e2bb4/ : SUCCESS in 32m 13s\n- swift-dsvm-functional http://logs.openstack.org/99/662799/2/gate/swift-dsvm-functional/a271c68/ : SUCCESS in 24m 44s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/99/662799/2/gate/swift-dsvm-functional-ipv6/5f94e7f/ : SUCCESS in 28m 39s\n- swift-tox-lower-constraints http://logs.openstack.org/99/662799/2/gate/swift-tox-lower-constraints/0b20904/ : SUCCESS in 12m 15s\n- openstack-tox-pep8 http://logs.openstack.org/99/662799/2/gate/openstack-tox-pep8/a4926a5/ : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":2},{"id":"66d6ab710f12687b92da07ba7ae6e9600a8e64e8","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2019-06-18 14:20:09.000000000","message":"Patch Set 2:\n\nretry","accounts_in_message":[],"_revision_number":2},{"id":"fb9956444b466c2eb89815718a35ef6c1252d380","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2019-06-18 14:20:16.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"20d4444d6ae484a229cf95f9f8c01843965978d8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 01:36:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/99/662799/2/check/openstack-tox-docs/d69cc34/html/ : SUCCESS in 4m 24s\n- swift-build-image http://logs.openstack.org/99/662799/2/check/swift-build-image/c7daa34/ : SUCCESS in 7m 30s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/99/662799/2/check/swift-tox-py27/8fc8f14/ : SUCCESS in 11m 52s\n- swift-tox-py36 http://logs.openstack.org/99/662799/2/check/swift-tox-py36/7e05667/ : SUCCESS in 10m 37s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/99/662799/2/check/swift-tox-py37/0d1a85f/ : SUCCESS in 13m 35s\n- swift-tox-func http://logs.openstack.org/99/662799/2/check/swift-tox-func/4105670/ : SUCCESS in 9m 58s\n- swift-tox-func-py37 http://logs.openstack.org/99/662799/2/check/swift-tox-func-py37/3a66ee4/ : SUCCESS in 10m 24s\n- swift-tox-func-encryption http://logs.openstack.org/99/662799/2/check/swift-tox-func-encryption/4dd2de8/ : SUCCESS in 8m 51s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/99/662799/2/check/swift-tox-func-domain-remap-staticweb/c632ec2/ : SUCCESS in 11m 12s\n- swift-tox-func-ec http://logs.openstack.org/99/662799/2/check/swift-tox-func-ec/83906c5/ : SUCCESS in 9m 35s\n- swift-tox-func-s3api http://logs.openstack.org/99/662799/2/check/swift-tox-func-s3api/1596d7b/ : SUCCESS in 12m 56s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/99/662799/2/check/swift-tox-func-s3api-ceph-s3tests-tempauth/fa288ce/ : SUCCESS in 13m 50s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/99/662799/2/check/swift-probetests-centos-7/2e7bc0f/ : SUCCESS in 32m 26s\n- swift-dsvm-functional http://logs.openstack.org/99/662799/2/check/swift-dsvm-functional/13e8670/ : SUCCESS in 26m 53s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/99/662799/2/check/swift-dsvm-functional-ipv6/7673dd0/ : SUCCESS in 26m 58s\n- swift-tox-lower-constraints http://logs.openstack.org/99/662799/2/check/swift-tox-lower-constraints/5cce1c1/ : SUCCESS in 14m 42s\n- openstack-tox-pep8 http://logs.openstack.org/99/662799/2/check/openstack-tox-pep8/8e6c3a1/ : SUCCESS in 4m 59s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/99/662799/2/check/swift-multinode-rolling-upgrade/1ee29c4/ : FAILURE in 9m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4c2e440f9cdb0497d74c786641b114a6bf842a39","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 01:37:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0fe08fec8c063e7941ea4da7e782bd17c054279a","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2019-06-19 02:44:42.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"719cc05bcd316c1c5319f6554acd1e1419a15624","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 03:21:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"3ca18e678f873b4e7e3a80ab7fcaa7becdf4e334","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 03:21:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/99/662799/2/gate/openstack-tox-docs/cb4deec/html/ : SUCCESS in 4m 58s\n- swift-upload-image http://logs.openstack.org/99/662799/2/gate/swift-upload-image/16d8c9a/ : SUCCESS in 9m 38s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/99/662799/2/gate/swift-tox-py27/cbdf6a6/ : SUCCESS in 11m 32s\n- swift-tox-py37 http://logs.openstack.org/99/662799/2/gate/swift-tox-py37/25a8355/ : SUCCESS in 11m 56s\n- swift-tox-func http://logs.openstack.org/99/662799/2/gate/swift-tox-func/a117f99/ : SUCCESS in 12m 42s\n- swift-tox-func-py37 http://logs.openstack.org/99/662799/2/gate/swift-tox-func-py37/3efea2b/ : SUCCESS in 10m 37s\n- swift-tox-func-encryption http://logs.openstack.org/99/662799/2/gate/swift-tox-func-encryption/3cf3dab/ : SUCCESS in 9m 54s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/99/662799/2/gate/swift-tox-func-domain-remap-staticweb/b76752a/ : SUCCESS in 9m 03s\n- swift-tox-func-ec http://logs.openstack.org/99/662799/2/gate/swift-tox-func-ec/b81e376/ : SUCCESS in 15m 43s\n- swift-tox-func-s3api http://logs.openstack.org/99/662799/2/gate/swift-tox-func-s3api/90a936f/ : SUCCESS in 12m 57s\n- swift-probetests-centos-7 http://logs.openstack.org/99/662799/2/gate/swift-probetests-centos-7/efc48a8/ : SUCCESS in 34m 27s\n- swift-dsvm-functional http://logs.openstack.org/99/662799/2/gate/swift-dsvm-functional/5244acd/ : SUCCESS in 27m 11s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/99/662799/2/gate/swift-dsvm-functional-ipv6/e8aafaa/ : SUCCESS in 30m 39s\n- swift-tox-lower-constraints http://logs.openstack.org/99/662799/2/gate/swift-tox-lower-constraints/34b9727/ : SUCCESS in 11m 46s\n- openstack-tox-pep8 http://logs.openstack.org/99/662799/2/gate/openstack-tox-pep8/9377445/ : SUCCESS in 6m 03s","accounts_in_message":[],"_revision_number":2},{"id":"d68a4225ffcbd1da47cb1041f8948a9266d2b73f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 03:23:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- swift-promote-image http://logs.openstack.org/99/662799/2/promote/swift-promote-image/939825a/ : SUCCESS in 51s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a122b720e9c61b38cfe502dcb63a9c119220c4c1","revisions":{"12302638cb8a70636614d595b6f3325cfc2a4885":{"kind":"REWORK","_number":1,"created":"2019-06-03 16:20:21.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/99/662799/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/99/662799/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/99/662799/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/99/662799/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/99/662799/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/99/662799/1"}}},"commit":{"parents":[{"commit":"3640338ffc96525ab614f862dd2a1674ad437e04","subject":"Installing liberasurecode from Alpine Linux repos instead of building from source.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3640338ffc96525ab614f862dd2a1674ad437e04"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-06-03 15:57:23.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-06-03 16:11:15.000000000","tz":-420},"subject":"Dockerfile cleanup","message":"Dockerfile cleanup\n\n- Remove some build packages that are no longer needed now that we\n  install liberasurecode from upstream.\n- Use `set -e` in all install scripts.\n\nChange-Id: I05f38c6366e3da0aeed2b6451a8bd9d7dc531119\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/12302638cb8a70636614d595b6f3325cfc2a4885"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/12302638cb8a70636614d595b6f3325cfc2a4885"}]},"branch":"refs/heads/master"},"a122b720e9c61b38cfe502dcb63a9c119220c4c1":{"kind":"REWORK","_number":2,"created":"2019-06-05 19:14:31.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/99/662799/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/99/662799/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/99/662799/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/99/662799/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/99/662799/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/99/662799/2"}}},"commit":{"parents":[{"commit":"23eca56ce00bd4421fcac02b0003fc76d335aeee","subject":"Merge \"Bump openstackdocstheme to 1.30.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/23eca56ce00bd4421fcac02b0003fc76d335aeee"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-06-03 15:57:23.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-06-05 19:12:43.000000000","tz":-420},"subject":"Dockerfile cleanup","message":"Dockerfile cleanup\n\n- Use `set -e` in all install scripts. Clean up swift_needs to still\n  work properly and exit cleanly under `set -e`.\n- Drop some redundant slashes.\n\nChange-Id: I05f38c6366e3da0aeed2b6451a8bd9d7dc531119\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a122b720e9c61b38cfe502dcb63a9c119220c4c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a122b720e9c61b38cfe502dcb63a9c119220c4c1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
