)]}'
{"id":"openstack%2Fswift~663098","triplet_id":"openstack%2Fswift~master~I12f6e4a5a880a2d5e0f4fff4618e4196be311267","project":"openstack/swift","branch":"master","topic":"s3-api-tests","hashtags":[],"change_id":"I12f6e4a5a880a2d5e0f4fff4618e4196be311267","subject":"Note s3api compatability requirements in sample.conf","status":"MERGED","created":"2019-06-04 16:47:18.000000000","updated":"2019-06-05 04:13:56.000000000","submitted":"2019-06-05 04:11:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663098-1559707915858-4bf1be21","meta_rev_id":"fcb5116d0ae85b37517bff3f22325af4857d7c3a","_number":663098,"virtual_id_number":663098,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-06-05 04:11:55.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-04 23:12:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-04 23:12:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-04 23:12:40.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2019-06-05 04:11:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0c8fb5413ca038389be31dad3308b4226e978957","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-06-04 16:47:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0851f0b7b83a7529b4342c537eac7767434fbf33","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 22:53:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/98/663098/1/check/openstack-tox-docs/784bc40/html/ : SUCCESS in 4m 21s\n- swift-build-image http://logs.openstack.org/98/663098/1/check/swift-build-image/0fbfd6e/ : SUCCESS in 10m 30s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/98/663098/1/check/swift-tox-py27/12eddba/ : SUCCESS in 10m 05s\n- swift-tox-py36 http://logs.openstack.org/98/663098/1/check/swift-tox-py36/d29ce2b/ : SUCCESS in 11m 52s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/98/663098/1/check/swift-tox-py37/645e9dc/ : SUCCESS in 14m 10s\n- swift-tox-func http://logs.openstack.org/98/663098/1/check/swift-tox-func/2d09347/ : SUCCESS in 8m 48s\n- swift-tox-func-py37 http://logs.openstack.org/98/663098/1/check/swift-tox-func-py37/4342393/ : SUCCESS in 8m 19s\n- swift-tox-func-encryption http://logs.openstack.org/98/663098/1/check/swift-tox-func-encryption/ec281a7/ : SUCCESS in 8m 59s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/98/663098/1/check/swift-tox-func-domain-remap-staticweb/a359875/ : SUCCESS in 8m 51s\n- swift-tox-func-ec http://logs.openstack.org/98/663098/1/check/swift-tox-func-ec/0c2acbb/ : SUCCESS in 11m 51s\n- swift-tox-func-s3api http://logs.openstack.org/98/663098/1/check/swift-tox-func-s3api/effc7b8/ : SUCCESS in 11m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/98/663098/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/c9373ba/ : SUCCESS in 19m 19s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/98/663098/1/check/swift-probetests-centos-7/0d3306c/ : SUCCESS in 39m 25s\n- swift-dsvm-functional http://logs.openstack.org/98/663098/1/check/swift-dsvm-functional/6304d0c/ : SUCCESS in 29m 03s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/98/663098/1/check/swift-dsvm-functional-ipv6/8e5ae19/ : SUCCESS in 39m 13s\n- swift-tox-lower-constraints http://logs.openstack.org/98/663098/1/check/swift-tox-lower-constraints/5567323/ : SUCCESS in 12m 09s\n- openstack-tox-pep8 http://logs.openstack.org/98/663098/1/check/openstack-tox-pep8/cd4b616/ : SUCCESS in 5m 26s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/98/663098/1/check/swift-multinode-rolling-upgrade/75360be/ : FAILURE in 8m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6a147150cd2aa00bd1a6c1211da1becc1cb2b936","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-04 23:12:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"81fa5e9cdcbff60607fd04e4630cc9703841cebc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 23:13:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4f819088e97cd644d4079a3aff11b63fe4dcc8f8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 04:11:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5eeba3fdb401967bcd15400001bf38d42669694a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 04:11:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/98/663098/1/gate/openstack-tox-docs/04e0be2/html/ : SUCCESS in 4m 29s\n- swift-upload-image http://logs.openstack.org/98/663098/1/gate/swift-upload-image/4c9fc76/ : SUCCESS in 9m 34s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/98/663098/1/gate/swift-tox-py27/96a5203/ : SUCCESS in 12m 28s\n- swift-tox-py37 http://logs.openstack.org/98/663098/1/gate/swift-tox-py37/aa08785/ : SUCCESS in 12m 07s\n- swift-tox-func http://logs.openstack.org/98/663098/1/gate/swift-tox-func/1578f4c/ : SUCCESS in 9m 07s\n- swift-tox-func-py37 http://logs.openstack.org/98/663098/1/gate/swift-tox-func-py37/f19db64/ : SUCCESS in 7m 51s\n- swift-tox-func-encryption http://logs.openstack.org/98/663098/1/gate/swift-tox-func-encryption/57c034d/ : SUCCESS in 9m 49s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/98/663098/1/gate/swift-tox-func-domain-remap-staticweb/968888f/ : SUCCESS in 9m 43s\n- swift-tox-func-ec http://logs.openstack.org/98/663098/1/gate/swift-tox-func-ec/dcd5561/ : SUCCESS in 9m 50s\n- swift-tox-func-s3api http://logs.openstack.org/98/663098/1/gate/swift-tox-func-s3api/2f8ed24/ : SUCCESS in 11m 16s\n- swift-probetests-centos-7 http://logs.openstack.org/98/663098/1/gate/swift-probetests-centos-7/5400271/ : SUCCESS in 31m 43s\n- swift-dsvm-functional http://logs.openstack.org/98/663098/1/gate/swift-dsvm-functional/8e3af6b/ : SUCCESS in 21m 46s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/98/663098/1/gate/swift-dsvm-functional-ipv6/b558b73/ : SUCCESS in 26m 00s\n- swift-tox-lower-constraints http://logs.openstack.org/98/663098/1/gate/swift-tox-lower-constraints/276fc20/ : SUCCESS in 11m 20s\n- openstack-tox-pep8 http://logs.openstack.org/98/663098/1/gate/openstack-tox-pep8/8020d13/ : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":1},{"id":"19c001106f15194a7f9a487ee087031e65e8908b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 04:13:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- swift-promote-image http://logs.openstack.org/98/663098/1/promote/swift-promote-image/1aab4ce/ : SUCCESS in 42s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"79ab3a490ada6aa4db28efa5b71a3f1a080920f7","revisions":{"79ab3a490ada6aa4db28efa5b71a3f1a080920f7":{"kind":"REWORK","_number":1,"created":"2019-06-04 16:47:18.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/98/663098/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/98/663098/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/663098/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/663098/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/663098/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/98/663098/1"}}},"commit":{"parents":[{"commit":"89c9c6f0b27fcc208996590610df8b04bec2ebc8","subject":"Have a separate s3api functional test suite","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/89c9c6f0b27fcc208996590610df8b04bec2ebc8"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2019-06-04 16:45:35.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2019-06-04 16:45:35.000000000","tz":-300},"subject":"Note s3api compatability requirements in sample.conf","message":"Note s3api compatability requirements in sample.conf\n\nChange-Id: I12f6e4a5a880a2d5e0f4fff4618e4196be311267\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/79ab3a490ada6aa4db28efa5b71a3f1a080920f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/79ab3a490ada6aa4db28efa5b71a3f1a080920f7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
