)]}'
{"id":"openstack%2Fswift~663230","triplet_id":"openstack%2Fswift~master~I166211b690b08521171b489582fa419d756b1972","project":"openstack/swift","branch":"master","topic":"bug#openstackdocstheme","hashtags":[],"change_id":"I166211b690b08521171b489582fa419d756b1972","subject":"Bump openstackdocstheme to 1.30.0","status":"MERGED","created":"2019-06-05 04:39:00.000000000","updated":"2019-07-01 07:39:12.000000000","submitted":"2019-06-05 15:18:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663230-1559747909379-6e0684d2","meta_rev_id":"ac56b3cc96054b411bc8d846ebf5f8ce25c90e79","_number":663230,"virtual_id_number":663230,"owner":{"_account_id":28239,"name":"zhaixiaojun","email":"zhaixiaojun@gohighsec.com","username":"zhaixiaojun"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},{"value":2,"date":"2019-06-05 15:18:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"all":[{"value":2,"date":"2019-06-05 14:30:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"all":[{"value":1,"date":"2019-06-05 14:30:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-05 14:30:26.000000000","updated_by":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"reviewer":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"state":"REVIEWER"},{"updated":"2019-06-05 15:18:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c2c342b39233fcfdf1d16c68a1f06d30a971e2ad","author":{"_account_id":28239,"name":"zhaixiaojun","email":"zhaixiaojun@gohighsec.com","username":"zhaixiaojun"},"date":"2019-06-05 04:39:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0bb7d2191b1062f5923b6e06321088b024477078","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 05:15:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/30/663230/1/check/build-openstack-api-ref/4ab7f2a/html/ : SUCCESS in 2m 55s\n- openstack-tox-docs http://logs.openstack.org/30/663230/1/check/openstack-tox-docs/424d336/html/ : SUCCESS in 4m 02s\n- requirements-check http://logs.openstack.org/30/663230/1/check/requirements-check/dbe6ffc/ : SUCCESS in 2m 41s\n- build-openstack-releasenotes http://logs.openstack.org/30/663230/1/check/build-openstack-releasenotes/c609230/html/ : SUCCESS in 7m 16s\n- swift-build-image http://logs.openstack.org/30/663230/1/check/swift-build-image/86d232d/ : SUCCESS in 8m 27s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/30/663230/1/check/swift-tox-py27/12d1010/ : SUCCESS in 10m 58s\n- swift-tox-py36 http://logs.openstack.org/30/663230/1/check/swift-tox-py36/6f56b29/ : SUCCESS in 10m 07s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/30/663230/1/check/swift-tox-py37/4cfa803/ : SUCCESS in 10m 07s\n- swift-tox-func http://logs.openstack.org/30/663230/1/check/swift-tox-func/b47a3e5/ : SUCCESS in 8m 26s\n- swift-tox-func-py37 http://logs.openstack.org/30/663230/1/check/swift-tox-func-py37/cb81da2/ : SUCCESS in 7m 02s\n- swift-tox-func-encryption http://logs.openstack.org/30/663230/1/check/swift-tox-func-encryption/f68763b/ : SUCCESS in 8m 38s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/30/663230/1/check/swift-tox-func-domain-remap-staticweb/32e9050/ : SUCCESS in 8m 22s\n- swift-tox-func-ec http://logs.openstack.org/30/663230/1/check/swift-tox-func-ec/878c852/ : SUCCESS in 9m 13s\n- swift-tox-func-s3api http://logs.openstack.org/30/663230/1/check/swift-tox-func-s3api/33bdf9f/ : SUCCESS in 11m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/30/663230/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/2484833/ : SUCCESS in 13m 01s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/30/663230/1/check/swift-probetests-centos-7/5b87f3d/ : SUCCESS in 30m 42s\n- swift-dsvm-functional http://logs.openstack.org/30/663230/1/check/swift-dsvm-functional/07e6bc0/ : SUCCESS in 25m 10s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/30/663230/1/check/swift-dsvm-functional-ipv6/3ad95d8/ : SUCCESS in 27m 15s\n- swift-tox-lower-constraints http://logs.openstack.org/30/663230/1/check/swift-tox-lower-constraints/46bd478/ : SUCCESS in 10m 35s\n- openstack-tox-pep8 http://logs.openstack.org/30/663230/1/check/openstack-tox-pep8/6e28c1a/ : SUCCESS in 5m 16s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/30/663230/1/check/swift-multinode-rolling-upgrade/0f46719/ : FAILURE in 8m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ba979b0cbea96baf0c09e092d96fe8a498b33608","author":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"date":"2019-06-05 14:30:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nthank you!","accounts_in_message":[],"_revision_number":1},{"id":"5f2774822c3af6c2a5ca2c2abafb1acd0bad1912","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 14:30:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7eb8471c5e8f95533110673468c6b4b0e181d353","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 15:18:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/30/663230/1/gate/build-openstack-api-ref/40ef90c/html/ : SUCCESS in 3m 19s\n- openstack-tox-docs http://logs.openstack.org/30/663230/1/gate/openstack-tox-docs/63dbf59/html/ : SUCCESS in 3m 47s\n- requirements-check http://logs.openstack.org/30/663230/1/gate/requirements-check/640e845/ : SUCCESS in 2m 11s\n- build-openstack-releasenotes http://logs.openstack.org/30/663230/1/gate/build-openstack-releasenotes/52bbea9/html/ : SUCCESS in 7m 12s\n- swift-upload-image http://logs.openstack.org/30/663230/1/gate/swift-upload-image/317463b/ : SUCCESS in 8m 39s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/30/663230/1/gate/swift-tox-py27/434c77d/ : SUCCESS in 10m 14s\n- swift-tox-py37 http://logs.openstack.org/30/663230/1/gate/swift-tox-py37/cbd21ca/ : SUCCESS in 10m 51s\n- swift-tox-func http://logs.openstack.org/30/663230/1/gate/swift-tox-func/e9ed56c/ : SUCCESS in 10m 38s\n- swift-tox-func-py37 http://logs.openstack.org/30/663230/1/gate/swift-tox-func-py37/1c19c66/ : SUCCESS in 10m 59s\n- swift-tox-func-encryption http://logs.openstack.org/30/663230/1/gate/swift-tox-func-encryption/9273e06/ : SUCCESS in 8m 49s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/30/663230/1/gate/swift-tox-func-domain-remap-staticweb/ab65957/ : SUCCESS in 9m 35s\n- swift-tox-func-ec http://logs.openstack.org/30/663230/1/gate/swift-tox-func-ec/4f380e5/ : SUCCESS in 9m 11s\n- swift-tox-func-s3api http://logs.openstack.org/30/663230/1/gate/swift-tox-func-s3api/db9bedd/ : SUCCESS in 11m 31s\n- swift-probetests-centos-7 http://logs.openstack.org/30/663230/1/gate/swift-probetests-centos-7/79139f8/ : SUCCESS in 31m 38s\n- swift-dsvm-functional http://logs.openstack.org/30/663230/1/gate/swift-dsvm-functional/11cc4a8/ : SUCCESS in 30m 57s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/30/663230/1/gate/swift-dsvm-functional-ipv6/5c5d4a5/ : SUCCESS in 27m 45s\n- swift-tox-lower-constraints http://logs.openstack.org/30/663230/1/gate/swift-tox-lower-constraints/3e20871/ : SUCCESS in 12m 17s\n- openstack-tox-pep8 http://logs.openstack.org/30/663230/1/gate/openstack-tox-pep8/665ea81/ : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":1},{"id":"04ae3e48b9395bf527802d5f964c6f538ea4cdbc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 15:18:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"431dc0607c53633fe284e959d22d9db6ee86db50","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 15:19:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- swift-promote-image http://logs.openstack.org/30/663230/1/promote/swift-promote-image/2a6b306/ : SUCCESS in 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cadbb180eaf38b8aaeab939cc302ddfabfb503d1","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-07-01 07:39:12.000000000","message":"Patch Set 1:\n\nFor the record, such a big bump wasn\u0027t really needed. See also the the discussion here: https://review.opendev.org/#/c/668081/","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"98637dc1e7a6ef5641079a6226d12bf106436b35","revisions":{"98637dc1e7a6ef5641079a6226d12bf106436b35":{"kind":"REWORK","_number":1,"created":"2019-06-05 04:39:00.000000000","uploader":{"_account_id":28239,"name":"zhaixiaojun","email":"zhaixiaojun@gohighsec.com","username":"zhaixiaojun"},"ref":"refs/changes/30/663230/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/30/663230/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/663230/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/663230/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/663230/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/30/663230/1"}}},"commit":{"parents":[{"commit":"4ae5ae100d18c9248bbc98d7f60b6c6b1a6ed219","subject":"Merge \"Have a separate s3api functional test suite\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ae5ae100d18c9248bbc98d7f60b6c6b1a6ed219"}]}],"author":{"name":"翟小君","email":"zhaixiaojun@gohighsec.com","date":"2019-06-05 04:35:00.000000000","tz":480},"committer":{"name":"翟小君","email":"zhaixiaojun@gohighsec.com","date":"2019-06-05 04:38:00.000000000","tz":480},"subject":"Bump openstackdocstheme to 1.30.0","message":"Bump openstackdocstheme to 1.30.0\n\n...to pick up many improvements, including the return of table borders.\n\nChange-Id: I166211b690b08521171b489582fa419d756b1972\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/98637dc1e7a6ef5641079a6226d12bf106436b35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/98637dc1e7a6ef5641079a6226d12bf106436b35"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
