)]}'
{"id":"openstack%2Fswift~663260","triplet_id":"openstack%2Fswift~feature%2Flosf~I3b1508c9555bcff9016fd7ba918cebcaa4eb484c","project":"openstack/swift","branch":"feature/losf","topic":"merge-master","hashtags":[],"change_id":"I3b1508c9555bcff9016fd7ba918cebcaa4eb484c","subject":"Merge remote-tracking branch \u0027remotes/origin/master\u0027 into merge-master","status":"MERGED","created":"2019-06-05 07:28:26.000000000","updated":"2019-06-05 23:23:31.000000000","submitted":"2019-06-05 23:22:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663260-1559776941553-2d8656ec","meta_rev_id":"df8e042a6f9aeee6866c122bd1d914cbd3ff1aee","_number":663260,"virtual_id_number":663260,"owner":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-06-05 23:22:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-05 21:42:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-05 21:42:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-05 21:42:31.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2019-06-05 23:22:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f6838f499ed44e4d852d309ae58f54504eb25eac","author":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"date":"2019-06-05 07:28:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"69008b45a4076924d337627542d8c511105f86b4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 08:12:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/60/663260/1/check/openstack-tox-docs/72afab9/html/ : SUCCESS in 4m 08s\n- swift-build-image http://logs.openstack.org/60/663260/1/check/swift-build-image/13bce20/ : SUCCESS in 13m 55s\n- swift-tox-py27 http://logs.openstack.org/60/663260/1/check/swift-tox-py27/c9dd871/ : SUCCESS in 10m 49s\n- swift-tox-py36 http://logs.openstack.org/60/663260/1/check/swift-tox-py36/2cd3bf3/ : SUCCESS in 9m 58s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/60/663260/1/check/swift-tox-py37/a415095/ : SUCCESS in 10m 26s\n- swift-tox-func http://logs.openstack.org/60/663260/1/check/swift-tox-func/cf2fc0b/ : SUCCESS in 8m 42s\n- swift-tox-func-encryption http://logs.openstack.org/60/663260/1/check/swift-tox-func-encryption/bc8bcc8/ : SUCCESS in 8m 53s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/60/663260/1/check/swift-tox-func-domain-remap-staticweb/55e5fa1/ : SUCCESS in 8m 44s\n- swift-tox-func-ec http://logs.openstack.org/60/663260/1/check/swift-tox-func-ec/ab88d1f/ : SUCCESS in 10m 06s\n- swift-tox-func-s3api http://logs.openstack.org/60/663260/1/check/swift-tox-func-s3api/bd5f35e/ : SUCCESS in 11m 32s\n- swift-tox-func-losf http://logs.openstack.org/60/663260/1/check/swift-tox-func-losf/e047eb3/ : SUCCESS in 14m 52s (non-voting)\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/60/663260/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/cb93941/ : SUCCESS in 12m 56s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/60/663260/1/check/swift-probetests-centos-7/9dd8372/ : SUCCESS in 41m 26s\n- swift-dsvm-functional http://logs.openstack.org/60/663260/1/check/swift-dsvm-functional/3adf418/ : SUCCESS in 25m 09s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/60/663260/1/check/swift-dsvm-functional-ipv6/a273084/ : SUCCESS in 30m 01s\n- swift-tox-lower-constraints http://logs.openstack.org/60/663260/1/check/swift-tox-lower-constraints/b2fa120/ : SUCCESS in 11m 23s\n- openstack-tox-pep8 http://logs.openstack.org/60/663260/1/check/openstack-tox-pep8/adecfd5/ : SUCCESS in 5m 15s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/60/663260/1/check/swift-multinode-rolling-upgrade/2534133/ : FAILURE in 8m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2b9722cd5025e3971a877e78f6f79561df205140","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-05 21:42:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e3f10c4ce9e8a91c18c60e3631630b9ca7f543a2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 21:42:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"206b1c6c8427163863133cfe244c46df0d8e71b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 23:22:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/60/663260/1/gate/openstack-tox-docs/b88b9db/html/ : SUCCESS in 5m 02s\n- swift-upload-image http://logs.openstack.org/60/663260/1/gate/swift-upload-image/629689a/ : SUCCESS in 15m 39s\n- swift-tox-py27 http://logs.openstack.org/60/663260/1/gate/swift-tox-py27/13a2fc2/ : SUCCESS in 10m 48s\n- swift-tox-py37 http://logs.openstack.org/60/663260/1/gate/swift-tox-py37/91afbde/ : SUCCESS in 12m 13s\n- swift-tox-func http://logs.openstack.org/60/663260/1/gate/swift-tox-func/f94d3a5/ : SUCCESS in 9m 14s\n- swift-tox-func-encryption http://logs.openstack.org/60/663260/1/gate/swift-tox-func-encryption/8de8c90/ : SUCCESS in 9m 46s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/60/663260/1/gate/swift-tox-func-domain-remap-staticweb/3e075c2/ : SUCCESS in 9m 29s\n- swift-tox-func-ec http://logs.openstack.org/60/663260/1/gate/swift-tox-func-ec/4844ae7/ : SUCCESS in 10m 01s\n- swift-tox-func-s3api http://logs.openstack.org/60/663260/1/gate/swift-tox-func-s3api/e18f926/ : SUCCESS in 23m 12s\n- swift-probetests-centos-7 http://logs.openstack.org/60/663260/1/gate/swift-probetests-centos-7/ef6b472/ : SUCCESS in 35m 30s\n- swift-dsvm-functional http://logs.openstack.org/60/663260/1/gate/swift-dsvm-functional/b9308ff/ : SUCCESS in 26m 42s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/60/663260/1/gate/swift-dsvm-functional-ipv6/32255db/ : SUCCESS in 26m 07s\n- swift-tox-lower-constraints http://logs.openstack.org/60/663260/1/gate/swift-tox-lower-constraints/eb8bcab/ : SUCCESS in 12m 13s\n- openstack-tox-pep8 http://logs.openstack.org/60/663260/1/gate/openstack-tox-pep8/af46e3c/ : SUCCESS in 6m 01s","accounts_in_message":[],"_revision_number":1},{"id":"d730784584d261d3072c4feb08a2fc9cf0eba347","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 23:22:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ec9bcc98f428007cb25b1dfae4c635ff84a70373","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 23:23:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- swift-promote-image http://logs.openstack.org/60/663260/1/promote/swift-promote-image/d9c8119/ : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3ba65cb7eca6948d32b53ba91157037c48f06002","revisions":{"3ba65cb7eca6948d32b53ba91157037c48f06002":{"kind":"REWORK","_number":1,"created":"2019-06-05 07:28:26.000000000","uploader":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"ref":"refs/changes/60/663260/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/663260/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/663260/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/663260/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/663260/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/663260/1"}}},"commit":{"parents":[{"commit":"20b097818d7fa03a0bc70669676f26b5b40fd0b6","subject":"Merge \"Merge remote-tracking branch \u0027remotes/origin/master\u0027 into merg-master\" into feature/losf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/20b097818d7fa03a0bc70669676f26b5b40fd0b6"}]},{"commit":"4ffc5db5f23206a362c6695c6ea581f9787e6e95","subject":"Merge \"Note s3api compatability requirements in sample.conf\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ffc5db5f23206a362c6695c6ea581f9787e6e95"}]}],"author":{"name":"Kota Tsuyuzaki","email":"tsuyuzaki.kota@lab.ntt.co.jp","date":"2019-06-05 07:27:22.000000000","tz":540},"committer":{"name":"Kota Tsuyuzaki","email":"tsuyuzaki.kota@lab.ntt.co.jp","date":"2019-06-05 07:27:22.000000000","tz":540},"subject":"Merge remote-tracking branch \u0027remotes/origin/master\u0027 into merge-master","message":"Merge remote-tracking branch \u0027remotes/origin/master\u0027 into merge-master\n\nConflicts:\n    swift/obj/diskfile.py\n\nChange-Id: I3b1508c9555bcff9016fd7ba918cebcaa4eb484c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ba65cb7eca6948d32b53ba91157037c48f06002"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ba65cb7eca6948d32b53ba91157037c48f06002"}]},"branch":"refs/heads/feature/losf"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
