)]}'
{"id":"openstack%2Fswift~663456","triplet_id":"openstack%2Fswift~master~Ia438df2a5c9bdae83bd9b0e05e1cac564c638420","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"Ia438df2a5c9bdae83bd9b0e05e1cac564c638420","subject":"revert object-expirer pid to conf regression","status":"NEW","created":"2019-06-05 22:58:18.000000000","updated":"2019-06-17 19:46:11.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4c7c4ada08742971c62f561279bba7fe1ae498db","_number":663456,"virtual_id_number":663456,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-06-05 22:59:33.000000000","_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":1,"date":"2019-06-17 19:46:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-06-17 15:20:48.000000000","_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-17 15:20:48.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2019-06-17 19:46:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fec342814db563ed784bddde253fa3469d361e91","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-06-05 22:58:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2481f25d484a37196f8aba0245ab1ec3c965517f","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-06-05 22:59:33.000000000","message":"Patch Set 1:\n\nI\u0027m not really sure if this is the right thing to do.\n\nWe want operators to be able to configure object-expirer either way, but the in-ability to go from a pid file to the correct conf file makes some diagnostics more difficult...","accounts_in_message":[],"_revision_number":1},{"id":"7b7868a099374ae65809506fe39d4a683c1eff70","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-06 00:15:23.000000000","message":"Patch Set 1:\n\nI mean, the Server knows which conf files it found, right? Maybe it should go with object-server.conf if the expirer was configured there, and object-expirer.conf if it wasn\u0027t. If we really want to get fancy, get the pid out of the pid file and peek at its command line if its still running...","accounts_in_message":[],"_revision_number":1},{"id":"c15556d6f6b47b4f890aca5e3fd7698e95c83704","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 02:37:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/56/663456/1/check/openstack-tox-docs/a6fa642/html/ : SUCCESS in 5m 02s\n- swift-build-image http://logs.openstack.org/56/663456/1/check/swift-build-image/96526e2/ : SUCCESS in 9m 24s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/56/663456/1/check/swift-tox-py27/4cf7796/ : SUCCESS in 12m 40s\n- swift-tox-py36 http://logs.openstack.org/56/663456/1/check/swift-tox-py36/dea5ee4/ : SUCCESS in 11m 05s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/56/663456/1/check/swift-tox-py37/0e0f2c5/ : FAILURE in 13m 36s\n- swift-tox-func http://logs.openstack.org/56/663456/1/check/swift-tox-func/542749f/ : SUCCESS in 10m 50s\n- swift-tox-func-py37 http://logs.openstack.org/56/663456/1/check/swift-tox-func-py37/809fd55/ : SUCCESS in 8m 15s\n- swift-tox-func-encryption http://logs.openstack.org/56/663456/1/check/swift-tox-func-encryption/c255c30/ : SUCCESS in 8m 55s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/56/663456/1/check/swift-tox-func-domain-remap-staticweb/e52b08d/ : SUCCESS in 11m 11s\n- swift-tox-func-ec http://logs.openstack.org/56/663456/1/check/swift-tox-func-ec/1d9b6d6/ : SUCCESS in 11m 09s\n- swift-tox-func-s3api http://logs.openstack.org/56/663456/1/check/swift-tox-func-s3api/6dc2ebc/ : SUCCESS in 11m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/56/663456/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/75c4219/ : SUCCESS in 12m 23s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/56/663456/1/check/swift-probetests-centos-7/6a85e3c/ : SUCCESS in 33m 54s\n- swift-dsvm-functional http://logs.openstack.org/56/663456/1/check/swift-dsvm-functional/e3df519/ : SUCCESS in 26m 37s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/56/663456/1/check/swift-dsvm-functional-ipv6/33daba3/ : SUCCESS in 25m 41s\n- swift-tox-lower-constraints http://logs.openstack.org/56/663456/1/check/swift-tox-lower-constraints/2bddd6a/ : SUCCESS in 15m 22s\n- openstack-tox-pep8 http://logs.openstack.org/56/663456/1/check/openstack-tox-pep8/b139eff/ : SUCCESS in 5m 15s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/56/663456/1/check/swift-multinode-rolling-upgrade/8befefb/ : FAILURE in 8m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"faf39323d142d283d29af14a292d695b106b76c7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-15 04:28:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"71daaaf9c5458aafab4ce5062ae8e54b6197afb5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-15 08:35:25.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/56/663456/1/check/openstack-tox-docs/c588556/html/ : SUCCESS in 4m 56s\n- swift-build-image http://logs.openstack.org/56/663456/1/check/swift-build-image/c879af0/ : SUCCESS in 8m 53s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/56/663456/1/check/swift-tox-py27/5b9a14d/ : SUCCESS in 16m 49s\n- swift-tox-py36 http://logs.openstack.org/56/663456/1/check/swift-tox-py36/d45a636/ : SUCCESS in 12m 30s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/56/663456/1/check/swift-tox-py37/59bc354/ : SUCCESS in 11m 33s\n- swift-tox-func http://logs.openstack.org/56/663456/1/check/swift-tox-func/796e4b9/ : SUCCESS in 8m 27s\n- swift-tox-func-py37 http://logs.openstack.org/56/663456/1/check/swift-tox-func-py37/ff82265/ : SUCCESS in 8m 47s\n- swift-tox-func-encryption http://logs.openstack.org/56/663456/1/check/swift-tox-func-encryption/a6f8bf4/ : SUCCESS in 9m 15s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/56/663456/1/check/swift-tox-func-domain-remap-staticweb/36f6b52/ : SUCCESS in 14m 07s\n- swift-tox-func-ec http://logs.openstack.org/56/663456/1/check/swift-tox-func-ec/4177ded/ : SUCCESS in 10m 15s\n- swift-tox-func-s3api http://logs.openstack.org/56/663456/1/check/swift-tox-func-s3api/4f08948/ : SUCCESS in 13m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/56/663456/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/3225f38/ : SUCCESS in 13m 11s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/56/663456/1/check/swift-probetests-centos-7/8940592/ : FAILURE in 32m 54s\n- swift-dsvm-functional http://logs.openstack.org/56/663456/1/check/swift-dsvm-functional/ad45aa4/ : SUCCESS in 35m 29s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/56/663456/1/check/swift-dsvm-functional-ipv6/d11349d/ : SUCCESS in 37m 01s\n- swift-tox-lower-constraints http://logs.openstack.org/56/663456/1/check/swift-tox-lower-constraints/17051ee/ : SUCCESS in 12m 53s\n- openstack-tox-pep8 http://logs.openstack.org/56/663456/1/check/openstack-tox-pep8/454f50b/ : SUCCESS in 6m 00s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/56/663456/1/check/swift-multinode-rolling-upgrade/90a8311/ : FAILURE in 9m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"50af7ba5688d80c1449643e0234757a0237be8a0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-17 15:20:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4c7c4ada08742971c62f561279bba7fe1ae498db","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 19:46:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/56/663456/1/check/openstack-tox-docs/3a60a8a/html/ : SUCCESS in 4m 25s\n- swift-build-image http://logs.openstack.org/56/663456/1/check/swift-build-image/9f29429/ : SUCCESS in 7m 21s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/56/663456/1/check/swift-tox-py27/c21791c/ : SUCCESS in 13m 02s\n- swift-tox-py36 http://logs.openstack.org/56/663456/1/check/swift-tox-py36/e3c4e50/ : SUCCESS in 16m 13s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/56/663456/1/check/swift-tox-py37/5efae75/ : SUCCESS in 11m 51s\n- swift-tox-func http://logs.openstack.org/56/663456/1/check/swift-tox-func/3752089/ : SUCCESS in 9m 00s\n- swift-tox-func-py37 http://logs.openstack.org/56/663456/1/check/swift-tox-func-py37/cc31484/ : SUCCESS in 8m 40s\n- swift-tox-func-encryption http://logs.openstack.org/56/663456/1/check/swift-tox-func-encryption/a784e1b/ : SUCCESS in 9m 15s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/56/663456/1/check/swift-tox-func-domain-remap-staticweb/6eb7832/ : SUCCESS in 9m 15s\n- swift-tox-func-ec http://logs.openstack.org/56/663456/1/check/swift-tox-func-ec/0aab84f/ : SUCCESS in 10m 16s\n- swift-tox-func-s3api http://logs.openstack.org/56/663456/1/check/swift-tox-func-s3api/37145a9/ : SUCCESS in 23m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/56/663456/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/dbd5d28/ : FAILURE in 18m 49s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/56/663456/1/check/swift-probetests-centos-7/69be89d/ : SUCCESS in 33m 36s\n- swift-dsvm-functional http://logs.openstack.org/56/663456/1/check/swift-dsvm-functional/e8bd351/ : SUCCESS in 26m 47s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/56/663456/1/check/swift-dsvm-functional-ipv6/f5b1028/ : SUCCESS in 28m 17s\n- swift-tox-lower-constraints http://logs.openstack.org/56/663456/1/check/swift-tox-lower-constraints/570713c/ : SUCCESS in 11m 58s\n- openstack-tox-pep8 http://logs.openstack.org/56/663456/1/check/openstack-tox-pep8/bc09488/ : SUCCESS in 5m 38s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/56/663456/1/check/swift-multinode-rolling-upgrade/f358dba/ : FAILURE in 9m 18s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4b530b7f640a46692431e05bfd3b9e2699fae3ae","revisions":{"4b530b7f640a46692431e05bfd3b9e2699fae3ae":{"kind":"REWORK","_number":1,"created":"2019-06-05 22:58:18.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/56/663456/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/663456/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/663456/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/663456/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/663456/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/663456/1"}}},"commit":{"parents":[{"commit":"23eca56ce00bd4421fcac02b0003fc76d335aeee","subject":"Merge \"Bump openstackdocstheme to 1.30.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/23eca56ce00bd4421fcac02b0003fc76d335aeee"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2019-06-05 22:58:12.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2019-06-05 22:58:12.000000000","tz":-300},"subject":"revert object-expirer pid to conf regression","message":"revert object-expirer pid to conf regression\n\nChange-Id: Ia438df2a5c9bdae83bd9b0e05e1cac564c638420\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4b530b7f640a46692431e05bfd3b9e2699fae3ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4b530b7f640a46692431e05bfd3b9e2699fae3ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
