)]}'
{"id":"openstack%2Fswift~663751","triplet_id":"openstack%2Fswift~master~I0d33864f4bffa401082548ee9a52f6eb50cb1f39","project":"openstack/swift","branch":"master","topic":"bug/1831932","hashtags":[],"change_id":"I0d33864f4bffa401082548ee9a52f6eb50cb1f39","subject":"Fix swift with python \u003c2.7.9","status":"MERGED","created":"2019-06-06 19:26:07.000000000","updated":"2019-06-08 09:54:04.000000000","submitted":"2019-06-08 09:53:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663751-1559987588711-811cd442","meta_rev_id":"b76a1a2adb356526147bb61c2f28a2bdf8360661","_number":663751,"virtual_id_number":663751,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":2,"date":"2019-06-08 09:53:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2019-06-07 22:17:56.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-07 15:41:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-07 15:41:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-07 15:41:13.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2019-06-07 22:17:56.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2019-06-08 09:53:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"879a00b06ba692ed0747d7ce29539894dc295b84","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-06-06 19:26:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e0365a5f2d1c88f9145b599611a29b51c3512c55","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-06 19:37:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0aade0d57e40041beee97ecd1e54881248a4fbe2","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-06-06 19:43:58.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"01a86230696f6abdaa4e074bad00a53abf6c6640","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-06-06 19:52:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"23f3a098db1730051661c7ca869a438da92b763c","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-06-06 19:53:57.000000000","message":"Patch Set 2: Code-Review+1\n\ni was able to verify that if block short-circut shit definitely DOES work on trusty exactly how you\u0027d expect - dunno how well my test demonstrates that or protects us from breaking it later...\n\nit\u0027ll probably be fine :shipit:","accounts_in_message":[],"_revision_number":2},{"id":"b52b8a0156482e7ba5b95a6c6af7564bb4017981","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 02:04:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/51/663751/2/check/openstack-tox-docs/4c00364/html/ : SUCCESS in 4m 16s\n- swift-build-image http://logs.openstack.org/51/663751/2/check/swift-build-image/df22bf4/ : SUCCESS in 7m 56s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/51/663751/2/check/swift-tox-py27/7c6fbde/ : SUCCESS in 10m 15s\n- swift-tox-py36 http://logs.openstack.org/51/663751/2/check/swift-tox-py36/0ffd2f1/ : FAILURE in 11m 35s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/51/663751/2/check/swift-tox-py37/4069fb7/ : FAILURE in 11m 38s\n- swift-tox-func http://logs.openstack.org/51/663751/2/check/swift-tox-func/3e7b5ea/ : SUCCESS in 8m 54s\n- swift-tox-func-py37 http://logs.openstack.org/51/663751/2/check/swift-tox-func-py37/a3a30a9/ : SUCCESS in 9m 35s\n- swift-tox-func-encryption http://logs.openstack.org/51/663751/2/check/swift-tox-func-encryption/fb718b0/ : SUCCESS in 8m 48s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/51/663751/2/check/swift-tox-func-domain-remap-staticweb/120a312/ : SUCCESS in 9m 21s\n- swift-tox-func-ec http://logs.openstack.org/51/663751/2/check/swift-tox-func-ec/6ebb30a/ : SUCCESS in 9m 46s\n- swift-tox-func-s3api http://logs.openstack.org/51/663751/2/check/swift-tox-func-s3api/2aa7456/ : SUCCESS in 12m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/51/663751/2/check/swift-tox-func-s3api-ceph-s3tests-tempauth/154c898/ : FAILURE in 15m 33s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/51/663751/2/check/swift-probetests-centos-7/9dae4aa/ : FAILURE in 35m 20s\n- swift-dsvm-functional http://logs.openstack.org/51/663751/2/check/swift-dsvm-functional/416b8a5/ : SUCCESS in 26m 33s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/51/663751/2/check/swift-dsvm-functional-ipv6/1c17d3b/ : SUCCESS in 26m 49s\n- swift-tox-lower-constraints http://logs.openstack.org/51/663751/2/check/swift-tox-lower-constraints/2a586d1/ : FAILURE in 12m 23s\n- openstack-tox-pep8 http://logs.openstack.org/51/663751/2/check/openstack-tox-pep8/52cd778/ : SUCCESS in 5m 38s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/51/663751/2/check/swift-multinode-rolling-upgrade/a251fcc/ : FAILURE in 8m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3d38c568c4a5de490074a4e3e4a37fa9fd84cbc0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-07 15:25:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"841f5702a1e362ef395e1f22eebae7cdfcbdbf3e","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-06-07 15:39:09.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6ec35712063e7c7a7fb25e3e7b5a7241119ab68a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-07 15:41:13.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"7f10caf1f10d6d12aa9fb0e2a2cd0a048f692ef4","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-06-07 22:17:56.000000000","message":"Patch Set 3: Code-Review+2\n\nnice, a regression","accounts_in_message":[],"_revision_number":3},{"id":"edd63ddf04d5242bf21efa27d3ef502aff1e9799","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 00:59:42.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/51/663751/3/check/openstack-tox-docs/7c42aa9/html/ : SUCCESS in 5m 44s\n- swift-build-image http://logs.openstack.org/51/663751/3/check/swift-build-image/fac1958/ : SUCCESS in 9m 48s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/51/663751/3/check/swift-tox-py27/9564f0a/ : SUCCESS in 10m 10s\n- swift-tox-py36 http://logs.openstack.org/51/663751/3/check/swift-tox-py36/791754f/ : SUCCESS in 10m 58s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/51/663751/3/check/swift-tox-py37/604350e/ : FAILURE in 3m 20s\n- swift-tox-func http://logs.openstack.org/51/663751/3/check/swift-tox-func/52620ec/ : SUCCESS in 11m 15s\n- swift-tox-func-py37 http://logs.openstack.org/51/663751/3/check/swift-tox-func-py37/1441dc8/ : SUCCESS in 9m 14s\n- swift-tox-func-encryption http://logs.openstack.org/51/663751/3/check/swift-tox-func-encryption/f16b10f/ : SUCCESS in 9m 01s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/51/663751/3/check/swift-tox-func-domain-remap-staticweb/1603ac1/ : SUCCESS in 13m 36s\n- swift-tox-func-ec http://logs.openstack.org/51/663751/3/check/swift-tox-func-ec/0beee31/ : SUCCESS in 18m 53s\n- swift-tox-func-s3api http://logs.openstack.org/51/663751/3/check/swift-tox-func-s3api/de07cf4/ : SUCCESS in 11m 57s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/51/663751/3/check/swift-tox-func-s3api-ceph-s3tests-tempauth/da1c9e7/ : SUCCESS in 13m 42s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/51/663751/3/check/swift-probetests-centos-7/1231e4d/ : SUCCESS in 33m 00s\n- swift-dsvm-functional http://logs.openstack.org/51/663751/3/check/swift-dsvm-functional/3ce8d7e/ : SUCCESS in 28m 38s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/51/663751/3/check/swift-dsvm-functional-ipv6/f0674fe/ : SUCCESS in 28m 14s\n- swift-tox-lower-constraints http://logs.openstack.org/51/663751/3/check/swift-tox-lower-constraints/32fa289/ : SUCCESS in 11m 49s\n- openstack-tox-pep8 http://logs.openstack.org/51/663751/3/check/openstack-tox-pep8/4adb165/ : SUCCESS in 5m 43s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/51/663751/3/check/swift-multinode-rolling-upgrade/c5debad/ : FAILURE in 8m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4d0ff47826405e9479f24211f316e535491507a5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-08 01:02:09.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"e36c7584928dddb137bec790933f79d8dbb5aa24","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 07:26:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/51/663751/3/check/openstack-tox-docs/5cd99cc/html/ : SUCCESS in 3m 56s\n- swift-build-image http://logs.openstack.org/51/663751/3/check/swift-build-image/d8a5115/ : SUCCESS in 8m 51s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/51/663751/3/check/swift-tox-py27/f694181/ : SUCCESS in 12m 06s\n- swift-tox-py36 http://logs.openstack.org/51/663751/3/check/swift-tox-py36/112a95f/ : SUCCESS in 16m 31s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/51/663751/3/check/swift-tox-py37/0930be4/ : SUCCESS in 10m 56s\n- swift-tox-func http://logs.openstack.org/51/663751/3/check/swift-tox-func/d3850b7/ : SUCCESS in 10m 21s\n- swift-tox-func-py37 http://logs.openstack.org/51/663751/3/check/swift-tox-func-py37/64e5e01/ : SUCCESS in 9m 31s\n- swift-tox-func-encryption http://logs.openstack.org/51/663751/3/check/swift-tox-func-encryption/fa6862c/ : SUCCESS in 9m 03s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/51/663751/3/check/swift-tox-func-domain-remap-staticweb/96bbbe1/ : SUCCESS in 16m 29s\n- swift-tox-func-ec http://logs.openstack.org/51/663751/3/check/swift-tox-func-ec/cd7be9d/ : SUCCESS in 9m 35s\n- swift-tox-func-s3api http://logs.openstack.org/51/663751/3/check/swift-tox-func-s3api/7796fbc/ : SUCCESS in 11m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/51/663751/3/check/swift-tox-func-s3api-ceph-s3tests-tempauth/ea8ec27/ : SUCCESS in 13m 30s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/51/663751/3/check/swift-probetests-centos-7/03cd844/ : SUCCESS in 32m 24s\n- swift-dsvm-functional http://logs.openstack.org/51/663751/3/check/swift-dsvm-functional/be85325/ : SUCCESS in 27m 36s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/51/663751/3/check/swift-dsvm-functional-ipv6/927e8fd/ : SUCCESS in 26m 58s\n- swift-tox-lower-constraints http://logs.openstack.org/51/663751/3/check/swift-tox-lower-constraints/a8388e5/ : SUCCESS in 11m 45s\n- openstack-tox-pep8 http://logs.openstack.org/51/663751/3/check/openstack-tox-pep8/efc50e9/ : SUCCESS in 5m 51s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/51/663751/3/check/swift-multinode-rolling-upgrade/18c2dcb/ : FAILURE in 8m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"765adb3ed18115b2eba6658078f73ea019a489b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 07:27:17.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"bb7a515a510927351f6b56093061352a5ac2e912","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 09:53:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"8388bf25249ff7b3c12c0b6cf1489ec0374359c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 09:53:08.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/51/663751/3/gate/openstack-tox-docs/a9d0b2b/html/ : SUCCESS in 5m 09s\n- swift-upload-image http://logs.openstack.org/51/663751/3/gate/swift-upload-image/fa9f94e/ : SUCCESS in 9m 36s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/51/663751/3/gate/swift-tox-py27/dd33202/ : SUCCESS in 15m 21s\n- swift-tox-py37 http://logs.openstack.org/51/663751/3/gate/swift-tox-py37/4b31a24/ : SUCCESS in 17m 11s\n- swift-tox-func http://logs.openstack.org/51/663751/3/gate/swift-tox-func/6585def/ : SUCCESS in 22m 31s\n- swift-tox-func-py37 http://logs.openstack.org/51/663751/3/gate/swift-tox-func-py37/72333ba/ : SUCCESS in 13m 49s\n- swift-tox-func-encryption http://logs.openstack.org/51/663751/3/gate/swift-tox-func-encryption/9bc00f5/ : SUCCESS in 24m 24s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/51/663751/3/gate/swift-tox-func-domain-remap-staticweb/3f5fccb/ : SUCCESS in 8m 23s\n- swift-tox-func-ec http://logs.openstack.org/51/663751/3/gate/swift-tox-func-ec/769ebbf/ : SUCCESS in 10m 12s\n- swift-tox-func-s3api http://logs.openstack.org/51/663751/3/gate/swift-tox-func-s3api/feade9d/ : SUCCESS in 13m 20s\n- swift-probetests-centos-7 http://logs.openstack.org/51/663751/3/gate/swift-probetests-centos-7/e59ec74/ : SUCCESS in 31m 09s\n- swift-dsvm-functional http://logs.openstack.org/51/663751/3/gate/swift-dsvm-functional/fd0b4d2/ : SUCCESS in 24m 57s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/51/663751/3/gate/swift-dsvm-functional-ipv6/ffcd8bd/ : SUCCESS in 25m 12s\n- swift-tox-lower-constraints http://logs.openstack.org/51/663751/3/gate/swift-tox-lower-constraints/7f26768/ : SUCCESS in 12m 43s\n- openstack-tox-pep8 http://logs.openstack.org/51/663751/3/gate/openstack-tox-pep8/8a87f51/ : SUCCESS in 5m 43s","accounts_in_message":[],"_revision_number":3},{"id":"04b870573da4761d7f017311937e417aac70d4e2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 09:54:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- swift-promote-image http://logs.openstack.org/51/663751/3/promote/swift-promote-image/16d06c2/ : SUCCESS in 37s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"dca658103a63d212bdf9195fcde6038557c13401","revisions":{"8cf20617a4997684d9c46818cfba78807f68ba77":{"kind":"REWORK","_number":1,"created":"2019-06-06 19:26:07.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/51/663751/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/51/663751/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/663751/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/663751/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/663751/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/51/663751/1"}}},"commit":{"parents":[{"commit":"05bfaf7ae1667b7a3f59543d17280715a5b33969","subject":"Merge \"py3: port ssync\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/05bfaf7ae1667b7a3f59543d17280715a5b33969"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2019-06-06 19:25:22.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2019-06-06 19:25:40.000000000","tz":-300},"subject":"Fix swift with python \u003c2.7.9","message":"Fix swift with python \u003c2.7.9\n\nCloses-Bug: #1831932\n\nChange-Id: I0d33864f4bffa401082548ee9a52f6eb50cb1f39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8cf20617a4997684d9c46818cfba78807f68ba77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8cf20617a4997684d9c46818cfba78807f68ba77"}]},"branch":"refs/heads/master"},"5dbf71eef215ee95fef6c076e6c00409963d2980":{"kind":"REWORK","_number":2,"created":"2019-06-06 19:52:40.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/51/663751/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/51/663751/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/663751/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/663751/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/663751/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/51/663751/2"}}},"commit":{"parents":[{"commit":"05bfaf7ae1667b7a3f59543d17280715a5b33969","subject":"Merge \"py3: port ssync\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/05bfaf7ae1667b7a3f59543d17280715a5b33969"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2019-06-06 19:25:22.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2019-06-06 19:52:34.000000000","tz":-300},"subject":"Fix swift with python \u003c2.7.9","message":"Fix swift with python \u003c2.7.9\n\nCloses-Bug: #1831932\n\nChange-Id: I0d33864f4bffa401082548ee9a52f6eb50cb1f39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5dbf71eef215ee95fef6c076e6c00409963d2980"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5dbf71eef215ee95fef6c076e6c00409963d2980"}]},"branch":"refs/heads/master"},"dca658103a63d212bdf9195fcde6038557c13401":{"kind":"REWORK","_number":3,"created":"2019-06-07 15:39:09.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/51/663751/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/51/663751/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/663751/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/663751/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/663751/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/51/663751/3"}}},"commit":{"parents":[{"commit":"05bfaf7ae1667b7a3f59543d17280715a5b33969","subject":"Merge \"py3: port ssync\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/05bfaf7ae1667b7a3f59543d17280715a5b33969"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2019-06-06 19:25:22.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2019-06-07 15:39:00.000000000","tz":-300},"subject":"Fix swift with python \u003c2.7.9","message":"Fix swift with python \u003c2.7.9\n\nCloses-Bug: #1831932\n\nChange-Id: I0d33864f4bffa401082548ee9a52f6eb50cb1f39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dca658103a63d212bdf9195fcde6038557c13401"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dca658103a63d212bdf9195fcde6038557c13401"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
