)]}'
{"id":"openstack%2Fswift~663794","triplet_id":"openstack%2Fswift~master~Ic2b7b870ec3d45c2f1e25ccd38d521d56ede2c18","project":"openstack/swift","branch":"master","topic":"py2-tests-against-py3-servers","hashtags":[],"change_id":"Ic2b7b870ec3d45c2f1e25ccd38d521d56ede2c18","subject":"py3: get func tests passing","status":"ABANDONED","created":"2019-06-06 23:01:18.000000000","updated":"2019-06-21 17:19:57.000000000","total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"21f0abec8b8a1f06ad629995821be0d25a01fa11","_number":663794,"virtual_id_number":663794,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-06-07 03:40:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-06-21 17:19:39.000000000","_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-07 03:40:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e4442a29e407e99f1353ba9bbd4f40550c6d3da6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-06 23:01:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a17462764bd2e6f9abae03c1b5d5bca23fe0de04","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 03:40:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/94/663794/1/check/openstack-tox-docs/e5de11c/html/ : SUCCESS in 3m 44s\n- swift-build-image swift-build-image : FAILURE (non-voting)\n- swift-tox-py27 http://logs.openstack.org/94/663794/1/check/swift-tox-py27/3832bdb/ : FAILURE in 12m 40s\n- swift-tox-py36 http://logs.openstack.org/94/663794/1/check/swift-tox-py36/807016b/ : FAILURE in 12m 39s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/94/663794/1/check/swift-tox-py37/358e668/ : FAILURE in 12m 49s\n- swift-tox-func http://logs.openstack.org/94/663794/1/check/swift-tox-func/bb8cd7d/ : SUCCESS in 9m 03s\n- swift-tox-func-py37 http://logs.openstack.org/94/663794/1/check/swift-tox-func-py37/3e9b28f/ : SUCCESS in 9m 15s\n- swift-tox-func-encryption http://logs.openstack.org/94/663794/1/check/swift-tox-func-encryption/f11854e/ : SUCCESS in 11m 13s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/94/663794/1/check/swift-tox-func-domain-remap-staticweb/490459a/ : FAILURE in 14m 59s\n- swift-tox-func-ec http://logs.openstack.org/94/663794/1/check/swift-tox-func-ec/c0c1182/ : SUCCESS in 9m 37s\n- swift-tox-func-s3api http://logs.openstack.org/94/663794/1/check/swift-tox-func-s3api/a3dba98/ : TIMED_OUT in 40m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/94/663794/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/c0110a1/ : SUCCESS in 13m 56s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/94/663794/1/check/swift-probetests-centos-7/340d27e/ : FAILURE in 27m 24s\n- swift-dsvm-functional http://logs.openstack.org/94/663794/1/check/swift-dsvm-functional/5e79688/ : TIMED_OUT in 47m 30s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/94/663794/1/check/swift-dsvm-functional-ipv6/d6b579e/ : SUCCESS in 26m 51s\n- swift-tox-lower-constraints http://logs.openstack.org/94/663794/1/check/swift-tox-lower-constraints/f0a1456/ : FAILURE in 12m 05s\n- openstack-tox-pep8 http://logs.openstack.org/94/663794/1/check/openstack-tox-pep8/def3ed5/ : FAILURE in 4m 59s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/94/663794/1/check/swift-multinode-rolling-upgrade/841cf4d/ : FAILURE in 8m 58s (non-voting)\n\nWarning:\n  Requirements [\u0027docker-image\u0027] not met by build 4c03b612a7e04899a12513651ee16c12","accounts_in_message":[],"_revision_number":1},{"id":"62bc56e80871faa47dafb9b598bbb860ba9d8f83","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-21 17:19:39.000000000","message":"Patch Set 1:\n\n(10 comments)","accounts_in_message":[],"_revision_number":1},{"id":"49c55cf8b555a614bda160dc862a50a4d92b3f81","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-21 17:19:57.000000000","message":"Abandoned\n\nBroken up into separate patches that actually tell a story.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"265824c64f256a72669cb769c861b8c611855dbc","revisions":{"265824c64f256a72669cb769c861b8c611855dbc":{"kind":"REWORK","_number":1,"created":"2019-06-06 23:01:18.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/94/663794/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/94/663794/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/663794/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/663794/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/663794/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/94/663794/1"}}},"commit":{"parents":[{"commit":"3cd558989fc7f292b2341af2eb0faa6abf563e5f","subject":"py3: Stop truncating response headers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3cd558989fc7f292b2341af2eb0faa6abf563e5f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-06-01 17:46:54.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-06-06 23:00:58.000000000","tz":-420},"subject":"py3: get func tests passing","message":"py3: get func tests passing\n\n...when running services under py3 but tests under py2\n\nChange-Id: Ic2b7b870ec3d45c2f1e25ccd38d521d56ede2c18\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/265824c64f256a72669cb769c861b8c611855dbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/265824c64f256a72669cb769c861b8c611855dbc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
