)]}'
{"id":"openstack%2Fswift~664200","triplet_id":"openstack%2Fswift~master~I0c1f77425d9cefb2e6340ffb4c65cdea956c216f","project":"openstack/swift","branch":"master","topic":"update-mailmap","hashtags":[],"change_id":"I0c1f77425d9cefb2e6340ffb4c65cdea956c216f","subject":"Update mailmap for Kota Tsuyuzaki","status":"MERGED","created":"2019-06-10 06:33:58.000000000","updated":"2019-06-14 03:42:27.000000000","submitted":"2019-06-14 03:41:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664200-1560483678054-2902cfdb","meta_rev_id":"952f7fbf38f66810f53d09afea212ef222af1a8d","_number":664200,"virtual_id_number":664200,"owner":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"value":2,"date":"2019-06-14 03:41:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-13 16:34:42.000000000","_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"all":[{"value":2,"date":"2019-06-11 04:29:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"all":[{"value":1,"date":"2019-06-11 04:29:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-13 16:34:42.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2019-06-14 03:41:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"648ecee10ea2d4ad45a5fb091973ffed81ab5bb9","author":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"date":"2019-06-10 06:33:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2e1393c25545c1fd27af7212e1589bfd85032a86","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 07:04:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/00/664200/1/check/openstack-tox-docs/60dbb38/html/ : SUCCESS in 5m 46s\n- swift-build-image http://logs.openstack.org/00/664200/1/check/swift-build-image/a3d61ff/ : SUCCESS in 7m 49s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/00/664200/1/check/swift-tox-py27/d9f46f3/ : SUCCESS in 12m 04s\n- swift-tox-py36 http://logs.openstack.org/00/664200/1/check/swift-tox-py36/818d497/ : SUCCESS in 15m 41s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/00/664200/1/check/swift-tox-py37/e6841f1/ : SUCCESS in 13m 47s\n- swift-tox-lower-constraints http://logs.openstack.org/00/664200/1/check/swift-tox-lower-constraints/de533cc/ : SUCCESS in 11m 44s\n- openstack-tox-pep8 http://logs.openstack.org/00/664200/1/check/openstack-tox-pep8/795e8cc/ : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":1},{"id":"1887a091252d2df60887d7a9109a6304fee7b612","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-10 15:14:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"df04f1fad50f5b5c1baf0d8e1ff85be9ad65628a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-10 15:14:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f3059cbf02d9a873828b9fc291dcfe5cf1e2e482","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 18:55:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/00/664200/2/check/openstack-tox-docs/196dff7/html/ : SUCCESS in 4m 25s\n- swift-build-image http://logs.openstack.org/00/664200/2/check/swift-build-image/19993f1/ : SUCCESS in 8m 50s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/00/664200/2/check/swift-tox-py27/9cecb4f/ : FAILURE in 12m 58s\n- swift-tox-py36 http://logs.openstack.org/00/664200/2/check/swift-tox-py36/ff2c1cf/ : SUCCESS in 14m 13s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/00/664200/2/check/swift-tox-py37/b86ea97/ : SUCCESS in 12m 26s\n- swift-tox-lower-constraints http://logs.openstack.org/00/664200/2/check/swift-tox-lower-constraints/98d8d81/ : SUCCESS in 11m 55s\n- openstack-tox-pep8 http://logs.openstack.org/00/664200/2/check/openstack-tox-pep8/6a02914/ : SUCCESS in 6m 04s","accounts_in_message":[],"_revision_number":2},{"id":"f899dea7fb0ab4c2dfb2deda2ad2c80f8bb6ce9b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-10 19:31:51.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"868f7a742f7b138ef9f7c35ad38afb589a629707","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 00:20:32.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/00/664200/2/check/openstack-tox-docs/152d3d8/html/ : SUCCESS in 3m 45s\n- swift-build-image http://logs.openstack.org/00/664200/2/check/swift-build-image/1d4475c/ : SUCCESS in 7m 39s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/00/664200/2/check/swift-tox-py27/a869b36/ : SUCCESS in 15m 51s\n- swift-tox-py36 http://logs.openstack.org/00/664200/2/check/swift-tox-py36/2ce5013/ : SUCCESS in 10m 48s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/00/664200/2/check/swift-tox-py37/9e8668d/ : SUCCESS in 11m 39s\n- swift-tox-lower-constraints http://logs.openstack.org/00/664200/2/check/swift-tox-lower-constraints/eea3c00/ : FAILURE in 5m 14s\n- openstack-tox-pep8 http://logs.openstack.org/00/664200/2/check/openstack-tox-pep8/1fcf212/ : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":2},{"id":"dbc69f2ad52f50a3de2164422873b4a2592e7106","author":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"date":"2019-06-11 04:27:34.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5560c0fb198697ed4f2d8e0dda6860a92f16d091","author":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"date":"2019-06-11 04:28:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0c60394ccd1b64b6fec63a10a190f3e29f4f46d0","author":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"date":"2019-06-11 04:29:09.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"2a4120c52fa11d81715ebfd63f5f5a0f1038c6cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 08:57:44.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/00/664200/3/check/openstack-tox-docs/2011fb9/html/ : SUCCESS in 4m 08s\n- swift-build-image http://logs.openstack.org/00/664200/3/check/swift-build-image/3c01d5e/ : SUCCESS in 11m 19s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/00/664200/3/check/swift-tox-py27/5067ed8/ : SUCCESS in 21m 47s\n- swift-tox-py36 http://logs.openstack.org/00/664200/3/check/swift-tox-py36/3fc61e4/ : SUCCESS in 10m 45s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/00/664200/3/check/swift-tox-py37/89e0332/ : SUCCESS in 10m 55s\n- swift-tox-lower-constraints http://logs.openstack.org/00/664200/3/check/swift-tox-lower-constraints/55568cc/ : FAILURE in 4m 45s\n- openstack-tox-pep8 http://logs.openstack.org/00/664200/3/check/openstack-tox-pep8/cd19fcf/ : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":3},{"id":"83ac674e93752c4baf0a01e7e46e0c8b847a5f6f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-13 16:34:42.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"b9b752c6d50e9a8de2c9847030022d2b9a6d8567","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 03:22:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/00/664200/3/check/openstack-tox-docs/8aac775/html/ : SUCCESS in 3m 49s\n- swift-build-image http://logs.openstack.org/00/664200/3/check/swift-build-image/160294b/ : SUCCESS in 9m 39s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/00/664200/3/check/swift-tox-py27/91616a7/ : SUCCESS in 11m 38s\n- swift-tox-py36 http://logs.openstack.org/00/664200/3/check/swift-tox-py36/def33ef/ : FAILURE in 13m 45s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/00/664200/3/check/swift-tox-py37/739c970/ : SUCCESS in 10m 16s\n- swift-tox-lower-constraints http://logs.openstack.org/00/664200/3/check/swift-tox-lower-constraints/bb52955/ : SUCCESS in 12m 06s\n- openstack-tox-pep8 http://logs.openstack.org/00/664200/3/check/openstack-tox-pep8/4d8ce6e/ : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":3},{"id":"7b75db6e05f0c16e3070313326576858a71996ab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 03:22:33.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5c694be965d92325fddc9f3396c7023d421db863","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 03:41:17.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/00/664200/3/gate/openstack-tox-docs/88e741c/html/ : SUCCESS in 3m 54s\n- swift-upload-image http://logs.openstack.org/00/664200/3/gate/swift-upload-image/61ef0b1/ : SUCCESS in 8m 32s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/00/664200/3/gate/swift-tox-py27/0793e0b/ : SUCCESS in 10m 39s\n- swift-tox-py37 http://logs.openstack.org/00/664200/3/gate/swift-tox-py37/2c5c664/ : SUCCESS in 13m 18s\n- swift-tox-func http://logs.openstack.org/00/664200/3/gate/swift-tox-func/aeacf96/ : SUCCESS in 7m 53s\n- swift-tox-func-py37 http://logs.openstack.org/00/664200/3/gate/swift-tox-func-py37/49a1449/ : SUCCESS in 9m 12s\n- swift-tox-func-encryption http://logs.openstack.org/00/664200/3/gate/swift-tox-func-encryption/6f05d77/ : SUCCESS in 8m 42s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/00/664200/3/gate/swift-tox-func-domain-remap-staticweb/58a73c3/ : SUCCESS in 8m 58s\n- swift-tox-func-ec http://logs.openstack.org/00/664200/3/gate/swift-tox-func-ec/e9c3dca/ : SUCCESS in 9m 28s\n- swift-tox-func-s3api http://logs.openstack.org/00/664200/3/gate/swift-tox-func-s3api/f1cee4d/ : SUCCESS in 12m 39s\n- swift-tox-lower-constraints http://logs.openstack.org/00/664200/3/gate/swift-tox-lower-constraints/982f50d/ : SUCCESS in 11m 33s\n- openstack-tox-pep8 http://logs.openstack.org/00/664200/3/gate/openstack-tox-pep8/f01b648/ : SUCCESS in 5m 47s","accounts_in_message":[],"_revision_number":3},{"id":"54fb1924ee591c101ed139b2a891169bdb5f3b35","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 03:41:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"ca1efef9f36ab3d1ec0b8061a47d1a0b24b8f878","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 03:42:27.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- swift-promote-image http://logs.openstack.org/00/664200/3/promote/swift-promote-image/084a927/ : SUCCESS in 46s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"668e45af6dbb548682c46aaf9d4c515b22bc3ea7","revisions":{"1448febb66ce5cc003c413a77febdab668cb5ec7":{"kind":"REWORK","_number":1,"created":"2019-06-10 06:33:58.000000000","uploader":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"ref":"refs/changes/00/664200/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/664200/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/664200/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/664200/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/664200/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/664200/1"}}},"commit":{"parents":[{"commit":"dca658103a63d212bdf9195fcde6038557c13401","subject":"Fix swift with python \u003c2.7.9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dca658103a63d212bdf9195fcde6038557c13401"}]}],"author":{"name":"Kota Tsuyuzaki","email":"kota.tsuyuzaki.pc@hco.ntt.co.jp","date":"2019-06-10 06:27:49.000000000","tz":540},"committer":{"name":"Kota Tsuyuzaki","email":"kota.tsuyuzaki.pc@hco.ntt.co.jp","date":"2019-06-10 06:27:49.000000000","tz":540},"subject":"Update mailmap for Kota Tsuyuzaki","message":"Update mailmap for Kota Tsuyuzaki\n\nHis company changes the e-mail domain and the old one has been\nrecieve-only so to update the author\u0027s e-mail address, add the\nentry to the mailmap.\n\nChange-Id: I0c1f77425d9cefb2e6340ffb4c65cdea956c216f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1448febb66ce5cc003c413a77febdab668cb5ec7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1448febb66ce5cc003c413a77febdab668cb5ec7"}]},"branch":"refs/heads/master"},"efcf21cf53a8f4486069df6ff51337eb0e05bf70":{"kind":"REWORK","_number":2,"created":"2019-06-10 15:14:13.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/00/664200/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/664200/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/664200/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/664200/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/664200/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/664200/2"}}},"commit":{"parents":[{"commit":"dca658103a63d212bdf9195fcde6038557c13401","subject":"Fix swift with python \u003c2.7.9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dca658103a63d212bdf9195fcde6038557c13401"}]}],"author":{"name":"Kota Tsuyuzaki","email":"kota.tsuyuzaki.pc@hco.ntt.co.jp","date":"2019-06-10 06:27:49.000000000","tz":540},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-06-10 15:13:54.000000000","tz":-420},"subject":"Update mailmap for Kota Tsuyuzaki","message":"Update mailmap for Kota Tsuyuzaki\n\nHis company changes the e-mail domain and the old one has been\nrecieve-only so to update the author\u0027s e-mail address, add the\nentry to the mailmap.\n\nChange-Id: I0c1f77425d9cefb2e6340ffb4c65cdea956c216f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/efcf21cf53a8f4486069df6ff51337eb0e05bf70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/efcf21cf53a8f4486069df6ff51337eb0e05bf70"}]},"branch":"refs/heads/master"},"668e45af6dbb548682c46aaf9d4c515b22bc3ea7":{"kind":"REWORK","_number":3,"created":"2019-06-11 04:28:52.000000000","uploader":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"ref":"refs/changes/00/664200/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/664200/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/664200/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/664200/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/664200/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/664200/3"}}},"commit":{"parents":[{"commit":"dca658103a63d212bdf9195fcde6038557c13401","subject":"Fix swift with python \u003c2.7.9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dca658103a63d212bdf9195fcde6038557c13401"}]}],"author":{"name":"Kota Tsuyuzaki","email":"kota.tsuyuzaki.pc@hco.ntt.co.jp","date":"2019-06-10 06:27:49.000000000","tz":540},"committer":{"name":"Kota Tsuyuzaki","email":"kota.tsuyuzaki.pc@hco.ntt.co.jp","date":"2019-06-11 04:28:32.000000000","tz":540},"subject":"Update mailmap for Kota Tsuyuzaki","message":"Update mailmap for Kota Tsuyuzaki\n\nHis company changes the e-mail domain and the old one has been\nrecieve-only so to update the author\u0027s e-mail address, add the\nentry to the mailmap.\n\nChange-Id: I0c1f77425d9cefb2e6340ffb4c65cdea956c216f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/668e45af6dbb548682c46aaf9d4c515b22bc3ea7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/668e45af6dbb548682c46aaf9d4c515b22bc3ea7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
