)]}'
{"id":"openstack%2Fswift~664710","triplet_id":"openstack%2Fswift~master~I29bbea48ae38cfabf449a9f4cca1f5f27769405a","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I29bbea48ae38cfabf449a9f4cca1f5f27769405a","subject":"Ensure get_*_info keys are native strings","status":"MERGED","created":"2019-06-11 21:51:51.000000000","updated":"2019-06-13 06:42:56.000000000","submitted":"2019-06-13 06:41:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664710-1560408109986-b3e15709","meta_rev_id":"90c5bb73f2a0c365551f46306445d2aec422cccb","_number":664710,"virtual_id_number":664710,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":2,"date":"2019-06-13 06:41:49.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-12 05:38:13.000000000","_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"date":"2019-06-12 12:50:45.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":2,"date":"2019-06-11 22:59:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":1,"date":"2019-06-11 22:59:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-11 22:59:52.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2019-06-12 12:50:45.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2019-06-13 06:41:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0ef14ff0296675aa752d291d6e14cf30f471b73d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-11 21:51:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"194d1495f9e0a5489c313ff8644a867f2da7102a","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-06-11 22:59:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nI guess this at least keeps the type consistent across keys and values in python2 which is probably better than it was.\n\nit\u0027s definitely correct in that it works and makes external middleware from having to do this kind of translation for itself","accounts_in_message":[],"_revision_number":1},{"id":"ffeb4f03755cba6baa280400546ca08bb86a1a11","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-06-12 03:36:02.000000000","message":"Patch Set 1: Code-Review-1\n\nis this obsoleted by https://review.opendev.org/662115\n?","accounts_in_message":[],"_revision_number":1},{"id":"4fa6357f7f0fd388ceb272446cbebaea58f9a61e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-12 05:38:13.000000000","message":"Patch Set 1:\n\nNope -- https://review.opendev.org/#/c/662115/ covers how we get data *into* memcache; this covers how we get it *out*. I maybe should\u0027ve combined them, but we already had tempurl lined up on the other one...\n\nBleh. Gate\u0027s blocked on https://review.opendev.org/#/c/664459/ regardless, and I was careless enough to let these conflict...","accounts_in_message":[],"_revision_number":1},{"id":"1411655f3da5c49e1826fb8310bbae684ac819a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 08:28:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/10/664710/1/check/openstack-tox-docs/933eaf0/html/ : SUCCESS in 4m 13s\n- swift-build-image http://logs.openstack.org/10/664710/1/check/swift-build-image/06ffea1/ : SUCCESS in 8m 20s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/10/664710/1/check/swift-tox-py27/28296a5/ : SUCCESS in 10m 46s\n- swift-tox-py36 http://logs.openstack.org/10/664710/1/check/swift-tox-py36/300da1c/ : SUCCESS in 10m 37s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/10/664710/1/check/swift-tox-py37/dc5beea/ : SUCCESS in 10m 38s\n- swift-tox-func http://logs.openstack.org/10/664710/1/check/swift-tox-func/bb5a018/ : SUCCESS in 12m 08s\n- swift-tox-func-py37 http://logs.openstack.org/10/664710/1/check/swift-tox-func-py37/589697f/ : SUCCESS in 9m 12s\n- swift-tox-func-encryption http://logs.openstack.org/10/664710/1/check/swift-tox-func-encryption/e7aca41/ : SUCCESS in 9m 16s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/10/664710/1/check/swift-tox-func-domain-remap-staticweb/0d6411c/ : SUCCESS in 8m 51s\n- swift-tox-func-ec http://logs.openstack.org/10/664710/1/check/swift-tox-func-ec/51153bd/ : SUCCESS in 9m 14s\n- swift-tox-func-s3api http://logs.openstack.org/10/664710/1/check/swift-tox-func-s3api/5730772/ : SUCCESS in 14m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/10/664710/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/5435c3c/ : SUCCESS in 14m 34s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/10/664710/1/check/swift-probetests-centos-7/2bcc114/ : SUCCESS in 39m 21s\n- swift-dsvm-functional http://logs.openstack.org/10/664710/1/check/swift-dsvm-functional/512bd32/ : SUCCESS in 26m 30s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/10/664710/1/check/swift-dsvm-functional-ipv6/c3c92c9/ : SUCCESS in 27m 28s\n- swift-tox-lower-constraints http://logs.openstack.org/10/664710/1/check/swift-tox-lower-constraints/47f8bfb/ : SUCCESS in 11m 42s\n- openstack-tox-pep8 http://logs.openstack.org/10/664710/1/check/openstack-tox-pep8/543a234/ : SUCCESS in 5m 25s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/10/664710/1/check/swift-multinode-rolling-upgrade/e9d140e/ : FAILURE in 8m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f525384f47be01732f579ea7063b00f03b496b06","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-06-12 12:50:45.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"fb5c03076b001b3081addd6e024129880bae93e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 03:01:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7ad9928a578c74892d65b15985ce67fa47f10d8a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 06:41:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f257e5a233938eb7dd733c8d632831976be2964a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 06:41:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/10/664710/1/gate/openstack-tox-docs/3ce3593/html/ : SUCCESS in 4m 22s\n- swift-upload-image http://logs.openstack.org/10/664710/1/gate/swift-upload-image/24c4c30/ : SUCCESS in 9m 49s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/10/664710/1/gate/swift-tox-py27/11f7dd5/ : SUCCESS in 11m 25s\n- swift-tox-py37 http://logs.openstack.org/10/664710/1/gate/swift-tox-py37/5beb994/ : SUCCESS in 12m 44s\n- swift-tox-func http://logs.openstack.org/10/664710/1/gate/swift-tox-func/f9e23b8/ : SUCCESS in 9m 06s\n- swift-tox-func-py37 http://logs.openstack.org/10/664710/1/gate/swift-tox-func-py37/4aca7bc/ : SUCCESS in 9m 30s\n- swift-tox-func-encryption http://logs.openstack.org/10/664710/1/gate/swift-tox-func-encryption/428b6ef/ : SUCCESS in 9m 25s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/10/664710/1/gate/swift-tox-func-domain-remap-staticweb/4da61db/ : SUCCESS in 10m 57s\n- swift-tox-func-ec http://logs.openstack.org/10/664710/1/gate/swift-tox-func-ec/faf1452/ : SUCCESS in 9m 58s\n- swift-tox-func-s3api http://logs.openstack.org/10/664710/1/gate/swift-tox-func-s3api/577a0c9/ : SUCCESS in 14m 36s\n- swift-probetests-centos-7 http://logs.openstack.org/10/664710/1/gate/swift-probetests-centos-7/c5871f0/ : SUCCESS in 34m 48s\n- swift-dsvm-functional http://logs.openstack.org/10/664710/1/gate/swift-dsvm-functional/cc85a89/ : SUCCESS in 33m 20s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/10/664710/1/gate/swift-dsvm-functional-ipv6/c9b355a/ : SUCCESS in 27m 49s\n- swift-tox-lower-constraints http://logs.openstack.org/10/664710/1/gate/swift-tox-lower-constraints/65b643a/ : SUCCESS in 12m 01s\n- openstack-tox-pep8 http://logs.openstack.org/10/664710/1/gate/openstack-tox-pep8/973f24c/ : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":1},{"id":"90c5bb73f2a0c365551f46306445d2aec422cccb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 06:42:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- swift-promote-image http://logs.openstack.org/10/664710/1/promote/swift-promote-image/279da9a/ : SUCCESS in 41s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"aa2f1db1b71c1b2bf746b72515e3efd15598b6aa","revisions":{"aa2f1db1b71c1b2bf746b72515e3efd15598b6aa":{"kind":"REWORK","_number":1,"created":"2019-06-11 21:51:51.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/10/664710/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/10/664710/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/664710/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/664710/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/664710/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/10/664710/1"}}},"commit":{"parents":[{"commit":"b7b92b97b12f2a5c0e1beed59b6ffd4791cec896","subject":"Bump up minimum cryptography version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b7b92b97b12f2a5c0e1beed59b6ffd4791cec896"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-06-11 21:50:49.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-06-11 21:50:49.000000000","tz":-420},"subject":"Ensure get_*_info keys are native strings","message":"Ensure get_*_info keys are native strings\n\nChange-Id: I29bbea48ae38cfabf449a9f4cca1f5f27769405a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aa2f1db1b71c1b2bf746b72515e3efd15598b6aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aa2f1db1b71c1b2bf746b72515e3efd15598b6aa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
