)]}'
{"id":"openstack%2Fswift~665494","triplet_id":"openstack%2Fswift~master~Ib7e065c47871876786bcc9ff39737f5d1bb3c12c","project":"openstack/swift","branch":"master","topic":"py3-devstack","hashtags":[],"change_id":"Ib7e065c47871876786bcc9ff39737f5d1bb3c12c","subject":"Move call to global_conf_callback after loadapp()","status":"MERGED","created":"2019-06-14 23:20:19.000000000","updated":"2019-06-29 01:30:55.000000000","submitted":"2019-06-29 01:29:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665494-1561771791565-a45a0009","meta_rev_id":"b23e128160d8444463256623cfc7ac9e6e756647","_number":665494,"virtual_id_number":665494,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":2,"date":"2019-06-29 01:29:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"recommended":{"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},"all":[{"value":1,"date":"2019-06-27 14:57:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},{"value":2,"date":"2019-06-28 23:02:07.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":0,"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},{"value":1,"date":"2019-06-28 23:02:07.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-27 14:57:54.000000000","updated_by":{"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},"reviewer":{"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},"state":"REVIEWER"},{"updated":"2019-06-28 23:02:07.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2019-06-29 01:29:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"acdb3ac585f706bbfb52b67ec567587a7f5d9a62","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-14 23:20:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a47a32ce1b3630898b3320a60416e4322610da2d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-15 07:30:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/94/665494/1/check/openstack-tox-docs/bf5fc0c/html/ : SUCCESS in 4m 28s\n- swift-build-image http://logs.openstack.org/94/665494/1/check/swift-build-image/dc6943d/ : SUCCESS in 8m 26s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/94/665494/1/check/swift-tox-py27/f168e5b/ : FAILURE in 12m 31s\n- swift-tox-py36 http://logs.openstack.org/94/665494/1/check/swift-tox-py36/68cb75d/ : FAILURE in 11m 50s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/94/665494/1/check/swift-tox-py37/e8de160/ : FAILURE in 11m 03s\n- swift-tox-func http://logs.openstack.org/94/665494/1/check/swift-tox-func/f7ad45a/ : FAILURE in 13m 29s\n- swift-tox-func-py37 http://logs.openstack.org/94/665494/1/check/swift-tox-func-py37/ea64552/ : SUCCESS in 10m 04s\n- swift-tox-func-encryption http://logs.openstack.org/94/665494/1/check/swift-tox-func-encryption/ddebebf/ : SUCCESS in 9m 33s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/94/665494/1/check/swift-tox-func-domain-remap-staticweb/67abbdd/ : SUCCESS in 9m 34s\n- swift-tox-func-ec http://logs.openstack.org/94/665494/1/check/swift-tox-func-ec/179a00f/ : SUCCESS in 8m 55s\n- swift-tox-func-s3api http://logs.openstack.org/94/665494/1/check/swift-tox-func-s3api/660ec6d/ : SUCCESS in 13m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/94/665494/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/67ec9cb/ : SUCCESS in 13m 35s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/94/665494/1/check/swift-probetests-centos-7/c262d95/ : SUCCESS in 30m 05s\n- swift-dsvm-functional http://logs.openstack.org/94/665494/1/check/swift-dsvm-functional/5b74403/ : SUCCESS in 26m 36s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/94/665494/1/check/swift-dsvm-functional-ipv6/2a3d350/ : SUCCESS in 26m 20s\n- swift-tox-lower-constraints http://logs.openstack.org/94/665494/1/check/swift-tox-lower-constraints/44d1504/ : FAILURE in 11m 58s\n- openstack-tox-pep8 http://logs.openstack.org/94/665494/1/check/openstack-tox-pep8/796ad72/ : SUCCESS in 5m 22s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/94/665494/1/check/swift-multinode-rolling-upgrade/11047f9/ : FAILURE in 7m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"225c308626492cdb8175d2748697703f1f4331dd","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-17 20:07:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9c748e1c6b7b78b05b4fa2241b09c166d093c8b8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-17 20:32:46.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3452654fbf6fe472e581210b26b15aa27b7d26b2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 00:59:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/94/665494/3/check/openstack-tox-docs/8a9619c/html/ : SUCCESS in 4m 37s\n- swift-build-image http://logs.openstack.org/94/665494/3/check/swift-build-image/9932a8a/ : SUCCESS in 7m 57s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/94/665494/3/check/swift-tox-py27/9ebd92e/ : SUCCESS in 13m 57s\n- swift-tox-py36 http://logs.openstack.org/94/665494/3/check/swift-tox-py36/7130db4/ : SUCCESS in 11m 15s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/94/665494/3/check/swift-tox-py37/40acc7f/ : SUCCESS in 10m 33s\n- swift-tox-func http://logs.openstack.org/94/665494/3/check/swift-tox-func/dd27bef/ : SUCCESS in 8m 16s\n- swift-tox-func-py37 http://logs.openstack.org/94/665494/3/check/swift-tox-func-py37/9964f92/ : SUCCESS in 9m 05s\n- swift-tox-func-encryption http://logs.openstack.org/94/665494/3/check/swift-tox-func-encryption/1ed89e0/ : SUCCESS in 9m 38s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/94/665494/3/check/swift-tox-func-domain-remap-staticweb/fab8e53/ : SUCCESS in 8m 53s\n- swift-tox-func-ec http://logs.openstack.org/94/665494/3/check/swift-tox-func-ec/95d38fb/ : SUCCESS in 10m 00s\n- swift-tox-func-s3api http://logs.openstack.org/94/665494/3/check/swift-tox-func-s3api/7209a99/ : SUCCESS in 11m 25s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/94/665494/3/check/swift-tox-func-s3api-ceph-s3tests-tempauth/19c12ab/ : SUCCESS in 15m 00s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/94/665494/3/check/swift-probetests-centos-7/1059d64/ : SUCCESS in 31m 53s\n- swift-dsvm-functional http://logs.openstack.org/94/665494/3/check/swift-dsvm-functional/b38182d/ : SUCCESS in 32m 24s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/94/665494/3/check/swift-dsvm-functional-ipv6/b15b094/ : SUCCESS in 25m 34s\n- swift-tox-lower-constraints http://logs.openstack.org/94/665494/3/check/swift-tox-lower-constraints/5eeec25/ : SUCCESS in 11m 39s\n- openstack-tox-pep8 http://logs.openstack.org/94/665494/3/check/openstack-tox-pep8/5a08ece/ : SUCCESS in 5m 54s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/94/665494/3/check/swift-multinode-rolling-upgrade/dd5d678/ : FAILURE in 10m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"29823fefc43634cb528d31a43edd152bff2525a0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-22 05:38:43.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"33d61cd934565af5e16158d820313ed78b45b9fb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-22 06:15:08.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/94/665494/4/check/openstack-tox-docs/06240b0/html/ : SUCCESS in 3m 32s\n- swift-build-image http://logs.openstack.org/94/665494/4/check/swift-build-image/5595359/ : SUCCESS in 6m 56s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/94/665494/4/check/swift-tox-py27/981a19a/ : SUCCESS in 10m 55s\n- swift-tox-py36 http://logs.openstack.org/94/665494/4/check/swift-tox-py36/6a9d9cb/ : SUCCESS in 9m 54s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/94/665494/4/check/swift-tox-py37/4071766/ : SUCCESS in 11m 12s\n- swift-tox-func http://logs.openstack.org/94/665494/4/check/swift-tox-func/6a8b495/ : SUCCESS in 8m 30s\n- swift-tox-func-py37 http://logs.openstack.org/94/665494/4/check/swift-tox-func-py37/0dc6a1e/ : SUCCESS in 16m 11s\n- swift-tox-func-encryption http://logs.openstack.org/94/665494/4/check/swift-tox-func-encryption/e779296/ : SUCCESS in 17m 45s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/94/665494/4/check/swift-tox-func-domain-remap-staticweb/34e68f7/ : SUCCESS in 14m 33s\n- swift-tox-func-ec http://logs.openstack.org/94/665494/4/check/swift-tox-func-ec/c84b7ed/ : SUCCESS in 17m 54s\n- swift-tox-func-s3api http://logs.openstack.org/94/665494/4/check/swift-tox-func-s3api/9440a96/ : SUCCESS in 14m 46s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/94/665494/4/check/swift-tox-func-s3api-ceph-s3tests-tempauth/2af37a7/ : SUCCESS in 13m 01s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/94/665494/4/check/swift-probetests-centos-7/e7cfb0e/ : SUCCESS in 31m 08s\n- swift-dsvm-functional http://logs.openstack.org/94/665494/4/check/swift-dsvm-functional/4a89aed/ : SUCCESS in 28m 59s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/94/665494/4/check/swift-dsvm-functional-ipv6/166d99c/ : SUCCESS in 28m 59s\n- swift-tox-lower-constraints http://logs.openstack.org/94/665494/4/check/swift-tox-lower-constraints/b5a8ccb/ : SUCCESS in 10m 43s\n- openstack-tox-pep8 http://logs.openstack.org/94/665494/4/check/openstack-tox-pep8/25c43f4/ : SUCCESS in 5m 14s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/94/665494/4/check/swift-multinode-rolling-upgrade/cdbde30/ : FAILURE in 8m 06s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"985e20945141fe19312f185f8618128fa9579dba","author":{"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},"date":"2019-06-27 14:57:54.000000000","message":"Patch Set 4: Code-Review+1\n\nLooks good, and I\u0027ve checked it does work in SAIO while it would previously fail to start if replication_concurrency was \u003e 0","accounts_in_message":[],"_revision_number":4},{"id":"4b8d34f6a57845789f3948ad88c8e2fbcbaf5e08","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-06-28 23:02:07.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"6b9a8a8f466b9a173bc0e43fa2f1781f5afe7b77","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 23:03:13.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"06aeca2e25b3263c5960008af414196c931ffe04","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-29 01:29:51.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/94/665494/4/gate/openstack-tox-docs/0153c84/html/ : SUCCESS in 4m 06s\n- swift-upload-image http://logs.openstack.org/94/665494/4/gate/swift-upload-image/a080b9a/ : SUCCESS in 8m 04s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/94/665494/4/gate/swift-tox-py27/720910f/ : SUCCESS in 10m 29s\n- swift-tox-py37 http://logs.openstack.org/94/665494/4/gate/swift-tox-py37/9ce6ee4/ : SUCCESS in 10m 10s\n- swift-tox-func http://logs.openstack.org/94/665494/4/gate/swift-tox-func/94cfbf2/ : SUCCESS in 9m 28s\n- swift-tox-func-py37 http://logs.openstack.org/94/665494/4/gate/swift-tox-func-py37/cd6d6f5/ : SUCCESS in 11m 50s\n- swift-tox-func-encryption http://logs.openstack.org/94/665494/4/gate/swift-tox-func-encryption/8cb7d4f/ : SUCCESS in 8m 50s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/94/665494/4/gate/swift-tox-func-domain-remap-staticweb/eeb06fe/ : SUCCESS in 9m 29s\n- swift-tox-func-ec http://logs.openstack.org/94/665494/4/gate/swift-tox-func-ec/525b755/ : SUCCESS in 9m 48s\n- swift-tox-func-s3api http://logs.openstack.org/94/665494/4/gate/swift-tox-func-s3api/ec6ac49/ : SUCCESS in 11m 29s\n- swift-probetests-centos-7 http://logs.openstack.org/94/665494/4/gate/swift-probetests-centos-7/5e4275b/ : SUCCESS in 29m 34s\n- swift-dsvm-functional http://logs.openstack.org/94/665494/4/gate/swift-dsvm-functional/5bc146f/ : SUCCESS in 30m 22s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/94/665494/4/gate/swift-dsvm-functional-ipv6/6162277/ : SUCCESS in 25m 29s\n- swift-tox-lower-constraints http://logs.openstack.org/94/665494/4/gate/swift-tox-lower-constraints/ae0c4aa/ : SUCCESS in 11m 29s\n- openstack-tox-pep8 http://logs.openstack.org/94/665494/4/gate/openstack-tox-pep8/c65dd8f/ : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":4},{"id":"7a5990be3f42dc659b61bb4c0ea5f87e975a73b9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-29 01:29:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"c070a16463892a1a635cfd2bf7c0294a2a4e0f25","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-29 01:30:55.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- swift-promote-image http://logs.openstack.org/94/665494/4/promote/swift-promote-image/b0ac6d0/ : SUCCESS in 43s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"ff4459ed6bcaa26bc4df9873577ccad9578a89fc","revisions":{"61325fc25c8080834322d1938d853cd48db417c1":{"kind":"REWORK","_number":1,"created":"2019-06-14 23:20:19.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/94/665494/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/94/665494/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/665494/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/665494/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/665494/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/94/665494/1"}}},"commit":{"parents":[{"commit":"74e1f2e05339880b724161241f3acbb6af6ea769","subject":"Merge \"Update mailmap for Kota Tsuyuzaki\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/74e1f2e05339880b724161241f3acbb6af6ea769"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-06-14 23:20:03.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-06-14 23:20:03.000000000","tz":-420},"subject":"Move call to global_conf_callback after loadapp()","message":"Move call to global_conf_callback after loadapp()\n\nOtherwise, paste complains about not being able to interpolate values\ninto the replication_semaphore. As long as it gets dropped in before we\nfork(), I think it\u0027s OK?\n\nChange-Id: Ib7e065c47871876786bcc9ff39737f5d1bb3c12c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/61325fc25c8080834322d1938d853cd48db417c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/61325fc25c8080834322d1938d853cd48db417c1"}]},"branch":"refs/heads/master"},"52b248c6a4d9d266ad6428a1b9cf377f52dff77a":{"kind":"REWORK","_number":2,"created":"2019-06-17 20:07:02.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/94/665494/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/94/665494/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/665494/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/665494/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/665494/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/94/665494/2"}}},"commit":{"parents":[{"commit":"74e1f2e05339880b724161241f3acbb6af6ea769","subject":"Merge \"Update mailmap for Kota Tsuyuzaki\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/74e1f2e05339880b724161241f3acbb6af6ea769"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-06-14 23:20:03.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-06-17 20:04:54.000000000","tz":-420},"subject":"Move call to global_conf_callback after loadapp()","message":"Move call to global_conf_callback after loadapp()\n\nOtherwise, paste complains about not being able to interpolate values\ninto the replication_semaphore. As long as it gets dropped in before we\nfork(), I think it\u0027s OK?\n\nChange-Id: Ib7e065c47871876786bcc9ff39737f5d1bb3c12c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/52b248c6a4d9d266ad6428a1b9cf377f52dff77a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/52b248c6a4d9d266ad6428a1b9cf377f52dff77a"}]},"branch":"refs/heads/master"},"db27ad875ee8647e0629c2a95ffb2805be89ce1c":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2019-06-17 20:32:46.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/94/665494/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/94/665494/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/665494/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/665494/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/665494/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/94/665494/3"}}},"commit":{"parents":[{"commit":"bc19dd1649115d9c83e2c939bcca990acfe7867c","subject":"Merge \"Installing liberasurecode from Alpine Linux repos instead of building from source.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bc19dd1649115d9c83e2c939bcca990acfe7867c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-06-14 23:20:03.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-06-17 20:31:53.000000000","tz":-420},"subject":"Move call to global_conf_callback after loadapp()","message":"Move call to global_conf_callback after loadapp()\n\nOtherwise, paste complains about not being able to interpolate values\ninto the replication_semaphore. As long as it gets dropped in before we\nfork(), I think it\u0027s OK?\n\nCloses-Bug: 1691075\nChange-Id: Ib7e065c47871876786bcc9ff39737f5d1bb3c12c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/db27ad875ee8647e0629c2a95ffb2805be89ce1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/db27ad875ee8647e0629c2a95ffb2805be89ce1c"}]},"branch":"refs/heads/master"},"ff4459ed6bcaa26bc4df9873577ccad9578a89fc":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-06-22 05:38:43.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/94/665494/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/94/665494/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/665494/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/665494/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/665494/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/94/665494/4"}}},"commit":{"parents":[{"commit":"9f1ef35630ebf9da53cbd061f89889e534803725","subject":"py3: Port the tempurl middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f1ef35630ebf9da53cbd061f89889e534803725"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-06-14 23:20:03.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-06-22 05:30:47.000000000","tz":-420},"subject":"Move call to global_conf_callback after loadapp()","message":"Move call to global_conf_callback after loadapp()\n\nOtherwise, paste complains about not being able to interpolate values\ninto the replication_semaphore. As long as it gets dropped in before we\nfork(), I think it\u0027s OK?\n\nCloses-Bug: 1691075\nChange-Id: Ib7e065c47871876786bcc9ff39737f5d1bb3c12c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff4459ed6bcaa26bc4df9873577ccad9578a89fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff4459ed6bcaa26bc4df9873577ccad9578a89fc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
