)]}'
{"id":"openstack%2Fswift~665758","triplet_id":"openstack%2Fswift~master~If933b713bdd1dfc348959771d60c0756a9a94477","project":"openstack/swift","branch":"master","topic":"py3-eventlet","hashtags":[],"change_id":"If933b713bdd1dfc348959771d60c0756a9a94477","subject":"Bump up our minimum eventlet version","status":"MERGED","created":"2019-06-17 20:22:40.000000000","updated":"2019-07-23 17:53:40.000000000","submitted":"2019-07-23 17:53:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665758-1563904420495-61b424a5","meta_rev_id":"d4f03c3b5131ee7eed0cd3b5dcbbb17353e9fc85","_number":665758,"virtual_id_number":665758,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-07-23 17:53:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-23 15:36:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-23 15:36:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-23 17:53:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4bd33df34166cb0c4d72ad43936efe6a7db7d857","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-17 20:22:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d7c9f4ec9174913fa92d58ed319c3e4621a7ef04","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 00:55:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/58/665758/1/check/openstack-tox-docs/5b8a09c/html/ : SUCCESS in 5m 40s\n- requirements-check http://logs.openstack.org/58/665758/1/check/requirements-check/22cc775/ : SUCCESS in 3m 13s\n- swift-build-image http://logs.openstack.org/58/665758/1/check/swift-build-image/9d07627/ : SUCCESS in 9m 00s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/58/665758/1/check/swift-tox-py27/e8f4b92/ : SUCCESS in 13m 24s\n- swift-tox-py36 http://logs.openstack.org/58/665758/1/check/swift-tox-py36/054f3d6/ : SUCCESS in 13m 58s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/58/665758/1/check/swift-tox-py37/038076f/ : SUCCESS in 15m 54s\n- swift-tox-func http://logs.openstack.org/58/665758/1/check/swift-tox-func/829cd00/ : SUCCESS in 11m 45s\n- swift-tox-func-py37 http://logs.openstack.org/58/665758/1/check/swift-tox-func-py37/acfb921/ : SUCCESS in 9m 29s\n- swift-tox-func-encryption http://logs.openstack.org/58/665758/1/check/swift-tox-func-encryption/433a04c/ : SUCCESS in 9m 05s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/58/665758/1/check/swift-tox-func-domain-remap-staticweb/4aff7d5/ : SUCCESS in 9m 39s\n- swift-tox-func-ec http://logs.openstack.org/58/665758/1/check/swift-tox-func-ec/f39bc86/ : SUCCESS in 9m 30s\n- swift-tox-func-s3api http://logs.openstack.org/58/665758/1/check/swift-tox-func-s3api/438d527/ : SUCCESS in 13m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/58/665758/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/7d55a92/ : SUCCESS in 18m 03s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/58/665758/1/check/swift-probetests-centos-7/608c688/ : SUCCESS in 32m 54s\n- swift-dsvm-functional http://logs.openstack.org/58/665758/1/check/swift-dsvm-functional/4e35f9c/ : SUCCESS in 29m 33s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/58/665758/1/check/swift-dsvm-functional-ipv6/cd289be/ : SUCCESS in 26m 48s\n- swift-tox-lower-constraints http://logs.openstack.org/58/665758/1/check/swift-tox-lower-constraints/c74d7f2/ : FAILURE in 18m 49s\n- openstack-tox-pep8 http://logs.openstack.org/58/665758/1/check/openstack-tox-pep8/05f587e/ : SUCCESS in 5m 44s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/58/665758/1/check/swift-multinode-rolling-upgrade/b846471/ : FAILURE in 9m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bcc5f32ab3b046dd2233ca61fde3b968d1e06047","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-18 15:37:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ffc2925ea68924a356c4c917b0bb2bd36b5d3a65","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 02:57:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/58/665758/2/check/openstack-tox-docs/3e8f802/html/ : SUCCESS in 4m 04s\n- requirements-check http://logs.openstack.org/58/665758/2/check/requirements-check/1bb2ae1/ : FAILURE in 2m 33s\n- swift-build-image http://logs.openstack.org/58/665758/2/check/swift-build-image/39d9b82/ : SUCCESS in 8m 02s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/58/665758/2/check/swift-tox-py27/49577e9/ : SUCCESS in 11m 23s\n- swift-tox-py36 http://logs.openstack.org/58/665758/2/check/swift-tox-py36/5859c94/ : SUCCESS in 12m 04s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/58/665758/2/check/swift-tox-py37/03b5ee4/ : SUCCESS in 20m 40s\n- swift-tox-func http://logs.openstack.org/58/665758/2/check/swift-tox-func/1be9d39/ : SUCCESS in 12m 01s\n- swift-tox-func-py37 http://logs.openstack.org/58/665758/2/check/swift-tox-func-py37/73774dc/ : SUCCESS in 9m 01s\n- swift-tox-func-encryption http://logs.openstack.org/58/665758/2/check/swift-tox-func-encryption/1d0b4e4/ : SUCCESS in 8m 49s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/58/665758/2/check/swift-tox-func-domain-remap-staticweb/cf16a83/ : SUCCESS in 8m 17s\n- swift-tox-func-ec http://logs.openstack.org/58/665758/2/check/swift-tox-func-ec/8e4138e/ : SUCCESS in 12m 41s\n- swift-tox-func-s3api http://logs.openstack.org/58/665758/2/check/swift-tox-func-s3api/e806276/ : SUCCESS in 25m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/58/665758/2/check/swift-tox-func-s3api-ceph-s3tests-tempauth/335b77a/ : SUCCESS in 15m 00s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/58/665758/2/check/swift-probetests-centos-7/7bdf2cb/ : SUCCESS in 32m 36s\n- swift-dsvm-functional http://logs.openstack.org/58/665758/2/check/swift-dsvm-functional/edebaa3/ : SUCCESS in 25m 37s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/58/665758/2/check/swift-dsvm-functional-ipv6/cc7f3d2/ : SUCCESS in 28m 39s\n- swift-tox-lower-constraints http://logs.openstack.org/58/665758/2/check/swift-tox-lower-constraints/8c5af03/ : FAILURE in 9m 17s\n- openstack-tox-pep8 http://logs.openstack.org/58/665758/2/check/openstack-tox-pep8/39164e4/ : SUCCESS in 5m 59s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/58/665758/2/check/swift-multinode-rolling-upgrade/1b3a8a6/ : FAILURE in 9m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"12c8017daa1e957549748db4c675098703de08c6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-20 22:14:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ce566080a067c3ec170ab6fc338828bd141ca603","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 02:47:28.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/58/665758/3/check/openstack-tox-docs/fb3f3ef/html/ : SUCCESS in 4m 41s\n- requirements-check http://logs.openstack.org/58/665758/3/check/requirements-check/e321bd8/ : FAILURE in 2m 50s\n- swift-build-image http://logs.openstack.org/58/665758/3/check/swift-build-image/771f5c8/ : SUCCESS in 9m 24s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/58/665758/3/check/swift-tox-py27/1570310/ : SUCCESS in 10m 52s\n- swift-tox-py36 http://logs.openstack.org/58/665758/3/check/swift-tox-py36/b73fad5/ : SUCCESS in 10m 42s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/58/665758/3/check/swift-tox-py37/5c4b738/ : SUCCESS in 10m 50s\n- swift-tox-func http://logs.openstack.org/58/665758/3/check/swift-tox-func/998243d/ : SUCCESS in 8m 44s\n- swift-tox-func-py37 http://logs.openstack.org/58/665758/3/check/swift-tox-func-py37/8e484b1/ : SUCCESS in 9m 12s\n- swift-tox-func-encryption http://logs.openstack.org/58/665758/3/check/swift-tox-func-encryption/b5c68f3/ : SUCCESS in 9m 25s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/58/665758/3/check/swift-tox-func-domain-remap-staticweb/c92b1f1/ : SUCCESS in 11m 35s\n- swift-tox-func-ec http://logs.openstack.org/58/665758/3/check/swift-tox-func-ec/8d8f98f/ : SUCCESS in 16m 16s\n- swift-tox-func-s3api http://logs.openstack.org/58/665758/3/check/swift-tox-func-s3api/b039538/ : SUCCESS in 17m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/58/665758/3/check/swift-tox-func-s3api-ceph-s3tests-tempauth/73625d2/ : SUCCESS in 15m 42s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/58/665758/3/check/swift-probetests-centos-7/34d5943/ : SUCCESS in 31m 23s\n- swift-dsvm-functional http://logs.openstack.org/58/665758/3/check/swift-dsvm-functional/796b427/ : SUCCESS in 28m 47s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/58/665758/3/check/swift-dsvm-functional-ipv6/def7591/ : SUCCESS in 28m 38s\n- swift-tox-lower-constraints http://logs.openstack.org/58/665758/3/check/swift-tox-lower-constraints/82b3943/ : SUCCESS in 18m 07s\n- openstack-tox-pep8 http://logs.openstack.org/58/665758/3/check/openstack-tox-pep8/5c52c24/ : SUCCESS in 7m 17s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/58/665758/3/check/swift-multinode-rolling-upgrade/2a2a7b2/ : FAILURE in 10m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"af783594de972a1a596887014f3ea490ac661b88","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-21 17:45:04.000000000","message":"Patch Set 3:\n\nBesides fixing up the requirements-check, this patch should also address the concern at https://review.opendev.org/#/c/660542/4/swift/obj/ssync_receiver.py@143","accounts_in_message":[],"_revision_number":3},{"id":"9b311d18d330a809743c51e7776c4f719e3d98e3","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-24 15:23:00.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4f3fe58ec976c2ef1b9729a754415a410dde891c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-24 15:45:31.000000000","message":"Patch Set 4: Workflow-1\n\nI think this should be able to pass now -- but it should wait until after a 2.22.0 release.","accounts_in_message":[],"_revision_number":4},{"id":"51e51a44b7773355b35036e05327ea59f1710132","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 16:47:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/58/665758/4/check/openstack-tox-docs/fcab449/html/ : SUCCESS in 5m 48s\n- requirements-check http://logs.openstack.org/58/665758/4/check/requirements-check/daac204/ : SUCCESS in 4m 15s\n- swift-build-image http://logs.openstack.org/58/665758/4/check/swift-build-image/db02542/ : SUCCESS in 7m 20s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/58/665758/4/check/swift-tox-py27/85a73a0/ : SUCCESS in 10m 00s\n- swift-tox-py36 http://logs.openstack.org/58/665758/4/check/swift-tox-py36/b265466/ : SUCCESS in 10m 40s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/58/665758/4/check/swift-tox-py37/44cbc8e/ : SUCCESS in 20m 14s\n- swift-tox-func http://logs.openstack.org/58/665758/4/check/swift-tox-func/f4d320c/ : SUCCESS in 9m 29s\n- swift-tox-func-py37 http://logs.openstack.org/58/665758/4/check/swift-tox-func-py37/b1abc74/ : SUCCESS in 9m 37s\n- swift-tox-func-encryption http://logs.openstack.org/58/665758/4/check/swift-tox-func-encryption/376dce9/ : SUCCESS in 9m 09s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/58/665758/4/check/swift-tox-func-domain-remap-staticweb/4c00689/ : SUCCESS in 19m 43s\n- swift-tox-func-ec http://logs.openstack.org/58/665758/4/check/swift-tox-func-ec/e428244/ : SUCCESS in 20m 12s\n- swift-tox-func-s3api http://logs.openstack.org/58/665758/4/check/swift-tox-func-s3api/11671a2/ : SUCCESS in 11m 04s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/58/665758/4/check/swift-tox-func-s3api-ceph-s3tests-tempauth/18fec99/ : SUCCESS in 24m 32s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/58/665758/4/check/swift-probetests-centos-7/2207feb/ : SUCCESS in 38m 10s\n- swift-dsvm-functional http://logs.openstack.org/58/665758/4/check/swift-dsvm-functional/71685c4/ : SUCCESS in 28m 29s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/58/665758/4/check/swift-dsvm-functional-ipv6/3d60ecc/ : SUCCESS in 27m 40s\n- swift-tox-lower-constraints http://logs.openstack.org/58/665758/4/check/swift-tox-lower-constraints/8708991/ : SUCCESS in 14m 22s\n- openstack-tox-pep8 http://logs.openstack.org/58/665758/4/check/openstack-tox-pep8/09bfae7/ : SUCCESS in 5m 48s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/58/665758/4/check/swift-multinode-rolling-upgrade/5aae79f/ : FAILURE in 9m 40s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b173e3b24e74ea339bd90f7262845bb1dd50380b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-07-23 15:36:08.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"53a4ad4b29af65838b2293f2b9884d215cc294aa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 15:36:43.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"bb409ca94de960cb2df42cbe8475ef20c3287ffb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 17:53:40.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/58/665758/4/gate/openstack-tox-docs/bd1e92c/html/ : SUCCESS in 6m 13s\n- requirements-check http://logs.openstack.org/58/665758/4/gate/requirements-check/c10e54b/ : SUCCESS in 5m 22s\n- swift-upload-image http://logs.openstack.org/58/665758/4/gate/swift-upload-image/8942b62/ : SUCCESS in 11m 02s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/58/665758/4/gate/swift-tox-py27/3c8ca50/ : SUCCESS in 12m 25s\n- swift-tox-py36 http://logs.openstack.org/58/665758/4/gate/swift-tox-py36/f9cf4b1/ : SUCCESS in 11m 50s\n- swift-tox-py37 http://logs.openstack.org/58/665758/4/gate/swift-tox-py37/1c699ff/ : SUCCESS in 12m 01s\n- swift-tox-func http://logs.openstack.org/58/665758/4/gate/swift-tox-func/8314254/ : SUCCESS in 10m 25s\n- swift-tox-func-encryption http://logs.openstack.org/58/665758/4/gate/swift-tox-func-encryption/a9f190b/ : SUCCESS in 10m 54s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/58/665758/4/gate/swift-tox-func-domain-remap-staticweb/f9615bf/ : SUCCESS in 11m 47s\n- swift-tox-func-ec http://logs.openstack.org/58/665758/4/gate/swift-tox-func-ec/d291b7a/ : SUCCESS in 11m 11s\n- swift-tox-func-s3api http://logs.openstack.org/58/665758/4/gate/swift-tox-func-s3api/e608be4/ : SUCCESS in 12m 05s\n- swift-tox-func-py37 http://logs.openstack.org/58/665758/4/gate/swift-tox-func-py37/4b8eb7e/ : SUCCESS in 11m 23s\n- swift-tox-func-domain-remap-staticweb-py37 http://logs.openstack.org/58/665758/4/gate/swift-tox-func-domain-remap-staticweb-py37/e08b01d/ : SUCCESS in 10m 50s\n- swift-tox-func-ec-py37 http://logs.openstack.org/58/665758/4/gate/swift-tox-func-ec-py37/941af34/ : SUCCESS in 11m 16s\n- swift-probetests-centos-7 http://logs.openstack.org/58/665758/4/gate/swift-probetests-centos-7/ec13419/ : SUCCESS in 30m 30s\n- swift-dsvm-functional http://logs.openstack.org/58/665758/4/gate/swift-dsvm-functional/73fdb29/ : SUCCESS in 27m 26s\n- swift-dsvm-functional-py3 http://logs.openstack.org/58/665758/4/gate/swift-dsvm-functional-py3/1329d9d/ : SUCCESS in 38m 04s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/58/665758/4/gate/swift-dsvm-functional-ipv6/de4278b/ : SUCCESS in 30m 36s\n- swift-tox-lower-constraints http://logs.openstack.org/58/665758/4/gate/swift-tox-lower-constraints/bd82552/ : SUCCESS in 13m 26s\n- openstack-tox-pep8 http://logs.openstack.org/58/665758/4/gate/openstack-tox-pep8/e79333b/ : SUCCESS in 7m 10s","accounts_in_message":[],"_revision_number":4},{"id":"fbb3ddd5aa089e6341bdb4a3d8178c772f510f2f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 17:53:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"0ebfeddf65adb473fda2adfd1705534b6308cd2b","revisions":{"2521721a5166457faf7a8943cd0e70cc0096ca12":{"kind":"REWORK","_number":1,"created":"2019-06-17 20:22:40.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/58/665758/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/58/665758/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/665758/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/665758/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/665758/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/58/665758/1"}}},"commit":{"parents":[{"commit":"6374734d7d0f76d863afb167637d0b590316e306","subject":"Stop manually installing eventlet for our py3 tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6374734d7d0f76d863afb167637d0b590316e306"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-06-17 20:22:04.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-06-17 20:22:27.000000000","tz":-420},"subject":"Bump up our minimum eventlet version","message":"Bump up our minimum eventlet version\n\nWe need 0.25.0 for py3 support, and increasing it for both is simplest.\n\nChange-Id: If933b713bdd1dfc348959771d60c0756a9a94477\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2521721a5166457faf7a8943cd0e70cc0096ca12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2521721a5166457faf7a8943cd0e70cc0096ca12"}]},"branch":"refs/heads/master"},"ba04f5f31180416509854950e70661e26e6616b3":{"kind":"REWORK","_number":2,"created":"2019-06-18 15:37:18.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/58/665758/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/58/665758/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/665758/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/665758/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/665758/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/58/665758/2"}}},"commit":{"parents":[{"commit":"6374734d7d0f76d863afb167637d0b590316e306","subject":"Stop manually installing eventlet for our py3 tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6374734d7d0f76d863afb167637d0b590316e306"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-06-17 20:22:04.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-06-18 15:35:31.000000000","tz":-420},"subject":"Bump up our minimum eventlet version","message":"Bump up our minimum eventlet version\n\nWe need 0.25.0 for py3 support, and increasing it for both is simplest.\n\nAlso drag forward dnspython, as newer eventlet requires a newer\ndnspython. And explicitly list it as a dependency for py3 as well as 2.\n\nChange-Id: If933b713bdd1dfc348959771d60c0756a9a94477\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ba04f5f31180416509854950e70661e26e6616b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ba04f5f31180416509854950e70661e26e6616b3"}]},"branch":"refs/heads/master"},"ccbe620770d0237ce56f56ca62034f2fc6e41e40":{"kind":"REWORK","_number":3,"created":"2019-06-20 22:14:17.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/58/665758/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/58/665758/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/665758/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/665758/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/665758/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/58/665758/3"}}},"commit":{"parents":[{"commit":"d44a581cdd50c1158ac810f21202943b7d20624f","subject":"Merge \"py3: port the test of locale\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d44a581cdd50c1158ac810f21202943b7d20624f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-06-17 20:22:04.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-06-20 22:14:05.000000000","tz":-420},"subject":"Bump up our minimum eventlet version","message":"Bump up our minimum eventlet version\n\nWe need 0.25.0 for py3 support, and increasing it for both is simplest.\n\nAlso drag forward dnspython, as newer eventlet requires a newer\ndnspython. And explicitly list it as a dependency for py3 as well as 2.\n\nChange-Id: If933b713bdd1dfc348959771d60c0756a9a94477\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ccbe620770d0237ce56f56ca62034f2fc6e41e40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ccbe620770d0237ce56f56ca62034f2fc6e41e40"}]},"branch":"refs/heads/master"},"0ebfeddf65adb473fda2adfd1705534b6308cd2b":{"kind":"REWORK","_number":4,"created":"2019-06-24 15:23:00.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/58/665758/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/58/665758/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/665758/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/665758/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/665758/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/58/665758/4"}}},"commit":{"parents":[{"commit":"9f1ef35630ebf9da53cbd061f89889e534803725","subject":"py3: Port the tempurl middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f1ef35630ebf9da53cbd061f89889e534803725"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-06-17 20:22:04.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-06-24 15:22:37.000000000","tz":-420},"subject":"Bump up our minimum eventlet version","message":"Bump up our minimum eventlet version\n\nWe need 0.25.0 for py3 support, and increasing it for both is simplest.\n\nAlso drag forward dnspython, as newer eventlet requires a newer\ndnspython. Note that we still don\u0027t explicitly list it as a dependency\nfor py3; rather, it\u0027s been getting installed as a requirement for\neventlet.\n\nChange-Id: If933b713bdd1dfc348959771d60c0756a9a94477\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0ebfeddf65adb473fda2adfd1705534b6308cd2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0ebfeddf65adb473fda2adfd1705534b6308cd2b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
