)]}'
{"id":"openstack%2Fswift~665770","triplet_id":"openstack%2Fswift~master~Ie1f8808c04f975ccc7cea532c4a566d6984f1757","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"Ie1f8808c04f975ccc7cea532c4a566d6984f1757","subject":"py3: increase _MAXHEADERS for eventlet, too","status":"MERGED","created":"2019-06-17 21:42:52.000000000","updated":"2019-06-19 02:26:57.000000000","submitted":"2019-06-19 02:25:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665770-1560911146368-d473cd44","meta_rev_id":"3f548186d82ed1367ccf342e1a10438b08319c78","_number":665770,"virtual_id_number":665770,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":2,"date":"2019-06-19 02:25:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2019-06-18 23:44:20.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":1,"date":"2019-06-18 23:44:20.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-18 23:44:20.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2019-06-19 02:25:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"edc4a16a04c3794e706e29a53971770ab8f72eb3","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-17 21:42:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4cc7e8e2817be003feb640e1825726021e960b99","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 01:19:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/70/665770/1/check/openstack-tox-docs/16bff0f/html/ : SUCCESS in 5m 43s\n- swift-build-image http://logs.openstack.org/70/665770/1/check/swift-build-image/5b564bc/ : SUCCESS in 7m 24s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/70/665770/1/check/swift-tox-py27/18a8f2e/ : SUCCESS in 10m 31s\n- swift-tox-py36 http://logs.openstack.org/70/665770/1/check/swift-tox-py36/facfc33/ : SUCCESS in 23m 26s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/70/665770/1/check/swift-tox-py37/a07259b/ : SUCCESS in 15m 40s\n- swift-tox-func http://logs.openstack.org/70/665770/1/check/swift-tox-func/73f6b16/ : SUCCESS in 8m 57s\n- swift-tox-func-py37 http://logs.openstack.org/70/665770/1/check/swift-tox-func-py37/98ae226/ : SUCCESS in 9m 14s\n- swift-tox-func-encryption http://logs.openstack.org/70/665770/1/check/swift-tox-func-encryption/abba5be/ : SUCCESS in 10m 15s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/70/665770/1/check/swift-tox-func-domain-remap-staticweb/a877b2a/ : SUCCESS in 9m 37s\n- swift-tox-func-ec http://logs.openstack.org/70/665770/1/check/swift-tox-func-ec/17184e7/ : SUCCESS in 9m 05s\n- swift-tox-func-s3api http://logs.openstack.org/70/665770/1/check/swift-tox-func-s3api/4f00fbc/ : SUCCESS in 11m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/70/665770/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/fe63e85/ : SUCCESS in 21m 05s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/70/665770/1/check/swift-probetests-centos-7/84b7868/ : SUCCESS in 31m 04s\n- swift-dsvm-functional http://logs.openstack.org/70/665770/1/check/swift-dsvm-functional/f40db59/ : SUCCESS in 31m 51s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/70/665770/1/check/swift-dsvm-functional-ipv6/38bfa04/ : SUCCESS in 28m 12s\n- swift-tox-lower-constraints http://logs.openstack.org/70/665770/1/check/swift-tox-lower-constraints/1c7ba20/ : SUCCESS in 11m 46s\n- openstack-tox-pep8 http://logs.openstack.org/70/665770/1/check/openstack-tox-pep8/f3d8460/ : SUCCESS in 6m 57s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/70/665770/1/check/swift-multinode-rolling-upgrade/a89a63b/ : FAILURE in 8m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f5b7d1f4b79ade6f96a2babfcb5c5f47ffe10c62","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-06-18 23:44:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\ngreat, more magic","accounts_in_message":[],"_revision_number":1},{"id":"70699f78302128fcfe0ba55c50c2eb0822ab216e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 23:44:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e33452cf1ae167df51b4109aa12d472a8a17ade1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 02:25:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/70/665770/1/gate/openstack-tox-docs/034d12a/html/ : SUCCESS in 4m 42s\n- swift-upload-image http://logs.openstack.org/70/665770/1/gate/swift-upload-image/b5bbe5d/ : SUCCESS in 7m 55s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/70/665770/1/gate/swift-tox-py27/65a4481/ : SUCCESS in 10m 50s\n- swift-tox-py37 http://logs.openstack.org/70/665770/1/gate/swift-tox-py37/9ab85f7/ : SUCCESS in 11m 38s\n- swift-tox-func http://logs.openstack.org/70/665770/1/gate/swift-tox-func/deb0734/ : SUCCESS in 9m 36s\n- swift-tox-func-py37 http://logs.openstack.org/70/665770/1/gate/swift-tox-func-py37/0b8222a/ : SUCCESS in 12m 28s\n- swift-tox-func-encryption http://logs.openstack.org/70/665770/1/gate/swift-tox-func-encryption/dc7f941/ : SUCCESS in 14m 02s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/70/665770/1/gate/swift-tox-func-domain-remap-staticweb/40b975b/ : SUCCESS in 9m 32s\n- swift-tox-func-ec http://logs.openstack.org/70/665770/1/gate/swift-tox-func-ec/962de35/ : SUCCESS in 10m 24s\n- swift-tox-func-s3api http://logs.openstack.org/70/665770/1/gate/swift-tox-func-s3api/372fadb/ : SUCCESS in 12m 09s\n- swift-probetests-centos-7 http://logs.openstack.org/70/665770/1/gate/swift-probetests-centos-7/c20a6cf/ : SUCCESS in 31m 50s\n- swift-dsvm-functional http://logs.openstack.org/70/665770/1/gate/swift-dsvm-functional/0ee8094/ : SUCCESS in 25m 30s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/70/665770/1/gate/swift-dsvm-functional-ipv6/3ddf531/ : SUCCESS in 35m 15s\n- swift-tox-lower-constraints http://logs.openstack.org/70/665770/1/gate/swift-tox-lower-constraints/060b5a3/ : SUCCESS in 17m 58s\n- openstack-tox-pep8 http://logs.openstack.org/70/665770/1/gate/openstack-tox-pep8/e875bec/ : SUCCESS in 8m 27s","accounts_in_message":[],"_revision_number":1},{"id":"b0ae16406b2d03f39d8b9499b2efe20d11415acd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 02:25:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3f548186d82ed1367ccf342e1a10438b08319c78","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 02:26:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- swift-promote-image http://logs.openstack.org/70/665770/1/promote/swift-promote-image/91cb817/ : SUCCESS in 46s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"876ce231d616530e036521ba7cac28373e722ce2","revisions":{"876ce231d616530e036521ba7cac28373e722ce2":{"kind":"REWORK","_number":1,"created":"2019-06-17 21:42:52.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/70/665770/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/70/665770/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/665770/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/665770/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/665770/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/70/665770/1"}}},"commit":{"parents":[{"commit":"bc19dd1649115d9c83e2c939bcca990acfe7867c","subject":"Merge \"Installing liberasurecode from Alpine Linux repos instead of building from source.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bc19dd1649115d9c83e2c939bcca990acfe7867c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-06-17 21:41:19.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-06-17 21:41:19.000000000","tz":-420},"subject":"py3: increase _MAXHEADERS for eventlet, too","message":"py3: increase _MAXHEADERS for eventlet, too\n\neventlet.wsgi goes straight to eventlet.green.* instead of stdlib.\n\nChange-Id: Ie1f8808c04f975ccc7cea532c4a566d6984f1757\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/876ce231d616530e036521ba7cac28373e722ce2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/876ce231d616530e036521ba7cac28373e722ce2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
