)]}'
{"id":"openstack%2Fswift~666942","triplet_id":"openstack%2Fswift~master~Ia783ae10344380eb5afd9d43273969b433962945","project":"openstack/swift","branch":"master","topic":"py3-devstack","hashtags":[],"change_id":"Ia783ae10344380eb5afd9d43273969b433962945","subject":"Run all func tests under py2 against services running py3","status":"MERGED","created":"2019-06-22 05:38:43.000000000","updated":"2019-07-11 05:10:37.000000000","submitted":"2019-07-11 05:08:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"666942-1562821717415-3eb45556","meta_rev_id":"a1cebb6ddda929dd76a3b7d8da32f3f7066f1e64","_number":666942,"virtual_id_number":666942,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":2,"date":"2019-07-11 05:08:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-10 18:30:07.000000000","_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"all":[{"value":2,"date":"2019-07-10 15:52:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"all":[{"value":1,"date":"2019-07-10 15:52:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-10 15:52:09.000000000","updated_by":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"reviewer":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"state":"REVIEWER"},{"updated":"2019-07-11 05:08:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2c8e93ac58959ee57dd9b0344fe5d6a316f51224","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-22 05:38:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5b4d95f3882a649fcb6dad85cf355145870926c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-22 06:25:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/42/666942/1/check/openstack-tox-docs/37e130c/html/ : SUCCESS in 4m 16s\n- swift-build-image swift-build-image : FAILURE (non-voting)\n- swift-tox-py27 http://logs.openstack.org/42/666942/1/check/swift-tox-py27/4dc01a9/ : SUCCESS in 18m 17s\n- swift-tox-py36 http://logs.openstack.org/42/666942/1/check/swift-tox-py36/2ee0c64/ : FAILURE in 18m 15s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/42/666942/1/check/swift-tox-py37/2dfb852/ : SUCCESS in 18m 48s\n- swift-tox-func http://logs.openstack.org/42/666942/1/check/swift-tox-func/d262790/ : SUCCESS in 16m 27s\n- swift-tox-func-py37 http://logs.openstack.org/42/666942/1/check/swift-tox-func-py37/bbdc64c/ : SUCCESS in 11m 24s\n- swift-tox-func-encryption http://logs.openstack.org/42/666942/1/check/swift-tox-func-encryption/bed4de5/ : SUCCESS in 12m 03s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/42/666942/1/check/swift-tox-func-domain-remap-staticweb/5e0f13a/ : SUCCESS in 14m 32s\n- swift-tox-func-ec http://logs.openstack.org/42/666942/1/check/swift-tox-func-ec/694789d/ : SUCCESS in 17m 53s\n- swift-tox-func-s3api http://logs.openstack.org/42/666942/1/check/swift-tox-func-s3api/724a288/ : SUCCESS in 14m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/42/666942/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/71479cc/ : SUCCESS in 17m 20s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/42/666942/1/check/swift-probetests-centos-7/a71127e/ : SUCCESS in 37m 25s\n- swift-dsvm-functional http://logs.openstack.org/42/666942/1/check/swift-dsvm-functional/6acbe4b/ : SUCCESS in 25m 26s\n- swift-dsvm-functional-py3 http://logs.openstack.org/42/666942/1/check/swift-dsvm-functional-py3/dc8ef09/ : SUCCESS in 25m 01s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/42/666942/1/check/swift-dsvm-functional-ipv6/bed94ff/ : SUCCESS in 22m 06s\n- swift-tox-lower-constraints http://logs.openstack.org/42/666942/1/check/swift-tox-lower-constraints/a801ed1/ : SUCCESS in 12m 54s\n- openstack-tox-pep8 http://logs.openstack.org/42/666942/1/check/openstack-tox-pep8/6b60673/ : SUCCESS in 6m 31s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/42/666942/1/check/swift-multinode-rolling-upgrade/2e472d7/ : FAILURE in 10m 41s (non-voting)\n\nWarning:\n  Requirements [\u0027docker-image\u0027] not met by build 90efe41e0c994436a6dc0bf941bea981","accounts_in_message":[],"_revision_number":1},{"id":"aca29eb2df447eb6e1664533b0ab12cafc52a856","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2019-07-10 15:52:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bb1f94da54de29bee4e8498990c7d67eccc18f89","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 15:52:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"19f755427f01c442162be3887495dfcabd51d2ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 18:29:27.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"6d744c9c9dc3acb7d725b9c6e2593040325e246c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-07-10 18:30:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"59247bc15ecb3821a2155a98c328739a5ae0ef6c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 23:05:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/42/666942/1/check/openstack-tox-docs/fcc52f4/html/ : SUCCESS in 4m 19s\n- swift-build-image http://logs.openstack.org/42/666942/1/check/swift-build-image/ade2cf0/ : SUCCESS in 8m 05s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/42/666942/1/check/swift-tox-py27/3df8674/ : SUCCESS in 11m 54s\n- swift-tox-py36 http://logs.openstack.org/42/666942/1/check/swift-tox-py36/da4c1b5/ : SUCCESS in 10m 19s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/42/666942/1/check/swift-tox-py37/2702a68/ : SUCCESS in 10m 48s\n- swift-tox-func http://logs.openstack.org/42/666942/1/check/swift-tox-func/d60cd02/ : SUCCESS in 9m 04s\n- swift-tox-func-py37 http://logs.openstack.org/42/666942/1/check/swift-tox-func-py37/5ef418c/ : SUCCESS in 8m 58s\n- swift-tox-func-encryption http://logs.openstack.org/42/666942/1/check/swift-tox-func-encryption/e2f16ae/ : SUCCESS in 8m 50s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/42/666942/1/check/swift-tox-func-domain-remap-staticweb/4e97878/ : SUCCESS in 9m 55s\n- swift-tox-func-ec http://logs.openstack.org/42/666942/1/check/swift-tox-func-ec/7cf6803/ : SUCCESS in 12m 07s\n- swift-tox-func-s3api http://logs.openstack.org/42/666942/1/check/swift-tox-func-s3api/1697a14/ : SUCCESS in 11m 25s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/42/666942/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/f90eeb6/ : SUCCESS in 11m 44s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/42/666942/1/check/swift-probetests-centos-7/5ef3306/ : SUCCESS in 35m 41s\n- swift-dsvm-functional http://logs.openstack.org/42/666942/1/check/swift-dsvm-functional/28c0cc2/ : SUCCESS in 34m 02s\n- swift-dsvm-functional-py3 http://logs.openstack.org/42/666942/1/check/swift-dsvm-functional-py3/c9a422a/ : SUCCESS in 29m 02s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/42/666942/1/check/swift-dsvm-functional-ipv6/f013ea6/ : SUCCESS in 28m 25s\n- swift-tox-lower-constraints http://logs.openstack.org/42/666942/1/check/swift-tox-lower-constraints/0056fa4/ : SUCCESS in 11m 10s\n- openstack-tox-pep8 http://logs.openstack.org/42/666942/1/check/openstack-tox-pep8/3bdfb10/ : SUCCESS in 6m 07s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/42/666942/1/check/swift-multinode-rolling-upgrade/a477dd1/ : FAILURE in 9m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7f1d626cec2fc4c1a26ef8fcc3aa5197a295a714","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 01:50:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6465f3f1d4f94ff3a898666224752908874acfbc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 05:08:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/42/666942/1/gate/openstack-tox-docs/b34e8bd/html/ : SUCCESS in 5m 06s\n- swift-upload-image http://logs.openstack.org/42/666942/1/gate/swift-upload-image/8503079/ : SUCCESS in 8m 24s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/42/666942/1/gate/swift-tox-py27/3db87e3/ : SUCCESS in 10m 21s\n- swift-tox-py37 http://logs.openstack.org/42/666942/1/gate/swift-tox-py37/5dfd66d/ : SUCCESS in 12m 37s\n- swift-tox-func http://logs.openstack.org/42/666942/1/gate/swift-tox-func/d398f0b/ : SUCCESS in 8m 05s\n- swift-tox-func-py37 http://logs.openstack.org/42/666942/1/gate/swift-tox-func-py37/a10a1b5/ : SUCCESS in 9m 06s\n- swift-tox-func-encryption http://logs.openstack.org/42/666942/1/gate/swift-tox-func-encryption/194ac6a/ : SUCCESS in 8m 16s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/42/666942/1/gate/swift-tox-func-domain-remap-staticweb/8ca414e/ : SUCCESS in 8m 52s\n- swift-tox-func-ec http://logs.openstack.org/42/666942/1/gate/swift-tox-func-ec/211e772/ : SUCCESS in 9m 18s\n- swift-tox-func-s3api http://logs.openstack.org/42/666942/1/gate/swift-tox-func-s3api/32a3df0/ : SUCCESS in 11m 17s\n- swift-probetests-centos-7 http://logs.openstack.org/42/666942/1/gate/swift-probetests-centos-7/c29ce46/ : SUCCESS in 29m 35s\n- swift-dsvm-functional http://logs.openstack.org/42/666942/1/gate/swift-dsvm-functional/4739179/ : SUCCESS in 27m 22s\n- swift-dsvm-functional-py3 http://logs.openstack.org/42/666942/1/gate/swift-dsvm-functional-py3/23b974c/ : SUCCESS in 27m 31s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/42/666942/1/gate/swift-dsvm-functional-ipv6/4891865/ : SUCCESS in 26m 43s\n- swift-tox-lower-constraints http://logs.openstack.org/42/666942/1/gate/swift-tox-lower-constraints/8456898/ : SUCCESS in 10m 35s\n- openstack-tox-pep8 http://logs.openstack.org/42/666942/1/gate/openstack-tox-pep8/dab5e43/ : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":1},{"id":"686586ceea84be73240a8c765807ac46dad765b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 05:08:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6d185bde6e3ce6a180836dd1c10721df28e3f413","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 05:10:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- swift-promote-image http://logs.openstack.org/42/666942/1/promote/swift-promote-image/1fef6c7/ : SUCCESS in 59s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"55b04d16c4740fb731a2ffa97a84e1607ca94ef3","revisions":{"55b04d16c4740fb731a2ffa97a84e1607ca94ef3":{"kind":"REWORK","_number":1,"created":"2019-06-22 05:38:43.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/42/666942/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/666942/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/666942/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/666942/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/666942/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/666942/1"}}},"commit":{"parents":[{"commit":"39a54fecdc5d7f9a09f34fcd9e81ec60948b51d7","subject":"py3: add swift-dsvm-functional-py3 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/39a54fecdc5d7f9a09f34fcd9e81ec60948b51d7"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-06-22 05:34:07.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-06-22 05:37:32.000000000","tz":-420},"subject":"Run all func tests under py2 against services running py3","message":"Run all func tests under py2 against services running py3\n\n... at least, on the py3 dsvm job.\n\nDepends-On: I0f03c211f35a9a49e047a5718a9907b515ca88d7\nChange-Id: Ia783ae10344380eb5afd9d43273969b433962945\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/55b04d16c4740fb731a2ffa97a84e1607ca94ef3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/55b04d16c4740fb731a2ffa97a84e1607ca94ef3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
