)]}'
{"id":"openstack%2Fswift~667203","triplet_id":"openstack%2Fswift~master~Ic7a732146ea19a47669114ad5dbee0bacbe66919","project":"openstack/swift","branch":"master","topic":"bug/1781291","hashtags":[],"change_id":"Ic7a732146ea19a47669114ad5dbee0bacbe66919","subject":"sharding: Cache shard ranges for object writes","status":"MERGED","created":"2019-06-24 19:32:35.000000000","updated":"2019-07-13 00:35:27.000000000","submitted":"2019-07-13 00:34:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667203-1562978046217-8d65fae4","meta_rev_id":"61fa12336eeed3a7372b381c3cdce8c0db5bae4b","_number":667203,"virtual_id_number":667203,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":2,"date":"2019-07-13 00:34:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"all":[{"value":2,"date":"2019-07-12 23:55:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"all":[{"value":1,"date":"2019-07-12 23:55:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-24 20:36:17.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2019-07-12 23:55:53.000000000","updated_by":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"reviewer":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"state":"REVIEWER"},{"updated":"2019-07-13 00:34:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c11d92ea068db367c72bd74b426b5aa51afd0357","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-24 19:32:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9dffe9fce87e6f7777bf28b46da9af9a5e5b1732","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-06-24 20:36:17.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)\n\nthis looks surprisingly obvious","accounts_in_message":[],"_revision_number":1},{"id":"99e4a0dd560dfff629587d291dcf94199ded05c2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-24 21:23:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b07e682128d813209ecae5efe4286dacbd7b5f67","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-24 21:25:14.000000000","message":"Patch Set 1:\n\nOh yeah, and if we really start to get worried about it, we can always start gzipping what we\u0027re cache -- there\u0027s even a bit of plumbing for that sort of thing: https://github.com/openstack/swift/blob/master/swift/common/memcached.py#L260-L263","accounts_in_message":[],"_revision_number":1},{"id":"4ac96dc09c19f7d463adf5bc25ef4bd2c4398a56","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-24 21:34:34.000000000","message":"Patch Set 1:\n\nThis should probably have a reference to https://bugs.launchpad.net/swift/+bug/1781291 -- and maybe I should try some LRUCache thing that dangles off the proxy-server?","accounts_in_message":[],"_revision_number":1},{"id":"227afb1d5b895e493f3b3cdb499f8e54c93260bb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 22:04:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/03/667203/1/check/openstack-tox-docs/30e572f/html/ : SUCCESS in 4m 12s\n- swift-build-image http://logs.openstack.org/03/667203/1/check/swift-build-image/aacd038/ : SUCCESS in 8m 55s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/03/667203/1/check/swift-tox-py27/e22cdca/ : SUCCESS in 12m 15s\n- swift-tox-py36 http://logs.openstack.org/03/667203/1/check/swift-tox-py36/b9c6829/ : SUCCESS in 10m 12s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/03/667203/1/check/swift-tox-py37/cf58f15/ : SUCCESS in 11m 13s\n- swift-tox-func http://logs.openstack.org/03/667203/1/check/swift-tox-func/7c65769/ : SUCCESS in 8m 53s\n- swift-tox-func-py37 http://logs.openstack.org/03/667203/1/check/swift-tox-func-py37/d76ab08/ : SUCCESS in 9m 41s\n- swift-tox-func-encryption http://logs.openstack.org/03/667203/1/check/swift-tox-func-encryption/568b6c3/ : SUCCESS in 9m 45s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/03/667203/1/check/swift-tox-func-domain-remap-staticweb/a24a7be/ : SUCCESS in 9m 31s\n- swift-tox-func-ec http://logs.openstack.org/03/667203/1/check/swift-tox-func-ec/b70b914/ : SUCCESS in 10m 58s\n- swift-tox-func-s3api http://logs.openstack.org/03/667203/1/check/swift-tox-func-s3api/6ce6f94/ : SUCCESS in 11m 42s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/03/667203/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/faf3a14/ : SUCCESS in 13m 08s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/03/667203/1/check/swift-probetests-centos-7/bb8a861/ : FAILURE in 28m 42s\n- swift-dsvm-functional http://logs.openstack.org/03/667203/1/check/swift-dsvm-functional/bc9c1e5/ : SUCCESS in 27m 09s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/03/667203/1/check/swift-dsvm-functional-ipv6/98e3f9e/ : SUCCESS in 27m 40s\n- swift-tox-lower-constraints http://logs.openstack.org/03/667203/1/check/swift-tox-lower-constraints/e9906aa/ : SUCCESS in 13m 41s\n- openstack-tox-pep8 http://logs.openstack.org/03/667203/1/check/openstack-tox-pep8/3c17cb1/ : FAILURE in 5m 36s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/03/667203/1/check/swift-multinode-rolling-upgrade/fd42b63/ : FAILURE in 9m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"03069f3fe7d61d27cc1b7f65ea57462c87a8f366","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-25 16:16:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"feff6c549844c70a3cf7acb58117a46320c3e71f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-25 23:17:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"42b5dff2cd24fafb63f1c5e89ddce3cbb3fd9127","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-25 23:17:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"98c4e3284ac26051921645f69ef289d2e66a082b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 11:20:34.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/03/667203/3/check/openstack-tox-docs/5fce2c5/html/ : SUCCESS in 4m 07s\n- swift-build-image http://logs.openstack.org/03/667203/3/check/swift-build-image/5dd625f/ : RETRY_LIMIT in 4m 03s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/03/667203/3/check/swift-tox-py27/7bc9a1f/ : SUCCESS in 11m 08s\n- swift-tox-py36 http://logs.openstack.org/03/667203/3/check/swift-tox-py36/8fb8f5e/ : SUCCESS in 11m 08s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/03/667203/3/check/swift-tox-py37/f821fec/ : FAILURE in 10m 54s\n- swift-tox-func http://logs.openstack.org/03/667203/3/check/swift-tox-func/83d1cb1/ : SUCCESS in 8m 52s\n- swift-tox-func-py37 http://logs.openstack.org/03/667203/3/check/swift-tox-func-py37/c851fc1/ : SUCCESS in 9m 59s\n- swift-tox-func-encryption http://logs.openstack.org/03/667203/3/check/swift-tox-func-encryption/9b7b7db/ : SUCCESS in 9m 46s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/03/667203/3/check/swift-tox-func-domain-remap-staticweb/fed4c5a/ : SUCCESS in 10m 19s\n- swift-tox-func-ec http://logs.openstack.org/03/667203/3/check/swift-tox-func-ec/859ea9e/ : RETRY_LIMIT in 2m 35s\n- swift-tox-func-s3api http://logs.openstack.org/03/667203/3/check/swift-tox-func-s3api/a39c382/ : SUCCESS in 11m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/03/667203/3/check/swift-tox-func-s3api-ceph-s3tests-tempauth/380769e/ : RETRY_LIMIT in 7m 06s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/03/667203/3/check/swift-probetests-centos-7/2d369ab/ : SUCCESS in 30m 09s\n- swift-dsvm-functional http://logs.openstack.org/03/667203/3/check/swift-dsvm-functional/8c067d4/ : SUCCESS in 38m 41s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/03/667203/3/check/swift-dsvm-functional-ipv6/3344ad7/ : SUCCESS in 26m 48s\n- swift-tox-lower-constraints http://logs.openstack.org/03/667203/3/check/swift-tox-lower-constraints/d5b334e/ : RETRY_LIMIT in 2m 43s\n- openstack-tox-pep8 http://logs.openstack.org/03/667203/3/check/openstack-tox-pep8/8b2d3c1/ : SUCCESS in 7m 21s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/03/667203/3/check/swift-multinode-rolling-upgrade/15a9083/ : FAILURE in 10m 36s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7f4dd1691207f2f1e970710fe0244ec3e85b5cbf","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-27 00:11:02.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"148d437eac67329154a9951b0af9a4e786bfdd1f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 05:37:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/03/667203/3/check/openstack-tox-docs/d2651cd/html/ : SUCCESS in 3m 49s\n- swift-build-image http://logs.openstack.org/03/667203/3/check/swift-build-image/df5145e/ : SUCCESS in 7m 27s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/03/667203/3/check/swift-tox-py27/565b8cd/ : SUCCESS in 10m 20s\n- swift-tox-py36 http://logs.openstack.org/03/667203/3/check/swift-tox-py36/ae65145/ : SUCCESS in 10m 27s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/03/667203/3/check/swift-tox-py37/b3d1f26/ : SUCCESS in 11m 12s\n- swift-tox-func http://logs.openstack.org/03/667203/3/check/swift-tox-func/7d0ed1b/ : SUCCESS in 8m 20s\n- swift-tox-func-py37 http://logs.openstack.org/03/667203/3/check/swift-tox-func-py37/b2c2054/ : SUCCESS in 8m 49s\n- swift-tox-func-encryption http://logs.openstack.org/03/667203/3/check/swift-tox-func-encryption/fea661a/ : SUCCESS in 11m 30s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/03/667203/3/check/swift-tox-func-domain-remap-staticweb/02f1677/ : SUCCESS in 8m 32s\n- swift-tox-func-ec http://logs.openstack.org/03/667203/3/check/swift-tox-func-ec/3bd0051/ : SUCCESS in 10m 56s\n- swift-tox-func-s3api http://logs.openstack.org/03/667203/3/check/swift-tox-func-s3api/6777e99/ : SUCCESS in 15m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/03/667203/3/check/swift-tox-func-s3api-ceph-s3tests-tempauth/5bb6426/ : SUCCESS in 12m 58s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/03/667203/3/check/swift-probetests-centos-7/3f918d6/ : SUCCESS in 31m 47s\n- swift-dsvm-functional http://logs.openstack.org/03/667203/3/check/swift-dsvm-functional/4f92269/ : SUCCESS in 26m 56s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/03/667203/3/check/swift-dsvm-functional-ipv6/f0ca456/ : SUCCESS in 25m 55s\n- swift-tox-lower-constraints http://logs.openstack.org/03/667203/3/check/swift-tox-lower-constraints/03a49d6/ : SUCCESS in 17m 38s\n- openstack-tox-pep8 http://logs.openstack.org/03/667203/3/check/openstack-tox-pep8/5aa758e/ : SUCCESS in 5m 01s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/03/667203/3/check/swift-multinode-rolling-upgrade/2959e22/ : FAILURE in 9m 10s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0c8f5e48a848adceadabe1b9945d16a2dd4e19e5","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2019-07-10 16:49:00.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"d26e5df0d47db7072b4d49fc39e7cafa6684d177","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-07-11 17:40:53.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6b417b56892be34083016a011074208088ed6402","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-07-11 17:41:34.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"ebb990570ee2b06663156077647e28ff0bcad7ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 21:18:12.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/03/667203/4/check/openstack-tox-docs/1512498/html/ : SUCCESS in 4m 36s\n- swift-build-image http://logs.openstack.org/03/667203/4/check/swift-build-image/67cd205/ : SUCCESS in 7m 56s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/03/667203/4/check/swift-tox-py27/a8649c9/ : SUCCESS in 11m 54s\n- swift-tox-py36 http://logs.openstack.org/03/667203/4/check/swift-tox-py36/2a53ce0/ : SUCCESS in 10m 35s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/03/667203/4/check/swift-tox-py37/22ccdef/ : SUCCESS in 11m 53s\n- swift-tox-func http://logs.openstack.org/03/667203/4/check/swift-tox-func/6a2cf04/ : SUCCESS in 8m 57s\n- swift-tox-func-py37 http://logs.openstack.org/03/667203/4/check/swift-tox-func-py37/6bdb9e0/ : SUCCESS in 9m 47s\n- swift-tox-func-encryption http://logs.openstack.org/03/667203/4/check/swift-tox-func-encryption/b81978c/ : SUCCESS in 9m 28s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/03/667203/4/check/swift-tox-func-domain-remap-staticweb/df17ef0/ : SUCCESS in 10m 25s\n- swift-tox-func-ec http://logs.openstack.org/03/667203/4/check/swift-tox-func-ec/609c93a/ : SUCCESS in 8m 47s\n- swift-tox-func-s3api http://logs.openstack.org/03/667203/4/check/swift-tox-func-s3api/e1e1105/ : SUCCESS in 13m 09s\n- swift-tox-func-domain-remap-staticweb-py37 http://logs.openstack.org/03/667203/4/check/swift-tox-func-domain-remap-staticweb-py37/71756c3/ : SUCCESS in 7m 56s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/03/667203/4/check/swift-tox-func-s3api-ceph-s3tests-tempauth/ea8cf8e/ : SUCCESS in 13m 08s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/03/667203/4/check/swift-probetests-centos-7/70516ae/ : SUCCESS in 31m 05s\n- swift-dsvm-functional http://logs.openstack.org/03/667203/4/check/swift-dsvm-functional/75199d4/ : SUCCESS in 29m 37s\n- swift-dsvm-functional-py3 http://logs.openstack.org/03/667203/4/check/swift-dsvm-functional-py3/522b782/ : SUCCESS in 31m 20s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/03/667203/4/check/swift-dsvm-functional-ipv6/395327f/ : SUCCESS in 29m 28s\n- swift-tox-lower-constraints http://logs.openstack.org/03/667203/4/check/swift-tox-lower-constraints/62b0a4e/ : SUCCESS in 12m 24s\n- openstack-tox-pep8 http://logs.openstack.org/03/667203/4/check/openstack-tox-pep8/d214fc6/ : SUCCESS in 5m 41s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/03/667203/4/check/swift-multinode-rolling-upgrade/ff92aec/ : FAILURE in 9m 53s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"15900fbbffc9d9cbff34abf15606c94b3278d0a7","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2019-07-12 23:55:53.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nlooks good. The code change is straight forward.\n\nThe tests show that requests are forwarded to container layer when shards are not cached and *not* forwarded when the shards are cached.","accounts_in_message":[],"_revision_number":4},{"id":"7d936cedaa0a95750b392801ea7e7818aeecb17c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 23:56:05.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"538bb7e05325cb3a7f3df99e4fcca0f442cd8eb1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-13 00:34:05.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/03/667203/4/gate/openstack-tox-docs/c92acc3/html/ : SUCCESS in 4m 09s\n- swift-upload-image http://logs.openstack.org/03/667203/4/gate/swift-upload-image/b7b0dce/ : SUCCESS in 10m 20s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/03/667203/4/gate/swift-tox-py27/c2c9581/ : SUCCESS in 14m 53s\n- swift-tox-py37 http://logs.openstack.org/03/667203/4/gate/swift-tox-py37/d364d4f/ : SUCCESS in 13m 02s\n- swift-tox-func http://logs.openstack.org/03/667203/4/gate/swift-tox-func/2761f2d/ : SUCCESS in 8m 36s\n- swift-tox-func-py37 http://logs.openstack.org/03/667203/4/gate/swift-tox-func-py37/f131694/ : SUCCESS in 12m 18s\n- swift-tox-func-encryption http://logs.openstack.org/03/667203/4/gate/swift-tox-func-encryption/9107ed9/ : SUCCESS in 9m 09s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/03/667203/4/gate/swift-tox-func-domain-remap-staticweb/189e466/ : SUCCESS in 11m 25s\n- swift-tox-func-ec http://logs.openstack.org/03/667203/4/gate/swift-tox-func-ec/85dbaa8/ : SUCCESS in 13m 55s\n- swift-tox-func-s3api http://logs.openstack.org/03/667203/4/gate/swift-tox-func-s3api/271d90c/ : SUCCESS in 13m 54s\n- swift-tox-func-domain-remap-staticweb-py37 http://logs.openstack.org/03/667203/4/gate/swift-tox-func-domain-remap-staticweb-py37/aabf05d/ : SUCCESS in 7m 37s\n- swift-probetests-centos-7 http://logs.openstack.org/03/667203/4/gate/swift-probetests-centos-7/fbb0cea/ : SUCCESS in 37m 03s\n- swift-dsvm-functional http://logs.openstack.org/03/667203/4/gate/swift-dsvm-functional/2d2db19/ : SUCCESS in 29m 34s\n- swift-dsvm-functional-py3 http://logs.openstack.org/03/667203/4/gate/swift-dsvm-functional-py3/a8087f4/ : SUCCESS in 31m 19s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/03/667203/4/gate/swift-dsvm-functional-ipv6/3ff2245/ : SUCCESS in 25m 17s\n- swift-tox-lower-constraints http://logs.openstack.org/03/667203/4/gate/swift-tox-lower-constraints/c788689/ : SUCCESS in 14m 27s\n- openstack-tox-pep8 http://logs.openstack.org/03/667203/4/gate/openstack-tox-pep8/cb68be3/ : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":4},{"id":"3870d6f6e66bc1d8a196129299ece54598fa69ab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-13 00:34:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"541c5196e987dc6f5716765d6633b6176bd12f4e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-13 00:35:27.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- swift-promote-image http://logs.openstack.org/03/667203/4/promote/swift-promote-image/90faea9/ : SUCCESS in 59s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a1af3811a79963e2e5d1db3e5588cbc9748f9d57","revisions":{"767b6758317ac1b88f40162e7f653b911fb7371b":{"kind":"REWORK","_number":1,"created":"2019-06-24 19:32:35.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/03/667203/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/03/667203/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/667203/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/667203/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/667203/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/03/667203/1"}}},"commit":{"parents":[{"commit":"9f1ef35630ebf9da53cbd061f89889e534803725","subject":"py3: Port the tempurl middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f1ef35630ebf9da53cbd061f89889e534803725"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-06-24 19:25:33.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-06-24 19:31:22.000000000","tz":-420},"subject":"sharding: Cache shard ranges for object writes","message":"sharding: Cache shard ranges for object writes\n\nPreviously, we issued a GET to the root container for every object PUT,\nPOST, and DELETE. This puts load on the container server, potentially\nleading to timeouts, error limiting, and erroneous 404s (!).\n\nNow, cache the complete set of \u0027updating\u0027 shards, and find the shard for\nthis particular update in the proxy. Add a new config option,\nrecheck_updating_shard_ranges, to control the cache time; it defaults to\none hour. Set to 0 to fall back to previous behavior.\n\nNote that we should be able to tolerate stale shard data just fine; we\nalready have to worry about async pendings that got written down with\none shard but may not get processed until that shard has itself sharded\nor shrunk into another shard.\n\nChange-Id: Ic7a732146ea19a47669114ad5dbee0bacbe66919\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/767b6758317ac1b88f40162e7f653b911fb7371b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/767b6758317ac1b88f40162e7f653b911fb7371b"}]},"branch":"refs/heads/master"},"844b6fa06e1f8724a02b292af23cd176790667e4":{"kind":"REWORK","_number":2,"created":"2019-06-25 16:16:07.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/03/667203/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/03/667203/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/667203/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/667203/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/667203/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/03/667203/2"}}},"commit":{"parents":[{"commit":"82169ead1cc1c60318507fc3f0fbd7deff0f15e9","subject":"Don\u0027t handle object without container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/82169ead1cc1c60318507fc3f0fbd7deff0f15e9"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-06-24 19:25:33.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-06-25 16:14:35.000000000","tz":-420},"subject":"sharding: Cache shard ranges for object writes","message":"sharding: Cache shard ranges for object writes\n\nPreviously, we issued a GET to the root container for every object PUT,\nPOST, and DELETE. This puts load on the container server, potentially\nleading to timeouts, error limiting, and erroneous 404s (!).\n\nNow, cache the complete set of \u0027updating\u0027 shards, and find the shard for\nthis particular update in the proxy. Add a new config option,\nrecheck_updating_shard_ranges, to control the cache time; it defaults to\none hour. Set to 0 to fall back to previous behavior.\n\nNote that we should be able to tolerate stale shard data just fine; we\nalready have to worry about async pendings that got written down with\none shard but may not get processed until that shard has itself sharded\nor shrunk into another shard.\n\nAlso note that memcache has a default value limit of 1MiB, which may be\nexceeded if a container has thousands of shards. In that case, set()\nwill act like a delete(), causing increased memcache churn but otherwise\npreserving existing behavior. In the future, we may want to add support\nfor gzipping the cached shard ranges as they should compress well.\n\nChange-Id: Ic7a732146ea19a47669114ad5dbee0bacbe66919\nCloses-Bug: 1781291\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/844b6fa06e1f8724a02b292af23cd176790667e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/844b6fa06e1f8724a02b292af23cd176790667e4"}]},"branch":"refs/heads/master"},"30fb94fb0a32a07ebd888b7f598e1e0fb04c0c8b":{"kind":"REWORK","_number":3,"created":"2019-06-25 23:17:29.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/03/667203/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/03/667203/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/667203/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/667203/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/667203/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/03/667203/3"}}},"commit":{"parents":[{"commit":"82169ead1cc1c60318507fc3f0fbd7deff0f15e9","subject":"Don\u0027t handle object without container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/82169ead1cc1c60318507fc3f0fbd7deff0f15e9"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-06-24 19:25:33.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-06-25 23:17:11.000000000","tz":-420},"subject":"sharding: Cache shard ranges for object writes","message":"sharding: Cache shard ranges for object writes\n\nPreviously, we issued a GET to the root container for every object PUT,\nPOST, and DELETE. This puts load on the container server, potentially\nleading to timeouts, error limiting, and erroneous 404s (!).\n\nNow, cache the complete set of \u0027updating\u0027 shards, and find the shard for\nthis particular update in the proxy. Add a new config option,\nrecheck_updating_shard_ranges, to control the cache time; it defaults to\none hour. Set to 0 to fall back to previous behavior.\n\nNote that we should be able to tolerate stale shard data just fine; we\nalready have to worry about async pendings that got written down with\none shard but may not get processed until that shard has itself sharded\nor shrunk into another shard.\n\nAlso note that memcache has a default value limit of 1MiB, which may be\nexceeded if a container has thousands of shards. In that case, set()\nwill act like a delete(), causing increased memcache churn but otherwise\npreserving existing behavior. In the future, we may want to add support\nfor gzipping the cached shard ranges as they should compress well.\n\nChange-Id: Ic7a732146ea19a47669114ad5dbee0bacbe66919\nCloses-Bug: 1781291\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/30fb94fb0a32a07ebd888b7f598e1e0fb04c0c8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/30fb94fb0a32a07ebd888b7f598e1e0fb04c0c8b"}]},"branch":"refs/heads/master"},"a1af3811a79963e2e5d1db3e5588cbc9748f9d57":{"kind":"REWORK","_number":4,"created":"2019-07-11 17:40:53.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/03/667203/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/03/667203/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/667203/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/667203/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/667203/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/03/667203/4"}}},"commit":{"parents":[{"commit":"e62f07d9884c584233fd55f46c2dd79257c23bd9","subject":"Merge \"py3: port staticweb and domain_remap func tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e62f07d9884c584233fd55f46c2dd79257c23bd9"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-06-24 19:25:33.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-07-11 17:40:38.000000000","tz":-420},"subject":"sharding: Cache shard ranges for object writes","message":"sharding: Cache shard ranges for object writes\n\nPreviously, we issued a GET to the root container for every object PUT,\nPOST, and DELETE. This puts load on the container server, potentially\nleading to timeouts, error limiting, and erroneous 404s (!).\n\nNow, cache the complete set of \u0027updating\u0027 shards, and find the shard for\nthis particular update in the proxy. Add a new config option,\nrecheck_updating_shard_ranges, to control the cache time; it defaults to\none hour. Set to 0 to fall back to previous behavior.\n\nNote that we should be able to tolerate stale shard data just fine; we\nalready have to worry about async pendings that got written down with\none shard but may not get processed until that shard has itself sharded\nor shrunk into another shard.\n\nAlso note that memcache has a default value limit of 1MiB, which may be\nexceeded if a container has thousands of shards. In that case, set()\nwill act like a delete(), causing increased memcache churn but otherwise\npreserving existing behavior. In the future, we may want to add support\nfor gzipping the cached shard ranges as they should compress well.\n\nChange-Id: Ic7a732146ea19a47669114ad5dbee0bacbe66919\nCloses-Bug: 1781291\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a1af3811a79963e2e5d1db3e5588cbc9748f9d57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a1af3811a79963e2e5d1db3e5588cbc9748f9d57"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
