)]}'
{"id":"openstack%2Fswift~667411","triplet_id":"openstack%2Fswift~master~I53ed04b5de20c261ddd79c98c629580472e09961","project":"openstack/swift","branch":"master","topic":"bug/1833612","hashtags":[],"change_id":"I53ed04b5de20c261ddd79c98c629580472e09961","subject":"Return 503 when primary containers can\u0027t respond","status":"MERGED","created":"2019-06-25 16:57:59.000000000","updated":"2019-06-27 07:36:58.000000000","submitted":"2019-06-27 07:35:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667411-1561620940950-31ea3c29","meta_rev_id":"c5f134bed09be2dcb143c6519c5e1f30d331369c","_number":667411,"virtual_id_number":667411,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":2,"date":"2019-06-27 07:35:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-25 18:08:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-25 19:49:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-25 19:49:41.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2019-06-27 07:35:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"950fcd73cec23dbbe7a3261ee008ec648f21db46","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-06-25 16:57:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c4296ea22011df4232972466ae9bf59a51a42504","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-25 17:06:14.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"a42682a4827a65650d8188271bc872695bbc1909","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-06-25 17:22:50.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)\n\n#willfix","accounts_in_message":[],"_revision_number":1},{"id":"292a22e0dff23152aa3ed68b9372406815b57af1","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-06-25 17:23:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1cdd08cd4f9ef9a15ddbc57b5a72b540b95abb34","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-25 18:08:23.000000000","message":"Patch Set 2: Code-Review+2\n\n(3 comments)\n\nThere\u0027s one more case I\u0027d like to think about (primary 404 because of rebalance), but better to cover a fair number of cases *now* than all the cases eventually.","accounts_in_message":[],"_revision_number":2},{"id":"50845366952462197bef12bd37b62e05cb9a8f62","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-25 18:51:16.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"5a30dad80477b133b4e64cae04ad0e65a763801b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-25 19:49:41.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"aba0dbe66d5de6ec13031a5231e48268f47130e6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 05:42:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/11/667411/2/check/openstack-tox-docs/198c8ca/html/ : SUCCESS in 4m 03s\n- swift-build-image http://logs.openstack.org/11/667411/2/check/swift-build-image/af60a47/ : SUCCESS in 7m 03s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/11/667411/2/check/swift-tox-py27/bf84e22/ : SUCCESS in 9m 45s\n- swift-tox-py36 http://logs.openstack.org/11/667411/2/check/swift-tox-py36/483588f/ : SUCCESS in 10m 30s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/11/667411/2/check/swift-tox-py37/fab1725/ : SUCCESS in 11m 16s\n- swift-tox-func http://logs.openstack.org/11/667411/2/check/swift-tox-func/449f8bf/ : SUCCESS in 8m 41s\n- swift-tox-func-py37 http://logs.openstack.org/11/667411/2/check/swift-tox-func-py37/c82a7ba/ : SUCCESS in 9m 40s\n- swift-tox-func-encryption http://logs.openstack.org/11/667411/2/check/swift-tox-func-encryption/6273211/ : SUCCESS in 10m 26s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/11/667411/2/check/swift-tox-func-domain-remap-staticweb/bb95bc3/ : SUCCESS in 8m 42s\n- swift-tox-func-ec http://logs.openstack.org/11/667411/2/check/swift-tox-func-ec/559f354/ : SUCCESS in 9m 40s\n- swift-tox-func-s3api http://logs.openstack.org/11/667411/2/check/swift-tox-func-s3api/a18ca57/ : SUCCESS in 11m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/11/667411/2/check/swift-tox-func-s3api-ceph-s3tests-tempauth/12c72e2/ : SUCCESS in 13m 18s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/11/667411/2/check/swift-probetests-centos-7/8e04b68/ : SUCCESS in 37m 23s\n- swift-dsvm-functional http://logs.openstack.org/11/667411/2/check/swift-dsvm-functional/5a8f12a/ : SUCCESS in 25m 19s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/11/667411/2/check/swift-dsvm-functional-ipv6/5efb61a/ : SUCCESS in 25m 53s\n- swift-tox-lower-constraints http://logs.openstack.org/11/667411/2/check/swift-tox-lower-constraints/2544727/ : SUCCESS in 11m 03s\n- openstack-tox-pep8 http://logs.openstack.org/11/667411/2/check/openstack-tox-pep8/ea62214/ : SUCCESS in 5m 28s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/11/667411/2/check/swift-multinode-rolling-upgrade/141b546/ : FAILURE in 7m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6547db21600e5c9a7cbfc4f72acad01376268935","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 05:43:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d9467ed762a1209718eac25e57dd9cc4bdace93e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 06:43:49.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/11/667411/2/gate/openstack-tox-docs/ef18276/html/ : SUCCESS in 4m 34s\n- swift-upload-image http://logs.openstack.org/11/667411/2/gate/swift-upload-image/3e3064f/ : SUCCESS in 8m 41s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/11/667411/2/gate/swift-tox-py27/0fae952/ : SUCCESS in 10m 30s\n- swift-tox-py37 http://logs.openstack.org/11/667411/2/gate/swift-tox-py37/c98052e/ : SUCCESS in 11m 04s\n- swift-tox-func http://logs.openstack.org/11/667411/2/gate/swift-tox-func/387219e/ : SUCCESS in 13m 40s\n- swift-tox-func-py37 http://logs.openstack.org/11/667411/2/gate/swift-tox-func-py37/dad87e6/ : SUCCESS in 11m 04s\n- swift-tox-func-encryption http://logs.openstack.org/11/667411/2/gate/swift-tox-func-encryption/6bd46da/ : SUCCESS in 10m 06s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/11/667411/2/gate/swift-tox-func-domain-remap-staticweb/e05a74f/ : FAILURE in 14m 33s\n- swift-tox-func-ec http://logs.openstack.org/11/667411/2/gate/swift-tox-func-ec/509099f/ : SUCCESS in 9m 37s\n- swift-tox-func-s3api http://logs.openstack.org/11/667411/2/gate/swift-tox-func-s3api/515a741/ : SUCCESS in 18m 00s\n- swift-probetests-centos-7 http://logs.openstack.org/11/667411/2/gate/swift-probetests-centos-7/4d84f7f/ : FAILURE in 38m 50s\n- swift-dsvm-functional http://logs.openstack.org/11/667411/2/gate/swift-dsvm-functional/2c9eae9/ : SUCCESS in 29m 46s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/11/667411/2/gate/swift-dsvm-functional-ipv6/7be1a36/ : SUCCESS in 30m 47s\n- swift-tox-lower-constraints http://logs.openstack.org/11/667411/2/gate/swift-tox-lower-constraints/68c666b/ : FAILURE in 13m 45s\n- openstack-tox-pep8 http://logs.openstack.org/11/667411/2/gate/openstack-tox-pep8/faaadbd/ : SUCCESS in 6m 56s","accounts_in_message":[],"_revision_number":2},{"id":"bc4368bd6917181f10ed90c7806759c06734ecc6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-26 22:08:18.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"91c3690007f14da83f952e4f1345c5421d16d752","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 04:57:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/11/667411/2/check/openstack-tox-docs/a6a740d/html/ : SUCCESS in 3m 54s\n- swift-build-image http://logs.openstack.org/11/667411/2/check/swift-build-image/a7413c8/ : SUCCESS in 7m 26s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/11/667411/2/check/swift-tox-py27/676f9b6/ : SUCCESS in 10m 31s\n- swift-tox-py36 http://logs.openstack.org/11/667411/2/check/swift-tox-py36/9f9ffdb/ : SUCCESS in 10m 15s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/11/667411/2/check/swift-tox-py37/fb26aac/ : SUCCESS in 10m 09s\n- swift-tox-func http://logs.openstack.org/11/667411/2/check/swift-tox-func/46a4a1e/ : SUCCESS in 9m 04s\n- swift-tox-func-py37 http://logs.openstack.org/11/667411/2/check/swift-tox-func-py37/da423a5/ : SUCCESS in 9m 52s\n- swift-tox-func-encryption http://logs.openstack.org/11/667411/2/check/swift-tox-func-encryption/5d01a2e/ : SUCCESS in 10m 06s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/11/667411/2/check/swift-tox-func-domain-remap-staticweb/219a1b1/ : SUCCESS in 9m 04s\n- swift-tox-func-ec http://logs.openstack.org/11/667411/2/check/swift-tox-func-ec/3e95c70/ : SUCCESS in 10m 00s\n- swift-tox-func-s3api http://logs.openstack.org/11/667411/2/check/swift-tox-func-s3api/83392b6/ : SUCCESS in 11m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/11/667411/2/check/swift-tox-func-s3api-ceph-s3tests-tempauth/c2180ce/ : SUCCESS in 12m 27s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/11/667411/2/check/swift-probetests-centos-7/f68b668/ : SUCCESS in 31m 53s\n- swift-dsvm-functional http://logs.openstack.org/11/667411/2/check/swift-dsvm-functional/e695037/ : SUCCESS in 29m 24s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/11/667411/2/check/swift-dsvm-functional-ipv6/02c1de7/ : SUCCESS in 30m 41s\n- swift-tox-lower-constraints http://logs.openstack.org/11/667411/2/check/swift-tox-lower-constraints/b148d7d/ : SUCCESS in 11m 20s\n- openstack-tox-pep8 http://logs.openstack.org/11/667411/2/check/openstack-tox-pep8/dd01b4f/ : SUCCESS in 5m 05s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/11/667411/2/check/swift-multinode-rolling-upgrade/709c005/ : FAILURE in 8m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8292163c2fcfb0f113240bfaeea6d468c3109874","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 04:57:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6274e93bd2942489cc6f41b2a0b45140769daffd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 07:35:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"fb344d230537f3888f0424536ed601a0e5fe47b8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 07:35:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/11/667411/2/gate/openstack-tox-docs/fe23981/html/ : SUCCESS in 6m 23s\n- swift-upload-image http://logs.openstack.org/11/667411/2/gate/swift-upload-image/fb617c2/ : SUCCESS in 11m 00s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/11/667411/2/gate/swift-tox-py27/b37201f/ : SUCCESS in 19m 48s\n- swift-tox-py37 http://logs.openstack.org/11/667411/2/gate/swift-tox-py37/c9c0811/ : SUCCESS in 19m 23s\n- swift-tox-func http://logs.openstack.org/11/667411/2/gate/swift-tox-func/f7ca0ea/ : SUCCESS in 17m 59s\n- swift-tox-func-py37 http://logs.openstack.org/11/667411/2/gate/swift-tox-func-py37/58bcd1a/ : SUCCESS in 10m 31s\n- swift-tox-func-encryption http://logs.openstack.org/11/667411/2/gate/swift-tox-func-encryption/411efa4/ : SUCCESS in 10m 01s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/11/667411/2/gate/swift-tox-func-domain-remap-staticweb/cd5fba5/ : SUCCESS in 10m 33s\n- swift-tox-func-ec http://logs.openstack.org/11/667411/2/gate/swift-tox-func-ec/087226b/ : SUCCESS in 11m 30s\n- swift-tox-func-s3api http://logs.openstack.org/11/667411/2/gate/swift-tox-func-s3api/5d21c2d/ : SUCCESS in 18m 03s\n- swift-probetests-centos-7 http://logs.openstack.org/11/667411/2/gate/swift-probetests-centos-7/abb8bcc/ : SUCCESS in 36m 13s\n- swift-dsvm-functional http://logs.openstack.org/11/667411/2/gate/swift-dsvm-functional/f29c40a/ : SUCCESS in 28m 17s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/11/667411/2/gate/swift-dsvm-functional-ipv6/b586066/ : SUCCESS in 26m 47s\n- swift-tox-lower-constraints http://logs.openstack.org/11/667411/2/gate/swift-tox-lower-constraints/1374fc1/ : SUCCESS in 11m 05s\n- openstack-tox-pep8 http://logs.openstack.org/11/667411/2/gate/openstack-tox-pep8/697f8e9/ : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":2},{"id":"f6f5f51ccf82a1e4b76b69ef93d794c305a9980f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 07:36:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- swift-promote-image http://logs.openstack.org/11/667411/2/promote/swift-promote-image/9d584cd/ : SUCCESS in 52s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"563e1671cf0583767d652a72ebc4d73d7cc07435","revisions":{"8d44c23f1ede85dcc36028237bcce1a01cf04831":{"kind":"REWORK","_number":1,"created":"2019-06-25 16:57:59.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/11/667411/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/667411/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/667411/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/667411/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/667411/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/667411/1"}}},"commit":{"parents":[{"commit":"82169ead1cc1c60318507fc3f0fbd7deff0f15e9","subject":"Don\u0027t handle object without container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/82169ead1cc1c60318507fc3f0fbd7deff0f15e9"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2019-06-25 16:53:32.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2019-06-25 16:57:46.000000000","tz":-300},"subject":"Return 503 when primary containers can\u0027t respond","message":"Return 503 when primary containers can\u0027t respond\n\nCloses-Bug: #1833612\n\nChange-Id: I53ed04b5de20c261ddd79c98c629580472e09961\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8d44c23f1ede85dcc36028237bcce1a01cf04831"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8d44c23f1ede85dcc36028237bcce1a01cf04831"}]},"branch":"refs/heads/master"},"563e1671cf0583767d652a72ebc4d73d7cc07435":{"kind":"REWORK","_number":2,"created":"2019-06-25 17:23:16.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/11/667411/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/667411/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/667411/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/667411/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/667411/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/667411/2"}}},"commit":{"parents":[{"commit":"82169ead1cc1c60318507fc3f0fbd7deff0f15e9","subject":"Don\u0027t handle object without container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/82169ead1cc1c60318507fc3f0fbd7deff0f15e9"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2019-06-25 16:53:32.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2019-06-25 17:23:12.000000000","tz":-300},"subject":"Return 503 when primary containers can\u0027t respond","message":"Return 503 when primary containers can\u0027t respond\n\nCloses-Bug: #1833612\n\nChange-Id: I53ed04b5de20c261ddd79c98c629580472e09961\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/563e1671cf0583767d652a72ebc4d73d7cc07435"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/563e1671cf0583767d652a72ebc4d73d7cc07435"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
