)]}'
{"id":"openstack%2Fswift~667438","triplet_id":"openstack%2Fswift~master~I30f5af07e2e1fc7cbb6bdb1c334a0a161caf0906","project":"openstack/swift","branch":"master","topic":"lp1833616","hashtags":[],"change_id":"I30f5af07e2e1fc7cbb6bdb1c334a0a161caf0906","subject":"More tests for 404 handoff skipping","status":"MERGED","created":"2019-06-25 18:43:32.000000000","updated":"2019-06-27 07:37:56.000000000","submitted":"2019-06-27 07:35:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667438-1561620942913-e8d46088","meta_rev_id":"85528ce641b0e6187bda318ca384f381308a35e6","_number":667438,"virtual_id_number":667438,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-06-27 07:35:42.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-25 18:49:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-25 18:49:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-25 18:49:49.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2019-06-27 07:35:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4d0e816a2644cfeb633fe12b9972716cd1c02e61","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-06-25 18:43:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c7ce25261746bbc01a43fc4346ccbe074a08b808","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-25 18:49:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)\n\nMore tests are good.","accounts_in_message":[],"_revision_number":1},{"id":"6d7721d5347fdd5f7c2a5566d7da28432a0d3818","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 08:37:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/38/667438/1/check/openstack-tox-docs/52be382/html/ : SUCCESS in 4m 14s\n- swift-build-image http://logs.openstack.org/38/667438/1/check/swift-build-image/712a548/ : SUCCESS in 8m 10s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/38/667438/1/check/swift-tox-py27/bb2ceb7/ : SUCCESS in 10m 54s\n- swift-tox-py36 http://logs.openstack.org/38/667438/1/check/swift-tox-py36/16db446/ : SUCCESS in 16m 43s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/38/667438/1/check/swift-tox-py37/49808e4/ : SUCCESS in 11m 10s\n- swift-tox-func http://logs.openstack.org/38/667438/1/check/swift-tox-func/7132368/ : SUCCESS in 9m 01s\n- swift-tox-func-py37 http://logs.openstack.org/38/667438/1/check/swift-tox-func-py37/af9ea78/ : SUCCESS in 9m 18s\n- swift-tox-func-encryption http://logs.openstack.org/38/667438/1/check/swift-tox-func-encryption/70a581b/ : SUCCESS in 9m 20s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/38/667438/1/check/swift-tox-func-domain-remap-staticweb/1b3bdae/ : SUCCESS in 10m 49s\n- swift-tox-func-ec http://logs.openstack.org/38/667438/1/check/swift-tox-func-ec/7409040/ : SUCCESS in 9m 13s\n- swift-tox-func-s3api http://logs.openstack.org/38/667438/1/check/swift-tox-func-s3api/bfebfd0/ : SUCCESS in 12m 22s\n- swift-dsvm-functional http://logs.openstack.org/38/667438/1/check/swift-dsvm-functional/39ba3ac/ : SUCCESS in 33m 59s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/38/667438/1/check/swift-dsvm-functional-ipv6/daf4b22/ : SUCCESS in 23m 40s\n- swift-tox-lower-constraints http://logs.openstack.org/38/667438/1/check/swift-tox-lower-constraints/0e23e5a/ : SUCCESS in 13m 11s\n- openstack-tox-pep8 http://logs.openstack.org/38/667438/1/check/openstack-tox-pep8/0ed1e2a/ : SUCCESS in 5m 48s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/38/667438/1/check/swift-multinode-rolling-upgrade/23f3c32/ : FAILURE in 7m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"775a605435072067d18577f3da24de504201c265","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 04:58:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4347c7baff213da8336c55ad2191f84cb57579a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 07:35:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"84df9520ea8f83d7c64fa1b3b7132470ef751d0e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 07:35:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/38/667438/1/gate/openstack-tox-docs/bbf882f/html/ : SUCCESS in 5m 16s\n- swift-upload-image http://logs.openstack.org/38/667438/1/gate/swift-upload-image/8250ca4/ : SUCCESS in 9m 52s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/38/667438/1/gate/swift-tox-py27/1a6699e/ : SUCCESS in 10m 40s\n- swift-tox-py37 http://logs.openstack.org/38/667438/1/gate/swift-tox-py37/814dac6/ : SUCCESS in 10m 55s\n- swift-tox-func http://logs.openstack.org/38/667438/1/gate/swift-tox-func/251d318/ : SUCCESS in 8m 44s\n- swift-tox-func-py37 http://logs.openstack.org/38/667438/1/gate/swift-tox-func-py37/c0d79e5/ : SUCCESS in 10m 33s\n- swift-tox-func-encryption http://logs.openstack.org/38/667438/1/gate/swift-tox-func-encryption/8fe866b/ : SUCCESS in 9m 32s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/38/667438/1/gate/swift-tox-func-domain-remap-staticweb/8b90951/ : SUCCESS in 9m 11s\n- swift-tox-func-ec http://logs.openstack.org/38/667438/1/gate/swift-tox-func-ec/d018ba6/ : SUCCESS in 8m 56s\n- swift-tox-func-s3api http://logs.openstack.org/38/667438/1/gate/swift-tox-func-s3api/eedb1c8/ : SUCCESS in 12m 53s\n- swift-dsvm-functional http://logs.openstack.org/38/667438/1/gate/swift-dsvm-functional/dded290/ : SUCCESS in 25m 44s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/38/667438/1/gate/swift-dsvm-functional-ipv6/1057740/ : SUCCESS in 29m 03s\n- swift-tox-lower-constraints http://logs.openstack.org/38/667438/1/gate/swift-tox-lower-constraints/f12a389/ : SUCCESS in 11m 25s\n- openstack-tox-pep8 http://logs.openstack.org/38/667438/1/gate/openstack-tox-pep8/c381fba/ : SUCCESS in 6m 01s","accounts_in_message":[],"_revision_number":1},{"id":"cdd0ca9bc0e4aaefeaf8c8b20a756a603f9aeabc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 07:37:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- swift-promote-image http://logs.openstack.org/38/667438/1/promote/swift-promote-image/2cb385d/ : SUCCESS in 47s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"044c919871f80e1df91ff9a40318a5afd6a2960d","revisions":{"044c919871f80e1df91ff9a40318a5afd6a2960d":{"kind":"REWORK","_number":1,"created":"2019-06-25 18:43:32.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/38/667438/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/38/667438/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/667438/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/667438/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/667438/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/38/667438/1"}}},"commit":{"parents":[{"commit":"563e1671cf0583767d652a72ebc4d73d7cc07435","subject":"Return 503 when primary containers can\u0027t respond","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/563e1671cf0583767d652a72ebc4d73d7cc07435"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2019-06-25 18:42:26.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2019-06-25 18:42:26.000000000","tz":-300},"subject":"More tests for 404 handoff skipping","message":"More tests for 404 handoff skipping\n\nSometimes we want 404, sometimes we want 503 - it\u0027s tricky\n\nChange-Id: I30f5af07e2e1fc7cbb6bdb1c334a0a161caf0906\nRelated-Change-Id: I53ed04b5de20c261ddd79c98c629580472e09961\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/044c919871f80e1df91ff9a40318a5afd6a2960d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/044c919871f80e1df91ff9a40318a5afd6a2960d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
