)]}'
{"id":"openstack%2Fswift~668007","triplet_id":"openstack%2Fswift~feature%2Flosf~I2aa34aa57196d27ca325750e967e76c47b799bf6","project":"openstack/swift","branch":"feature/losf","topic":"merge-from-master","hashtags":[],"change_id":"I2aa34aa57196d27ca325750e967e76c47b799bf6","subject":"Merge branch \u0027master\u0027 into feature/losf","status":"MERGED","created":"2019-06-27 17:55:53.000000000","updated":"2019-07-11 23:21:53.000000000","submitted":"2019-07-11 23:19:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668007-1562887177002-dffa1978","meta_rev_id":"b2b9aa98dbede1fc123ccf27bf250e9ff09f689a","_number":668007,"virtual_id_number":668007,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-07-11 23:19:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-11 16:17:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-11 16:17:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-11 23:19:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3a64f8cb4f03a9066d984a85e51535f4dd7e3ac9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-27 17:55:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2a8e8c8a508a198f00e17275a6677b74771223d7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 01:27:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/07/668007/1/check/openstack-tox-docs/666459a/html/ : SUCCESS in 4m 40s\n- build-openstack-releasenotes http://logs.openstack.org/07/668007/1/check/build-openstack-releasenotes/c309113/html/ : SUCCESS in 8m 39s\n- swift-build-image http://logs.openstack.org/07/668007/1/check/swift-build-image/513bab6/ : SUCCESS in 13m 30s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/07/668007/1/check/swift-tox-py27/37c45cb/ : SUCCESS in 15m 05s\n- swift-tox-py36 http://logs.openstack.org/07/668007/1/check/swift-tox-py36/a20db04/ : SUCCESS in 10m 09s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/07/668007/1/check/swift-tox-py37/10948ad/ : SUCCESS in 12m 02s\n- swift-tox-func http://logs.openstack.org/07/668007/1/check/swift-tox-func/a30fea4/ : SUCCESS in 9m 22s\n- swift-tox-func-py37 http://logs.openstack.org/07/668007/1/check/swift-tox-func-py37/78b484e/ : SUCCESS in 10m 28s\n- swift-tox-func-encryption http://logs.openstack.org/07/668007/1/check/swift-tox-func-encryption/c096de5/ : SUCCESS in 9m 02s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/07/668007/1/check/swift-tox-func-domain-remap-staticweb/ef6eb20/ : SUCCESS in 10m 23s\n- swift-tox-func-ec http://logs.openstack.org/07/668007/1/check/swift-tox-func-ec/29081bc/ : SUCCESS in 9m 43s\n- swift-tox-func-s3api http://logs.openstack.org/07/668007/1/check/swift-tox-func-s3api/2ed0b1e/ : SUCCESS in 11m 44s\n- swift-tox-func-losf http://logs.openstack.org/07/668007/1/check/swift-tox-func-losf/9e504ea/ : SUCCESS in 15m 03s (non-voting)\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/07/668007/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/af29c69/ : SUCCESS in 15m 28s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/07/668007/1/check/swift-probetests-centos-7/3d353e8/ : SUCCESS in 35m 04s\n- swift-dsvm-functional http://logs.openstack.org/07/668007/1/check/swift-dsvm-functional/741dac7/ : RETRY_LIMIT in 17m 05s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/07/668007/1/check/swift-dsvm-functional-ipv6/70fed86/ : RETRY_LIMIT in 16m 47s\n- swift-tox-lower-constraints http://logs.openstack.org/07/668007/1/check/swift-tox-lower-constraints/ca6243b/ : SUCCESS in 12m 06s\n- openstack-tox-pep8 http://logs.openstack.org/07/668007/1/check/openstack-tox-pep8/328773c/ : SUCCESS in 6m 47s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/07/668007/1/check/swift-multinode-rolling-upgrade/6d8978e/ : FAILURE in 9m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7a0094d1c13550aa3e8cb8ba43c88e5ba5bd27b0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-07-11 06:00:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e2bcb88cb71639b013f00dca3af950fa1b02f892","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 07:21:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/07/668007/2/check/openstack-tox-docs/fce994a/html/ : SUCCESS in 5m 38s\n- build-openstack-releasenotes http://logs.openstack.org/07/668007/2/check/build-openstack-releasenotes/cc445ad/html/ : SUCCESS in 7m 36s\n- swift-build-image http://logs.openstack.org/07/668007/2/check/swift-build-image/5e41bb9/ : SUCCESS in 14m 53s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/07/668007/2/check/swift-tox-py27/030b177/ : SUCCESS in 14m 39s\n- swift-tox-py36 http://logs.openstack.org/07/668007/2/check/swift-tox-py36/c838799/ : SUCCESS in 15m 09s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/07/668007/2/check/swift-tox-py37/4c4fb20/ : SUCCESS in 16m 16s\n- swift-tox-func http://logs.openstack.org/07/668007/2/check/swift-tox-func/b25c1a4/ : SUCCESS in 13m 22s\n- swift-tox-func-py37 http://logs.openstack.org/07/668007/2/check/swift-tox-func-py37/370f1e2/ : SUCCESS in 8m 59s\n- swift-tox-func-encryption http://logs.openstack.org/07/668007/2/check/swift-tox-func-encryption/2693729/ : SUCCESS in 10m 33s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/07/668007/2/check/swift-tox-func-domain-remap-staticweb/16c1dcd/ : SUCCESS in 9m 45s\n- swift-tox-func-ec http://logs.openstack.org/07/668007/2/check/swift-tox-func-ec/bb8f7d4/ : SUCCESS in 11m 31s\n- swift-tox-func-s3api http://logs.openstack.org/07/668007/2/check/swift-tox-func-s3api/d39c6cd/ : SUCCESS in 13m 20s\n- swift-tox-func-losf http://logs.openstack.org/07/668007/2/check/swift-tox-func-losf/9aaa1cf/ : SUCCESS in 13m 24s (non-voting)\n- swift-tox-func-domain-remap-staticweb-py37 http://logs.openstack.org/07/668007/2/check/swift-tox-func-domain-remap-staticweb-py37/378edf3/ : SUCCESS in 8m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/07/668007/2/check/swift-tox-func-s3api-ceph-s3tests-tempauth/c5715cd/ : SUCCESS in 14m 53s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/07/668007/2/check/swift-probetests-centos-7/3fdb4b8/ : SUCCESS in 35m 15s\n- swift-dsvm-functional http://logs.openstack.org/07/668007/2/check/swift-dsvm-functional/1c8fff6/ : SUCCESS in 29m 49s\n- swift-dsvm-functional-py3 http://logs.openstack.org/07/668007/2/check/swift-dsvm-functional-py3/edbf1ce/ : SUCCESS in 34m 25s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/07/668007/2/check/swift-dsvm-functional-ipv6/146c786/ : SUCCESS in 32m 43s\n- swift-tox-lower-constraints http://logs.openstack.org/07/668007/2/check/swift-tox-lower-constraints/2a95aab/ : SUCCESS in 12m 06s\n- openstack-tox-pep8 http://logs.openstack.org/07/668007/2/check/openstack-tox-pep8/b756bd3/ : SUCCESS in 6m 22s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/07/668007/2/check/swift-multinode-rolling-upgrade/f5ed17e/ : FAILURE in 12m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"498e7cfd0a80960661aa1d5b3317a45c636fcd4c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-07-11 16:17:24.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5e0362e4e22c630f1ebd0e0701ccfe318506a48c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 16:17:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0ea816bcbd08bca131c4214196d829915f083007","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 19:00:07.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/07/668007/2/gate/openstack-tox-docs/0f34336/html/ : SUCCESS in 6m 22s\n- build-openstack-releasenotes http://logs.openstack.org/07/668007/2/gate/build-openstack-releasenotes/7bff476/html/ : SUCCESS in 8m 47s\n- swift-upload-image http://logs.openstack.org/07/668007/2/gate/swift-upload-image/7babeb6/ : SUCCESS in 15m 14s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/07/668007/2/gate/swift-tox-py27/e9934ea/ : SUCCESS in 10m 49s\n- swift-tox-py37 http://logs.openstack.org/07/668007/2/gate/swift-tox-py37/a7c720e/ : SUCCESS in 12m 39s\n- swift-tox-func http://logs.openstack.org/07/668007/2/gate/swift-tox-func/cc732c7/ : SUCCESS in 9m 59s\n- swift-tox-func-py37 http://logs.openstack.org/07/668007/2/gate/swift-tox-func-py37/da3ae06/ : SUCCESS in 10m 45s\n- swift-tox-func-encryption http://logs.openstack.org/07/668007/2/gate/swift-tox-func-encryption/0e76b55/ : SUCCESS in 10m 28s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/07/668007/2/gate/swift-tox-func-domain-remap-staticweb/ef0f8b6/ : SUCCESS in 8m 42s\n- swift-tox-func-ec http://logs.openstack.org/07/668007/2/gate/swift-tox-func-ec/174db3f/ : SUCCESS in 9m 39s\n- swift-tox-func-s3api http://logs.openstack.org/07/668007/2/gate/swift-tox-func-s3api/444cab8/ : SUCCESS in 15m 32s\n- swift-tox-func-domain-remap-staticweb-py37 http://logs.openstack.org/07/668007/2/gate/swift-tox-func-domain-remap-staticweb-py37/0b53df1/ : SUCCESS in 8m 55s\n- swift-probetests-centos-7 http://logs.openstack.org/07/668007/2/gate/swift-probetests-centos-7/73ed7da/ : FAILURE in 33m 24s\n- swift-dsvm-functional http://logs.openstack.org/07/668007/2/gate/swift-dsvm-functional/c8d3c15/ : SUCCESS in 30m 30s\n- swift-dsvm-functional-py3 http://logs.openstack.org/07/668007/2/gate/swift-dsvm-functional-py3/4d3ae3c/ : SUCCESS in 32m 21s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/07/668007/2/gate/swift-dsvm-functional-ipv6/98f9879/ : SUCCESS in 30m 17s\n- swift-tox-lower-constraints http://logs.openstack.org/07/668007/2/gate/swift-tox-lower-constraints/a3f1e69/ : SUCCESS in 12m 17s\n- openstack-tox-pep8 http://logs.openstack.org/07/668007/2/gate/openstack-tox-pep8/6afd5fc/ : SUCCESS in 6m 21s","accounts_in_message":[],"_revision_number":2},{"id":"90ae3c2abe1d57677a6d69910d17a13ee4748689","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-07-11 19:16:19.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7273c1ef61e3471211059f1b7a2f458e03a23030","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 21:58:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/07/668007/2/check/test-release-openstack/e37e6a5/ : SUCCESS in 3m 21s\n- openstack-tox-docs http://logs.openstack.org/07/668007/2/check/openstack-tox-docs/02a458b/html/ : SUCCESS in 4m 38s\n- build-openstack-releasenotes http://logs.openstack.org/07/668007/2/check/build-openstack-releasenotes/f7df510/html/ : SUCCESS in 8m 18s\n- swift-build-image http://logs.openstack.org/07/668007/2/check/swift-build-image/2356d09/ : SUCCESS in 13m 16s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/07/668007/2/check/swift-tox-py27/91a61b7/ : SUCCESS in 10m 30s\n- swift-tox-py36 http://logs.openstack.org/07/668007/2/check/swift-tox-py36/bc7b65c/ : SUCCESS in 10m 32s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/07/668007/2/check/swift-tox-py37/1d01afe/ : SUCCESS in 11m 51s\n- swift-tox-func http://logs.openstack.org/07/668007/2/check/swift-tox-func/0fda1a7/ : SUCCESS in 12m 40s\n- swift-tox-func-py37 http://logs.openstack.org/07/668007/2/check/swift-tox-func-py37/2a97268/ : SUCCESS in 10m 23s\n- swift-tox-func-encryption http://logs.openstack.org/07/668007/2/check/swift-tox-func-encryption/cb79c7b/ : SUCCESS in 8m 30s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/07/668007/2/check/swift-tox-func-domain-remap-staticweb/668bcfb/ : SUCCESS in 8m 28s\n- swift-tox-func-ec http://logs.openstack.org/07/668007/2/check/swift-tox-func-ec/5b45548/ : SUCCESS in 9m 28s\n- swift-tox-func-s3api http://logs.openstack.org/07/668007/2/check/swift-tox-func-s3api/8015dfc/ : SUCCESS in 11m 06s\n- swift-tox-func-losf http://logs.openstack.org/07/668007/2/check/swift-tox-func-losf/8b34b3a/ : SUCCESS in 15m 48s (non-voting)\n- swift-tox-func-domain-remap-staticweb-py37 http://logs.openstack.org/07/668007/2/check/swift-tox-func-domain-remap-staticweb-py37/544b140/ : SUCCESS in 8m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/07/668007/2/check/swift-tox-func-s3api-ceph-s3tests-tempauth/e02335d/ : SUCCESS in 13m 34s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/07/668007/2/check/swift-probetests-centos-7/6d3254f/ : SUCCESS in 41m 36s\n- swift-dsvm-functional http://logs.openstack.org/07/668007/2/check/swift-dsvm-functional/0be111c/ : SUCCESS in 32m 03s\n- swift-dsvm-functional-py3 http://logs.openstack.org/07/668007/2/check/swift-dsvm-functional-py3/ff9f7a9/ : SUCCESS in 32m 10s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/07/668007/2/check/swift-dsvm-functional-ipv6/3720825/ : SUCCESS in 31m 11s\n- swift-tox-lower-constraints http://logs.openstack.org/07/668007/2/check/swift-tox-lower-constraints/cadaacd/ : SUCCESS in 11m 23s\n- openstack-tox-pep8 http://logs.openstack.org/07/668007/2/check/openstack-tox-pep8/95e1e9e/ : SUCCESS in 5m 48s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/07/668007/2/check/swift-multinode-rolling-upgrade/113dc2d/ : FAILURE in 8m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4f63566e8785eb042e230eaa73777c9226807f4a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 21:58:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2cfbc332f79293a08febab8527d24ea1beca1fab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 23:19:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/07/668007/2/gate/openstack-tox-docs/730f3bf/html/ : SUCCESS in 4m 30s\n- build-openstack-releasenotes http://logs.openstack.org/07/668007/2/gate/build-openstack-releasenotes/b576c2f/html/ : SUCCESS in 7m 49s\n- swift-upload-image http://logs.openstack.org/07/668007/2/gate/swift-upload-image/5ae21e4/ : SUCCESS in 16m 34s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/07/668007/2/gate/swift-tox-py27/bdc24e7/ : SUCCESS in 10m 56s\n- swift-tox-py37 http://logs.openstack.org/07/668007/2/gate/swift-tox-py37/beec324/ : SUCCESS in 12m 52s\n- swift-tox-func http://logs.openstack.org/07/668007/2/gate/swift-tox-func/08932ad/ : SUCCESS in 10m 47s\n- swift-tox-func-py37 http://logs.openstack.org/07/668007/2/gate/swift-tox-func-py37/49c301e/ : SUCCESS in 9m 38s\n- swift-tox-func-encryption http://logs.openstack.org/07/668007/2/gate/swift-tox-func-encryption/dcef8ed/ : SUCCESS in 10m 19s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/07/668007/2/gate/swift-tox-func-domain-remap-staticweb/b98cf63/ : SUCCESS in 8m 48s\n- swift-tox-func-ec http://logs.openstack.org/07/668007/2/gate/swift-tox-func-ec/93f3cf8/ : SUCCESS in 10m 57s\n- swift-tox-func-s3api http://logs.openstack.org/07/668007/2/gate/swift-tox-func-s3api/473e3c2/ : SUCCESS in 13m 29s\n- swift-tox-func-domain-remap-staticweb-py37 http://logs.openstack.org/07/668007/2/gate/swift-tox-func-domain-remap-staticweb-py37/f8138f8/ : SUCCESS in 8m 14s\n- swift-probetests-centos-7 http://logs.openstack.org/07/668007/2/gate/swift-probetests-centos-7/a7dfecb/ : SUCCESS in 35m 02s\n- swift-dsvm-functional http://logs.openstack.org/07/668007/2/gate/swift-dsvm-functional/14b4942/ : SUCCESS in 26m 53s\n- swift-dsvm-functional-py3 http://logs.openstack.org/07/668007/2/gate/swift-dsvm-functional-py3/a2d6a1d/ : SUCCESS in 36m 21s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/07/668007/2/gate/swift-dsvm-functional-ipv6/2edfcbe/ : SUCCESS in 24m 44s\n- swift-tox-lower-constraints http://logs.openstack.org/07/668007/2/gate/swift-tox-lower-constraints/ff4695f/ : SUCCESS in 12m 42s\n- openstack-tox-pep8 http://logs.openstack.org/07/668007/2/gate/openstack-tox-pep8/0d71a11/ : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":2},{"id":"79089fa4abb30e0a67af37c4efd4d9e43c871ea6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 23:19:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"c89db4da6feb114e7e618ac33d395d8132e4db74","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 23:21:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- swift-promote-image http://logs.openstack.org/07/668007/2/promote/swift-promote-image/68ae207/ : SUCCESS in 55s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"aca7474ca3c7bba54473fca35b6d88a3a1efdca4","revisions":{"0e7617e3af84d1066db3b9e1a096393222573289":{"kind":"REWORK","_number":1,"created":"2019-06-27 17:55:53.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/07/668007/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/07/668007/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/07/668007/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/07/668007/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/07/668007/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/07/668007/1"}}},"commit":{"parents":[{"commit":"a353d420dd9c80d5a211d1d9172f3961767fd75c","subject":"Get new unit tests running under py3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a353d420dd9c80d5a211d1d9172f3961767fd75c"}]},{"commit":"044c919871f80e1df91ff9a40318a5afd6a2960d","subject":"More tests for 404 handoff skipping","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/044c919871f80e1df91ff9a40318a5afd6a2960d"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-06-27 17:54:59.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-06-27 17:54:59.000000000","tz":-420},"subject":"Merge remote-tracking branch \u0027gerrit/master\u0027 into review/tim_burke/py3","message":"Merge remote-tracking branch \u0027gerrit/master\u0027 into review/tim_burke/py3\n\nChange-Id: I2aa34aa57196d27ca325750e967e76c47b799bf6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0e7617e3af84d1066db3b9e1a096393222573289"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0e7617e3af84d1066db3b9e1a096393222573289"}]},"branch":"refs/heads/feature/losf"},"aca7474ca3c7bba54473fca35b6d88a3a1efdca4":{"kind":"REWORK","_number":2,"created":"2019-07-11 06:00:05.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/07/668007/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/07/668007/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/07/668007/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/07/668007/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/07/668007/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/07/668007/2"}}},"commit":{"parents":[{"commit":"52cf2a852e8c9fd8236f92abcf5e5241c9bebff5","subject":"Merge \"Get new unit tests running under py3\" into feature/losf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/52cf2a852e8c9fd8236f92abcf5e5241c9bebff5"}]},{"commit":"e62f07d9884c584233fd55f46c2dd79257c23bd9","subject":"Merge \"py3: port staticweb and domain_remap func tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e62f07d9884c584233fd55f46c2dd79257c23bd9"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-07-11 05:57:41.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-07-11 05:59:30.000000000","tz":-420},"subject":"Merge branch \u0027master\u0027 into feature/losf","message":"Merge branch \u0027master\u0027 into feature/losf\n\nChange-Id: I2aa34aa57196d27ca325750e967e76c47b799bf6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aca7474ca3c7bba54473fca35b6d88a3a1efdca4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aca7474ca3c7bba54473fca35b6d88a3a1efdca4"}]},"branch":"refs/heads/feature/losf"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
