)]}'
{"id":"openstack%2Fswift~668031","triplet_id":"openstack%2Fswift~master~I08e9f7c0643ae0469bd1937a30c1c4477b67c31f","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I08e9f7c0643ae0469bd1937a30c1c4477b67c31f","subject":"s3api: Put more information in the str() for InternalError","status":"MERGED","created":"2019-06-27 19:32:24.000000000","updated":"2019-07-01 21:26:23.000000000","submitted":"2019-07-01 21:24:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668031-1562016245754-a98d8c21","meta_rev_id":"ca8b04f0edc16db4646d9304442bcc6812cec1fa","_number":668031,"virtual_id_number":668031,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":2,"date":"2019-07-01 21:24:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2019-07-01 20:24:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2019-07-01 20:24:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-01 20:24:49.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2019-07-01 21:24:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dd479d2d8e15847d3e7ea28aab1ccf3f6d136803","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-27 19:32:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4bd6eb9d9eb4d0483b7d1401a17770aff15581b2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 01:10:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/31/668031/1/check/openstack-tox-docs/bb0a01d/html/ : SUCCESS in 4m 13s\n- swift-build-image http://logs.openstack.org/31/668031/1/check/swift-build-image/c89bc97/ : SUCCESS in 7m 45s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/31/668031/1/check/swift-tox-py27/e07ea8c/ : SUCCESS in 10m 23s\n- swift-tox-py36 http://logs.openstack.org/31/668031/1/check/swift-tox-py36/3cc0486/ : SUCCESS in 14m 42s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/31/668031/1/check/swift-tox-py37/c3961b8/ : SUCCESS in 14m 35s\n- swift-tox-func http://logs.openstack.org/31/668031/1/check/swift-tox-func/245d3ef/ : SUCCESS in 8m 34s\n- swift-tox-func-py37 http://logs.openstack.org/31/668031/1/check/swift-tox-func-py37/7275012/ : SUCCESS in 8m 48s\n- swift-tox-func-encryption http://logs.openstack.org/31/668031/1/check/swift-tox-func-encryption/7b117e0/ : SUCCESS in 8m 42s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/31/668031/1/check/swift-tox-func-domain-remap-staticweb/acb4022/ : SUCCESS in 7m 53s\n- swift-tox-func-ec http://logs.openstack.org/31/668031/1/check/swift-tox-func-ec/0c3887c/ : SUCCESS in 8m 16s\n- swift-tox-func-s3api http://logs.openstack.org/31/668031/1/check/swift-tox-func-s3api/64f9d29/ : SUCCESS in 11m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/31/668031/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/e5fecb9/ : SUCCESS in 12m 27s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/31/668031/1/check/swift-probetests-centos-7/45c8573/ : SUCCESS in 30m 52s\n- swift-dsvm-functional http://logs.openstack.org/31/668031/1/check/swift-dsvm-functional/b329bac/ : SUCCESS in 27m 48s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/31/668031/1/check/swift-dsvm-functional-ipv6/e5ab0ff/ : SUCCESS in 29m 20s\n- swift-tox-lower-constraints http://logs.openstack.org/31/668031/1/check/swift-tox-lower-constraints/0afe927/ : SUCCESS in 10m 54s\n- openstack-tox-pep8 http://logs.openstack.org/31/668031/1/check/openstack-tox-pep8/f116def/ : SUCCESS in 5m 34s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/31/668031/1/check/swift-multinode-rolling-upgrade/c43d858/ : FAILURE in 10m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0a251a10b3e87b664910e69ea4ca8de35f767dc0","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-07-01 20:24:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nall good here, plenty of coverage in tests, render looks great:\n\nInternalError: InternalError: 500 Internal Server Error (unexpected status code 500)\n\nHaving the status code will help a ton!","accounts_in_message":[],"_revision_number":1},{"id":"e7d4b67160e629909f2d9e70df61185baa03af86","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 20:25:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0600fcd28506a332411a94dd5faf9b8dfce4ac54","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 21:24:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/31/668031/1/gate/openstack-tox-docs/54d3deb/html/ : SUCCESS in 5m 33s\n- swift-upload-image http://logs.openstack.org/31/668031/1/gate/swift-upload-image/9ffa961/ : SUCCESS in 8m 47s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/31/668031/1/gate/swift-tox-py27/a0e13e7/ : SUCCESS in 11m 25s\n- swift-tox-py37 http://logs.openstack.org/31/668031/1/gate/swift-tox-py37/a3cc186/ : SUCCESS in 11m 48s\n- swift-tox-func http://logs.openstack.org/31/668031/1/gate/swift-tox-func/b2b8940/ : SUCCESS in 9m 42s\n- swift-tox-func-py37 http://logs.openstack.org/31/668031/1/gate/swift-tox-func-py37/5a3b0a9/ : SUCCESS in 9m 20s\n- swift-tox-func-encryption http://logs.openstack.org/31/668031/1/gate/swift-tox-func-encryption/55e86dd/ : SUCCESS in 10m 29s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/31/668031/1/gate/swift-tox-func-domain-remap-staticweb/6348f4d/ : SUCCESS in 11m 21s\n- swift-tox-func-ec http://logs.openstack.org/31/668031/1/gate/swift-tox-func-ec/8937971/ : SUCCESS in 10m 06s\n- swift-tox-func-s3api http://logs.openstack.org/31/668031/1/gate/swift-tox-func-s3api/e12b65a/ : SUCCESS in 11m 56s\n- swift-probetests-centos-7 http://logs.openstack.org/31/668031/1/gate/swift-probetests-centos-7/71d00f7/ : SUCCESS in 37m 20s\n- swift-dsvm-functional http://logs.openstack.org/31/668031/1/gate/swift-dsvm-functional/c93310c/ : SUCCESS in 30m 15s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/31/668031/1/gate/swift-dsvm-functional-ipv6/f285938/ : SUCCESS in 25m 11s\n- swift-tox-lower-constraints http://logs.openstack.org/31/668031/1/gate/swift-tox-lower-constraints/871a16f/ : SUCCESS in 11m 04s\n- openstack-tox-pep8 http://logs.openstack.org/31/668031/1/gate/openstack-tox-pep8/769ef52/ : SUCCESS in 7m 17s","accounts_in_message":[],"_revision_number":1},{"id":"9ddf38b7e10f43da624caed413ce7acbb80262b5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 21:24:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ca8b04f0edc16db4646d9304442bcc6812cec1fa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 21:26:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- swift-promote-image http://logs.openstack.org/31/668031/1/promote/swift-promote-image/240da1e/ : SUCCESS in 58s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"88cec2c3966080fa79e5e623c7603c0eed4b94e3","revisions":{"88cec2c3966080fa79e5e623c7603c0eed4b94e3":{"kind":"REWORK","_number":1,"created":"2019-06-27 19:32:24.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/668031/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/668031/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/668031/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/668031/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/668031/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/668031/1"}}},"commit":{"parents":[{"commit":"044c919871f80e1df91ff9a40318a5afd6a2960d","subject":"More tests for 404 handoff skipping","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/044c919871f80e1df91ff9a40318a5afd6a2960d"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-06-27 19:29:04.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-06-27 19:32:09.000000000","tz":-420},"subject":"s3api: Put more information in the str() for InternalError","message":"s3api: Put more information in the str() for InternalError\n\nThere are few things more frustrating when digging through a log than\nfinding a traceback that ends with\n\n    raise InternalError(\u0027unexpected status code %d\u0027 % status)\n\nwhile having no indication of what the status code was.\n\nChange-Id: I08e9f7c0643ae0469bd1937a30c1c4477b67c31f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/88cec2c3966080fa79e5e623c7603c0eed4b94e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/88cec2c3966080fa79e5e623c7603c0eed4b94e3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
