)]}'
{"id":"openstack%2Fswift~668117","triplet_id":"openstack%2Fswift~master~I7b07f0833d675d2939c910f679b54da2b8cda482","project":"openstack/swift","branch":"master","topic":"assertTrue","hashtags":[],"change_id":"I7b07f0833d675d2939c910f679b54da2b8cda482","subject":"Fix invalid assert states","status":"MERGED","created":"2019-06-28 09:00:42.000000000","updated":"2019-07-01 03:47:03.000000000","submitted":"2019-07-01 03:45:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668117-1561952755885-a45a0009","meta_rev_id":"6b345104a1da6b64d9ca2dd1b1b553279354f4a2","_number":668117,"virtual_id_number":668117,"owner":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-07-01 03:45:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-01 03:09:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-01 03:09:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-01 03:09:37.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2019-07-01 03:45:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"90041d1e345a45c71c5c732b12b57ccba5c66bb6","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2019-06-28 09:00:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5fbc020ea8e6af4385e3829a9d934fa27265af7c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 10:26:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/17/668117/1/check/openstack-tox-docs/92d7c77/html/ : SUCCESS in 5m 49s\n- swift-build-image http://logs.openstack.org/17/668117/1/check/swift-build-image/ed3dc7b/ : SUCCESS in 7m 37s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/17/668117/1/check/swift-tox-py27/4bc1419/ : FAILURE in 10m 39s\n- swift-tox-py36 http://logs.openstack.org/17/668117/1/check/swift-tox-py36/9913598/ : FAILURE in 10m 14s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/17/668117/1/check/swift-tox-py37/145e350/ : FAILURE in 11m 38s\n- swift-tox-func http://logs.openstack.org/17/668117/1/check/swift-tox-func/085b07c/ : SUCCESS in 8m 43s\n- swift-tox-func-py37 http://logs.openstack.org/17/668117/1/check/swift-tox-func-py37/f8fe929/ : SUCCESS in 9m 31s\n- swift-tox-func-encryption http://logs.openstack.org/17/668117/1/check/swift-tox-func-encryption/15062bc/ : SUCCESS in 8m 54s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/17/668117/1/check/swift-tox-func-domain-remap-staticweb/8ce28c8/ : SUCCESS in 8m 50s\n- swift-tox-func-ec http://logs.openstack.org/17/668117/1/check/swift-tox-func-ec/0d9c947/ : SUCCESS in 10m 50s\n- swift-tox-func-s3api http://logs.openstack.org/17/668117/1/check/swift-tox-func-s3api/4489683/ : SUCCESS in 11m 59s\n- swift-dsvm-functional http://logs.openstack.org/17/668117/1/check/swift-dsvm-functional/e223440/ : SUCCESS in 26m 57s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/17/668117/1/check/swift-dsvm-functional-ipv6/23ab020/ : SUCCESS in 28m 22s\n- swift-tox-lower-constraints http://logs.openstack.org/17/668117/1/check/swift-tox-lower-constraints/f1c5087/ : FAILURE in 12m 19s\n- openstack-tox-pep8 http://logs.openstack.org/17/668117/1/check/openstack-tox-pep8/513cbc0/ : SUCCESS in 5m 46s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/17/668117/1/check/swift-multinode-rolling-upgrade/9b8d04d/ : FAILURE in 9m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"afe5769958939fb969bd46741263c7a4cc77a2f9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-28 19:08:30.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nYeah, those didn\u0027t make sense, did they :-)","accounts_in_message":[],"_revision_number":1},{"id":"25d4017df21d8363513720e724351e4f6d03cb0f","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2019-07-01 01:25:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2cda60fb957ba9cb2237f6abbd6198e1e33aa0e2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 01:56:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/17/668117/2/check/openstack-tox-docs/ab22dba/html/ : SUCCESS in 3m 59s\n- swift-build-image http://logs.openstack.org/17/668117/2/check/swift-build-image/52401bc/ : SUCCESS in 7m 14s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/17/668117/2/check/swift-tox-py27/bf0903a/ : SUCCESS in 12m 00s\n- swift-tox-py36 http://logs.openstack.org/17/668117/2/check/swift-tox-py36/8374f98/ : SUCCESS in 9m 46s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/17/668117/2/check/swift-tox-py37/aa550fe/ : SUCCESS in 12m 17s\n- swift-tox-func http://logs.openstack.org/17/668117/2/check/swift-tox-func/6b9174a/ : SUCCESS in 7m 55s\n- swift-tox-func-py37 http://logs.openstack.org/17/668117/2/check/swift-tox-func-py37/7d9ceb6/ : SUCCESS in 9m 57s\n- swift-tox-func-encryption http://logs.openstack.org/17/668117/2/check/swift-tox-func-encryption/1a364d9/ : SUCCESS in 9m 41s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/17/668117/2/check/swift-tox-func-domain-remap-staticweb/564e164/ : SUCCESS in 8m 02s\n- swift-tox-func-ec http://logs.openstack.org/17/668117/2/check/swift-tox-func-ec/e5d4bd9/ : SUCCESS in 8m 52s\n- swift-tox-func-s3api http://logs.openstack.org/17/668117/2/check/swift-tox-func-s3api/447d799/ : SUCCESS in 14m 15s\n- swift-dsvm-functional http://logs.openstack.org/17/668117/2/check/swift-dsvm-functional/5690dbb/ : SUCCESS in 25m 10s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/17/668117/2/check/swift-dsvm-functional-ipv6/12a9e50/ : SUCCESS in 25m 51s\n- swift-tox-lower-constraints http://logs.openstack.org/17/668117/2/check/swift-tox-lower-constraints/4328e62/ : SUCCESS in 11m 05s\n- openstack-tox-pep8 http://logs.openstack.org/17/668117/2/check/openstack-tox-pep8/2c74ed5/ : SUCCESS in 5m 15s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/17/668117/2/check/swift-multinode-rolling-upgrade/2e1fd0a/ : FAILURE in 8m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"49b9275392989f5ce20bd14d0e5c5b43fc95eeb8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-07-01 03:09:37.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nThanks!","accounts_in_message":[],"_revision_number":2},{"id":"bc773891c846ce5a6638847953c631394369a67a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 03:09:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b14149a3c34266e69332fcdf5a51b4699afaf888","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 03:45:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"b331c3b7721b398232c271a25b909f7d8e53f41a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 03:45:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/17/668117/2/gate/openstack-tox-docs/ba41c42/html/ : SUCCESS in 4m 04s\n- swift-upload-image http://logs.openstack.org/17/668117/2/gate/swift-upload-image/0adbb6f/ : SUCCESS in 8m 23s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/17/668117/2/gate/swift-tox-py27/9a647ec/ : SUCCESS in 10m 19s\n- swift-tox-py37 http://logs.openstack.org/17/668117/2/gate/swift-tox-py37/88f5127/ : SUCCESS in 10m 20s\n- swift-tox-func http://logs.openstack.org/17/668117/2/gate/swift-tox-func/e802596/ : SUCCESS in 8m 38s\n- swift-tox-func-py37 http://logs.openstack.org/17/668117/2/gate/swift-tox-func-py37/ba3d621/ : SUCCESS in 8m 52s\n- swift-tox-func-encryption http://logs.openstack.org/17/668117/2/gate/swift-tox-func-encryption/3e98b9d/ : SUCCESS in 8m 46s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/17/668117/2/gate/swift-tox-func-domain-remap-staticweb/bdd7126/ : SUCCESS in 8m 58s\n- swift-tox-func-ec http://logs.openstack.org/17/668117/2/gate/swift-tox-func-ec/37989fc/ : SUCCESS in 9m 25s\n- swift-tox-func-s3api http://logs.openstack.org/17/668117/2/gate/swift-tox-func-s3api/26a39c0/ : SUCCESS in 11m 28s\n- swift-dsvm-functional http://logs.openstack.org/17/668117/2/gate/swift-dsvm-functional/94ebb51/ : SUCCESS in 30m 46s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/17/668117/2/gate/swift-dsvm-functional-ipv6/eeaaaa7/ : SUCCESS in 28m 50s\n- swift-tox-lower-constraints http://logs.openstack.org/17/668117/2/gate/swift-tox-lower-constraints/fc50181/ : SUCCESS in 10m 50s\n- openstack-tox-pep8 http://logs.openstack.org/17/668117/2/gate/openstack-tox-pep8/6e07167/ : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":2},{"id":"3e8df5be78aecd801da35e2c84a094091dc35237","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 03:47:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- swift-promote-image http://logs.openstack.org/17/668117/2/promote/swift-promote-image/5efb905/ : SUCCESS in 47s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c46b88ab742761c8918013fa9cc40a169e015474","revisions":{"79950478e847013e87bcddc7f1c1e9e473ca4729":{"kind":"REWORK","_number":1,"created":"2019-06-28 09:00:42.000000000","uploader":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"ref":"refs/changes/17/668117/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/668117/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/668117/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/668117/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/668117/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/668117/1"}}},"commit":{"parents":[{"commit":"82169ead1cc1c60318507fc3f0fbd7deff0f15e9","subject":"Don\u0027t handle object without container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/82169ead1cc1c60318507fc3f0fbd7deff0f15e9"}]}],"author":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2019-06-28 08:54:27.000000000","tz":480},"committer":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2019-06-28 08:54:27.000000000","tz":480},"subject":"Fix invalid assert states","message":"Fix invalid assert states\n\n\"self.assertTrue(policies[1].is_deprecated, True)\" and\n\"self.assertTrue(crashy_calls[0], 1)\" are not correct, this is\nto fix them.\n\nChange-Id: I7b07f0833d675d2939c910f679b54da2b8cda482\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/79950478e847013e87bcddc7f1c1e9e473ca4729"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/79950478e847013e87bcddc7f1c1e9e473ca4729"}]},"branch":"refs/heads/master"},"c46b88ab742761c8918013fa9cc40a169e015474":{"kind":"REWORK","_number":2,"created":"2019-07-01 01:25:48.000000000","uploader":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"ref":"refs/changes/17/668117/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/668117/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/668117/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/668117/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/668117/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/668117/2"}}},"commit":{"parents":[{"commit":"82169ead1cc1c60318507fc3f0fbd7deff0f15e9","subject":"Don\u0027t handle object without container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/82169ead1cc1c60318507fc3f0fbd7deff0f15e9"}]}],"author":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2019-06-28 08:54:27.000000000","tz":480},"committer":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2019-07-01 01:20:02.000000000","tz":480},"subject":"Fix invalid assert states","message":"Fix invalid assert states\n\n\"self.assertTrue(policies[1].is_deprecated, True)\" and\n\"self.assertTrue(crashy_calls[0], 1)\" are not correct, this is\nto fix them.\n\nChange-Id: I7b07f0833d675d2939c910f679b54da2b8cda482\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c46b88ab742761c8918013fa9cc40a169e015474"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c46b88ab742761c8918013fa9cc40a169e015474"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
