)]}'
{"id":"openstack%2Fswift~668578","triplet_id":"openstack%2Fswift~master~I5dd57aad794c050c44e328c43346be0063170492","project":"openstack/swift","branch":"master","topic":"setup","hashtags":[],"change_id":"I5dd57aad794c050c44e328c43346be0063170492","subject":"Add python3 to setup.cfg","status":"MERGED","created":"2019-07-02 03:04:20.000000000","updated":"2019-09-18 18:02:35.000000000","submitted":"2019-09-18 18:01:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668578-1568829680931-d2cb535d","meta_rev_id":"5b5ce8af60f17df9971c8f7a77989577220d724d","_number":668578,"virtual_id_number":668578,"owner":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":2,"date":"2019-09-18 18:01:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"recommended":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"all":[{"value":1,"date":"2019-09-16 05:15:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-18 16:49:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-18 16:49:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-16 05:15:16.000000000","updated_by":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"reviewer":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"state":"REVIEWER"},{"updated":"2019-09-18 16:49:33.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2019-09-18 18:01:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"235fe379b0190ecdea362d41aba3f2dd6d1dd654","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-07-02 03:04:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7cdb58bfaea211ad81f351f3b53657335fee1194","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 03:40:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/78/668578/1/check/test-release-openstack/76baf56/ : SUCCESS in 2m 34s\n- openstack-tox-docs http://logs.openstack.org/78/668578/1/check/openstack-tox-docs/c183b50/html/ : SUCCESS in 4m 04s\n- swift-build-image http://logs.openstack.org/78/668578/1/check/swift-build-image/e4eedb3/ : SUCCESS in 7m 57s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/78/668578/1/check/swift-tox-py27/4e5425f/ : SUCCESS in 11m 03s\n- swift-tox-py36 http://logs.openstack.org/78/668578/1/check/swift-tox-py36/b3d8bac/ : SUCCESS in 13m 12s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/78/668578/1/check/swift-tox-py37/71cd6da/ : SUCCESS in 9m 54s\n- swift-tox-func http://logs.openstack.org/78/668578/1/check/swift-tox-func/f6d281d/ : SUCCESS in 10m 15s\n- swift-tox-func-py37 http://logs.openstack.org/78/668578/1/check/swift-tox-func-py37/277221f/ : SUCCESS in 8m 36s\n- swift-tox-func-encryption http://logs.openstack.org/78/668578/1/check/swift-tox-func-encryption/97125d9/ : SUCCESS in 8m 57s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/78/668578/1/check/swift-tox-func-domain-remap-staticweb/f8e06fe/ : SUCCESS in 12m 53s\n- swift-tox-func-ec http://logs.openstack.org/78/668578/1/check/swift-tox-func-ec/fcfcf0f/ : SUCCESS in 11m 15s\n- swift-tox-func-s3api http://logs.openstack.org/78/668578/1/check/swift-tox-func-s3api/b291705/ : SUCCESS in 11m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/78/668578/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/fe30297/ : SUCCESS in 18m 15s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/78/668578/1/check/swift-probetests-centos-7/9da9a06/ : SUCCESS in 30m 45s\n- swift-dsvm-functional http://logs.openstack.org/78/668578/1/check/swift-dsvm-functional/f2f73f5/ : SUCCESS in 26m 46s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/78/668578/1/check/swift-dsvm-functional-ipv6/bfc4261/ : SUCCESS in 25m 06s\n- swift-tox-lower-constraints http://logs.openstack.org/78/668578/1/check/swift-tox-lower-constraints/905a8f0/ : SUCCESS in 10m 34s\n- openstack-tox-pep8 http://logs.openstack.org/78/668578/1/check/openstack-tox-pep8/df915e4/ : SUCCESS in 5m 19s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/78/668578/1/check/swift-multinode-rolling-upgrade/f303cee/ : FAILURE in 9m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"182a7570e5b6810559c6756a97f08d34c4eb2777","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-09-16 05:15:16.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3cec3e626ac09a4da202e570614bfe1a25594aab","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-09-18 16:49:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2f5c713faed2b49c55e91d3e55e9c7de9bcb4644","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 16:50:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"61cc71124c6e46d573a02a0cb251f1afa4cb26f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 18:01:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b385871dfefc50c28fca25d19e6e3d9913c0e551","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 18:01:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b4c9098e0dc24bbf981ce120a2c4f202 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/541fb2a0cc724741a4a973e1407d553c : SUCCESS in 6m 20s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/05f41448ebdb42ec815e62603b0393e2 : SUCCESS in 1h 03m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4bee057f70cc47fab2f2089ceabd67f5 : SUCCESS in 48m 12s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/e931053b902648148fbbfe6e125d3563 : SUCCESS in 8m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/aa59d116a5c34e0e8165b8ce79cb5d0b : SUCCESS in 12m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a9919501b92c4ea3bf7cc517ac629c32 : SUCCESS in 11m 40s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/89112b45ba0e408fbc3159ff5c6db5bd : SUCCESS in 13m 16s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/f5646e1dfb0d42f9a93de7fa4c44fab6 : SUCCESS in 10m 29s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/6ff068bf2f8d4474b68eb39accc4606d : SUCCESS in 9m 16s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/1b0717197ea54ee99672809470ef6980 : SUCCESS in 12m 07s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/799c4f7876244750a0eb5fd29b7d9f71 : SUCCESS in 11m 45s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/fdf216c364fb4e7788e719e06b8bcfad : SUCCESS in 17m 40s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/ed11b9fa28934939a332efb3e6d0743f : SUCCESS in 23m 30s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/9d322ea370e84ec88387e16d96b50b2e : SUCCESS in 11m 20s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/28a5dbbcb1794233984272e5805d5826 : SUCCESS in 10m 23s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/28ba4c36f7c2495f95e0c5a5fd09c4d0 : SUCCESS in 14m 07s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b086ff0d768f4ce98f5df9bdb7fea272 : SUCCESS in 29m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f5f36389eae64f31b6cefd8ee8ef410f : SUCCESS in 30m 08s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/03f10e455e00446a901623db362d56e5 : SUCCESS in 29m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ee497a99c9764460ae8b465b26181f94 : SUCCESS in 31m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0036a85b61ec4740aaf9f7b3ac8c32db : SUCCESS in 12m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e71f5a43cade46a39956c451dbb47b2c : SUCCESS in 6m 43s","accounts_in_message":[],"_revision_number":1},{"id":"95e8cc5b6efa09f47c01941b822826aa4838074d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 18:02:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8a4491d0afc412dbb6f945ec1d1d634 : SUCCESS in 52s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/e998225cbb184bc59d2d15f595696f98 : SUCCESS in 40s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5270da86e6eead273f58a24cba65b951550e3037","revisions":{"5270da86e6eead273f58a24cba65b951550e3037":{"kind":"REWORK","_number":1,"created":"2019-07-02 03:04:20.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/78/668578/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/78/668578/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/668578/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/668578/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/668578/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/78/668578/1"}}},"commit":{"parents":[{"commit":"bf3e2548b337ad149994eca596e6b67904cdf00a","subject":"Merge \"s3api: add unit test case for s3acl\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bf3e2548b337ad149994eca596e6b67904cdf00a"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-07-02 02:59:28.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-07-02 03:04:20.000000000","tz":480},"subject":"Add python3 to setup.cfg","message":"Add python3 to setup.cfg\n\nChange-Id: I5dd57aad794c050c44e328c43346be0063170492\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5270da86e6eead273f58a24cba65b951550e3037"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5270da86e6eead273f58a24cba65b951550e3037"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
