)]}'
{"id":"openstack%2Fswift~668919","triplet_id":"openstack%2Fswift~master~I71f411a2e99fa8259b86f11ed29d1b816ff469cb","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I71f411a2e99fa8259b86f11ed29d1b816ff469cb","subject":"s3token: fix conf option name","status":"MERGED","created":"2019-07-03 14:28:51.000000000","updated":"2019-07-03 19:39:21.000000000","submitted":"2019-07-03 19:38:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668919-1562182683739-354c625b","meta_rev_id":"7f2760b804525b6abf22e5bf17a5cc095411ea37","_number":668919,"virtual_id_number":668919,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-07-03 19:38:03.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-03 14:29:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-03 14:29:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-03 19:38:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"66866e3ddd949b8a310699735ae6025a360c0ef7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-07-03 14:28:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cea703caaa3d1f503d620507ecd1283166013dbb","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-07-03 14:29:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"03343ee2897607f89f270b3616ba39b129e46b91","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 18:00:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/19/668919/1/check/openstack-tox-docs/57fd354/html/ : SUCCESS in 4m 22s\n- swift-build-image http://logs.openstack.org/19/668919/1/check/swift-build-image/d918468/ : SUCCESS in 8m 10s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/19/668919/1/check/swift-tox-py27/0caeb59/ : SUCCESS in 14m 04s\n- swift-tox-py36 http://logs.openstack.org/19/668919/1/check/swift-tox-py36/a15c915/ : SUCCESS in 15m 54s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/19/668919/1/check/swift-tox-py37/e51cd47/ : SUCCESS in 13m 22s\n- swift-tox-func http://logs.openstack.org/19/668919/1/check/swift-tox-func/eaa4d88/ : SUCCESS in 8m 57s\n- swift-tox-func-py37 http://logs.openstack.org/19/668919/1/check/swift-tox-func-py37/506ee6f/ : SUCCESS in 9m 26s\n- swift-tox-func-encryption http://logs.openstack.org/19/668919/1/check/swift-tox-func-encryption/1f14f0f/ : SUCCESS in 9m 37s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/19/668919/1/check/swift-tox-func-domain-remap-staticweb/aa10b35/ : SUCCESS in 9m 35s\n- swift-tox-func-ec http://logs.openstack.org/19/668919/1/check/swift-tox-func-ec/65775d2/ : SUCCESS in 10m 30s\n- swift-tox-func-s3api http://logs.openstack.org/19/668919/1/check/swift-tox-func-s3api/12ba953/ : SUCCESS in 12m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/19/668919/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/be4cb4c/ : SUCCESS in 12m 55s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/19/668919/1/check/swift-probetests-centos-7/45563d6/ : SUCCESS in 31m 12s\n- swift-dsvm-functional http://logs.openstack.org/19/668919/1/check/swift-dsvm-functional/561e6f5/ : SUCCESS in 28m 48s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/19/668919/1/check/swift-dsvm-functional-ipv6/2350c59/ : SUCCESS in 26m 36s\n- swift-tox-lower-constraints http://logs.openstack.org/19/668919/1/check/swift-tox-lower-constraints/5d60c65/ : SUCCESS in 10m 36s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/19/668919/1/check/swift-multinode-rolling-upgrade/1a745fa/ : FAILURE in 13m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0624cd3fb5207f1eae8d483bdae67c4b33da8b5f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 18:01:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"505ec460429422156dad37b0fccf967f24f492e7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 19:38:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"cc8301861c724d8bd58c54982e5afc6d1d70d3cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 19:38:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/19/668919/1/gate/openstack-tox-docs/0cee356/html/ : SUCCESS in 4m 12s\n- swift-upload-image http://logs.openstack.org/19/668919/1/gate/swift-upload-image/4518921/ : SUCCESS in 8m 24s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/19/668919/1/gate/swift-tox-py27/50e4fac/ : SUCCESS in 10m 54s\n- swift-tox-py37 http://logs.openstack.org/19/668919/1/gate/swift-tox-py37/31b1343/ : SUCCESS in 9m 25s\n- swift-tox-func http://logs.openstack.org/19/668919/1/gate/swift-tox-func/0c047e4/ : SUCCESS in 8m 36s\n- swift-tox-func-py37 http://logs.openstack.org/19/668919/1/gate/swift-tox-func-py37/2fe16f2/ : SUCCESS in 9m 15s\n- swift-tox-func-encryption http://logs.openstack.org/19/668919/1/gate/swift-tox-func-encryption/daf821c/ : SUCCESS in 9m 43s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/19/668919/1/gate/swift-tox-func-domain-remap-staticweb/243741f/ : SUCCESS in 9m 06s\n- swift-tox-func-ec http://logs.openstack.org/19/668919/1/gate/swift-tox-func-ec/c44f234/ : SUCCESS in 9m 49s\n- swift-tox-func-s3api http://logs.openstack.org/19/668919/1/gate/swift-tox-func-s3api/de0d601/ : SUCCESS in 14m 45s\n- swift-probetests-centos-7 http://logs.openstack.org/19/668919/1/gate/swift-probetests-centos-7/959664b/ : SUCCESS in 37m 03s\n- swift-dsvm-functional http://logs.openstack.org/19/668919/1/gate/swift-dsvm-functional/4eda291/ : SUCCESS in 28m 08s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/19/668919/1/gate/swift-dsvm-functional-ipv6/c55791d/ : SUCCESS in 27m 51s\n- swift-tox-lower-constraints http://logs.openstack.org/19/668919/1/gate/swift-tox-lower-constraints/2ba70d3/ : SUCCESS in 12m 06s","accounts_in_message":[],"_revision_number":1},{"id":"7f2760b804525b6abf22e5bf17a5cc095411ea37","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 19:39:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- swift-promote-image http://logs.openstack.org/19/668919/1/promote/swift-promote-image/7507d7a/ : SUCCESS in 53s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"345f577ff191ee01c3cb4626805338028815c2b4","revisions":{"345f577ff191ee01c3cb4626805338028815c2b4":{"kind":"REWORK","_number":1,"created":"2019-07-03 14:28:51.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/19/668919/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/668919/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/668919/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/668919/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/668919/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/668919/1"}}},"commit":{"parents":[{"commit":"dc84c69c652d899bf6b8177fa352cf980ae5ff8a","subject":"Merge \"Add information about secret_cache_duration to sample config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dc84c69c652d899bf6b8177fa352cf980ae5ff8a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-07-03 14:28:14.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-07-03 14:28:36.000000000","tz":-420},"subject":"s3token: fix conf option name","message":"s3token: fix conf option name\n\nRelated-Change: Ica740c28b47aa3f3b38dbfed4a7f5662ec46c2c4\nChange-Id: I71f411a2e99fa8259b86f11ed29d1b816ff469cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/345f577ff191ee01c3cb4626805338028815c2b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/345f577ff191ee01c3cb4626805338028815c2b4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
