)]}'
{"id":"openstack%2Fswift~669511","triplet_id":"openstack%2Fswift~master~I29f2951053fa6714de46ed757116d99dd5c896cf","project":"openstack/swift","branch":"master","topic":"python3-train","hashtags":[],"change_id":"I29f2951053fa6714de46ed757116d99dd5c896cf","subject":"Add Python 3 Train unit tests","status":"MERGED","created":"2019-07-05 20:23:25.000000000","updated":"2019-07-23 17:58:29.000000000","submitted":"2019-07-23 17:53:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669511-1563904423102-39e60a20","meta_rev_id":"ed19d29edcd893a22ccae67c73b2776eebcfd108","_number":669511,"virtual_id_number":669511,"owner":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":2,"date":"2019-07-23 17:53:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2019-07-10 01:23:06.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":1,"date":"2019-07-14 02:44:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":1,"date":"2019-07-10 01:25:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":1,"date":"2019-07-13 18:42:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-23 15:36:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-10 01:23:06.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2019-07-10 01:25:18.000000000","updated_by":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"reviewer":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"state":"REVIEWER"},{"updated":"2019-07-13 18:42:16.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2019-07-14 02:44:32.000000000","updated_by":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"reviewer":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"state":"REVIEWER"},{"updated":"2019-07-23 15:36:19.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2019-07-23 17:53:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"35bfc6aa5d0c76e7fd337b2ff62dc5881f2b2a4d","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-07-05 20:23:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e7efe5d95a5dd60805eba4c326d3e377ccfc7c37","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 21:01:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/11/669511/1/check/test-release-openstack/1236c95/ : SUCCESS in 2m 46s\n- openstack-tox-docs http://logs.openstack.org/11/669511/1/check/openstack-tox-docs/49cb2ea/html/ : SUCCESS in 4m 21s\n- swift-build-image http://logs.openstack.org/11/669511/1/check/swift-build-image/6d6d131/ : SUCCESS in 7m 18s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/11/669511/1/check/swift-tox-py27/55ec7b2/ : SUCCESS in 11m 42s\n- swift-tox-py36 http://logs.openstack.org/11/669511/1/check/swift-tox-py36/b247e61/ : SUCCESS in 10m 46s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/11/669511/1/check/swift-tox-py37/b02e0db/ : SUCCESS in 10m 51s\n- swift-tox-func http://logs.openstack.org/11/669511/1/check/swift-tox-func/b2d8f5b/ : SUCCESS in 8m 48s\n- swift-tox-func-py37 http://logs.openstack.org/11/669511/1/check/swift-tox-func-py37/4aa80fd/ : SUCCESS in 10m 14s\n- swift-tox-func-encryption http://logs.openstack.org/11/669511/1/check/swift-tox-func-encryption/8a17bdc/ : SUCCESS in 9m 30s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/11/669511/1/check/swift-tox-func-domain-remap-staticweb/fc6c9a9/ : SUCCESS in 8m 52s\n- swift-tox-func-ec http://logs.openstack.org/11/669511/1/check/swift-tox-func-ec/939a3c1/ : SUCCESS in 13m 28s\n- swift-tox-func-s3api http://logs.openstack.org/11/669511/1/check/swift-tox-func-s3api/6d4ddbd/ : SUCCESS in 15m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/11/669511/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/720acc3/ : SUCCESS in 16m 32s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/11/669511/1/check/swift-probetests-centos-7/8098990/ : SUCCESS in 31m 28s\n- swift-dsvm-functional http://logs.openstack.org/11/669511/1/check/swift-dsvm-functional/4b49db6/ : SUCCESS in 27m 26s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/11/669511/1/check/swift-dsvm-functional-ipv6/3d4f14e/ : SUCCESS in 34m 55s\n- swift-tox-lower-constraints http://logs.openstack.org/11/669511/1/check/swift-tox-lower-constraints/f3d3cee/ : SUCCESS in 12m 38s\n- openstack-tox-pep8 http://logs.openstack.org/11/669511/1/check/openstack-tox-pep8/4f78c0b/ : SUCCESS in 7m 31s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/11/669511/1/check/swift-multinode-rolling-upgrade/27ca5b3/ : FAILURE in 9m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c3db058d29b77570be4460e0fa85bdd2b749d2c0","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-07-10 01:23:06.000000000","message":"Patch Set 1: Code-Review+2\n\nI don\u0027t see why not... We have unit tests passing, right?","accounts_in_message":[],"_revision_number":1},{"id":"75e9adf288846491760e88e08a0b49b7e8a0a40d","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-07-10 01:25:18.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5097f026289947fc2ab6689147b52c28419781d1","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-07-13 18:42:16.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"540b45808288fdd2ea1f4204461f852ca8d7f8a5","author":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"date":"2019-07-14 02:44:32.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"65b22611f9ada5acbabd9293851f9d366b49a6f4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-07-23 15:36:19.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d5ebe69b2aa0f474a811b16ced424bc9ecdf6c2d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 15:43:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a56c5968e58b474b527fca461942047a00fce10d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 17:53:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/11/669511/1/gate/test-release-openstack/19add6f/ : SUCCESS in 4m 14s\n- openstack-tox-docs http://logs.openstack.org/11/669511/1/gate/openstack-tox-docs/743c6d6/html/ : SUCCESS in 6m 17s\n- swift-upload-image http://logs.openstack.org/11/669511/1/gate/swift-upload-image/5f6b880/ : SUCCESS in 10m 08s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/11/669511/1/gate/swift-tox-py27/60f1dbb/ : SUCCESS in 13m 19s\n- swift-tox-py36 http://logs.openstack.org/11/669511/1/gate/swift-tox-py36/635978c/ : SUCCESS in 12m 13s\n- swift-tox-py37 http://logs.openstack.org/11/669511/1/gate/swift-tox-py37/466a0f2/ : SUCCESS in 12m 27s\n- swift-tox-func http://logs.openstack.org/11/669511/1/gate/swift-tox-func/768ee0a/ : SUCCESS in 9m 06s\n- swift-tox-func-encryption http://logs.openstack.org/11/669511/1/gate/swift-tox-func-encryption/6bf6141/ : SUCCESS in 10m 02s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/11/669511/1/gate/swift-tox-func-domain-remap-staticweb/2a94bf7/ : SUCCESS in 15m 23s\n- swift-tox-func-ec http://logs.openstack.org/11/669511/1/gate/swift-tox-func-ec/9d8505f/ : SUCCESS in 13m 18s\n- swift-tox-func-s3api http://logs.openstack.org/11/669511/1/gate/swift-tox-func-s3api/29c82e0/ : SUCCESS in 13m 11s\n- swift-tox-func-py37 http://logs.openstack.org/11/669511/1/gate/swift-tox-func-py37/f11da58/ : SUCCESS in 10m 56s\n- swift-tox-func-domain-remap-staticweb-py37 http://logs.openstack.org/11/669511/1/gate/swift-tox-func-domain-remap-staticweb-py37/40464d7/ : SUCCESS in 8m 42s\n- swift-tox-func-ec-py37 http://logs.openstack.org/11/669511/1/gate/swift-tox-func-ec-py37/a6c77e2/ : SUCCESS in 10m 13s\n- swift-probetests-centos-7 http://logs.openstack.org/11/669511/1/gate/swift-probetests-centos-7/07adc8b/ : SUCCESS in 34m 10s\n- swift-dsvm-functional http://logs.openstack.org/11/669511/1/gate/swift-dsvm-functional/6dc8613/ : SUCCESS in 31m 31s\n- swift-dsvm-functional-py3 http://logs.openstack.org/11/669511/1/gate/swift-dsvm-functional-py3/981e634/ : SUCCESS in 32m 00s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/11/669511/1/gate/swift-dsvm-functional-ipv6/ce99843/ : SUCCESS in 30m 53s\n- swift-tox-lower-constraints http://logs.openstack.org/11/669511/1/gate/swift-tox-lower-constraints/37ce320/ : SUCCESS in 18m 08s\n- openstack-tox-pep8 http://logs.openstack.org/11/669511/1/gate/openstack-tox-pep8/9be988e/ : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":1},{"id":"630c32d0dc092d0b357586422110fe32c0f00a56","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 17:53:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4e1e8b4fc16bd58d92778820cbf13b6a705226c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 17:58:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- swift-promote-image http://logs.openstack.org/11/669511/1/promote/swift-promote-image/9a07cdf/ : SUCCESS in 1m 02s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3532df0fd3dc91af5534b1cb3d4de4bfec101b8f","revisions":{"3532df0fd3dc91af5534b1cb3d4de4bfec101b8f":{"kind":"REWORK","_number":1,"created":"2019-07-05 20:23:25.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/11/669511/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/669511/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/669511/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/669511/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/669511/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/669511/1"}}},"commit":{"parents":[{"commit":"345f577ff191ee01c3cb4626805338028815c2b4","subject":"s3token: fix conf option name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/345f577ff191ee01c3cb4626805338028815c2b4"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 20:22:56.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 20:23:24.000000000","tz":-240},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: I29f2951053fa6714de46ed757116d99dd5c896cf\nStory: #2005924\nTask: #34249\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3532df0fd3dc91af5534b1cb3d4de4bfec101b8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3532df0fd3dc91af5534b1cb3d4de4bfec101b8f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
