)]}'
{"id":"openstack%2Fswift~670005","triplet_id":"openstack%2Fswift~master~Id00cfd2c921da352abdbbbb6668b921f3cb31a1a","project":"openstack/swift","branch":"master","topic":"update_auth_url","hashtags":[],"change_id":"Id00cfd2c921da352abdbbbb6668b921f3cb31a1a","subject":"Update auth_url in install docs","status":"MERGED","created":"2019-07-10 07:53:36.000000000","updated":"2019-07-11 19:01:27.000000000","submitted":"2019-07-11 19:00:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670005-1562871607927-5870c8eb","meta_rev_id":"7357057f2f80a3348b6c260fc00adc0679735450","_number":670005,"virtual_id_number":670005,"owner":{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-07-11 07:08:21.000000000","_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},{"value":2,"date":"2019-07-11 19:00:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-11 16:22:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-11 16:22:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-11 16:22:25.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2019-07-11 19:00:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"db17e74ed17fd29c0d2286e7b001d6aae1904b94","author":{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},"date":"2019-07-10 07:53:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"333c692836d78d560bf68615171ec67f1c9f56ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 07:59:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/05/670005/1/check/openstack-tox-docs/95230ff/html/ : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":1},{"id":"d329f00933f7fe7373c3ca32e31cbb5a9c16b780","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-07-11 00:08:52.000000000","message":"Patch Set 1:\n\nShould we also update the sample configs?\n- https://github.com/openstack/swift/blob/2.21.0/etc/proxy-server.conf-sample#L383\n- https://github.com/openstack/swift/blob/2.21.0/etc/proxy-server.conf-sample#L594\n- Looks like we\u0027ve also got a sample config in https://github.com/openstack/swift/blob/2.21.0/doc/source/overview_auth.rst\n\nMan pages, too? https://github.com/openstack/swift/blob/2.21.0/doc/manpages/proxy-server.conf.5#L308","accounts_in_message":[],"_revision_number":1},{"id":"22f56f6e04536acc18f7a25305d07ed9625d733a","author":{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},"date":"2019-07-11 07:04:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"30884d7e2a8f2125a57ef270886da4b793bc28f0","author":{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},"date":"2019-07-11 07:08:21.000000000","message":"Patch Set 2:\n\n\u003e Should we also update the sample configs?\n \u003e - https://github.com/openstack/swift/blob/2.21.0/etc/proxy-server.conf-sample#L383\n \u003e - https://github.com/openstack/swift/blob/2.21.0/etc/proxy-server.conf-sample#L594\n \u003e - Looks like we\u0027ve also got a sample config in https://github.com/openstack/swift/blob/2.21.0/doc/source/overview_auth.rst\n \u003e \n \u003e Man pages, too? https://github.com/openstack/swift/blob/2.21.0/doc/manpages/proxy-server.conf.5#L308\n\nYes, it needs to be update","accounts_in_message":[],"_revision_number":2},{"id":"9e0f349b9e39af15c528d300603be5875697a08a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 07:52:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/05/670005/2/check/openstack-tox-docs/8b5aea2/html/ : SUCCESS in 4m 20s\n- swift-build-image http://logs.openstack.org/05/670005/2/check/swift-build-image/5b3727d/ : SUCCESS in 9m 23s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/05/670005/2/check/swift-tox-py27/550edd1/ : SUCCESS in 11m 19s\n- swift-tox-py36 http://logs.openstack.org/05/670005/2/check/swift-tox-py36/ef72353/ : SUCCESS in 12m 28s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/05/670005/2/check/swift-tox-py37/275dde0/ : SUCCESS in 11m 03s\n- swift-tox-func http://logs.openstack.org/05/670005/2/check/swift-tox-func/9f4c9f9/ : SUCCESS in 10m 14s\n- swift-tox-func-py37 http://logs.openstack.org/05/670005/2/check/swift-tox-func-py37/0e243ed/ : SUCCESS in 9m 15s\n- swift-tox-func-encryption http://logs.openstack.org/05/670005/2/check/swift-tox-func-encryption/9229509/ : SUCCESS in 11m 03s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/05/670005/2/check/swift-tox-func-domain-remap-staticweb/42d8042/ : SUCCESS in 9m 01s\n- swift-tox-func-ec http://logs.openstack.org/05/670005/2/check/swift-tox-func-ec/40775e7/ : SUCCESS in 9m 42s\n- swift-tox-func-s3api http://logs.openstack.org/05/670005/2/check/swift-tox-func-s3api/c30dc79/ : SUCCESS in 13m 24s\n- swift-tox-func-domain-remap-staticweb-py37 http://logs.openstack.org/05/670005/2/check/swift-tox-func-domain-remap-staticweb-py37/ba061f9/ : SUCCESS in 8m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/05/670005/2/check/swift-tox-func-s3api-ceph-s3tests-tempauth/30e19c0/ : SUCCESS in 13m 45s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/05/670005/2/check/swift-probetests-centos-7/ba9ac1c/ : SUCCESS in 32m 20s\n- swift-dsvm-functional http://logs.openstack.org/05/670005/2/check/swift-dsvm-functional/3a8bf7e/ : SUCCESS in 26m 22s\n- swift-dsvm-functional-py3 http://logs.openstack.org/05/670005/2/check/swift-dsvm-functional-py3/fe759ec/ : SUCCESS in 26m 56s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/05/670005/2/check/swift-dsvm-functional-ipv6/a1dd251/ : SUCCESS in 25m 03s\n- swift-tox-lower-constraints http://logs.openstack.org/05/670005/2/check/swift-tox-lower-constraints/9a3dda4/ : SUCCESS in 12m 49s\n- openstack-tox-pep8 http://logs.openstack.org/05/670005/2/check/openstack-tox-pep8/b4a3e39/ : SUCCESS in 5m 36s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/05/670005/2/check/swift-multinode-rolling-upgrade/43eef8a/ : FAILURE in 8m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"68b0ad53b7a603437a68ca79aa1ef93551788659","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-07-11 16:22:25.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nLooks good, thanks!","accounts_in_message":[],"_revision_number":2},{"id":"e7e76f542d77b9a56b75813b315ee883677f3e75","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 16:23:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b42c490bb12980aadd787db62f0c74104d01203d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 19:00:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"f62bfda5cb01392e12552991dfe6e3eda042609d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 19:00:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/05/670005/2/gate/openstack-tox-docs/183ee7b/html/ : SUCCESS in 5m 59s\n- swift-upload-image http://logs.openstack.org/05/670005/2/gate/swift-upload-image/8b9fef6/ : SUCCESS in 11m 13s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/05/670005/2/gate/swift-tox-py27/d9236aa/ : SUCCESS in 19m 26s\n- swift-tox-py37 http://logs.openstack.org/05/670005/2/gate/swift-tox-py37/355288d/ : SUCCESS in 13m 57s\n- swift-tox-func http://logs.openstack.org/05/670005/2/gate/swift-tox-func/90b6b73/ : SUCCESS in 8m 32s\n- swift-tox-func-py37 http://logs.openstack.org/05/670005/2/gate/swift-tox-func-py37/d0d485f/ : SUCCESS in 10m 22s\n- swift-tox-func-encryption http://logs.openstack.org/05/670005/2/gate/swift-tox-func-encryption/38e5507/ : SUCCESS in 9m 39s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/05/670005/2/gate/swift-tox-func-domain-remap-staticweb/1fd210b/ : SUCCESS in 8m 49s\n- swift-tox-func-ec http://logs.openstack.org/05/670005/2/gate/swift-tox-func-ec/0fc41b2/ : SUCCESS in 8m 51s\n- swift-tox-func-s3api http://logs.openstack.org/05/670005/2/gate/swift-tox-func-s3api/c36f489/ : SUCCESS in 11m 26s\n- swift-tox-func-domain-remap-staticweb-py37 http://logs.openstack.org/05/670005/2/gate/swift-tox-func-domain-remap-staticweb-py37/301272d/ : SUCCESS in 8m 04s\n- swift-probetests-centos-7 http://logs.openstack.org/05/670005/2/gate/swift-probetests-centos-7/e526883/ : SUCCESS in 37m 18s\n- swift-dsvm-functional http://logs.openstack.org/05/670005/2/gate/swift-dsvm-functional/0e074a2/ : SUCCESS in 31m 56s\n- swift-dsvm-functional-py3 http://logs.openstack.org/05/670005/2/gate/swift-dsvm-functional-py3/a6943eb/ : SUCCESS in 30m 36s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/05/670005/2/gate/swift-dsvm-functional-ipv6/0fe63a0/ : SUCCESS in 27m 32s\n- swift-tox-lower-constraints http://logs.openstack.org/05/670005/2/gate/swift-tox-lower-constraints/c60fa81/ : SUCCESS in 11m 59s\n- openstack-tox-pep8 http://logs.openstack.org/05/670005/2/gate/openstack-tox-pep8/707b10e/ : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":2},{"id":"720dc81b776a7ca58bafaaad012ca4b7573eb4bc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 19:01:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- swift-promote-image http://logs.openstack.org/05/670005/2/promote/swift-promote-image/3e1d0e4/ : SUCCESS in 59s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0ae1ad63c10ea8643fc0998dca159b844286c414","revisions":{"9da7b1fdcba06e5cd84a3013790e2c090d4ac808":{"kind":"REWORK","_number":1,"created":"2019-07-10 07:53:36.000000000","uploader":{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},"ref":"refs/changes/05/670005/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/670005/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/670005/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/670005/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/670005/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/670005/1"}}},"commit":{"parents":[{"commit":"4ae412e13b1b16e7a043235427229ead75186ce7","subject":"Merge \"Increase some middleware test coverage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ae412e13b1b16e7a043235427229ead75186ce7"}]}],"author":{"name":"zengjia","email":"zengjia@awcloud.com","date":"2019-07-10 07:46:32.000000000","tz":480},"committer":{"name":"zengjia","email":"zengjia@awcloud.com","date":"2019-07-10 07:46:32.000000000","tz":480},"subject":"Update auth_url in install docs","message":"Update auth_url in install docs\n\nBeginning with the Queens release, the keystone install guide\nrecommends running all interfaces on the same port.This patch\nupdates the swift install guide to reflect that change\n\nChange-Id: Id00cfd2c921da352abdbbbb6668b921f3cb31a1a\nCloses-bug: #1754104\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9da7b1fdcba06e5cd84a3013790e2c090d4ac808"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9da7b1fdcba06e5cd84a3013790e2c090d4ac808"}]},"branch":"refs/heads/master"},"0ae1ad63c10ea8643fc0998dca159b844286c414":{"kind":"REWORK","_number":2,"created":"2019-07-11 07:04:26.000000000","uploader":{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},"ref":"refs/changes/05/670005/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/670005/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/670005/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/670005/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/670005/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/670005/2"}}},"commit":{"parents":[{"commit":"e62f07d9884c584233fd55f46c2dd79257c23bd9","subject":"Merge \"py3: port staticweb and domain_remap func tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e62f07d9884c584233fd55f46c2dd79257c23bd9"}]}],"author":{"name":"zengjia","email":"zengjia@awcloud.com","date":"2019-07-10 07:46:32.000000000","tz":480},"committer":{"name":"zengjia","email":"zengjia@awcloud.com","date":"2019-07-11 07:03:16.000000000","tz":480},"subject":"Update auth_url in install docs","message":"Update auth_url in install docs\n\nBeginning with the Queens release, the keystone install guide\nrecommends running all interfaces on the same port.This patch\nupdates the swift install guide to reflect that change\n\nChange-Id: Id00cfd2c921da352abdbbbb6668b921f3cb31a1a\nCloses-bug: #1754104\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0ae1ad63c10ea8643fc0998dca159b844286c414"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0ae1ad63c10ea8643fc0998dca159b844286c414"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
