)]}'
{"id":"openstack%2Fswift~670111","triplet_id":"openstack%2Fswift~master~If62d82beb202dea553776920a95c177518b162ab","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"If62d82beb202dea553776920a95c177518b162ab","subject":"functests: make container creation less flakey in test_object","status":"MERGED","created":"2019-07-10 16:13:52.000000000","updated":"2019-07-10 21:29:20.000000000","submitted":"2019-07-10 21:26:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670111-1562794008363-0a1ee91c","meta_rev_id":"92a3ca7b78f8db8523ff02b9850ad008759fb658","_number":670111,"virtual_id_number":670111,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":2,"date":"2019-07-10 21:26:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-10 16:14:11.000000000","_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2019-07-10 16:50:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2019-07-10 16:50:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-10 16:50:32.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2019-07-10 21:26:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"63c3f82e06573d68511c2405fc66d53f106eb6ee","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-07-10 16:13:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0823d3a0522393b299bbf0c5891bd4fad33c3de5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-07-10 16:14:11.000000000","message":"Patch Set 1:\n\nSeen in the gate: http://logs.openstack.org/48/653548/14/gate/swift-tox-func-encryption/0d5b728/job-output.txt.gz#_2019-07-10_16_08_23_988565","accounts_in_message":[],"_revision_number":1},{"id":"8731066e2572de800b9fc571d2275ae2f65f7bd7","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-07-10 16:50:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nyeah for sure!","accounts_in_message":[],"_revision_number":1},{"id":"6ca6a880c8537ca1de708b9e6db9bd32401eb15a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 20:24:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/11/670111/1/check/openstack-tox-docs/8394218/html/ : SUCCESS in 5m 15s\n- swift-tox-func http://logs.openstack.org/11/670111/1/check/swift-tox-func/0377c80/ : SUCCESS in 11m 31s\n- swift-tox-func-py37 http://logs.openstack.org/11/670111/1/check/swift-tox-func-py37/cf44504/ : SUCCESS in 8m 20s\n- swift-tox-func-encryption http://logs.openstack.org/11/670111/1/check/swift-tox-func-encryption/9d399e3/ : SUCCESS in 11m 39s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/11/670111/1/check/swift-tox-func-domain-remap-staticweb/ebb0f9e/ : SUCCESS in 12m 55s\n- swift-tox-func-ec http://logs.openstack.org/11/670111/1/check/swift-tox-func-ec/36825e0/ : SUCCESS in 13m 04s\n- swift-tox-func-s3api http://logs.openstack.org/11/670111/1/check/swift-tox-func-s3api/e073bc6/ : SUCCESS in 11m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/11/670111/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/4244ffc/ : SUCCESS in 13m 09s (non-voting)\n- swift-dsvm-functional http://logs.openstack.org/11/670111/1/check/swift-dsvm-functional/e201e2f/ : SUCCESS in 29m 48s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/11/670111/1/check/swift-dsvm-functional-ipv6/0cecc9a/ : SUCCESS in 26m 57s\n- openstack-tox-pep8 http://logs.openstack.org/11/670111/1/check/openstack-tox-pep8/7fff588/ : SUCCESS in 5m 08s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/11/670111/1/check/swift-multinode-rolling-upgrade/6c9a190/ : FAILURE in 8m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2769309f6c75ab5efcdb665ba37f321e653f4d32","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 20:25:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c5dfdeffcf64d349badc221c3bd3758df5a1a2c0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 21:26:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/11/670111/1/gate/openstack-tox-docs/587a7ec/html/ : SUCCESS in 4m 02s\n- swift-upload-image http://logs.openstack.org/11/670111/1/gate/swift-upload-image/f42dacb/ : SUCCESS in 8m 31s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/11/670111/1/gate/swift-tox-py27/c76e092/ : SUCCESS in 11m 40s\n- swift-tox-py37 http://logs.openstack.org/11/670111/1/gate/swift-tox-py37/00f7084/ : SUCCESS in 9m 35s\n- swift-tox-func http://logs.openstack.org/11/670111/1/gate/swift-tox-func/e7c2204/ : SUCCESS in 8m 53s\n- swift-tox-func-py37 http://logs.openstack.org/11/670111/1/gate/swift-tox-func-py37/a0331f9/ : SUCCESS in 10m 02s\n- swift-tox-func-encryption http://logs.openstack.org/11/670111/1/gate/swift-tox-func-encryption/306252f/ : SUCCESS in 9m 07s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/11/670111/1/gate/swift-tox-func-domain-remap-staticweb/f646b48/ : SUCCESS in 9m 08s\n- swift-tox-func-ec http://logs.openstack.org/11/670111/1/gate/swift-tox-func-ec/2bb2520/ : SUCCESS in 9m 27s\n- swift-tox-func-s3api http://logs.openstack.org/11/670111/1/gate/swift-tox-func-s3api/17af784/ : SUCCESS in 16m 23s\n- swift-dsvm-functional http://logs.openstack.org/11/670111/1/gate/swift-dsvm-functional/4545cb9/ : SUCCESS in 25m 03s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/11/670111/1/gate/swift-dsvm-functional-ipv6/44aba17/ : SUCCESS in 29m 53s\n- openstack-tox-pep8 http://logs.openstack.org/11/670111/1/gate/openstack-tox-pep8/5173cec/ : SUCCESS in 6m 08s","accounts_in_message":[],"_revision_number":1},{"id":"9cfaf217a62ec3ca99e2c5182f5dc1ddd6e75f42","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 21:26:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"92a3ca7b78f8db8523ff02b9850ad008759fb658","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 21:29:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- swift-promote-image http://logs.openstack.org/11/670111/1/promote/swift-promote-image/fc4fded/ : SUCCESS in 1m 04s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"38a24571ad5c192bacaf60a5634ea66164dbbb71","revisions":{"38a24571ad5c192bacaf60a5634ea66164dbbb71":{"kind":"REWORK","_number":1,"created":"2019-07-10 16:13:52.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/11/670111/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/670111/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/670111/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/670111/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/670111/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/670111/1"}}},"commit":{"parents":[{"commit":"392c967af53a7dd95c725fa8a765b3882a87f471","subject":"Merge \"container-replicator: Add a timeout for get_shard_ranges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/392c967af53a7dd95c725fa8a765b3882a87f471"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-07-10 16:13:44.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-07-10 16:13:44.000000000","tz":-420},"subject":"functests: make container creation less flakey in test_object","message":"functests: make container creation less flakey in test_object\n\nChange-Id: If62d82beb202dea553776920a95c177518b162ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38a24571ad5c192bacaf60a5634ea66164dbbb71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38a24571ad5c192bacaf60a5634ea66164dbbb71"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
