)]}'
{"id":"openstack%2Fswift~670932","triplet_id":"openstack%2Fswift~master~Id1280abd92e8bb02fcaa4701a0e9d211d9d6e33e","project":"openstack/swift","branch":"master","topic":"670932","hashtags":[],"change_id":"Id1280abd92e8bb02fcaa4701a0e9d211d9d6e33e","subject":"py3: fix up swift-orphans","status":"MERGED","created":"2019-07-16 01:08:29.000000000","updated":"2019-08-20 04:58:15.000000000","submitted":"2019-08-20 04:57:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670932-1566277024164-0fc9d827","meta_rev_id":"5de7b085a3800164f81e5452b4e8819b76303595","_number":670932,"virtual_id_number":670932,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":2,"date":"2019-08-20 04:57:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2019-08-17 01:05:27.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-19 23:17:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-17 01:05:27.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2019-08-20 04:57:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2a5c5ff8bd03402e073308b1d3e1893cfff24063","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-07-16 01:08:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a66a00a134c895a49de812a133427a5a04c6da9e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 01:45:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/32/670932/1/check/openstack-tox-docs/ed2cc37/html/ : SUCCESS in 3m 52s\n- swift-build-image http://logs.openstack.org/32/670932/1/check/swift-build-image/01548f7/ : SUCCESS in 7m 18s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/32/670932/1/check/swift-tox-py27/8e950bc/ : SUCCESS in 12m 51s\n- swift-tox-py36 http://logs.openstack.org/32/670932/1/check/swift-tox-py36/2b08099/ : SUCCESS in 10m 17s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/32/670932/1/check/swift-tox-py37/42a3fcd/ : SUCCESS in 10m 49s\n- swift-tox-func http://logs.openstack.org/32/670932/1/check/swift-tox-func/41ff88d/ : SUCCESS in 8m 42s\n- swift-tox-func-py37 http://logs.openstack.org/32/670932/1/check/swift-tox-func-py37/2b17d1f/ : SUCCESS in 9m 10s\n- swift-tox-func-encryption http://logs.openstack.org/32/670932/1/check/swift-tox-func-encryption/ce57105/ : SUCCESS in 9m 11s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/32/670932/1/check/swift-tox-func-domain-remap-staticweb/d83abaa/ : SUCCESS in 9m 00s\n- swift-tox-func-ec http://logs.openstack.org/32/670932/1/check/swift-tox-func-ec/fa4fd9f/ : SUCCESS in 9m 17s\n- swift-tox-func-s3api http://logs.openstack.org/32/670932/1/check/swift-tox-func-s3api/741b278/ : SUCCESS in 12m 22s\n- swift-tox-func-domain-remap-staticweb-py37 http://logs.openstack.org/32/670932/1/check/swift-tox-func-domain-remap-staticweb-py37/7de62f0/ : SUCCESS in 7m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/32/670932/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/462317b/ : SUCCESS in 12m 10s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/32/670932/1/check/swift-probetests-centos-7/279de62/ : SUCCESS in 30m 16s\n- swift-dsvm-functional http://logs.openstack.org/32/670932/1/check/swift-dsvm-functional/c157de6/ : SUCCESS in 25m 58s\n- swift-dsvm-functional-py3 http://logs.openstack.org/32/670932/1/check/swift-dsvm-functional-py3/8417322/ : SUCCESS in 31m 46s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/32/670932/1/check/swift-dsvm-functional-ipv6/bf6e6d9/ : SUCCESS in 31m 33s\n- swift-tox-lower-constraints http://logs.openstack.org/32/670932/1/check/swift-tox-lower-constraints/3f3d7e6/ : SUCCESS in 11m 13s\n- openstack-tox-pep8 http://logs.openstack.org/32/670932/1/check/openstack-tox-pep8/24c069f/ : SUCCESS in 5m 38s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/32/670932/1/check/swift-multinode-rolling-upgrade/57fe8fd/ : SUCCESS in 20m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7a7ca7914573b3d5e1c17e1f8bf421b7d00844ed","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-07-25 15:57:17.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"feaf128f7aeda5e3e1e8d1d55f9fac0b823ab866","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-14 23:52:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6593a137971cccc0c66eee3675c375f0c84a6cb0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 01:28:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37b8fab7b45c4b3fb6de58638a9c269b : SUCCESS in 4m 00s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e2d4142345c84c09bc27c71cea8fc145 : SUCCESS in 1h 00m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7f1a5440a61448478ec0b9639ff4743b : SUCCESS in 59m 18s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5e26a352f22c47ac860c5c29fe44ac5d : SUCCESS in 8m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/12a5db247add4a03b5ae3b1be29b6621 : SUCCESS in 12m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/91c43eb589b74007b68c1407b1948312 : SUCCESS in 11m 08s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/06b6e8866b2040188f9b21efeb905136 : SUCCESS in 10m 55s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/b45bf9754fe14decb6839b8c98cabcb8 : SUCCESS in 10m 37s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/e76668d3aff74ed598f84a3c8bf4779a : SUCCESS in 9m 17s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/3ba44b17a3ac4ef4b6adb68279239514 : SUCCESS in 8m 53s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/dd9edf47a75d4b6d8a7e26a579517fd0 : SUCCESS in 8m 37s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/421a044a51dd41dc9283c4864a341da6 : SUCCESS in 11m 56s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/f6a49b232bd4473abb312c53f9f0dff0 : SUCCESS in 13m 01s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/c55d30de06ad49da9576a22bbe0d006e : SUCCESS in 12m 55s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/398f6e5a5dc34b3c85d55c622d522671 : SUCCESS in 9m 53s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/59ddb7cf5a26403a92ea0492386aa152 : SUCCESS in 9m 43s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/4b12591dcc8545658c62ee66205c72e0 : FAILURE in 3m 18s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/425d2e6e415d4ce6bfc5af1c6b897571 : SUCCESS in 16m 12s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/341452eabf8b48dbbdf630df75791664 : SUCCESS in 33m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bbb4e65a2d6d4e33826c70bc67881ab2 : SUCCESS in 28m 17s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/9e01194ca52040c696b3fa4106fd27b2 : SUCCESS in 34m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/eb5a3e81004946a2a09f27febef9c775 : SUCCESS in 34m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a45aad8d9af245b9b20ebfa6f5de6460 : SUCCESS in 11m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53908f37f361402cad37b0426321046d : FAILURE in 5m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/aed676f4e69e4910ab6b1fced261a616 : SUCCESS in 17m 30s","accounts_in_message":[],"_revision_number":2},{"id":"041730dc90b64b31f9d85d7222774dfc421150f4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-15 17:35:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"456c327d2e3c3e1f268bae0bd1c7c2e082185a1a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 19:27:10.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/213ceecf37e943d696ceae29bd3dd6b3 : SUCCESS in 4m 23s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/6935415ba86d4e8fb5a0653aca09c81f : SUCCESS in 1h 04m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d5cec2dd60e949a788edfaea32fd3ef8 : SUCCESS in 57m 41s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/91189b0ac67f4dcd9c850105b32a0b49 : SUCCESS in 9m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/86925b0cc72742189c381c83f9f4d42a : SUCCESS in 15m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b4d2ad7315c6429ab9b5922f9873a390 : SUCCESS in 12m 23s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/98b5866281f34b4cabe267bd6312f891 : SUCCESS in 14m 11s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/dcbeae2248964532b5153f0b701607b6 : SUCCESS in 9m 59s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/951a559f31724809988484238ff55ea9 : SUCCESS in 9m 42s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/ef996943abfb43b8b30e0c291fc1fc67 : SUCCESS in 12m 34s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/d62aed48e67a4ba08569fec27ee992ed : SUCCESS in 12m 29s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/8b951de164a94123a4c17169d0695931 : SUCCESS in 14m 53s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/fcbf4b80c8124a4d82d6d18abc2e6728 : SUCCESS in 12m 44s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/07a2040ee8224aa6bcd98d0529a02124 : SUCCESS in 8m 47s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/29a2a293726e44e2a720a1872b0ae5b4 : SUCCESS in 10m 18s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/b54393474c0c467898136f2e6898c5c3 : SUCCESS in 12m 47s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/f1b73b78d6954daf9b5b19a657b8e07f : SUCCESS in 11m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/91bd4b8432de4338addf196ffa239b99 : SUCCESS in 13m 45s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5406cea8d97042ed98b123aba75ce35d : FAILURE in 42m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/54f0406e71dc4ecd857f055e9812bce3 : SUCCESS in 31m 58s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/dcf8c38a0884486bb2f4b64c6d4cdf11 : SUCCESS in 31m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/809807b003364a39bdeb93214628bfa1 : SUCCESS in 31m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e5d42126705c4c4080d80b79aaa75614 : SUCCESS in 13m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c68e4a0da1bd4c9b985fc5be5a0ea0c1 : SUCCESS in 6m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f5b2860a041e48fbbb515a5ed3fa36a6 : SUCCESS in 23m 14s","accounts_in_message":[],"_revision_number":3},{"id":"3492f9efa1856f42715cfc18ba55c0debc5e5b70","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-15 22:47:24.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"33e6140c800e1181a09cf5b99fc15db831a82346","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 23:55:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97c3d8f7775940b6bf875a079a9ac230 : SUCCESS in 5m 15s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1e81a4e28cc14d81bce85baa77310e96 : SUCCESS in 58m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d358d7d038034dd0b42927ec1abd89b3 : SUCCESS in 1h 01m 39s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/aacb8c7f6f254079a6dde461fe6a6d25 : SUCCESS in 9m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ffae0609a4d3493c84fce03022008831 : SUCCESS in 16m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/df94c2d932d744fd8df5bcda508f9a2a : SUCCESS in 12m 36s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/03bbf38250174d428bc6bcd9f42c2853 : SUCCESS in 12m 06s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/83468cb7f0414705bd53e90c4a8953f4 : SUCCESS in 8m 55s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/573033610ac641e3943f8fe2c4b36aba : SUCCESS in 10m 05s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/f26e5924f78e4e5f8492f6bdf1d3a33b : SUCCESS in 11m 26s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/360298b9f07d4abd98da0a859951412b : SUCCESS in 10m 57s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/50d90d482ff246d4b2ed105f660097ab : SUCCESS in 13m 41s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/0b6a3018bbd64ccd87dab95241c20ec5 : SUCCESS in 11m 21s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/887d1cdbdb174f3b9a3fd414ec0ddf75 : SUCCESS in 9m 02s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/d8b62e74e41b4678bc64443adb864509 : SUCCESS in 10m 51s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/ba3e880e0c6445079c4c1243e0513cc2 : SUCCESS in 10m 28s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/f56c8682bf6646659942454e5f18bb70 : SUCCESS in 11m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1e84b786a0bd4b59b6ad65ff7765860a : SUCCESS in 15m 09s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e5015b5efa6c46d2bbb0f1fe1aaa58e0 : SUCCESS in 31m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/62beae326fa74be5bd8bb217d3fe7815 : SUCCESS in 28m 26s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/a020ce9494b34fc3b47f90cb44c5d77e : SUCCESS in 31m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/53169b4339c1484b8c0b49ff9e2b10b2 : SUCCESS in 29m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f342525d9961473e9b631a649f5576cc : SUCCESS in 12m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/048ca7ca3d544e78a7aa04529b569d9f : SUCCESS in 7m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/32f1b3c876754280b7ed268eeef87265 : SUCCESS in 19m 54s","accounts_in_message":[],"_revision_number":3},{"id":"0963d6135ebf72024d230c807d7f87912b13847d","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-08-17 01:02:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7aa5b08768b35023236f0ee59dec205137273349","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-08-17 01:05:27.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"e6347c29aa56e05fc6ac037f04afeaeaf3f703b2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-19 23:17:07.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b7bbde0c7201fc873fe4e889bfecb3a02848f5db","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 23:17:38.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"44a74828851228bb50bb42462aa65acd3c16db2f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 04:57:03.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70332f3092ca40c483dc5d8f475e1e39 : SUCCESS in 5m 03s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9883647e950e4944bad7ef7ff465e66a : SUCCESS in 1h 03m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/54868316c6824f8d934accb75f414c5c : SUCCESS in 55m 12s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/86b6398ff3ae46a390840e07c1f95739 : SUCCESS in 11m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8b2ebc9b272e493093de27e105408f11 : SUCCESS in 11m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/066466a6273d4482bd8c3dce996fb25a : SUCCESS in 12m 10s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a072d340cafb42328fb7ca7b5243f0ef : SUCCESS in 13m 36s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/d4030b7f48864b1e9c3b55f30095aff1 : SUCCESS in 10m 19s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/c6c5b0d45e5b400f95434f0b1ea86cf2 : SUCCESS in 10m 55s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/def43e4bd1b3406da6bec7ab24debcea : SUCCESS in 11m 29s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/c53d22adf62a4c10becd879be3484e49 : SUCCESS in 12m 11s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/b2b9c4c7d9114d81b6e52aea830a7558 : SUCCESS in 14m 38s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/750425067ac94976a9c821848cf3e8af : SUCCESS in 15m 28s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/008c0043779146a685ccce258b47fc53 : SUCCESS in 11m 17s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/2f771ba913af41b4af0778541dcfd335 : SUCCESS in 25m 28s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/57aa1bd526de40f2ba023b12b7a072f5 : SUCCESS in 11m 18s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3a362a335e3f48c6946e103773b2728d : SUCCESS in 34m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/68d160972db14d08a3c89e2a1d21614b : SUCCESS in 33m 26s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/bc706607d051436f975c7243dc16cdc6 : SUCCESS in 48m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bcfc1b2bade84595b57599df72fa835b : SUCCESS in 32m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/628c68284e2345cb9cdd7d024bc694de : SUCCESS in 13m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96fb55f2325047b5bcfe0650a72d8634 : SUCCESS in 7m 22s","accounts_in_message":[],"_revision_number":3},{"id":"94cdc1a88aeb7cc8c7b9f795d62b7703f66e939f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 04:57:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"0756083da55fc2a4af09f5b631d3e032a767a366","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 04:58:15.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/f9ba16cafc7542858e177b306bf899b3 : SUCCESS in 51s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"27e7e80e922a51f3f4d21aa4d58f84fa9b5c6621","revisions":{"b91807c2a2455bed2ab2c18d7bf66050babc00b7":{"kind":"REWORK","_number":1,"created":"2019-07-16 01:08:29.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/32/670932/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/32/670932/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/670932/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/670932/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/670932/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/32/670932/1"}}},"commit":{"parents":[{"commit":"4643412bd11a7151bf56a590c260ccab9e1725bf","subject":"Merge \"func tests: tolerate more 404s when deleting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4643412bd11a7151bf56a590c260ccab9e1725bf"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-07-16 01:07:01.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-07-16 01:08:23.000000000","tz":-420},"subject":"py3: fix up swift-orphans","message":"py3: fix up swift-orphans\n\nChange-Id: Id1280abd92e8bb02fcaa4701a0e9d211d9d6e33e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b91807c2a2455bed2ab2c18d7bf66050babc00b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b91807c2a2455bed2ab2c18d7bf66050babc00b7"}]},"branch":"refs/heads/master"},"f3789d4497ccc0138a72165246a904065e5c45a2":{"kind":"REWORK","_number":2,"created":"2019-08-14 23:52:35.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/32/670932/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/32/670932/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/670932/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/670932/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/670932/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/32/670932/2"}}},"commit":{"parents":[{"commit":"4643412bd11a7151bf56a590c260ccab9e1725bf","subject":"Merge \"func tests: tolerate more 404s when deleting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4643412bd11a7151bf56a590c260ccab9e1725bf"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-07-16 01:07:01.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-14 23:52:24.000000000","tz":-420},"subject":"py3: fix up swift-orphans","message":"py3: fix up swift-orphans\n\nChange-Id: Id1280abd92e8bb02fcaa4701a0e9d211d9d6e33e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f3789d4497ccc0138a72165246a904065e5c45a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f3789d4497ccc0138a72165246a904065e5c45a2"}]},"branch":"refs/heads/master"},"27e7e80e922a51f3f4d21aa4d58f84fa9b5c6621":{"kind":"REWORK","_number":3,"created":"2019-08-15 17:35:01.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/32/670932/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/32/670932/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/670932/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/670932/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/670932/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/32/670932/3"}}},"commit":{"parents":[{"commit":"4643412bd11a7151bf56a590c260ccab9e1725bf","subject":"Merge \"func tests: tolerate more 404s when deleting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4643412bd11a7151bf56a590c260ccab9e1725bf"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-07-16 01:07:01.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-15 17:34:54.000000000","tz":-420},"subject":"py3: fix up swift-orphans","message":"py3: fix up swift-orphans\n\nChange-Id: Id1280abd92e8bb02fcaa4701a0e9d211d9d6e33e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27e7e80e922a51f3f4d21aa4d58f84fa9b5c6621"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27e7e80e922a51f3f4d21aa4d58f84fa9b5c6621"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
