)]}'
{"id":"openstack%2Fswift~671167","triplet_id":"openstack%2Fswift~master~Iddf9e089ef95dc709ab76dc58952a776246991fd","project":"openstack/swift","branch":"master","topic":"671167","hashtags":[],"change_id":"Iddf9e089ef95dc709ab76dc58952a776246991fd","subject":"py3: fix up listings on sharded containers","status":"MERGED","created":"2019-07-17 00:02:31.000000000","updated":"2019-08-28 07:39:27.000000000","submitted":"2019-08-28 07:37:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671167-1566977836229-2a78a105","meta_rev_id":"0b2529c0ed11117b813feb8832d9d381c3f580a8","_number":671167,"virtual_id_number":671167,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":2,"date":"2019-08-28 07:37:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-21 05:27:30.000000000","_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2019-08-28 06:28:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":2,"date":"2019-08-20 22:15:13.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2019-08-28 06:28:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-20 22:15:13.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2019-08-28 06:28:49.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2019-08-28 07:37:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a42964359b2f2550d233272eb78eec308c8ec519","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-07-17 00:02:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c58e8c3e7fe9027400ed4feeddeab5cef4149310","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 00:59:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/67/671167/1/check/openstack-tox-docs/68cef8e/html/ : SUCCESS in 4m 45s\n- swift-build-image http://logs.openstack.org/67/671167/1/check/swift-build-image/f17260e/ : SUCCESS in 9m 29s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/67/671167/1/check/swift-tox-py27/e7bc726/ : SUCCESS in 18m 09s\n- swift-tox-py36 http://logs.openstack.org/67/671167/1/check/swift-tox-py36/3b0cc40/ : SUCCESS in 18m 33s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/67/671167/1/check/swift-tox-py37/7e0b405/ : SUCCESS in 18m 47s\n- swift-tox-func http://logs.openstack.org/67/671167/1/check/swift-tox-func/fec7624/ : SUCCESS in 14m 16s\n- swift-tox-func-py37 http://logs.openstack.org/67/671167/1/check/swift-tox-func-py37/d9c04b2/ : SUCCESS in 14m 34s\n- swift-tox-func-encryption http://logs.openstack.org/67/671167/1/check/swift-tox-func-encryption/b7f3371/ : SUCCESS in 11m 33s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/67/671167/1/check/swift-tox-func-domain-remap-staticweb/5cf42c3/ : SUCCESS in 13m 56s\n- swift-tox-func-ec http://logs.openstack.org/67/671167/1/check/swift-tox-func-ec/b7f4732/ : SUCCESS in 11m 50s\n- swift-tox-func-s3api http://logs.openstack.org/67/671167/1/check/swift-tox-func-s3api/4a64fa9/ : SUCCESS in 18m 16s\n- swift-tox-func-domain-remap-staticweb-py37 http://logs.openstack.org/67/671167/1/check/swift-tox-func-domain-remap-staticweb-py37/62f1466/ : SUCCESS in 10m 45s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/67/671167/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/9a953ba/ : SUCCESS in 21m 30s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/67/671167/1/check/swift-probetests-centos-7/5f05807/ : SUCCESS in 31m 10s\n- swift-dsvm-functional http://logs.openstack.org/67/671167/1/check/swift-dsvm-functional/eee81ac/ : SUCCESS in 39m 38s\n- swift-dsvm-functional-py3 http://logs.openstack.org/67/671167/1/check/swift-dsvm-functional-py3/c95914b/ : SUCCESS in 31m 45s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/67/671167/1/check/swift-dsvm-functional-ipv6/9ff9c7e/ : SUCCESS in 38m 33s\n- swift-tox-lower-constraints http://logs.openstack.org/67/671167/1/check/swift-tox-lower-constraints/c288b68/ : FAILURE in 13m 32s\n- openstack-tox-pep8 http://logs.openstack.org/67/671167/1/check/openstack-tox-pep8/90c22e9/ : SUCCESS in 6m 54s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/67/671167/1/check/swift-multinode-rolling-upgrade/22627c7/ : SUCCESS in 16m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"227caeb54d9615466e5a0079471d473c6d8444a9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-07-24 23:14:25.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)\n\nstill needs tests","accounts_in_message":[],"_revision_number":1},{"id":"0cff20c89f273a192e9f5596f2ea9eee57c6ecf0","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2019-07-30 07:09:41.000000000","message":"Patch Set 1:\n\noh yeah nice. Looks clean... now just those tests :)","accounts_in_message":[],"_revision_number":1},{"id":"b08fe11c7da51817e499930850ba9f9c049b6a46","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-14 23:47:16.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nSince I\u0027d forgotten what the actual trouble was...","accounts_in_message":[],"_revision_number":1},{"id":"70150a2a75b9b5f0dc231052198ee4b09f9925e7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-14 23:47:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b4ef99ff819325ec7174ae7bb7e1183047bcf78f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 01:04:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3577bb478d9e4e6e9752621648ab4a2e : SUCCESS in 3m 57s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5702dc5809284c3abda32509147b7624 : SUCCESS in 50m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/079d1c3d93794690a193ad032c8bcb6a : SUCCESS in 53m 35s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8929e47a77bb42c2804379e7f98668f0 : SUCCESS in 7m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/815d6234b8a447739e8f756fce9449ff : FAILURE in 10m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c6c2b8fa9f5d41adba4c9bae56658439 : SUCCESS in 11m 06s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e49da1789e5d403e9faf629a8a4f2842 : SUCCESS in 10m 56s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/cba25970b43b4b82b814436a132f7589 : SUCCESS in 8m 17s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/2b3585d711c448b2ac9d61e353415c81 : SUCCESS in 8m 47s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/268f64b245dd4e19b3f95134655351e6 : SUCCESS in 9m 09s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/8e8caff74a40447b8a7796c30f2b1e38 : SUCCESS in 8m 43s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/ca65a70b601042259062727651f5ac8f : SUCCESS in 11m 27s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/50c6dcd46c164458b36545a6f0752cdf : SUCCESS in 10m 45s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/45313958876a40b0adb54aed388fdb52 : SUCCESS in 10m 36s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/4bc65b55e6d64d9e8227c4d78fbf7b22 : SUCCESS in 8m 56s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/7e531e405cc84d8d900dff29f9c321ea : SUCCESS in 9m 18s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cbe1f7a322ea442896f8d3da21400c2a : SUCCESS in 12m 54s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4cb3778329e844f493dffe8a48e0c8b5 : SUCCESS in 36m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f875963f6ddb46e1bf78263da6cc3582 : SUCCESS in 28m 55s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/9b75161794004ee0942533d2f5ec1418 : SUCCESS in 28m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7b122023a6be4085a8811ee3e739265e : SUCCESS in 27m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bb22fb5617bf4bef814e7d09ddf75607 : FAILURE in 12m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2df5514e2054cc79d5f5b33b0aec7c3 : FAILURE in 5m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0da6c43838494b6ca995f3b0d170df97 : SUCCESS in 18m 16s","accounts_in_message":[],"_revision_number":2},{"id":"dd8a258b453233ba56cd22d67f7496c4598d23cf","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-15 17:31:35.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"73b431200649e5869237c535cfe8eba2bcae70e5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 19:20:01.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88cf6b40d7c94836968f7c11d768154a : SUCCESS in 4m 52s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/016d0d201b39424bb504e2202697bcda : SUCCESS in 1h 01m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6cc27b1d237d4c1f87b92cf09af13484 : SUCCESS in 52m 59s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/817666a672b94cc88233d3b5f79eec48 : SUCCESS in 9m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/158e8d8f18ac48489a1264dd1b746f99 : FAILURE in 11m 22s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5148b3f037794c1688b9c4eb8ba17a24 : SUCCESS in 13m 37s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d09253751ce545c9b5898a5e7dfb6b06 : SUCCESS in 16m 59s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/0f9f3d57cb024034aa4837748b6a42be : SUCCESS in 8m 53s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/5e60037fdf154e2dbcab0c96a6101171 : SUCCESS in 9m 34s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/c5c7d5e256704cdc8c5363da0d46c8c7 : SUCCESS in 9m 50s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/c04b4ab5768549ff8b1361de65b09f71 : SUCCESS in 10m 03s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/e0ffab197f9a4bd48befcd4d7c81b254 : SUCCESS in 12m 07s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/82ea3f3ce390420490ec846bd5790fb8 : SUCCESS in 13m 53s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/4f611b800fbb4fa395a91950f002eb06 : SUCCESS in 13m 27s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/d67b8a85c6e740faaf975c05880e32a9 : SUCCESS in 11m 36s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3320ebdfedaa44379fb6c03bf5aa396f : SUCCESS in 13m 14s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/c005cb240816438a9fd77c2717a31701 : SUCCESS in 15m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1e59f0895ffc44c19ae2df6b207c7cb5 : SUCCESS in 15m 45s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/eda58d6893d64ae3ad35383d755c1776 : SUCCESS in 33m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e014677677b74e93ae3f94713dae26bd : SUCCESS in 32m 11s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/29b9b90bef7a40f89a05d0587108171d : SUCCESS in 35m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/eb19e64ab02d4e749ef76515666cc0a7 : SUCCESS in 29m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ecccc17526fa473082d41bfe6b9fe45d : FAILURE in 13m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/051d491bf52b4c26b8943459490fdaf0 : SUCCESS in 6m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8f4dc142d7c344eb93d57e3799810c9c : SUCCESS in 21m 37s","accounts_in_message":[],"_revision_number":3},{"id":"b50c3e1b3ef7ce9f77789f54c352533772fe7b9b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-15 19:34:19.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"bd24436b33ae64a23b5883ffbb5298c7c67657b2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 20:36:16.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04305958d7dd4ecb97de40f9d73cf93f : SUCCESS in 5m 59s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9b4aa953e7e74b85813ce8763fad4b9c : SUCCESS in 59m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/45f5cb3b842d4ce1a4b753bd4466444f : SUCCESS in 54m 21s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/35c3355b34ce4ab09fd2fd855d8bf83c : SUCCESS in 13m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/dd532300a1bf460590ad40eb9123da28 : SUCCESS in 14m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d2e911029378476186118199a05d938f : FAILURE in 14m 56s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5461c3d2c5c246cea43b5e1bd177ced7 : SUCCESS in 12m 47s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/15cbd4b6846f43d79b137ea691a6c3a0 : SUCCESS in 15m 13s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/1afd522e50bd49f4bfbb2ebede08b0b2 : SUCCESS in 14m 59s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/a319e640ca0d4eb09287e972c920647d : SUCCESS in 10m 32s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/cfffe283432949bcbfda75ad15560735 : SUCCESS in 16m 54s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/e77b902e3b0545eaaac7ece1840f536e : SUCCESS in 12m 48s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/8be716c9659c4176b98b4512adb93b52 : SUCCESS in 13m 08s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/356dce29aa694602b0ffad331082b1b1 : SUCCESS in 15m 27s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/3d4a2c19b6d44738bbd3a45b779b5b1c : SUCCESS in 10m 00s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/44d1503eacaf451eafca04efd6f50238 : SUCCESS in 13m 32s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/61cde81e5a1949c7a1d9e49429df0063 : SUCCESS in 11m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0230a327bcc94ef49953b21dece2ab22 : SUCCESS in 23m 23s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d22e0768b8d7442d8d9b0421ddc92aa0 : SUCCESS in 36m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/46bbd4d6298d460bafe49f1aa0ce0d8b : SUCCESS in 26m 25s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/cf90da1318ac400799a1df4e36ffa0a8 : SUCCESS in 36m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c0fd9a2d0d6840ee8df6221d2e12bfae : SUCCESS in 28m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9bf47c969024470fa269adfe03787a85 : SUCCESS in 13m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e6d2f7a6e0440e1afd695c32b7faca7 : SUCCESS in 7m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dcc3a4b54f3a459e86fa562e748ff5d8 : SUCCESS in 20m 13s","accounts_in_message":[],"_revision_number":4},{"id":"c9b0c31b482ca6356bd0ab1576ee64e31a265859","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-15 22:46:39.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"aa21470cbb7e9638a26e594967b348add0f16666","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 23:56:03.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bae206275d94ecbacbadd248610f0dc : SUCCESS in 6m 00s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/0c868f3c17d642bb9b85b4c9d317bf8e : SUCCESS in 1h 01m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/de0613b3f95b4e36b0561e48438f6e29 : SUCCESS in 1h 03m 12s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c0ec84810da44893920e8c3771efb9c0 : SUCCESS in 9m 43s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/40f08c9df37e4a0abab8709cc5cfeb15 : SUCCESS in 13m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/07fe244fabaf4cc7a38d9f568f3b04e6 : SUCCESS in 14m 49s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ae8276639274479e8c3e939aeef23140 : SUCCESS in 13m 45s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/8a4cb7d7f44f4a3b89103e6e8cff209d : SUCCESS in 10m 27s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/d9fb6d9139c84a0eac0b6fe456afaf74 : SUCCESS in 10m 56s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/da02fd55ccf24144b771d89cebae004a : SUCCESS in 9m 58s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/5b7662f3e9aa428bb94b282bf09ba59e : SUCCESS in 11m 00s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/027381bb8b37413b8cdf7b81d5fefb3c : SUCCESS in 12m 49s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/920eb853450e44cfb0ec7bba19f17c13 : SUCCESS in 12m 00s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/646638609909476db0510cc09cce1f4c : SUCCESS in 11m 17s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/8db6a8b2bc254092a8913f6eee02a08b : SUCCESS in 10m 34s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/d87ac752ac5541619f9e2d6f67190fc6 : SUCCESS in 9m 34s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/67b44fa59ed9416a95e1008f494d3880 : SUCCESS in 10m 36s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7383aadc5a734b9c93715edfee3f3dc7 : SUCCESS in 14m 49s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0c595917ba3f450491102916d1b01400 : SUCCESS in 32m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1c8eca5106df4767a86aafb3f72a19f8 : SUCCESS in 33m 57s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/f64fef6c764d4eb692ed41b99934cb59 : SUCCESS in 37m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/599153c49eba45cc888f7241f69b47ce : SUCCESS in 31m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/38a62b5b959c485a9d4e971b070ade09 : SUCCESS in 14m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1d3b5d560264406a4faf06dcea4fb60 : SUCCESS in 6m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3aeffe057a2e4c6cad1fe03a976f71be : SUCCESS in 22m 58s","accounts_in_message":[],"_revision_number":4},{"id":"90826ca14c78b31fe29185b76ef0d2997e558552","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-08-20 22:15:13.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3ea6caa797b84d14cc5d9a5960cb41b9dbc7c484","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-21 05:27:30.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f6b2e99ab6ca1fd73e3a2f594c2778f686bf346f","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2019-08-28 06:28:49.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"32512cfdc6ebcdd64a39e7a7c1a6a7e9b1539aa5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 06:29:02.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"23cce01b81e9d93de648bb8cf1e717a694a0d29b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 07:37:15.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e782565e1bed4bdfb6c163547d4431aa : SUCCESS in 6m 29s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3daf78dd3761443d9b4ab32483e7e4dd : SUCCESS in 1h 02m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1edee28e0316466c8f091da87f61f422 : SUCCESS in 52m 24s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/cc596cac889647ad813efba274152100 : SUCCESS in 10m 47s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e1e08b8fd5cc4254b5b4587a8ccd276e : SUCCESS in 16m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/628f3d0916154c71acdc025a56716f4a : SUCCESS in 12m 59s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e98379a5e3684c6cafcd2a068676b72c : SUCCESS in 13m 39s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/979b3a4abea8402d960669f41c0ecca4 : SUCCESS in 9m 51s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/1b8c4e188ce64085a28a603b2330466a : SUCCESS in 10m 11s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/554f7c90efe94a91a5ad514c4ef637a1 : SUCCESS in 10m 59s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/b51abddbcb41459cb66edce0bc51df05 : SUCCESS in 10m 58s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/f4f187b3c4bc4da9befd4c85aab24434 : SUCCESS in 12m 19s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/225c8672cd6a403d807774f447a3d92c : SUCCESS in 11m 29s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/c01f6c1ba568437187750a98e2607693 : SUCCESS in 11m 18s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/9ce78422c6f64448b50a92cad396a3ac : SUCCESS in 12m 09s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/85325d7db0504f849d8e495ddb9fc55a : SUCCESS in 13m 29s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f397a68e346a48ecbeae7ac29869f74f : SUCCESS in 32m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a98bf03358144b308f70a9c065c8bdd2 : SUCCESS in 28m 34s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/079a44b384a84301988bac0b95e5150a : SUCCESS in 30m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/138d19e8ebde410c9b1f589dacb53298 : SUCCESS in 33m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6f9d706358fb4f349ca4268627b000d8 : SUCCESS in 23m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9101e505a82d4937b8a795cac43688db : SUCCESS in 6m 51s","accounts_in_message":[],"_revision_number":4},{"id":"f4f6190371ecb25099e514123e7ee645ffac4059","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 07:37:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"8bbdf6ae5b370c084b5372ad15bf510730e01b9f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 07:39:27.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c7bd47a1eb646a696cb041d7e2a1498 : SUCCESS in 1m 46s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/20be227a62b3490dabf001f7040e12a5 : SUCCESS in 59s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"3750285bc863f8b6b56ba9526b028ee9cddcf04b","revisions":{"c167a8d926f11c7d3902d02d2586d25c7639bbdb":{"kind":"REWORK","_number":1,"created":"2019-07-17 00:02:31.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/67/671167/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/671167/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/671167/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/671167/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/671167/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/671167/1"}}},"commit":{"parents":[{"commit":"4643412bd11a7151bf56a590c260ccab9e1725bf","subject":"Merge \"func tests: tolerate more 404s when deleting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4643412bd11a7151bf56a590c260ccab9e1725bf"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-07-16 23:24:14.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-07-17 00:02:25.000000000","tz":-420},"subject":"py3: fix up listings on sharded containers","message":"py3: fix up listings on sharded containers\n\nWe were playing a little fast \u0026 losse with types before; as a result,\nmarker/end_marker weren\u0027t quite working right.\n\nNow, get everything to native strings before making comparisons, and\nget them back to wsgi when we shove them in the params dict.\n\nChange-Id: Iddf9e089ef95dc709ab76dc58952a776246991fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c167a8d926f11c7d3902d02d2586d25c7639bbdb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c167a8d926f11c7d3902d02d2586d25c7639bbdb"}]},"branch":"refs/heads/master"},"c37cc344da596a8571f4d22ea1b4c63def26a7cf":{"kind":"REWORK","_number":2,"created":"2019-08-14 23:47:22.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/67/671167/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/671167/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/671167/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/671167/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/671167/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/671167/2"}}},"commit":{"parents":[{"commit":"4643412bd11a7151bf56a590c260ccab9e1725bf","subject":"Merge \"func tests: tolerate more 404s when deleting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4643412bd11a7151bf56a590c260ccab9e1725bf"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-07-16 23:24:14.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-14 23:44:08.000000000","tz":-420},"subject":"py3: fix up listings on sharded containers","message":"py3: fix up listings on sharded containers\n\nWe were playing a little fast \u0026 loose with types before; as a result,\nmarker/end_marker weren\u0027t quite working right. In particular, we were\nchecking whether a WSGI string was contained in a shard range, while\nShardRange assumes all comparisons are against native strings.\n\nNow, get everything to native strings before making comparisons, and\nget them back to wsgi when we shove them in the params dict.\n\nChange-Id: Iddf9e089ef95dc709ab76dc58952a776246991fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c37cc344da596a8571f4d22ea1b4c63def26a7cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c37cc344da596a8571f4d22ea1b4c63def26a7cf"}]},"branch":"refs/heads/master"},"13bb9f017ca5d336b2a590ef129b6788dec21859":{"kind":"REWORK","_number":3,"created":"2019-08-15 17:31:35.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/67/671167/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/671167/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/671167/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/671167/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/671167/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/671167/3"}}},"commit":{"parents":[{"commit":"4643412bd11a7151bf56a590c260ccab9e1725bf","subject":"Merge \"func tests: tolerate more 404s when deleting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4643412bd11a7151bf56a590c260ccab9e1725bf"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-07-16 23:24:14.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-15 17:31:29.000000000","tz":-420},"subject":"py3: fix up listings on sharded containers","message":"py3: fix up listings on sharded containers\n\nWe were playing a little fast \u0026 loose with types before; as a result,\nmarker/end_marker weren\u0027t quite working right. In particular, we were\nchecking whether a WSGI string was contained in a shard range, while\nShardRange assumes all comparisons are against native strings.\n\nNow, get everything to native strings before making comparisons, and\nget them back to wsgi when we shove them in the params dict.\n\nChange-Id: Iddf9e089ef95dc709ab76dc58952a776246991fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/13bb9f017ca5d336b2a590ef129b6788dec21859"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/13bb9f017ca5d336b2a590ef129b6788dec21859"}]},"branch":"refs/heads/master"},"3750285bc863f8b6b56ba9526b028ee9cddcf04b":{"kind":"REWORK","_number":4,"created":"2019-08-15 19:34:19.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/67/671167/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/671167/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/671167/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/671167/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/671167/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/671167/4"}}},"commit":{"parents":[{"commit":"4643412bd11a7151bf56a590c260ccab9e1725bf","subject":"Merge \"func tests: tolerate more 404s when deleting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4643412bd11a7151bf56a590c260ccab9e1725bf"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-07-16 23:24:14.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-15 19:34:02.000000000","tz":-420},"subject":"py3: fix up listings on sharded containers","message":"py3: fix up listings on sharded containers\n\nWe were playing a little fast \u0026 loose with types before; as a result,\nmarker/end_marker weren\u0027t quite working right. In particular, we were\nchecking whether a WSGI string was contained in a shard range, while\nShardRange assumes all comparisons are against native strings.\n\nNow, get everything to native strings before making comparisons, and\nget them back to wsgi when we shove them in the params dict.\n\nChange-Id: Iddf9e089ef95dc709ab76dc58952a776246991fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3750285bc863f8b6b56ba9526b028ee9cddcf04b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3750285bc863f8b6b56ba9526b028ee9cddcf04b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
