)]}'
{"id":"openstack%2Fswift~671168","triplet_id":"openstack%2Fswift~master~I196b4e886942eccd2cb3aff3819da1558ddb19d4","project":"openstack/swift","branch":"master","topic":"errno","hashtags":[],"change_id":"I196b4e886942eccd2cb3aff3819da1558ddb19d4","subject":"Fix up errno checking","status":"MERGED","created":"2019-07-17 00:03:24.000000000","updated":"2019-07-24 13:18:46.000000000","submitted":"2019-07-24 13:16:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671168-1563974173557-6eba4f17","meta_rev_id":"81cc77c7ca95f4b5401bd9e335ba02301f468e13","_number":671168,"virtual_id_number":671168,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":2,"date":"2019-07-24 13:16:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"all":[{"value":2,"date":"2019-07-24 12:21:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"all":[{"value":1,"date":"2019-07-24 12:21:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-24 12:21:13.000000000","updated_by":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"reviewer":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"state":"REVIEWER"},{"updated":"2019-07-24 13:16:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5e64aad1d46fecd7a34ff37af4d96263a7239615","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-07-17 00:03:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f849d128c278a756183001918f56ed47431c6e4c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 00:45:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/68/671168/1/check/openstack-tox-docs/0ab1a2b/html/ : SUCCESS in 3m 57s\n- swift-build-image http://logs.openstack.org/68/671168/1/check/swift-build-image/998c6d3/ : SUCCESS in 7m 28s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/68/671168/1/check/swift-tox-py27/d28a828/ : SUCCESS in 10m 38s\n- swift-tox-py36 http://logs.openstack.org/68/671168/1/check/swift-tox-py36/f0d5e0a/ : SUCCESS in 10m 17s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/68/671168/1/check/swift-tox-py37/fcf64c4/ : SUCCESS in 9m 51s\n- swift-tox-func http://logs.openstack.org/68/671168/1/check/swift-tox-func/30ee4ad/ : SUCCESS in 8m 26s\n- swift-tox-func-py37 http://logs.openstack.org/68/671168/1/check/swift-tox-func-py37/4f6b548/ : SUCCESS in 9m 08s\n- swift-tox-func-encryption http://logs.openstack.org/68/671168/1/check/swift-tox-func-encryption/8e9c4d6/ : SUCCESS in 8m 53s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/68/671168/1/check/swift-tox-func-domain-remap-staticweb/ab0f991/ : SUCCESS in 8m 23s\n- swift-tox-func-ec http://logs.openstack.org/68/671168/1/check/swift-tox-func-ec/dab9beb/ : SUCCESS in 8m 50s\n- swift-tox-func-s3api http://logs.openstack.org/68/671168/1/check/swift-tox-func-s3api/16be9ea/ : SUCCESS in 11m 51s\n- swift-tox-func-domain-remap-staticweb-py37 http://logs.openstack.org/68/671168/1/check/swift-tox-func-domain-remap-staticweb-py37/e53fc8e/ : SUCCESS in 7m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/68/671168/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/7088c32/ : SUCCESS in 13m 01s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/68/671168/1/check/swift-probetests-centos-7/450d3ca/ : SUCCESS in 30m 37s\n- swift-dsvm-functional http://logs.openstack.org/68/671168/1/check/swift-dsvm-functional/6635dc1/ : SUCCESS in 27m 16s\n- swift-dsvm-functional-py3 http://logs.openstack.org/68/671168/1/check/swift-dsvm-functional-py3/8082481/ : SUCCESS in 29m 45s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/68/671168/1/check/swift-dsvm-functional-ipv6/f3af12b/ : SUCCESS in 25m 44s\n- swift-tox-lower-constraints http://logs.openstack.org/68/671168/1/check/swift-tox-lower-constraints/a8d177e/ : SUCCESS in 11m 03s\n- openstack-tox-pep8 http://logs.openstack.org/68/671168/1/check/openstack-tox-pep8/9c47177/ : SUCCESS in 5m 19s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/68/671168/1/check/swift-multinode-rolling-upgrade/72c1d2d/ : SUCCESS in 15m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c6b97300ef0e03a24ca069c70e7c417498acb5ee","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2019-07-24 12:21:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"eb2c6ac881e938bb85f48b02aabce36d6a82afa7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 12:21:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"00bf089f0e5216dd96d2dd63a253a7313e89c284","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 13:16:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/68/671168/1/gate/openstack-tox-docs/87a193f/html/ : SUCCESS in 5m 56s\n- swift-upload-image http://logs.openstack.org/68/671168/1/gate/swift-upload-image/c13b095/ : SUCCESS in 8m 45s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/68/671168/1/gate/swift-tox-py27/000bc65/ : SUCCESS in 11m 23s\n- swift-tox-py36 http://logs.openstack.org/68/671168/1/gate/swift-tox-py36/311c451/ : SUCCESS in 12m 07s\n- swift-tox-py37 http://logs.openstack.org/68/671168/1/gate/swift-tox-py37/75db0c0/ : SUCCESS in 13m 15s\n- swift-tox-func http://logs.openstack.org/68/671168/1/gate/swift-tox-func/ac1fb2a/ : SUCCESS in 8m 58s\n- swift-tox-func-encryption http://logs.openstack.org/68/671168/1/gate/swift-tox-func-encryption/0f3eb2b/ : SUCCESS in 10m 12s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/68/671168/1/gate/swift-tox-func-domain-remap-staticweb/6d24b10/ : SUCCESS in 10m 13s\n- swift-tox-func-ec http://logs.openstack.org/68/671168/1/gate/swift-tox-func-ec/453151a/ : SUCCESS in 10m 51s\n- swift-tox-func-s3api http://logs.openstack.org/68/671168/1/gate/swift-tox-func-s3api/20b99c9/ : SUCCESS in 12m 25s\n- swift-tox-func-py37 http://logs.openstack.org/68/671168/1/gate/swift-tox-func-py37/45c68e9/ : SUCCESS in 11m 01s\n- swift-tox-func-domain-remap-staticweb-py37 http://logs.openstack.org/68/671168/1/gate/swift-tox-func-domain-remap-staticweb-py37/7e1d3f5/ : SUCCESS in 9m 01s\n- swift-tox-func-ec-py37 http://logs.openstack.org/68/671168/1/gate/swift-tox-func-ec-py37/1909399/ : SUCCESS in 10m 03s\n- swift-probetests-centos-7 http://logs.openstack.org/68/671168/1/gate/swift-probetests-centos-7/ea44256/ : SUCCESS in 33m 29s\n- swift-dsvm-functional http://logs.openstack.org/68/671168/1/gate/swift-dsvm-functional/7a42764/ : SUCCESS in 30m 17s\n- swift-dsvm-functional-py3 http://logs.openstack.org/68/671168/1/gate/swift-dsvm-functional-py3/9349c7c/ : SUCCESS in 33m 32s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/68/671168/1/gate/swift-dsvm-functional-ipv6/86d263d/ : SUCCESS in 31m 13s\n- swift-tox-lower-constraints http://logs.openstack.org/68/671168/1/gate/swift-tox-lower-constraints/a3a0a9d/ : SUCCESS in 12m 09s\n- openstack-tox-pep8 http://logs.openstack.org/68/671168/1/gate/openstack-tox-pep8/51d687a/ : SUCCESS in 6m 23s","accounts_in_message":[],"_revision_number":1},{"id":"d2ee020eee2517adb034ab4c1afc21f448886c8e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 13:16:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c16215d3232ba9c4286c46e70a7ceb421ec5beff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 13:18:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- swift-promote-image http://logs.openstack.org/68/671168/1/promote/swift-promote-image/0918553/ : SUCCESS in 57s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"424a2603d8d4a454f269c9df5b4c7dc0d389b9da","revisions":{"424a2603d8d4a454f269c9df5b4c7dc0d389b9da":{"kind":"REWORK","_number":1,"created":"2019-07-17 00:03:24.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/68/671168/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/68/671168/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/671168/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/671168/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/671168/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/68/671168/1"}}},"commit":{"parents":[{"commit":"4643412bd11a7151bf56a590c260ccab9e1725bf","subject":"Merge \"func tests: tolerate more 404s when deleting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4643412bd11a7151bf56a590c260ccab9e1725bf"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-07-16 23:59:38.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-07-17 00:02:57.000000000","tz":-420},"subject":"Fix up errno checking","message":"Fix up errno checking\n\nChange-Id: I196b4e886942eccd2cb3aff3819da1558ddb19d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/424a2603d8d4a454f269c9df5b4c7dc0d389b9da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/424a2603d8d4a454f269c9df5b4c7dc0d389b9da"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
