)]}'
{"id":"openstack%2Fswift~671351","triplet_id":"openstack%2Fswift~master~Ieed88646d97ab2a1c7323a147ee97dae6fc86a0f","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"Ieed88646d97ab2a1c7323a147ee97dae6fc86a0f","subject":"Make rolling-upgrade job voting again","status":"MERGED","created":"2019-07-17 19:10:20.000000000","updated":"2019-07-24 00:49:56.000000000","submitted":"2019-07-24 00:48:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671351-1563929291670-7c4e2a43","meta_rev_id":"2f3ce4d91dfb311663bb721845cf1c49b1e5df5e","_number":671351,"virtual_id_number":671351,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-07-24 00:48:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-24 00:11:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-24 00:11:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-24 00:48:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fc50559cc8502d7ecd9c1694195c01a061655bda","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-07-17 19:10:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"440d23dd656b7266fd83960c6cc4feb2c509d325","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 19:49:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/51/671351/1/check/openstack-tox-docs/db39543/html/ : SUCCESS in 4m 36s\n- swift-build-image http://logs.openstack.org/51/671351/1/check/swift-build-image/1de04ac/ : SUCCESS in 7m 58s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/51/671351/1/check/swift-tox-py27/a8a74b5/ : SUCCESS in 11m 23s\n- swift-tox-py36 http://logs.openstack.org/51/671351/1/check/swift-tox-py36/932adf3/ : FAILURE in 10m 37s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/51/671351/1/check/swift-tox-py37/9b0f9e2/ : SUCCESS in 12m 25s\n- swift-tox-func http://logs.openstack.org/51/671351/1/check/swift-tox-func/bcb0f06/ : SUCCESS in 8m 59s\n- swift-tox-func-py37 http://logs.openstack.org/51/671351/1/check/swift-tox-func-py37/691221b/ : SUCCESS in 10m 11s\n- swift-tox-func-encryption http://logs.openstack.org/51/671351/1/check/swift-tox-func-encryption/202013f/ : FAILURE in 9m 58s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/51/671351/1/check/swift-tox-func-domain-remap-staticweb/d3db20f/ : SUCCESS in 10m 03s\n- swift-tox-func-ec http://logs.openstack.org/51/671351/1/check/swift-tox-func-ec/9416562/ : SUCCESS in 10m 46s\n- swift-tox-func-s3api http://logs.openstack.org/51/671351/1/check/swift-tox-func-s3api/dc7a020/ : FAILURE in 13m 43s\n- swift-tox-func-domain-remap-staticweb-py37 http://logs.openstack.org/51/671351/1/check/swift-tox-func-domain-remap-staticweb-py37/de35279/ : SUCCESS in 9m 24s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/51/671351/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/f56fbc4/ : SUCCESS in 17m 23s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/51/671351/1/check/swift-probetests-centos-7/c75e58b/ : SUCCESS in 31m 44s\n- swift-dsvm-functional http://logs.openstack.org/51/671351/1/check/swift-dsvm-functional/f818f9f/ : SUCCESS in 24m 35s\n- swift-dsvm-functional-py3 http://logs.openstack.org/51/671351/1/check/swift-dsvm-functional-py3/912ba85/ : SUCCESS in 27m 50s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/51/671351/1/check/swift-dsvm-functional-ipv6/0ddba33/ : SUCCESS in 26m 18s\n- swift-tox-lower-constraints http://logs.openstack.org/51/671351/1/check/swift-tox-lower-constraints/486674e/ : SUCCESS in 12m 14s\n- openstack-tox-pep8 http://logs.openstack.org/51/671351/1/check/openstack-tox-pep8/8992662/ : SUCCESS in 6m 44s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/51/671351/1/check/swift-multinode-rolling-upgrade/44ded36/ : SUCCESS in 18m 49s","accounts_in_message":[],"_revision_number":1},{"id":"2add84c4dc24890e3f341134b4dcbf8013a342b6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-07-17 21:54:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"af412993ffb8384b624c394987810abfddb44ab5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 22:30:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/51/671351/1/check/openstack-tox-docs/e7f46e1/html/ : SUCCESS in 4m 17s\n- swift-build-image http://logs.openstack.org/51/671351/1/check/swift-build-image/d9cb040/ : SUCCESS in 7m 31s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/51/671351/1/check/swift-tox-py27/cf57cdc/ : SUCCESS in 10m 14s\n- swift-tox-py36 http://logs.openstack.org/51/671351/1/check/swift-tox-py36/9838d76/ : SUCCESS in 13m 03s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/51/671351/1/check/swift-tox-py37/bd72ef8/ : SUCCESS in 10m 18s\n- swift-tox-func http://logs.openstack.org/51/671351/1/check/swift-tox-func/9cff7f7/ : SUCCESS in 10m 42s\n- swift-tox-func-py37 http://logs.openstack.org/51/671351/1/check/swift-tox-func-py37/ca93e50/ : SUCCESS in 8m 47s\n- swift-tox-func-encryption http://logs.openstack.org/51/671351/1/check/swift-tox-func-encryption/6cf1422/ : SUCCESS in 12m 14s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/51/671351/1/check/swift-tox-func-domain-remap-staticweb/1d78710/ : SUCCESS in 8m 34s\n- swift-tox-func-ec http://logs.openstack.org/51/671351/1/check/swift-tox-func-ec/b91276a/ : SUCCESS in 9m 26s\n- swift-tox-func-s3api http://logs.openstack.org/51/671351/1/check/swift-tox-func-s3api/42d11a7/ : SUCCESS in 14m 03s\n- swift-tox-func-domain-remap-staticweb-py37 http://logs.openstack.org/51/671351/1/check/swift-tox-func-domain-remap-staticweb-py37/2479075/ : SUCCESS in 7m 06s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/51/671351/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/8d9cbfa/ : SUCCESS in 18m 13s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/51/671351/1/check/swift-probetests-centos-7/dc9d8f5/ : SUCCESS in 31m 38s\n- swift-dsvm-functional http://logs.openstack.org/51/671351/1/check/swift-dsvm-functional/73d3bd6/ : SUCCESS in 29m 13s\n- swift-dsvm-functional-py3 http://logs.openstack.org/51/671351/1/check/swift-dsvm-functional-py3/7c6c6b9/ : SUCCESS in 31m 38s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/51/671351/1/check/swift-dsvm-functional-ipv6/fceee76/ : SUCCESS in 27m 12s\n- swift-tox-lower-constraints http://logs.openstack.org/51/671351/1/check/swift-tox-lower-constraints/d559f8a/ : SUCCESS in 15m 17s\n- openstack-tox-pep8 http://logs.openstack.org/51/671351/1/check/openstack-tox-pep8/1c8eb18/ : SUCCESS in 6m 53s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/51/671351/1/check/swift-multinode-rolling-upgrade/c0f6022/ : SUCCESS in 15m 35s","accounts_in_message":[],"_revision_number":1},{"id":"4060f1a79b1db3279840d149aee00b3fe91e12f1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-07-24 00:11:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"97e9218be67a10bc7d5c22653087dac5b7374ce3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 00:11:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1bc9c5b94629e7c6a3ab5c8c460a47100157b664","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 00:48:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/51/671351/1/gate/openstack-tox-docs/18f4ddb/html/ : SUCCESS in 4m 35s\n- swift-upload-image http://logs.openstack.org/51/671351/1/gate/swift-upload-image/d19707f/ : SUCCESS in 10m 26s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/51/671351/1/gate/swift-tox-py27/6ee379c/ : SUCCESS in 12m 58s\n- swift-tox-py36 http://logs.openstack.org/51/671351/1/gate/swift-tox-py36/330f578/ : SUCCESS in 15m 47s\n- swift-tox-py37 http://logs.openstack.org/51/671351/1/gate/swift-tox-py37/e205f22/ : SUCCESS in 15m 00s\n- swift-tox-func http://logs.openstack.org/51/671351/1/gate/swift-tox-func/b4a65db/ : SUCCESS in 9m 52s\n- swift-tox-func-encryption http://logs.openstack.org/51/671351/1/gate/swift-tox-func-encryption/38dfb96/ : SUCCESS in 13m 06s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/51/671351/1/gate/swift-tox-func-domain-remap-staticweb/034865e/ : SUCCESS in 13m 03s\n- swift-tox-func-ec http://logs.openstack.org/51/671351/1/gate/swift-tox-func-ec/041b9c0/ : SUCCESS in 12m 51s\n- swift-tox-func-s3api http://logs.openstack.org/51/671351/1/gate/swift-tox-func-s3api/43e0aa3/ : SUCCESS in 12m 13s\n- swift-tox-func-py37 http://logs.openstack.org/51/671351/1/gate/swift-tox-func-py37/6e48af5/ : SUCCESS in 13m 56s\n- swift-tox-func-domain-remap-staticweb-py37 http://logs.openstack.org/51/671351/1/gate/swift-tox-func-domain-remap-staticweb-py37/4c09f72/ : SUCCESS in 7m 37s\n- swift-tox-func-ec-py37 http://logs.openstack.org/51/671351/1/gate/swift-tox-func-ec-py37/2e5cbff/ : SUCCESS in 8m 36s\n- swift-probetests-centos-7 http://logs.openstack.org/51/671351/1/gate/swift-probetests-centos-7/6e31ee5/ : SUCCESS in 29m 54s\n- swift-dsvm-functional http://logs.openstack.org/51/671351/1/gate/swift-dsvm-functional/7cd75aa/ : SUCCESS in 29m 37s\n- swift-dsvm-functional-py3 http://logs.openstack.org/51/671351/1/gate/swift-dsvm-functional-py3/5148570/ : SUCCESS in 26m 07s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/51/671351/1/gate/swift-dsvm-functional-ipv6/7ec19a8/ : SUCCESS in 27m 38s\n- swift-tox-lower-constraints http://logs.openstack.org/51/671351/1/gate/swift-tox-lower-constraints/e93dcc7/ : SUCCESS in 13m 22s\n- openstack-tox-pep8 http://logs.openstack.org/51/671351/1/gate/openstack-tox-pep8/eda5220/ : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":1},{"id":"6177165aaf363e5dfa8ab8a46f52dde049a2eaa5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 00:48:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2f3ce4d91dfb311663bb721845cf1c49b1e5df5e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 00:49:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- swift-promote-image http://logs.openstack.org/51/671351/1/promote/swift-promote-image/0109a14/ : SUCCESS in 44s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a65350e887a53460ac18cc6c1c4fbb445aeee7e7","revisions":{"a65350e887a53460ac18cc6c1c4fbb445aeee7e7":{"kind":"REWORK","_number":1,"created":"2019-07-17 19:10:20.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/51/671351/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/51/671351/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/671351/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/671351/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/671351/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/51/671351/1"}}},"commit":{"parents":[{"commit":"a48ac28efe5c133b439ab0e525d315271bf370da","subject":"Authors/changelog for 2.22.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a48ac28efe5c133b439ab0e525d315271bf370da"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-07-17 19:10:05.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-07-17 19:10:05.000000000","tz":-420},"subject":"Make rolling-upgrade job voting again","message":"Make rolling-upgrade job voting again\n\nNow that os-testr has had a release, it passes again.\n\nChange-Id: Ieed88646d97ab2a1c7323a147ee97dae6fc86a0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a65350e887a53460ac18cc6c1c4fbb445aeee7e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a65350e887a53460ac18cc6c1c4fbb445aeee7e7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
