)]}'
{"id":"openstack%2Fswift~671882","triplet_id":"openstack%2Fswift~master~I106b94c77da220c762869aa800c31b87c3dffeeb","project":"openstack/swift","branch":"master","topic":"slo-416","hashtags":[],"change_id":"I106b94c77da220c762869aa800c31b87c3dffeeb","subject":"slo: Better handle non-manifest responses when refetching manifest","status":"MERGED","created":"2019-07-20 04:45:13.000000000","updated":"2019-07-22 17:44:35.000000000","submitted":"2019-07-22 17:43:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671882-1563817393530-7c4e2a43","meta_rev_id":"4440c9355840d1f877ede9501d90c2cc219493ff","_number":671882,"virtual_id_number":671882,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":2,"date":"2019-07-22 17:43:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-22 16:51:20.000000000","_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"all":[{"value":2,"date":"2019-07-22 16:19:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"all":[{"value":1,"date":"2019-07-22 16:43:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-22 16:43:06.000000000","updated_by":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"reviewer":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"state":"REVIEWER"},{"updated":"2019-07-22 17:43:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"aa324ce6cc9b7c1d62c7dfe672acc5b813055ab7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-07-20 04:45:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9697d31733cdb715c8f21a1282d0b31fd4135f5c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-07-20 04:45:58.000000000","message":"Topic changed from py3-func-tests to slo-416","accounts_in_message":[],"_revision_number":1},{"id":"e58bb44f0ce06e9f89cc6827979cebc0d3db4aab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-20 05:26:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/82/671882/1/check/openstack-tox-docs/d3c779d/html/ : SUCCESS in 4m 08s\n- swift-build-image http://logs.openstack.org/82/671882/1/check/swift-build-image/6a2dbe6/ : SUCCESS in 7m 55s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/82/671882/1/check/swift-tox-py27/8878d54/ : SUCCESS in 10m 33s\n- swift-tox-py36 http://logs.openstack.org/82/671882/1/check/swift-tox-py36/1d14bc2/ : SUCCESS in 11m 31s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/82/671882/1/check/swift-tox-py37/a9248a1/ : FAILURE in 12m 03s\n- swift-tox-func http://logs.openstack.org/82/671882/1/check/swift-tox-func/fa4aa4e/ : SUCCESS in 9m 58s\n- swift-tox-func-encryption http://logs.openstack.org/82/671882/1/check/swift-tox-func-encryption/149353c/ : SUCCESS in 9m 55s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/82/671882/1/check/swift-tox-func-domain-remap-staticweb/35a711a/ : SUCCESS in 9m 29s\n- swift-tox-func-ec http://logs.openstack.org/82/671882/1/check/swift-tox-func-ec/e7ee785/ : SUCCESS in 11m 21s\n- swift-tox-func-s3api http://logs.openstack.org/82/671882/1/check/swift-tox-func-s3api/17439dc/ : SUCCESS in 13m 04s\n- swift-tox-func-py37 http://logs.openstack.org/82/671882/1/check/swift-tox-func-py37/9553983/ : SUCCESS in 13m 37s\n- swift-tox-func-domain-remap-staticweb-py37 http://logs.openstack.org/82/671882/1/check/swift-tox-func-domain-remap-staticweb-py37/38531df/ : SUCCESS in 7m 56s\n- swift-tox-func-ec-py37 http://logs.openstack.org/82/671882/1/check/swift-tox-func-ec-py37/3d5ffe1/ : SUCCESS in 7m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/82/671882/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/cb9ec6a/ : SUCCESS in 11m 20s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/82/671882/1/check/swift-probetests-centos-7/d229e07/ : FAILURE in 26m 50s\n- swift-dsvm-functional http://logs.openstack.org/82/671882/1/check/swift-dsvm-functional/4909d1c/ : SUCCESS in 25m 19s\n- swift-dsvm-functional-py3 http://logs.openstack.org/82/671882/1/check/swift-dsvm-functional-py3/37896fd/ : SUCCESS in 28m 09s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/82/671882/1/check/swift-dsvm-functional-ipv6/a87247b/ : SUCCESS in 27m 18s\n- swift-tox-lower-constraints http://logs.openstack.org/82/671882/1/check/swift-tox-lower-constraints/508198d/ : SUCCESS in 9m 52s\n- openstack-tox-pep8 http://logs.openstack.org/82/671882/1/check/openstack-tox-pep8/0b450bf/ : SUCCESS in 5m 15s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/82/671882/1/check/swift-multinode-rolling-upgrade/3224bb3/ : SUCCESS in 16m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"359f901115460b4d35186ff5bd23a881deeccc23","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-07-20 15:42:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"780a2a688707effe8615d008d1bb04bcf929b744","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-20 16:18:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/82/671882/1/check/openstack-tox-docs/f88eecc/html/ : SUCCESS in 4m 51s\n- swift-build-image http://logs.openstack.org/82/671882/1/check/swift-build-image/0755649/ : SUCCESS in 7m 42s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/82/671882/1/check/swift-tox-py27/09d91d5/ : SUCCESS in 10m 06s\n- swift-tox-py36 http://logs.openstack.org/82/671882/1/check/swift-tox-py36/d1518e7/ : SUCCESS in 9m 46s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/82/671882/1/check/swift-tox-py37/8a7f8e9/ : SUCCESS in 10m 16s\n- swift-tox-func http://logs.openstack.org/82/671882/1/check/swift-tox-func/76b2854/ : SUCCESS in 8m 01s\n- swift-tox-func-encryption http://logs.openstack.org/82/671882/1/check/swift-tox-func-encryption/76220c8/ : SUCCESS in 7m 58s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/82/671882/1/check/swift-tox-func-domain-remap-staticweb/2e57282/ : SUCCESS in 9m 57s\n- swift-tox-func-ec http://logs.openstack.org/82/671882/1/check/swift-tox-func-ec/ef06479/ : SUCCESS in 8m 51s\n- swift-tox-func-s3api http://logs.openstack.org/82/671882/1/check/swift-tox-func-s3api/01d9963/ : SUCCESS in 11m 04s\n- swift-tox-func-py37 http://logs.openstack.org/82/671882/1/check/swift-tox-func-py37/1b9f330/ : SUCCESS in 8m 58s\n- swift-tox-func-domain-remap-staticweb-py37 http://logs.openstack.org/82/671882/1/check/swift-tox-func-domain-remap-staticweb-py37/ea89a84/ : SUCCESS in 7m 11s\n- swift-tox-func-ec-py37 http://logs.openstack.org/82/671882/1/check/swift-tox-func-ec-py37/13907c4/ : SUCCESS in 7m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/82/671882/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/1098e9b/ : SUCCESS in 11m 59s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/82/671882/1/check/swift-probetests-centos-7/5024f81/ : SUCCESS in 29m 56s\n- swift-dsvm-functional http://logs.openstack.org/82/671882/1/check/swift-dsvm-functional/6a28b60/ : SUCCESS in 24m 43s\n- swift-dsvm-functional-py3 http://logs.openstack.org/82/671882/1/check/swift-dsvm-functional-py3/eb6f9d7/ : SUCCESS in 29m 30s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/82/671882/1/check/swift-dsvm-functional-ipv6/1b06003/ : SUCCESS in 28m 10s\n- swift-tox-lower-constraints http://logs.openstack.org/82/671882/1/check/swift-tox-lower-constraints/bb0a2e9/ : SUCCESS in 10m 46s\n- openstack-tox-pep8 http://logs.openstack.org/82/671882/1/check/openstack-tox-pep8/abf0236/ : SUCCESS in 5m 59s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/82/671882/1/check/swift-multinode-rolling-upgrade/d85054b/ : SUCCESS in 15m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"276a5e43f77167c357e4a5fef97c23e3b7da5b51","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2019-07-22 14:52:06.000000000","message":"Patch Set 1: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"906a81397608f5544aa4fe24e09e187affa08814","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-07-22 15:42:29.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"dbe42c91bfa3a6b8d44e92917443727ef6c230cc","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2019-07-22 16:19:06.000000000","message":"Patch Set 1: Code-Review+2\n\non IRC we discussed some of the opinionated decisions made on this patch. main points are. After a re-fetch this is the new behavior:\n\n - If we find a newer 404 (with tombstone) then that is returned\n - If we get a 404 without timestamp (probably a handoff) we return a 503\n - if we get older (stale) SLO manifest, we serve that to client.","accounts_in_message":[],"_revision_number":1},{"id":"626df3cd59cd661f84b580fefc628a11890af060","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2019-07-22 16:43:06.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d5a1a4769e9b092d67b015f4ddd51285dbaf82b1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 16:45:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fa0871a6a0cd0ffdaebd03a83683fb7fc3f87b8e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-07-22 16:51:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"825c19c0f090283c422a9fbf279fffe5b7b8e96d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 17:43:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/82/671882/1/gate/openstack-tox-docs/96eae2c/html/ : SUCCESS in 6m 10s\n- swift-upload-image http://logs.openstack.org/82/671882/1/gate/swift-upload-image/e10b92f/ : SUCCESS in 8m 48s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/82/671882/1/gate/swift-tox-py27/db6a920/ : SUCCESS in 10m 51s\n- swift-tox-py37 http://logs.openstack.org/82/671882/1/gate/swift-tox-py37/2326976/ : SUCCESS in 11m 18s\n- swift-tox-func http://logs.openstack.org/82/671882/1/gate/swift-tox-func/2e9d951/ : SUCCESS in 11m 27s\n- swift-tox-func-encryption http://logs.openstack.org/82/671882/1/gate/swift-tox-func-encryption/cfd9799/ : SUCCESS in 12m 25s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/82/671882/1/gate/swift-tox-func-domain-remap-staticweb/ead2004/ : SUCCESS in 11m 30s\n- swift-tox-func-ec http://logs.openstack.org/82/671882/1/gate/swift-tox-func-ec/d2e7467/ : SUCCESS in 13m 50s\n- swift-tox-func-s3api http://logs.openstack.org/82/671882/1/gate/swift-tox-func-s3api/1478063/ : SUCCESS in 12m 34s\n- swift-tox-func-py37 http://logs.openstack.org/82/671882/1/gate/swift-tox-func-py37/07c1181/ : SUCCESS in 12m 04s\n- swift-tox-func-domain-remap-staticweb-py37 http://logs.openstack.org/82/671882/1/gate/swift-tox-func-domain-remap-staticweb-py37/20504a6/ : SUCCESS in 10m 21s\n- swift-tox-func-ec-py37 http://logs.openstack.org/82/671882/1/gate/swift-tox-func-ec-py37/1b5d04a/ : SUCCESS in 10m 45s\n- swift-probetests-centos-7 http://logs.openstack.org/82/671882/1/gate/swift-probetests-centos-7/bfa74e7/ : SUCCESS in 43m 32s\n- swift-dsvm-functional http://logs.openstack.org/82/671882/1/gate/swift-dsvm-functional/1c6f57a/ : SUCCESS in 31m 30s\n- swift-dsvm-functional-py3 http://logs.openstack.org/82/671882/1/gate/swift-dsvm-functional-py3/14c8d62/ : SUCCESS in 31m 34s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/82/671882/1/gate/swift-dsvm-functional-ipv6/b87df77/ : SUCCESS in 32m 13s\n- swift-tox-lower-constraints http://logs.openstack.org/82/671882/1/gate/swift-tox-lower-constraints/7da4845/ : SUCCESS in 11m 52s\n- openstack-tox-pep8 http://logs.openstack.org/82/671882/1/gate/openstack-tox-pep8/981a693/ : SUCCESS in 6m 54s","accounts_in_message":[],"_revision_number":1},{"id":"577147ced9a93d958eae7dfb94846487e098890e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 17:43:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4440c9355840d1f877ede9501d90c2cc219493ff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 17:44:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- swift-promote-image http://logs.openstack.org/82/671882/1/promote/swift-promote-image/4109bff/ : SUCCESS in 51s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ef5a37c2bf5a32b0fa6404974aa76332a88c0841","revisions":{"ef5a37c2bf5a32b0fa6404974aa76332a88c0841":{"kind":"REWORK","_number":1,"created":"2019-07-20 04:45:13.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/82/671882/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/82/671882/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/671882/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/671882/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/671882/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/82/671882/1"}}},"commit":{"parents":[{"commit":"2926fc96a36a573380eca18d60440a2337442015","subject":"py3: Bring functional/test_object.py under test; add func-ec-py37 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2926fc96a36a573380eca18d60440a2337442015"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-07-20 04:24:47.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-07-20 04:42:43.000000000","tz":-420},"subject":"slo: Better handle non-manifest responses when refetching manifest","message":"slo: Better handle non-manifest responses when refetching manifest\n\nPreviously, we never checked whether the response we get when refetching\nis even successful, much less whether it\u0027s still coming from an SLO.\n\nNow, if the refetched data is newer, act on it. If it\u0027s older, 503.\n\nCloses-Bug: #1837270\nChange-Id: I106b94c77da220c762869aa800c31b87c3dffeeb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ef5a37c2bf5a32b0fa6404974aa76332a88c0841"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ef5a37c2bf5a32b0fa6404974aa76332a88c0841"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
